)]}'
{"id":"openstack%2Fmanila-tempest-plugin~744851","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I60bf3bd1bd2cc22b5bb6be11c8174b7529d0a066","project":"openstack/manila-tempest-plugin","branch":"master","topic":"ping_export_location","hashtags":[],"change_id":"I60bf3bd1bd2cc22b5bb6be11c8174b7529d0a066","subject":"Ping to export location until it is reachable","status":"MERGED","created":"2020-08-05 08:28:15.000000000","updated":"2020-09-08 13:51:55.000000000","submitted":"2020-09-08 13:51:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744851-1599573115850-210ac784","meta_rev_id":"f2c0fa0a69cb699bde9fc7298e7a6e6e6cf80d32","_number":744851,"virtual_id_number":744851,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-07 20:26:55.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-08 13:51:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":2,"date":"2020-09-08 12:09:45.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-08 03:57:52.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":1,"date":"2020-09-08 06:22:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":1,"date":"2020-09-08 12:09:45.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-05 08:29:11.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2020-08-06 17:50:22.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-08-06 19:29:03.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-06 20:24:27.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-07 20:26:55.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-09-08 03:57:52.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-09-08 06:22:36.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-09-08 12:09:45.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-09-08 13:51:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fdd48e8a1f8eba728586f47012ca102ecc5e1e68","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-05 08:28:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec798fa004345a0be2267018e0557bdda114bc60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 10:06:36.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd82f6e3cab342bb9f383cb703806305 : FAILURE in 5m 22s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/da5a1ac3ff294ceab75a343b55262e6c : SUCCESS in 33m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/14118b5af40b4808badbfc63f4d34cb7 : SUCCESS in 29m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f65342d963eb4a7ca85784771877065d : FAILURE in 1h 02m 08s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/84f1a2cdd1a647f496c3ce1b44dbaa31 : SUCCESS in 1h 10m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/66841f23ed874f47a1fc804b6e476250 : SUCCESS in 1h 03m 44s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7384ca1865c840f1bb7fe29afdb49ae6 : FAILURE in 1h 03m 32s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e768d95438684ae6891308bbd65017fe : SUCCESS in 32m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/42897a8d53b84b868465eb388c281fa8 : FAILURE in 1h 35m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a7852bceede24c3488861fe63e3c6693 : SUCCESS in 24m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c4478c6750d9f4d1a2f4f79725da8e111e66affb","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-05 10:31:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fd120c8e3de67778a9f2c9b58844494292fa3c5e","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-08-05 10:46:35.000000000","message":"Patch Set 2:\n\n(4 comments)\n\nThanks, Liron, a waiter seems like a good idea -- some comments inline.","accounts_in_message":[],"_revision_number":2},{"id":"984183ff35138fcc615676046beaedca5ceb8fd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 12:46:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19ec20b2b2214017b2bebb66498cf8c1 : SUCCESS in 4m 29s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0dbcf8886b58471a9bcc749d4138beba : SUCCESS in 30m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0a1f94625fdb46efa177ac4930b30314 : SUCCESS in 24m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6fb6ac6bb8a345949e7c62a99efa33ea : SUCCESS in 1h 02m 25s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ba12c762716b452f9ba1268140884236 : SUCCESS in 1h 08m 36s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/232dc52c54bb4b5eb35fdc63726233bb : SUCCESS in 1h 05m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2f30b29eebc7466d99af5becd9150d8b : SUCCESS in 56m 21s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/dcb042b39d004bef82e16d7f4b93a5d6 : SUCCESS in 41m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a7c98b7be0104980afa308a73bf9f659 : FAILURE in 2h 04m 53s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/447dc18880da4ad98cefffcfcf5cdede : SUCCESS in 27m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"be1ff4b227c3323d99e139d1b90510f99164d5d8","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-08-05 13:52:44.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/51/744851/2/check/DellEMC_VNX_Manila/7d3f140/DellEMC_VNX_Manila/134 : SUCCESS in 1h 06m 04s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"442af2fdf663e46e8a8a7999a1ada5e934f14e0f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-05 15:02:49.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/51/744851/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/b71712b/ : SUCCESS in 2h 16m 05s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/51/744851/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/71d55ff/ : SUCCESS in 1h 41m 05s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"f97d3d5e024d3645ab87753189c47bb9be2f2322","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-05 21:41:25.000000000","message":"Patch Set 2:\n\nHi Liron, I still don\u0027t like this approach because it introduces two new configuration options... Why would we care if the IP address has been added or not, really?\n\nCurrently, before attempting to mount the share, we perform a ping test in most of the scenario tests (https://github.com/openstack/manila-tempest-plugin/blob/f8894b6f0b3d98394d6c573a065ac10baa7e621c/manila_tempest_tests/tests/scenario/test_share_basic_ops.py#L47-L52)\n\nThere\u0027s another validation ping checker that already exists in our scenario test manager code (https://github.com/openstack/manila-tempest-plugin/blob/7636a1fce4c2b49a03338097f110fadbb3f81206/manila_tempest_tests/tests/scenario/manager.py#L390-L425)\n\nIf you reuse this ping checker, you can make the tests wait until the export location IP is pingable with tempest.conf\u0027s config option: \"[validation]/ping_timeout\" (https://docs.openstack.org/tempest/latest/sampleconf.html#tempest-sampleconf)\n\nYou can then add this ping test to all the scenario tests; we only use it in the tests in the module: \"manila_tempest_tests/tests/scenario/test_share_basic_ops.py\" - this way you\u0027ll achieve what you need.","accounts_in_message":[],"_revision_number":2},{"id":"6502b0714f57bd72f2e9124d6addcfe6f01e2ca9","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-06 11:06:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7ccc2eb3e8dc3d2dfcb208ca36e4d31d86e3e1e5","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-06 11:10:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"08b87724c035d1dfc9e312596eeeb67477605758","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-06 11:16:21.000000000","message":"Patch Set 4:\n\n(2 comments)\n\n@Goutham\nTotally agree that it is better.\nBut we still need an one configuration option for the IP version of the StorageNFS network.","accounts_in_message":[],"_revision_number":4},{"id":"b84814c03b328bd64931743220d732cf17f59ed9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 13:28:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3211addf221a47d2be6368cbb525a20d : SUCCESS in 5m 11s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/61c9675a6d764132862f2faef908606d : SUCCESS in 28m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ced07f6b294f42ec82ff950eac3a780e : SUCCESS in 25m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0eac07f2d1be4c4584fe0cbba67e872e : FAILURE in 58m 04s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f54f4cbcf58a4e8cb6e61674dc77a6f8 : SUCCESS in 1h 00m 36s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b32854206a4340aeb4fe5ca0a234e365 : FAILURE in 1h 20m 52s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/085520fa6fcd490382c21b59a04102e1 : FAILURE in 1h 24m 30s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b7608c3fe09840578e8a0262b46e7571 : SUCCESS in 1h 03m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/aff6cbc4e5a44c4ab962d30e455afa9f : FAILURE in 2h 14m 02s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/76cbf0357bb44b0c8c859091c3e1d0c9 : SUCCESS in 22m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"114bdcf9300998083c19322bac6cb19808b3cd9f","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-06 16:13:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e48c05aa54759c595e23b0ace510a360b7d8e6a0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-06 17:50:22.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)\n\nThanks for the changes, some thoughts inline..","accounts_in_message":[],"_revision_number":5},{"id":"9700f35c2fc0f5096bccf1fe7c5450ff004918a2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-06 17:52:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1fdbb8d252780305afd4afb161a1660ced0e235b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 18:22:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0073c6cf2efd4118b10efbdc8d2d5559 : SUCCESS in 3m 46s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8985ade421f2413494e6f1069389a631 : SUCCESS in 25m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9b400fb728e24126aaa050f03be6a740 : SUCCESS in 33m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/176b5ecc567b400998c93b19a73ad38b : SUCCESS in 1h 09m 34s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cb943de26d0740dc8772cfe604d3d862 : FAILURE in 1h 06m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f4b6aa50f8514e21a47664ac031f7149 : FAILURE in 1h 19m 16s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3041bbf18f6b449189e1b3a460b82bc1 : SUCCESS in 1h 10m 27s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c1de8f88cfbe476fb8e15fa5ed5863a7 : SUCCESS in 41m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/99e27d5734c94e7f924c1ede6f8e9de7 : FAILURE in 2h 02m 52s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c4e3b66d54044d2e8ca150c2ff0d9ce5 : SUCCESS in 25m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cda93a6938af4b628002c9446de8038dc0be34ca","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-08-06 19:29:03.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/51/744851/5/check/DellEMC_VNX_Manila/d0a4431/DellEMC_VNX_Manila/137 : SUCCESS in 1h 06m 28s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"d1886849e2210f9ba663560d2b675ff2b791ffdc","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-06 20:24:27.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/51/744851/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/3a36526/ : SUCCESS in 2h 01m 50s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/51/744851/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/a7d9919/ : FAILURE in 1h 36m 33s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"b42732c0c268280cab7f9b86e5960a4d1de6d4b2","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-19 07:23:26.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"5fb679b800d37e79c0532574b98847fe7b365757","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-19 13:59:59.000000000","message":"Patch Set 5:\n\n(3 comments)\n\nHi liron, I\u0027ve responded - I feel we\u0027re talking past each other - I\u0027d be happy to discuss this synchronously and understand what I am missing..","accounts_in_message":[],"_revision_number":5},{"id":"3f360a9537eee580a618484e704e1e13c6c166fc","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-09-07 17:50:10.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"68f1a8f890a3d9aeca860d58e7241810d91eebab","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-09-07 17:57:28.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f06d437e1f8a4a34db453c8c8b9e261daa5cf32a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-07 18:47:01.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/51/744851/6/check/DellEMC_Unity_Manila_dhss_true/a1dc850/DellEMC_Unity_Manila_dhss_true/361 : FAILURE in 56m 40s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/51/744851/6/check/DellEMC_Unity_Manila_dhss_false/72508fa/DellEMC_Unity_Manila_dhss_false/343 : SUCCESS in 47m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"e2ea21f905caebd882dba7aa365e68838a0d9bd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 19:32:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58e2ebc102784125b54021beb43ea2bb : SUCCESS in 4m 26s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/824a563f3c764e1db5b038fb233a21a4 : SUCCESS in 20m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/70e4b9b721cc410fbd544df57b10eabc : SUCCESS in 22m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b7be3f8246a8459280cd0130eb8465c4 : SUCCESS in 46m 40s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/18511de59ca442108747a065ad6512c9 : SUCCESS in 44m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/500c42e3a6f44e9286f9164b6f837387 : SUCCESS in 37m 45s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/dbbced1d8c324b729f5ef4ba27178dc5 : SUCCESS in 40m 30s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a7a1f63c2e034960a6fc77b3cb0b53ee : SUCCESS in 32m 56s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/64bf9ab479a34e8ebf8e2b58b5c59cab : FAILURE in 1h 40m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/50605aebe4c946c392c071055e838ff0 : SUCCESS in 15m 33s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4d1d4274552dbc89782f12af6a7b701f94b1169e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-07 20:26:55.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/51/744851/6/check/DellEMC_Unity_Manila_dhss_true/95d9558/DellEMC_Unity_Manila_dhss_true/362 : FAILURE in 54m 00s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/51/744851/6/check/DellEMC_Unity_Manila_dhss_false/1d30be5/DellEMC_Unity_Manila_dhss_false/344 : SUCCESS in 47m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"a52ab307180ccc69caf803096c4bcabed504a120","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-09-08 03:57:52.000000000","message":"Patch Set 6: Code-Review+2\n\nLooks good Liron: quite simple and clean patch and CI passing ok. We might want to simplify the other ping checker (the one in manager.py), but that would be future work. Thanks!","accounts_in_message":[],"_revision_number":6},{"id":"be462b9f0009dccafd2250d8d3cf0164ef211266","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-09-08 06:22:36.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"1d4bd6453f97a7e083826689bd7757e98764eb40","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-09-08 12:09:45.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nLGTM, thanks!","accounts_in_message":[],"_revision_number":6},{"id":"9162e6c9aba551b5ddcdb52dee7f8fc431891f24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 12:09:56.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"9500e51c5d0ae481c2ed240c8aaa8a14c6b7e330","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 13:51:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"41e6df56e1b29ae00ca000a9a2350bc86b474454","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 13:51:55.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/480f51fd8cbe4803846c7835725a91fb : SUCCESS in 2m 58s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2cf0a5face2949ba94d5c5d3c8450f28 : SUCCESS in 30m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1a961ca246f448b0b2d571742eced614 : SUCCESS in 33m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/448affa563a94ef5b18914d4a37712d1 : SUCCESS in 1h 35m 22s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"4f0dbe042f2df75faf01f57a00caa1fb2f7462e2","revisions":{"d093bc986b4c1992e85c49655c329ae4e60e5968":{"kind":"REWORK","_number":1,"created":"2020-08-05 08:28:15.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/51/744851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/744851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/1"}}},"commit":{"parents":[{"commit":"7674f966482117ead95534875fa8116674f37f81","subject":"Merge \"Zuulv3 native \"standalone\" GlusterFS native, NFS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7674f966482117ead95534875fa8116674f37f81"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-05 08:09:29.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-05 08:28:12.000000000","tz":180},"subject":"Wait until Nova assigned IP is added to interface","message":"Wait until Nova assigned IP is added to interface\n\nIn some cases it takes time for the interface to acquire the\nIPv6 address.\nThis commit adds a new waiter called:\n\"_wait_until_ip_is_added_to_interface\" for waiting until the\nNova assigned IP will be added to interface.\n\nAdding a configuration option that will allow to handle in case the\nprivate network is IPv4 and the storage network is IPv6.\nCurrently, in such a case the scenario tests are skipped.\n\nChange-Id: I60bf3bd1bd2cc22b5bb6be11c8174b7529d0a066\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d093bc986b4c1992e85c49655c329ae4e60e5968"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d093bc986b4c1992e85c49655c329ae4e60e5968"}]},"branch":"refs/heads/master"},"be4443913ded00d5354f9006ce16079ed2a48f6d":{"kind":"REWORK","_number":2,"created":"2020-08-05 10:31:32.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/51/744851/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/744851/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/2"}}},"commit":{"parents":[{"commit":"7674f966482117ead95534875fa8116674f37f81","subject":"Merge \"Zuulv3 native \"standalone\" GlusterFS native, NFS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7674f966482117ead95534875fa8116674f37f81"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-05 08:09:29.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-05 10:31:19.000000000","tz":180},"subject":"Wait until Nova assigned IP is added to interface","message":"Wait until Nova assigned IP is added to interface\n\nIn some cases it takes time for the interface to acquire the\nIPv6 address.\nThis commit adds a new waiter called:\n\"_wait_until_ip_is_added_to_interface\" for waiting until the\nNova assigned IP will be added to interface.\n\nAdding a configuration option that will allow to handle in case the\nprivate network is IPv4 and the storage network is IPv6.\nCurrently, in such a case the scenario tests are skipped.\n\nChange-Id: I60bf3bd1bd2cc22b5bb6be11c8174b7529d0a066\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/be4443913ded00d5354f9006ce16079ed2a48f6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/be4443913ded00d5354f9006ce16079ed2a48f6d"}]},"branch":"refs/heads/master"},"ccb23f44f298cf64296b65fa88dc1215accd550e":{"kind":"REWORK","_number":3,"created":"2020-08-06 11:06:39.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/51/744851/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/744851/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/3"}}},"commit":{"parents":[{"commit":"7674f966482117ead95534875fa8116674f37f81","subject":"Merge \"Zuulv3 native \"standalone\" GlusterFS native, NFS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7674f966482117ead95534875fa8116674f37f81"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-05 08:09:29.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-06 10:44:26.000000000","tz":180},"subject":"Ping to export location until it is reachable","message":"Ping to export location until it is reachable\n\nIn some cases it takes time for the interface to acquire the\nIP address. This causes a failure in the ping operation.\nTherefore, we should ping until the network is reachable.\nThis is can be done by \"call_until_true\" method that allows\nreusing the ping method until the defined timeout.\n\nAdding a configuration option that will allow to handle in case the\nprivate network is IPv4 and the storage network is IPv6.\nCurrently, in such a case the scenario tests are skipped.\n\nChange-Id: I60bf3bd1bd2cc22b5bb6be11c8174b7529d0a066\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ccb23f44f298cf64296b65fa88dc1215accd550e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ccb23f44f298cf64296b65fa88dc1215accd550e"}]},"branch":"refs/heads/master"},"3493a8e50c1d5559eb3e98bca78b42cd61ec5a75":{"kind":"REWORK","_number":4,"created":"2020-08-06 11:10:53.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/51/744851/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/744851/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/4"}}},"commit":{"parents":[{"commit":"7674f966482117ead95534875fa8116674f37f81","subject":"Merge \"Zuulv3 native \"standalone\" GlusterFS native, NFS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7674f966482117ead95534875fa8116674f37f81"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-05 08:09:29.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-06 11:10:41.000000000","tz":180},"subject":"Ping to export location until it is reachable","message":"Ping to export location until it is reachable\n\nIn some cases it takes time for the interface to acquire the\nIP address. This causes a failure in the ping operation.\nTherefore, we should ping until the network is reachable.\nThis is can be done by \"call_until_true\" method that allows\nreusing the ping method until the defined timeout.\n\nAdding a configuration option that will allow to handle in case the\nprivate network is IPv4 and the storage network is IPv6.\nCurrently, in such a case the scenario tests are skipped.\n\nChange-Id: I60bf3bd1bd2cc22b5bb6be11c8174b7529d0a066\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3493a8e50c1d5559eb3e98bca78b42cd61ec5a75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3493a8e50c1d5559eb3e98bca78b42cd61ec5a75"}]},"branch":"refs/heads/master"},"3189d791f10999da9c6fed61d342e7e4bfc850a6":{"kind":"REWORK","_number":5,"created":"2020-08-06 16:13:53.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/51/744851/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/744851/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/5"}}},"commit":{"parents":[{"commit":"7674f966482117ead95534875fa8116674f37f81","subject":"Merge \"Zuulv3 native \"standalone\" GlusterFS native, NFS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7674f966482117ead95534875fa8116674f37f81"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-05 08:09:29.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-06 16:13:31.000000000","tz":180},"subject":"Ping to export location until it is reachable","message":"Ping to export location until it is reachable\n\nIn some cases it takes time for the interface to acquire the\nIP address. This causes a failure in the ping operation.\nTherefore, we should ping until the network is reachable.\nThis is can be done by \"call_until_true\" method that allows\nreusing the ping method until the defined timeout.\n\nAdding a configuration option that will allow to handle in case the\nprivate network is IPv4 and the storage network is IPv6.\nCurrently, in such a case the scenario tests are skipped.\n\nChange-Id: I60bf3bd1bd2cc22b5bb6be11c8174b7529d0a066\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3189d791f10999da9c6fed61d342e7e4bfc850a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3189d791f10999da9c6fed61d342e7e4bfc850a6"}]},"branch":"refs/heads/master"},"4f0dbe042f2df75faf01f57a00caa1fb2f7462e2":{"kind":"REWORK","_number":6,"created":"2020-09-07 17:50:10.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/51/744851/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/744851/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/744851/6"}}},"commit":{"parents":[{"commit":"2ab1bb66e0f7825a7afd6d85fe62ccba2de8a7af","subject":"Merge \"SSH over IPv4 when using a storage network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2ab1bb66e0f7825a7afd6d85fe62ccba2de8a7af"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-05 08:09:29.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-09-07 17:45:09.000000000","tz":180},"subject":"Ping to export location until it is reachable","message":"Ping to export location until it is reachable\n\nIn some cases it takes time for the interface to acquire the\nIP address. This causes a failure in the ping operation.\nTherefore, we should ping until the network is reachable.\nThis is can be done by \"call_until_true\" method that allows\nreusing the ping method until the defined timeout.\n\nThis patch also adds the other scenario tests for ipv6\nenvironmet.\n\nChange-Id: I60bf3bd1bd2cc22b5bb6be11c8174b7529d0a066\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4f0dbe042f2df75faf01f57a00caa1fb2f7462e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4f0dbe042f2df75faf01f57a00caa1fb2f7462e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
