)]}'
{"id":"openstack%2Fmanila-tempest-plugin~745609","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ic96b65d2caab788afca8bfc45575f3c05dc88008","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1890833","hashtags":[],"change_id":"Ic96b65d2caab788afca8bfc45575f3c05dc88008","subject":"Increase CephFS backing file size","status":"MERGED","created":"2020-08-11 00:43:23.000000000","updated":"2020-09-18 22:10:05.000000000","submitted":"2020-09-18 22:10:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745609-1600467005629-b51fe808","meta_rev_id":"9eeb5a31bfaf30454cd891b0be50596b5422f2a7","_number":745609,"virtual_id_number":745609,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-18 04:37:57.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"date":"2020-09-18 03:49:57.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-18 22:10:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2020-09-17 23:42:04.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-18 20:50:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-02 21:19:18.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-09-04 17:00:46.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-09-17 23:42:04.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-09-18 03:49:57.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-18 04:37:57.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-18 22:10:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9fae3097965c43631748ac01ede5ec3c29404278","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-11 00:43:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cdff6c97cc152e6f8e7a18d5a7cc9cc583416d4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-11 01:51:20.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/09/745609/1/check/DellEMC_Unity_Manila_dhss_true/12ed4ea/DellEMC_Unity_Manila_dhss_true/259 : FAILURE in 20m 40s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/09/745609/1/check/DellEMC_Unity_Manila_dhss_false/e0e9e89/DellEMC_Unity_Manila_dhss_false/243 : SUCCESS in 1h 07m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"be99e86f49158b60fe4da20a392eaa7cf01b84b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 01:55:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/24f6c41e4cb44db688957cff69e49296 : FAILURE in 1h 10m 25s\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e32592dd3f014f74ae3810c5e12f2077 : FAILURE in 1h 00m 29s","accounts_in_message":[],"_revision_number":1},{"id":"c94d96472bb4bc6bf65cad78668c6a32495c7cf3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-20 18:45:57.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"24979a43b354d0f0018be6453b95a86c970961fa","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-20 18:46:02.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"e5add23caf3fcdeaef121cac6edd8c3ddb20110f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-28 21:34:39.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"4c32e9eaa45c9a9c59bca383cd02a7b4c7e8813a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-28 21:34:50.000000000","message":"Topic changed from dnm to bug/1890833","accounts_in_message":[],"_revision_number":3},{"id":"7e02c31a2dde744e2c1db6c10bc3c4ab964a96be","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-28 23:26:41.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/09/745609/3/check/DellEMC_Unity_Manila_dhss_true/f31ccac/DellEMC_Unity_Manila_dhss_true/331 : SUCCESS in 1h 51m 53s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/09/745609/3/check/DellEMC_Unity_Manila_dhss_false/a9e5a4a/DellEMC_Unity_Manila_dhss_false/313 : SUCCESS in 47m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"c7db26aa6dca40bb907d8cd33acad9a63991b771","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 23:35:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b41366e48c445489e6e3bd2001123a6 : SUCCESS in 3m 55s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5feef7c740ac4d639a943fc42619c025 : SUCCESS in 26m 53s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/90ffb25cb2f64bb998e68dbed72ef032 : SUCCESS in 28m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4ba6e6252e5a49738f47b9871901edc0 : SUCCESS in 1h 00m 14s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2648c2c0ea604a4182d248f9b004cbe3 : SUCCESS in 1h 11m 11s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a2f68fbe6ebf4a08921cf40ede40e9d2 : FAILURE in 1h 24m 26s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/10527d8ffc8441e292b5235664ec99fb : SUCCESS in 50m 15s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/805d76a4b29e4fbbbf5552db8b820842 : SUCCESS in 48m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fb7b9b264866446da35c96f33e7b166c : FAILURE in 1h 57m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6f455c5e3d0b4820959ff7ae4961df5a : RETRY_LIMIT in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"18ad9bf9c7136af14962ea496648f39bc5b2b240","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-08-29 00:54:06.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/09/745609/3/check/DellEMC_VNX_Manila/035a7d9/DellEMC_VNX_Manila/189 : SUCCESS in 1h 18m 15s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"eab577cf752e22d8a6a1b93d5cb038d675dd26c7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-29 01:52:19.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/09/745609/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/b7f1a28/ : SUCCESS in 2h 16m 26s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/09/745609/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/91a09c4/ : FAILURE in 1h 30m 23s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"face2bd5767a0d5264b3057d13d69ee2fbe8220d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-29 03:13:36.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/09/745609/3/check/DellEMC_Unity_Manila_dhss_true/53012e8/DellEMC_Unity_Manila_dhss_true/333 : SUCCESS in 1h 53m 31s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/09/745609/3/check/DellEMC_Unity_Manila_dhss_false/c192256/DellEMC_Unity_Manila_dhss_false/315 : SUCCESS in 50m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"6aa342028a6062a887eb0940534aba86486c0325","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-08-29 11:45:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"21b2d32e4edeba30ead9964741a7335ae30f24b8","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-09-02 17:56:51.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c6ef4fca514fc7809d5fabd1961586d53510d9fe","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-02 19:38:04.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/09/745609/4/check/DellEMC_Unity_Manila_dhss_true/40881c8/DellEMC_Unity_Manila_dhss_true/353 : FAILURE in 58m 19s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/09/745609/4/check/DellEMC_Unity_Manila_dhss_false/502e351/DellEMC_Unity_Manila_dhss_false/335 : SUCCESS in 1h 01m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"c559381900d1ac7bce4bee10210ad2f245e7926e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 20:21:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5ac72b501df43f2ac69b0d53b576b0b : SUCCESS in 3m 56s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fc3d618c510f4267915a3d1491716d0e : SUCCESS in 37m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6c1e5950bf38439589572470c229e49e : SUCCESS in 33m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ba42c357261749f4901aaf28ca9a3974 : SUCCESS in 1h 37m 41s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2f02d4a161244b11bb0a619c3ae12f7f : SUCCESS in 1h 19m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/13f685e0e9ef4460846d5f90bd440972 : FAILURE in 14m 19s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c69fcb031b6547548f7f738406bf1f1d : FAILURE in 18m 44s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2db61653728b44edbffe358ed2b07a47 : SUCCESS in 52m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/66212bc50bb74de3bda086f1f98395bb : FAILURE in 2h 04m 07s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/fb1af6ece6be4899b4a15dd36dd830f6 : SUCCESS in 28m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"118f26c76941951f5d26340bc43b9ce67484542d","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-09-02 20:30:40.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/09/745609/4/check/DellEMC_VNX_Manila/97b062a/DellEMC_VNX_Manila/196 : FAILURE in 8m 25s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"d3d8a7c3ec3bb8515e21bc36153b132d79a14e80","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-02 21:19:18.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/09/745609/4/check/DellEMC_Unity_Manila_dhss_true/299b0e2/DellEMC_Unity_Manila_dhss_true/354 : FAILURE in 57m 11s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/09/745609/4/check/DellEMC_Unity_Manila_dhss_false/220843d/DellEMC_Unity_Manila_dhss_false/336 : SUCCESS in 50m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"77a538f76a61bf6641b258e89e110f6d8dde82df","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-02 22:27:33.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/09/745609/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/742b9cd/ : SUCCESS in 2h 05m 17s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/09/745609/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/35e282b/ : SUCCESS in 1h 34m 11s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"3edfef5827ef39aa19f00824d79110c180b5f70a","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-09-04 14:52:48.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"d55b5354a2134e3f447db4b6c1ce28d86b0dc06f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 16:50:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13a08ee824df4b6ab92dafa3f55499a7 : SUCCESS in 5m 10s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/434ad48b922542c59ccb3ad72437f12d : SUCCESS in 33m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2f4425b79e4242b08aed5dbdc1e7a38b : SUCCESS in 35m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1a4155476f3644ecb981242eac78f625 : SUCCESS in 1h 05m 33s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/005a781785d146269fd0eb82c4e770c5 : FAILURE in 51m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1b7373e36f8d4d8da20992077023fa46 : SUCCESS in 1h 14m 56s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ae5e81c4c19e4ed0add024c39d2bbc5e : FAILURE in 1h 26m 25s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ac8e5bfffa934cf7a76dfd4e728fd97a : SUCCESS in 45m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6193452dd6a647fe9fdd650d5a69facd : FAILURE in 1h 51m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e566032ae9af4c01b31ca168a619c9ca : SUCCESS in 28m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"005ba1a107ce9dcb4f757a948b240c6f0281a609","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-09-04 17:00:46.000000000","message":"Patch Set 5: Code-Review+2\n\nThanks Goutham!","accounts_in_message":[],"_revision_number":5},{"id":"92532538a7634dfff8451e317bc7d2b616fcfdcd","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-09-04 17:30:34.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a6e60fe7553392bd156ee3534ac1f70fafc8a139","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-04 19:10:35.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/09/745609/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/19a79a8/ : SUCCESS in 2h 19m 00s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/09/745609/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/cc673c5/ : FAILURE in 1h 27m 20s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"a026fb90406ef1b482d26bf065aba3f083b24be3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-17 19:27:29.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"ab896c88104020a16412d6be72a8ae572d631f32","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-17 19:28:11.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"0bcd2174d65750a9927ecb972d5a5acffac8c056","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-17 19:28:13.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"e502514edccc5eab9925884efa217b5f664226f8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-17 19:29:38.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"71311afd5eb075486835f39d5e4c63fe1169818f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 22:27:15.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de2bd8b67cf546ffb300330dabe1ae63 : SUCCESS in 4m 02s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4b56d82b2c7145489f09a906c5190e03 : SUCCESS in 34m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2a8f9dea6eab403ba093b6e8e0a78bfd : FAILURE in 38m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5919b7f74c3d4085a1cff5c4c9f123cb : SUCCESS in 1h 11m 17s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/aa15c33a69de4cb69b70a0d3b6f67e82 : FAILURE in 1h 10m 15s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1c7f67c5f5004beca7601daaee3df308 : FAILURE in 1h 46m 54s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/88cd15ecd13e43a8824146651b92dd0f : FAILURE in 1h 05m 37s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a531a2b996f84d8d89adfd94dcddf7c1 : SUCCESS in 1h 15m 24s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b142887882294bdc99108c78376b9a2e : FAILURE in 2h 53m 16s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1ea57a01f0e94e6b952060fc48fa925a : FAILURE in 20m 33s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ca843469d89b4bb32a48e0d7c1d1f0bd8e69040d","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-09-17 23:42:04.000000000","message":"Patch Set 7: Code-Review+2\n\nrecheck\n\ndummy job failure is unrelated","accounts_in_message":[],"_revision_number":7},{"id":"c13656b25e5fb1958af5cfe5a55bb1c82943cccf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 01:31:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a50612f013144c62818509a1a8d8e8ad : SUCCESS in 3m 09s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/457ebf0f0f08439e98e3a1e26c30d0fc : SUCCESS in 34m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a46bf1f661204c878666960795b9420b : SUCCESS in 38m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/21b4aeebfb8744f3b1de5d5ad4c57c52 : SUCCESS in 1h 06m 56s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c3d2c519ad404555b8bdbf32a27e2007 : FAILURE in 1h 14m 55s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d9912dec3a0440599f52fcbaa0f08b0e : SUCCESS in 57m 34s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/49d7587b4f764921a67cc820a2478b0c : SUCCESS in 1h 37m 56s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0f9557f96f0946c78525936f44a45f1a : SUCCESS in 49m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/122dee2ac7d64a719e5e18a6e81db25e : FAILURE in 1h 43m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6c3ba14c63ce4ca699e78c25112af322 : SUCCESS in 23m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"eb9e0ae0e836d9d597044a4dc1954499a970f765","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-18 03:49:57.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/09/745609/7/upstream-check/manila-cDOT-manila-tempest-plugin-ss/827c51d/ : SUCCESS in 2h 17m 55s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/09/745609/7/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/dd45ae8/ : SUCCESS in 1h 39m 08s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"1e213123cc6b811e0ad7eb43e09099fda0417280","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-09-18 04:37:57.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/09/745609/7/check/DellEMC_VNX_Manila/4bb31f5/DellEMC_VNX_Manila/241 : SUCCESS in 1h 29m 00s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"adb1e39ba31d724258bcb3b55d236b4be87d411b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-18 20:50:18.000000000","message":"Patch Set 7: Workflow+1\n\nWas just a rebase, no need to hold this up for other reviews.","accounts_in_message":[],"_revision_number":7},{"id":"8af63791d22f0646ab7e463eac82a8eaa3ededf9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 20:50:34.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"9690940be9e6972ea46ba6ac05e55d2fdd8c4329","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 22:10:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"9eeb5a31bfaf30454cd891b0be50596b5422f2a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 22:10:05.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b2982c8002c4af3bda52a9b864bcb5f : SUCCESS in 3m 41s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2b40cf6d940449948e6f64a528c99e34 : SUCCESS in 35m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e5057091dae74e1aab804aef966dd15c : SUCCESS in 35m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/93e476e34a034e89a59759ee4b4cd9e6 : SUCCESS in 1h 11m 05s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"a6602ecf052c1a51c61e9a0e1829fb5013d33349","revisions":{"574a4cd185b429da76685645d995be1daa2094d1":{"kind":"REWORK","_number":1,"created":"2020-08-11 00:43:23.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/09/745609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/09/745609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/1"}}},"commit":{"parents":[{"commit":"58723ffbff12f19ff48f6d22f281e65d4d9d7cba","subject":"Fixed InvalidConfiguration() exception to print correct message","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/58723ffbff12f19ff48f6d22f281e65d4d9d7cba"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-11 00:43:12.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-11 00:43:12.000000000","tz":-420},"subject":"DNM: Increase CephFS backing file size","message":"DNM: Increase CephFS backing file size\n\nChange-Id: Ic96b65d2caab788afca8bfc45575f3c05dc88008\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/574a4cd185b429da76685645d995be1daa2094d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/574a4cd185b429da76685645d995be1daa2094d1"}]},"branch":"refs/heads/master"},"5d9f51cd1aeb8cde3404b0ebe061115da5c20c90":{"kind":"REWORK","_number":2,"created":"2020-08-20 18:45:57.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/09/745609/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/09/745609/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/2"}}},"commit":{"parents":[{"commit":"58723ffbff12f19ff48f6d22f281e65d4d9d7cba","subject":"Fixed InvalidConfiguration() exception to print correct message","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/58723ffbff12f19ff48f6d22f281e65d4d9d7cba"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-11 00:43:12.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-20 18:45:55.000000000","tz":0},"subject":"DNM: Increase CephFS backing file size","message":"DNM: Increase CephFS backing file size\n\nChange-Id: Ic96b65d2caab788afca8bfc45575f3c05dc88008\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5d9f51cd1aeb8cde3404b0ebe061115da5c20c90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5d9f51cd1aeb8cde3404b0ebe061115da5c20c90"}]},"branch":"refs/heads/master"},"65806117c2f95419b89b607ae6ab2ad76f14ac20":{"kind":"REWORK","_number":3,"created":"2020-08-28 21:34:39.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/09/745609/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/09/745609/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/3"}}},"commit":{"parents":[{"commit":"58723ffbff12f19ff48f6d22f281e65d4d9d7cba","subject":"Fixed InvalidConfiguration() exception to print correct message","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/58723ffbff12f19ff48f6d22f281e65d4d9d7cba"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-11 00:43:12.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-28 21:34:36.000000000","tz":0},"subject":"Increase CephFS backing file size","message":"Increase CephFS backing file size\n\nThe backing file constitutes to the disk space\non the fake OSDs created.\n\nIn an effort to fix bug #1890833 we realized\nthat more space is needed to run manila tempest\ntests at a high concurrency.\n \nNeeded-By: I6ab157d6d099fe910ec1d90193783b55053ce8f6\nChange-Id: Ic96b65d2caab788afca8bfc45575f3c05dc88008\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/65806117c2f95419b89b607ae6ab2ad76f14ac20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/65806117c2f95419b89b607ae6ab2ad76f14ac20"}]},"branch":"refs/heads/master"},"34a922548639e785fe58518bfe156869ca16f07e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-09-02 17:56:51.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/09/745609/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/09/745609/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/4"}}},"commit":{"parents":[{"commit":"2bf994168390ba8f5ce8a568c72d9889326e7d10","subject":"Merge \"Revert CephFS test image to Bionic Beaver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2bf994168390ba8f5ce8a568c72d9889326e7d10"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-11 00:43:12.000000000","tz":-420},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-09-02 17:56:51.000000000","tz":0},"subject":"Increase CephFS backing file size","message":"Increase CephFS backing file size\n\nThe backing file constitutes to the disk space\non the fake OSDs created.\n\nIn an effort to fix bug #1890833 we realized\nthat more space is needed to run manila tempest\ntests at a high concurrency.\n \nNeeded-By: I6ab157d6d099fe910ec1d90193783b55053ce8f6\nChange-Id: Ic96b65d2caab788afca8bfc45575f3c05dc88008\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/34a922548639e785fe58518bfe156869ca16f07e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/34a922548639e785fe58518bfe156869ca16f07e"}]},"branch":"refs/heads/master"},"06046e17275a59ca3d26ef5dd02d2facff74e4ee":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-09-04 14:52:48.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/09/745609/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/09/745609/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/5"}}},"commit":{"parents":[{"commit":"2bf994168390ba8f5ce8a568c72d9889326e7d10","subject":"Merge \"Revert CephFS test image to Bionic Beaver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2bf994168390ba8f5ce8a568c72d9889326e7d10"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-11 00:43:12.000000000","tz":-420},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-09-04 14:52:37.000000000","tz":0},"subject":"Increase CephFS backing file size","message":"Increase CephFS backing file size\n\nThe backing file constitutes to the disk space\non the fake OSDs created.\n\nIn an effort to fix bug #1890833 we realized\nthat more space is needed to run manila tempest\ntests at a high concurrency.\n\nDepends-On: I0eab8ad25112511b0c4dd4b7796b4cda821bca8b\nNeeded-By: I6ab157d6d099fe910ec1d90193783b55053ce8f6\nChange-Id: Ic96b65d2caab788afca8bfc45575f3c05dc88008\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/06046e17275a59ca3d26ef5dd02d2facff74e4ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/06046e17275a59ca3d26ef5dd02d2facff74e4ee"}]},"branch":"refs/heads/master"},"a83979fd0d58de86e1e6e10a8607b0e600913924":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-09-17 19:28:11.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/09/745609/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/09/745609/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/6"}}},"commit":{"parents":[{"commit":"2e12123d97d04113fdc0f23458fa27ee18978da0","subject":"Merge \"Add tests for share server migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e12123d97d04113fdc0f23458fa27ee18978da0"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-11 00:43:12.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-09-17 19:28:11.000000000","tz":0},"subject":"Increase CephFS backing file size","message":"Increase CephFS backing file size\n\nThe backing file constitutes to the disk space\non the fake OSDs created.\n\nIn an effort to fix bug #1890833 we realized\nthat more space is needed to run manila tempest\ntests at a high concurrency.\n\nDepends-On: I0eab8ad25112511b0c4dd4b7796b4cda821bca8b\nNeeded-By: I6ab157d6d099fe910ec1d90193783b55053ce8f6\nChange-Id: Ic96b65d2caab788afca8bfc45575f3c05dc88008\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a83979fd0d58de86e1e6e10a8607b0e600913924"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a83979fd0d58de86e1e6e10a8607b0e600913924"}]},"branch":"refs/heads/master"},"a6602ecf052c1a51c61e9a0e1829fb5013d33349":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2020-09-17 19:29:38.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/09/745609/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/09/745609/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/745609/7"}}},"commit":{"parents":[{"commit":"2e12123d97d04113fdc0f23458fa27ee18978da0","subject":"Merge \"Add tests for share server migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e12123d97d04113fdc0f23458fa27ee18978da0"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-11 00:43:12.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-09-17 19:29:36.000000000","tz":0},"subject":"Increase CephFS backing file size","message":"Increase CephFS backing file size\n\nThe backing file constitutes to the disk space\non the fake OSDs created.\n\nIn an effort to fix bug #1890833 we realized\nthat more space is needed to run manila tempest\ntests at a high concurrency.\n\nChange-Id: Ic96b65d2caab788afca8bfc45575f3c05dc88008\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a6602ecf052c1a51c61e9a0e1829fb5013d33349"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a6602ecf052c1a51c61e9a0e1829fb5013d33349"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
