)]}'
{"id":"openstack%2Fmanila-tempest-plugin~746172","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I3fb2260274d463986fe8453f0f64cccf4ddea47b","project":"openstack/manila-tempest-plugin","branch":"master","topic":"change-746172","attention_set":{},"removed_from_attention_set":{"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2024-11-05 20:53:47.000000000","reason":"Change was abandoned"},"30998":{"account":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"last_update":"2024-11-05 20:53:47.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3fb2260274d463986fe8453f0f64cccf4ddea47b","subject":"Enable share migration tests for CIFS protocol","status":"ABANDONED","created":"2020-08-13 15:56:22.000000000","updated":"2024-11-05 20:53:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a508989e900e20ea10335d15d6428b23131b4554","_number":746172,"virtual_id_number":746172,"owner":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-11-22 21:04:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-16 00:08:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-16 01:06:10.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-09-16 01:39:29.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-16 02:29:29.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-29 19:26:34.000000000","updated_by":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"reviewer":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"state":"REVIEWER"}],"messages":[{"id":"2aaa3a5fd1481503279658a93ca2a9b3d2331b0b","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-08-13 15:56:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e948f5e78c2958deb1dd361994dad0bda4dccf3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 18:02:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3ef529b4737427eb375c6f0dce1891b : SUCCESS in 3m 41s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2b5a575b3cbb4964bf622cb1c3e0a478 : SUCCESS in 23m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/178f5d5045514cbfbbd992ea6121c046 : SUCCESS in 26m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3ca12f2530c349da8ca358d422ed10fd : FAILURE in 46m 49s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/77371afda83041868921677772d35030 : SUCCESS in 47m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/274b4a0d97674dd1a2405731bcfb43d9 : FAILURE in 1h 17m 13s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7ae1a7b3be0c464abfbb0033c146f134 : FAILURE in 38m 59s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7d369ac7d3b1473b9abe682242054a8b : SUCCESS in 40m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/43a1961e14f94f8285bb64ef9d87b55f : FAILURE in 1h 55m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a89a9379b7c74fc286862a421149409b : FAILURE in 23m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5949d26830fae55877f449cbe1e9069272e65785","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-08-14 15:07:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"31374973c5328fa9c5944d05e91773c57435e25e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-14 16:24:15.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/72/746172/2/check/DellEMC_Unity_Manila_dhss_true/a08b7c8/DellEMC_Unity_Manila_dhss_true/265 : FAILURE in 1h 17m 04s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/72/746172/2/check/DellEMC_Unity_Manila_dhss_false/44287fc/DellEMC_Unity_Manila_dhss_false/249 : SUCCESS in 56m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"5bd2e469060a5e1ed6d4ae9e107e14499d47c01e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 17:03:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b84a08b49df4682b7e9be6752175bfa : SUCCESS in 2m 59s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4be8e970439c490299b4286267ac0457 : FAILURE in 37m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8a5c12355d794dc6b5d5a10bb3b54970 : FAILURE in 39m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/332c0c5136504d9aafa443c28b9bd8da : FAILURE in 1h 05m 18s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cc82436abfdc40938875875ab2a1eeb6 : FAILURE in 1h 08m 19s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ae0a24b4dcc3492da289a907725710de : FAILURE in 1h 41m 32s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9038357e89aa4988a24d8c305448b89a : SUCCESS in 1h 21m 11s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6a5894235d3d45cba53fbb8c04d8c01d : SUCCESS in 43m 02s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d44fbdf2f8c84398874c646be259c6b1 : FAILURE in 1h 50m 37s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3537aab98a1542c08138b47540c7a53d : FAILURE in 24m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e6f768a8cf8d4f108f963949b7f62cb97007979f","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-08-14 18:02:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2c62fc8490181d30316306f2e5ee373b9cf11fea","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-14 19:20:52.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/72/746172/3/check/DellEMC_Unity_Manila_dhss_true/323e2fb/DellEMC_Unity_Manila_dhss_true/266 : FAILURE in 1h 18m 17s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/72/746172/3/check/DellEMC_Unity_Manila_dhss_false/c1fbeed/DellEMC_Unity_Manila_dhss_false/250 : SUCCESS in 54m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"a71279e9790f32f4a7801c6c870ac2fcf9cd93a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 20:23:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/191ca88916824a6da4748ac290ccbe42 : SUCCESS in 2m 39s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a0c5d289aeba4519a99363581459ab3b : SUCCESS in 35m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/db67aec9b8c94145b31473a3dd8b1ffa : SUCCESS in 48m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/601db746f3f64e049ba947670fbe524b : SUCCESS in 1h 09m 00s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/bbd45cce2846431fb87a6f485081743b : SUCCESS in 1h 20m 42s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/bff6e5c0c37947a6a73387d256edaa7e : FAILURE in 2h 14m 09s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/cc1aab2982f1412e94fc5e0f393daff6 : FAILURE in 1h 48m 13s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5acb6d6814674bc8a371977714994d0b : SUCCESS in 1h 12m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6ad38d27ec674f88bd8507d4e7004ff9 : FAILURE in 2h 19m 54s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b2c69e6600e747a2b226dc1d7b76c524 : FAILURE in 36m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"324cdf9bd97d4e6e22e5f5f7c8ce0fa6810d94e1","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-08-14 21:30:19.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/72/746172/3/check/DellEMC_VNX_Manila/5c9327e/DellEMC_VNX_Manila/153 : SUCCESS in 1h 06m 12s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"880a79adadd03ac9164d963b9fc0124d33d9d6c3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-14 21:42:31.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/72/746172/3/check/DellEMC_Unity_Manila_dhss_true/6e7ee3d/DellEMC_Unity_Manila_dhss_true/267 : FAILURE in 1h 18m 30s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/72/746172/3/check/DellEMC_Unity_Manila_dhss_false/364edd1/DellEMC_Unity_Manila_dhss_false/251 : SUCCESS in 54m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"9dc079b9fbe9431076fbedec341afb088b2f5f7e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-14 22:40:57.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/72/746172/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/ee42c27/ : SUCCESS in 2h 16m 47s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/72/746172/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/55926d0/ : SUCCESS in 1h 30m 13s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"745a091ed2900c39a077acd42650c226d5bdd297","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-08-15 13:32:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"54c1a0f447dc4f9ebf2b16edb738590af92c78d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 15:15:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5db0cdfae9534fbea7b0488a8a92cbaf : SUCCESS in 3m 25s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0c940091eb4e4a2d8e7d7f1b75f49826 : FAILURE in 30m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d51fc6cf8dbc4864b62c3798725e7def : SUCCESS in 35m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/654f48524ae44c8ea7e6789a8807d2b0 : SUCCESS in 1h 09m 41s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8afcbb25524641239f043683ba4fb39f : SUCCESS in 1h 07m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/4ce4eca40c004975914bdd7c2bf4b01f : FAILURE in 1h 27m 08s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/87365439b0894a909def93db82688dd6 : FAILURE in 53m 48s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fa317174b1024073835c3dd61003d8df : SUCCESS in 41m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e91356705958480fa68f4dd58a907afb : FAILURE in 1h 42m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/fe0fe9f447384dcc889a2971b7c5b342 : SUCCESS in 23m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9fe0c05931073102f9cf9ed2b52c6118a7ee2e6d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-15 15:42:55.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/72/746172/4/check/DellEMC_Unity_Manila_dhss_true/e39f4cd/DellEMC_Unity_Manila_dhss_true/270 : FAILURE in 1h 06m 44s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/72/746172/4/check/DellEMC_Unity_Manila_dhss_false/66052ad/DellEMC_Unity_Manila_dhss_false/254 : SUCCESS in 57m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"8f887faa0831d5961efda5a609a35dce3afca452","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-08-15 17:29:39.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"578f5db3b652e0be034c064cf3bb97ffbe05469c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 18:32:40.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca214b15e02f4e76a33d73dab7fe6fd5 : SUCCESS in 2m 43s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/95e0baf5540c482eb2a15c7ace69e023 : SUCCESS in 31m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/535e0fc9c39c432796b7a04df68868ef : FAILURE in 41m 22s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/32d88b3ed1aa4564a1c2fbc8071323ca : SUCCESS in 1h 00m 52s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e75cd0732ff84275bff4ed52bdbc2a03 : SUCCESS in 1h 01m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/015bea2e2d164bc19a9f2234da7f2fbf : FAILURE in 23m 29s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c79e4ba50691463d9d4c54dbb1a0ac40 : FAILURE in 48m 13s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b4df9886c4394f259f9bff15ec2766c1 : SUCCESS in 41m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/97406fff332d4dce838d459574222fc8 : FAILURE in 41m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/72536ecb0a1746d1acab7aef19260244 : SUCCESS in 23m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2fbca63dae49e42e7bad185bdefa9c9c7495be60","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-08-20 19:21:36.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"ebdbff773692f136b52af39446735bb8f1501088","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 21:49:18.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcbd8445c1384de4a965a873ceed152b : SUCCESS in 3m 42s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a0d90c95606845c4b163894ad07adf72 : SUCCESS in 37m 47s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/05f1ee63682042e4819c72f3a7f8f4e8 : SUCCESS in 36m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f578ab2e0f324fa2969bf325dbc192b2 : FAILURE in 1h 21m 35s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7ca124c76aa04458ae4477c0ed5ea970 : SUCCESS in 1h 11m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/08114b4ae13d41cdaca2c825ab4c6159 : FAILURE in 1h 25m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/23f59037542044f1a25b112cbbc51e15 : SUCCESS in 1h 01m 02s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b6d809bbdebf43f7bff7ca912b41ed5e : SUCCESS in 37m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6f407619ee5645e7aa72077345c2cdce : FAILURE in 2h 24m 06s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e4524c7f29044b95bf88b64f7fc76b2e : SUCCESS in 35m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f3a02ae80bad450144984dae4cfd005e353ded2f","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-08-21 13:25:26.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"16242e0774786f0c284468753da11d56d4051d95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 15:40:15.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/935d5e46fb6543f48681a29b344a1b72 : SUCCESS in 4m 07s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4ea9e26ee83d4386aacdff95326735f8 : SUCCESS in 31m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b2a4baa43d994f7facf52a71e3830a1f : SUCCESS in 35m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d97d5f0e3792460da9efb640015a2a1f : FAILURE in 1h 23m 25s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c2813f06d4b241258b3b2ab3f052b490 : SUCCESS in 1h 13m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/001f8eb4c4254fe08496cb3813f0a3ae : FAILURE in 1h 25m 30s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b74098dd625b4d53a2f36575d539d3c7 : FAILURE in 48m 50s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/de0f50383b2444aaa7881d1634489b86 : SUCCESS in 49m 26s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/776f74e63ea946c299580a55a6062129 : FAILURE in 2h 02m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/10428fecbaea4a8d9d078a8db1dcaffb : SUCCESS in 21m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b2e86ab650052c4d33056d9726cd0330ffea2a79","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-09-15 22:02:03.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"b852135b407f6295c8e412a849fb958d1b9ab664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 00:08:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6329478887684caa8a7789bca29bf320 : SUCCESS in 2m 46s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ea2852a2e37e4001adb8291f6e8ab124 : SUCCESS in 35m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d34058325953490abf597aa7ac31b504 : SUCCESS in 38m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/58596b8e36354e078bc4179440072fbd : SUCCESS in 1h 07m 52s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/219e5576fa3941e5b74c3bcfa6b9c6f0 : SUCCESS in 1h 16m 53s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a32d038b13d84a3ea63c367dec47e120 : SUCCESS in 1h 11m 21s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/af947d1f90bd467a8c97ebb1cdf91c0f : FAILURE in 1h 08m 08s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e8e36da77eea4785843ada867cafda7a : SUCCESS in 46m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9d0e3f76ed324366b3279c1b4423f8c3 : FAILURE in 2h 00m 42s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/abfa941e2c8d4778af783fd9a91a233a : SUCCESS in 25m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"33d3f3d75fc51ace0d67489351e2baca04426378","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-16 01:06:10.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/72/746172/4/check/DellEMC_Unity_Manila_dhss_true/c5dd702/DellEMC_Unity_Manila_dhss_true/416 : FAILURE in 54m 46s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/72/746172/4/check/DellEMC_Unity_Manila_dhss_false/5359715/DellEMC_Unity_Manila_dhss_false/397 : SUCCESS in 57m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"981e627ad64a03ea9ac8bb492f06ac66c4c0cc3e","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-09-16 01:39:29.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/72/746172/4/check/DellEMC_VNX_Manila/ef7df61/DellEMC_VNX_Manila/227 : SUCCESS in 1h 30m 20s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"93cff51561d7fcd6ffe7f44c74e88a91cdf04c16","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-16 02:29:29.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/72/746172/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/ccad3ab/ : SUCCESS in 2h 20m 15s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/72/746172/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/4cc5e93/ : SUCCESS in 1h 41m 53s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"b1fff9b0181b95aeb6c7269b7672f7c559cc8655","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2021-11-22 18:47:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"821c96aa39d1b19bc4c1c49991251ed0490bad2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 21:04:28.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c22bc35cf2704d068c55957e716e8d2a : FAILURE in 4m 30s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1a2cbbd2eb654e26b3e184dac5b1fd40 : FAILURE in 38m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c98ae14ede85410ea6d391a4e2729a73 : FAILURE in 44m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/19c107971e15448e84dfbe089665d60f : FAILURE in 1h 22m 22s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/8fae4edc25e843cca06d6aca1429c16b : FAILURE in 1h 35m 06s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/b971c939eed54d7291df832ecd704079 : FAILURE in 1h 16m 32s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/410fe79daed141c28ae4a2d09d81f8bb : FAILURE in 1h 22m 42s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d5d6262491e54dfc98be0cd85dbee188 : FAILURE in 1h 30m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a04222f84d8c4e8f8aaa5f49cc510ae6 : SUCCESS in 1h 47m 56s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/58c7514f09b149fea6beeddf94f9d0d3 : FAILURE in 2h 09m 03s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2b44f40fd0724f34b690794f40e84099 : FAILURE in 48m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0587ac2df5aa47bba826b699befc243c : FAILURE in 1h 17m 02s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0f03556c6726445f9499257d853667ec : SUCCESS in 31m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a508989e900e20ea10335d15d6428b23131b4554","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2024-11-05 20:53:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b563e23e9313089dca6979444296695778c1d324","revisions":{"910f9f3ef2d0e744daefa6740ac668aa0a5885fe":{"kind":"REWORK","_number":1,"created":"2020-08-13 15:56:22.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/72/746172/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/72/746172/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/746172/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/746172/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/746172/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/746172/1"}}},"commit":{"parents":[{"commit":"58723ffbff12f19ff48f6d22f281e65d4d9d7cba","subject":"Fixed InvalidConfiguration() exception to print correct message","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/58723ffbff12f19ff48f6d22f281e65d4d9d7cba"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2020-08-13 13:01:13.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2020-08-13 15:54:33.000000000","tz":-180},"subject":"[DNM] Testing migration tests with CIFS","message":"[DNM] Testing migration tests with CIFS\n\n- Needed for testing share migration with container driver.\n\nChange-Id: I3fb2260274d463986fe8453f0f64cccf4ddea47b\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/910f9f3ef2d0e744daefa6740ac668aa0a5885fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/910f9f3ef2d0e744daefa6740ac668aa0a5885fe"}]},"branch":"refs/heads/master"},"35146a91e97d6c3e458a421fc82bf8615287818d":{"kind":"REWORK","_number":2,"created":"2020-08-14 15:07:01.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/72/746172/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/72/746172/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/746172/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/746172/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/746172/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/746172/2"}}},"commit":{"parents":[{"commit":"58723ffbff12f19ff48f6d22f281e65d4d9d7cba","subject":"Fixed InvalidConfiguration() exception to print correct message","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/58723ffbff12f19ff48f6d22f281e65d4d9d7cba"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2020-08-13 13:01:13.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2020-08-14 15:05:21.000000000","tz":-180},"subject":"Enable share migration tests for CIFS protocol","message":"Enable share migration tests for CIFS protocol\n\nThis patch enable share migration tests for CIFS protocol and\nadd a new configuration option to choose which protocol should be\ncovered by share migration tests.\n\nChange-Id: I3fb2260274d463986fe8453f0f64cccf4ddea47b\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/35146a91e97d6c3e458a421fc82bf8615287818d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/35146a91e97d6c3e458a421fc82bf8615287818d"}]},"branch":"refs/heads/master"},"0a1f7df737349ba74f65d5379a27e4724312c8de":{"kind":"REWORK","_number":3,"created":"2020-08-14 18:02:27.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/72/746172/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/72/746172/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/746172/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/746172/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/746172/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/746172/3"}}},"commit":{"parents":[{"commit":"58723ffbff12f19ff48f6d22f281e65d4d9d7cba","subject":"Fixed InvalidConfiguration() exception to print correct message","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/58723ffbff12f19ff48f6d22f281e65d4d9d7cba"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2020-08-13 13:01:13.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2020-08-14 18:02:01.000000000","tz":-180},"subject":"Enable share migration tests for CIFS protocol","message":"Enable share migration tests for CIFS protocol\n\nThis patch enable share migration tests for CIFS protocol and\nadd a new configuration option to choose which protocol should be\ncovered by share migration tests.\n\nChange-Id: I3fb2260274d463986fe8453f0f64cccf4ddea47b\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0a1f7df737349ba74f65d5379a27e4724312c8de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0a1f7df737349ba74f65d5379a27e4724312c8de"}]},"branch":"refs/heads/master"},"3a1230aa664134d9cba1814f8e6dbd75304ce5cb":{"kind":"REWORK","_number":4,"created":"2020-08-15 13:32:06.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/72/746172/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/72/746172/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/746172/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/746172/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/746172/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/746172/4"}}},"commit":{"parents":[{"commit":"58723ffbff12f19ff48f6d22f281e65d4d9d7cba","subject":"Fixed InvalidConfiguration() exception to print correct message","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/58723ffbff12f19ff48f6d22f281e65d4d9d7cba"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2020-08-13 13:01:13.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2020-08-15 13:28:45.000000000","tz":-180},"subject":"Enable share migration tests for CIFS protocol","message":"Enable share migration tests for CIFS protocol\n\n-This patch enables share migration tests for CIFS protocol and\nadd a new configuration option to choose which protocol should be\ncovered by share migration tests.\n- It enables driver assisted migration tests for container driver\nwhich will run these tests using CIFS protocol.\n\nDepends-On: I813801fd64d4d605c0a4fe06c7a130ad00865b6d\nChange-Id: I3fb2260274d463986fe8453f0f64cccf4ddea47b\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3a1230aa664134d9cba1814f8e6dbd75304ce5cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3a1230aa664134d9cba1814f8e6dbd75304ce5cb"}]},"branch":"refs/heads/master"},"b563e23e9313089dca6979444296695778c1d324":{"kind":"REWORK","_number":5,"created":"2021-11-22 18:47:51.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/72/746172/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/72/746172/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/746172/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/746172/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/746172/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/746172/5"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2020-08-13 13:01:13.000000000","tz":-180},"committer":{"name":"Andre Beltrami","email":"debeltrami@gmail.com","date":"2021-11-22 16:02:09.000000000","tz":-180},"subject":"Enable share migration tests for CIFS protocol","message":"Enable share migration tests for CIFS protocol\n\n-This patch enables share migration tests for CIFS protocol and\nadd a new configuration option to choose which protocol should be\ncovered by share migration tests.\n- It enables driver assisted migration tests for container driver\nwhich will run these tests using CIFS protocol.\n\nDepends-On: I813801fd64d4d605c0a4fe06c7a130ad00865b6d\nChange-Id: I3fb2260274d463986fe8453f0f64cccf4ddea47b\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b563e23e9313089dca6979444296695778c1d324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b563e23e9313089dca6979444296695778c1d324"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
