)]}'
{"id":"openstack%2Fmanila-tempest-plugin~746564","triplet_id":"openstack%2Fmanila-tempest-plugin~master~If5e21a05202c40464cf09f156c06768a63e6265c","project":"openstack/manila-tempest-plugin","branch":"master","topic":"thin","hashtags":[],"change_id":"If5e21a05202c40464cf09f156c06768a63e6265c","subject":"[WIP] test_over_subscription_free_space_reporting","status":"ABANDONED","created":"2020-08-17 16:44:16.000000000","updated":"2020-08-17 19:36:09.000000000","total_comment_count":23,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9fd5e86438e4868101857b014f44b836a558391d","_number":746564,"virtual_id_number":746564,"owner":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-08-17 18:55:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-08-17 18:40:01.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-17 18:40:01.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-08-17 18:55:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c059a3af9fd338bf3fda2c43aec172e9978f85f6","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2020-08-17 16:44:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6b03bd47481c20736857e5826f0de56f854152e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-17 18:40:01.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/64/746564/1/check/DellEMC_Unity_Manila_dhss_true/6063cf7/DellEMC_Unity_Manila_dhss_true/280 : FAILURE in 1h 15m 45s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/64/746564/1/check/DellEMC_Unity_Manila_dhss_false/93d7643/DellEMC_Unity_Manila_dhss_false/264 : FAILURE in 1h 01m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"80b407b8063f2e506e1468a642ed8e50fcef31a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 18:55:12.000000000","message":"Patch Set 1: Verified-1\n\n(23 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c067c6f076fd4a08850126f5c1eb4adf : FAILURE in 4m 07s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0253198f096a45888cdd50df2850381a : FAILURE in 30m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1c16e2243c2945bfbcc704d8bd86368e : FAILURE in 34m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9f345082ec1e441aad0349f2a9c10cc6 : SUCCESS in 1h 02m 54s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/979c625c42a947d5ba8b59daa61a41cb : SUCCESS in 1h 08m 57s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/761b5081327640c8bc9432fc9e5fb1b3 : FAILURE in 10m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/aa87887d82ce4b0c856375f7a3931de4 : FAILURE in 11m 30s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/72176a765fe6465f9d467851753b55fd : FAILURE in 40m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b62ad5cccfa44cdd9f7cb33c5e19fe57 : FAILURE in 1h 58m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/75b2e63f2cde46ac807ac5d9f17a1e6c : FAILURE in 14m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e1f7ba2d062e34c90e89fee885e3d2c483fca55d","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2020-08-17 19:36:09.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3cda7e137ab265599f88c4e3eb25d711cfc9adf7","revisions":{"3cda7e137ab265599f88c4e3eb25d711cfc9adf7":{"kind":"REWORK","_number":1,"created":"2020-08-17 16:44:16.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/64/746564/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/64/746564/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/746564/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/746564/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/746564/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/746564/1"}}},"commit":{"parents":[{"commit":"0660a47c76c0ff6f920b423dd47447830667e88b","subject":"[WIP] test_over_subscription_free_space_reporting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0660a47c76c0ff6f920b423dd47447830667e88b"}]}],"author":{"name":"V Haririan","email":"vhariria@redhat.com","date":"2020-08-17 16:42:28.000000000","tz":-240},"committer":{"name":"V Haririan","email":"vhariria@redhat.com","date":"2020-08-17 16:42:28.000000000","tz":-240},"subject":"[WIP] test_over_subscription_free_space_reporting","message":"[WIP] test_over_subscription_free_space_reporting\n\nChange-Id: If5e21a05202c40464cf09f156c06768a63e6265c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3cda7e137ab265599f88c4e3eb25d711cfc9adf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3cda7e137ab265599f88c4e3eb25d711cfc9adf7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
