)]}'
{"id":"openstack%2Fmanila-tempest-plugin~747157","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ice973bd2ad15bfb432e51da595b53c0480adba14","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bp/share-server-migration","hashtags":[],"change_id":"Ice973bd2ad15bfb432e51da595b53c0480adba14","subject":"Add tests for share server migration","status":"MERGED","created":"2020-08-20 13:08:15.000000000","updated":"2020-09-11 22:19:03.000000000","submitted":"2020-09-11 18:54:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":108,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747157-1599850452922-6df06640","meta_rev_id":"abb68918874bcb36bab18b9643e434012288520c","_number":747157,"virtual_id_number":747157,"owner":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-10 20:38:01.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},{"value":0,"date":"2020-09-10 22:24:56.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"date":"2020-09-10 23:39:14.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-11 18:54:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-11 13:04:03.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-11 14:07:45.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-11 17:42:38.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-22 13:56:10.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-09-09 04:54:45.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-09-10 20:38:01.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-10 22:24:56.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-10 23:39:14.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-09-11 13:04:03.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-09-11 17:42:38.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-09-11 18:54:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"822297af63ba7ceb81b7ae679657399f0830d25b","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-08-20 13:08:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"572164cf9bcdd08838a41045fd70d414fa8f507c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 15:09:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f2e28eb75754d75991365327f8aee86 : SUCCESS in 4m 26s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fe1f6a880c354f6999dd16b3085b688d : SUCCESS in 34m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/44990c6d8985442b92b9fa7a6d2b7695 : TIMED_OUT in 1h 03m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2c4d888e9b3c458b92c193ddd281e221 : SUCCESS in 1h 08m 07s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3d71171b5ae6486bad9baa46f7b058a4 : SUCCESS in 1h 00m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c4cbda4f9e8f493f9d037dbaec531642 : FAILURE in 1h 30m 41s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6f6a859127b6480eaa80ded509bca07d : FAILURE in 58m 08s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8840356b7b8f4757b86c9c734741669d : FAILURE in 43m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/877ff202bddb481097ca0aefb3cc9ee8 : FAILURE in 1h 54m 34s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/fa352ae8df2c4a9db21515cfa539c0c2 : SUCCESS in 30m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"44bf9160828436aa9cb4f6353567b66ed42abad1","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-08-20 17:56:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d792d6be6a4911600a95f8466f1d7a01cd6e40da","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-08-20 19:22:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"87721196c58b02d28329a0163d77926ab97f7843","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 21:48:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce65275fd0e34317991f9b1911e2dcba : SUCCESS in 3m 36s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3658aaa326a2477aa92bca326742344a : SUCCESS in 26m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/083f6123f4fc4dd2af84af37f89d09f2 : SUCCESS in 58m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b2bad7201c5f4defa7abec9f59becaf1 : SUCCESS in 1h 03m 33s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/efe6a490158e4c9d8535e37d345e2ad8 : SUCCESS in 1h 14m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f61f7eaa71ff47cbac1d48c34539fe86 : FAILURE in 1h 32m 56s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/46a98a60855143298828000eed2744d9 : SUCCESS in 1h 02m 23s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2341194d37574ae7acbdef17987317da : FAILURE in 54m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/12cbc5f5d1de48c2aa01bdae2a69b4a0 : FAILURE in 2h 17m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9aae6a9ee72f47c48ae71852a33dec2d : SUCCESS in 26m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a73bca8e3a37c161713a9e798445ffd396610834","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-08-20 23:02:32.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/57/747157/3/check/DellEMC_VNX_Manila/b650ef1/DellEMC_VNX_Manila/169 : SUCCESS in 1h 14m 02s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"1966a29fa2089bbd6fef454d3418a6ddbd15d7ec","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-20 23:58:11.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/57/747157/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/853365c/ : SUCCESS in 2h 09m 27s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/57/747157/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/fd8d95d/ : SUCCESS in 1h 40m 29s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"2af44c3b17c7013d7308a5de19d5b06685bb9d76","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-08-21 21:39:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"add0ce256c8426056b6154dfdf31af98a470dcbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 23:53:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/411cd80bc889457291df9ae6e816f2bb : SUCCESS in 5m 15s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b1f226ba1115435ea35043986fa16d28 : SUCCESS in 29m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/688f115a28384e5d883100dcdee95a6f : FAILURE in 59m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4426552d1ab64d53928d67e9cf4bf1f7 : SUCCESS in 1h 27m 49s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5318401acbb847f3ba7d11d2e9fcfde3 : SUCCESS in 1h 06m 16s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/dd2b8fc5bba747c183f4a90e215824f2 : FAILURE in 1h 30m 33s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4337b06817104f9fa1400ab293dffb4e : SUCCESS in 42m 28s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/aeafa1146026488cb5075222e41183b9 : SUCCESS in 32m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bf2e1afbbd4f45158e447695675c0ca9 : FAILURE in 2h 08m 01s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f5b83cf2184945d79a0fde2cdef26195 : FAILURE in 33m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3af42463b2ebdf2c702c0569f0cf9966f7316f67","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-22 03:27:49.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/57/747157/4/check/DellEMC_Unity_Manila_dhss_true/e557938/DellEMC_Unity_Manila_dhss_true/309 : FAILURE in 31m 09s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/57/747157/4/check/DellEMC_Unity_Manila_dhss_false/03ed066/DellEMC_Unity_Manila_dhss_false/292 : FAILURE in 34m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"3bb4bbb6dd4a37a84495a6b00add98d58a272399","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-08-22 13:56:10.000000000","message":"Patch Set 4: Code-Review-1\n\n(35 comments)\n\nThanks for adding more tests Andre. Please take a look on those comments.","accounts_in_message":[],"_revision_number":4},{"id":"7f4f43c6197b332b91f2fe966947690a55d5a2b6","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-08-27 13:44:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c9cff89d2bdf8233b8f0831efc3a426cb4287897","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-08-27 13:47:29.000000000","message":"Patch Set 4:\n\n(32 comments)\n\nHi Douglas thank you very much for the comments! Most of them are already addressed!","accounts_in_message":[],"_revision_number":4},{"id":"98cac783826c9dde1752719099824051e6c6755c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-27 14:13:29.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/57/747157/5/check/DellEMC_Unity_Manila_dhss_true/4d37af1/DellEMC_Unity_Manila_dhss_true/326 : FAILURE in 26m 22s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/57/747157/5/check/DellEMC_Unity_Manila_dhss_false/c24cfc2/DellEMC_Unity_Manila_dhss_false/308 : FAILURE in 29m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"f5e037fe39c4a6ca135ac2f8f133cc4db21e1bc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 15:59:43.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6030a75932b4483f97069a58045dbe82 : SUCCESS in 3m 39s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0aa65614a6674536ace27d5312366f64 : SUCCESS in 30m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/00b9224715f048b18fc0b7e36deb05cb : FAILURE in 36m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/510927f03f0c4238af4152f0808e47ed : SUCCESS in 1h 48m 01s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/84595a7962b742708d4b405476af9c14 : FAILURE in 1h 07m 53s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/75db2c2922db46dda6f59122698af891 : FAILURE in 1h 26m 45s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c0488f57b576413a83e9c975419d6873 : SUCCESS in 48m 14s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/43a657bfcbfb44cabfa316557f1248cb : SUCCESS in 47m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a054083f495d43dfb3e305882f3747f3 : FAILURE in 1h 57m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/76e50a66322c471a9afebb63fce8054f : SUCCESS in 23m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"53f7e7dea96ee10f017898f23d912e4cb6a05df6","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-08-27 23:15:07.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4ab5ab617ca2707e0251145d2873e18b4d236ed0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-27 23:43:16.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/57/747157/6/check/DellEMC_Unity_Manila_dhss_true/50c6e9c/DellEMC_Unity_Manila_dhss_true/330 : FAILURE in 26m 44s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/57/747157/6/check/DellEMC_Unity_Manila_dhss_false/a8f8b63/DellEMC_Unity_Manila_dhss_false/312 : FAILURE in 28m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"20616ec5ebbc25f781d73a4787605f9742462021","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 01:56:21.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/953b1935a6784397b4242891333f1b5f : SUCCESS in 3m 20s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e6297c0ce46040afb392a174137a26a5 : SUCCESS in 23m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8535f16e802048a0b77c83ec7d0a7049 : FAILURE in 38m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9c412bb671414436a482651d3c39cd66 : FAILURE in 48m 30s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a5b47d8adbc4469e918ff7afa2848b47 : FAILURE in 1h 06m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/800066de5ac14378ba2617154bededf1 : FAILURE in 1h 06m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/cc9892c5749041d081dddf0579ef374e : FAILURE in 52m 04s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5e48c3a3b70c4c5491dc7767ac86a7f5 : SUCCESS in 30m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5092ba29aa9947c9a32ecadbc7b6dd10 : FAILURE in 2h 27m 24s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9f60767e4c664c1f9fafea98614a2e23 : SUCCESS in 28m 00s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"057b4b51a13b0281b2b19954ed4510496930f04e","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-09-01 16:08:51.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"15ded4a9f3fae60e915331b0c2b960ba6f1bce2a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-01 16:37:39.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/57/747157/7/check/DellEMC_Unity_Manila_dhss_true/3462361/DellEMC_Unity_Manila_dhss_true/344 : FAILURE in 28m 40s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/57/747157/7/check/DellEMC_Unity_Manila_dhss_false/19419a8/DellEMC_Unity_Manila_dhss_false/326 : FAILURE in 28m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"2735c92fc88fb2a81c5fceee77d90ff2c72478e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 18:15:55.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/215c85c1e737456589442fa6f61bbaf7 : SUCCESS in 4m 14s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9bba17373dd14e2ba9d1172355d408e0 : SUCCESS in 36m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cd6241c570dc4fcc97031f2d979368ea : SUCCESS in 41m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f9451d4b4c134fcfa460714a4c00440d : SUCCESS in 1h 00m 30s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c57279fbcc5246b79aa9397c9c62866e : SUCCESS in 1h 04m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a9bf17782a76406ea33e4bfff95882ce : FAILURE in 1h 33m 09s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b24b29aa28bd430e96e9d8f4cf55de62 : FAILURE in 59m 05s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3f216ebd1e7c46c39a1fe26b76ad9188 : SUCCESS in 53m 58s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/62a0231504944d0ab7b4bcf5751f7c34 : FAILURE in 2h 00m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a5ec188e1c854ddc9093a5ea28d21819 : SUCCESS in 20m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"70e59541ac4cf31d47cac5a636cb8bc02a7d608d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-01 18:47:00.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/57/747157/7/check/DellEMC_Unity_Manila_dhss_true/a28edb5/DellEMC_Unity_Manila_dhss_true/346 : FAILURE in 30m 56s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/57/747157/7/check/DellEMC_Unity_Manila_dhss_false/2decfa1/DellEMC_Unity_Manila_dhss_false/328 : FAILURE in 30m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"5b30a193a1f2cf40ccea9d410c2d9051e3d5f0a2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-09 04:54:45.000000000","message":"Patch Set 7: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"fe549ab312962a9b1d8fd900e922e48fe64f5367","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-09 04:56:35.000000000","message":"Patch Set 7:\n\nAndre: Thanks for this change, I\u0027m Still reading through the tests - pointed out some concerns..","accounts_in_message":[],"_revision_number":7},{"id":"c92370da808fbee9e4a3bf43d7fc6c39128157fa","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-09-10 16:57:05.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"91eb6323392b4edbfb648d38475669ab52136d86","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-09-10 16:58:12.000000000","message":"Patch Set 7:\n\n(4 comments)\n\nHi Goutham, thank you for the review! The comments are addressed in PS 8...","accounts_in_message":[],"_revision_number":7},{"id":"199fbd5f73ea0eafa7c952346b6698e8a676acd6","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-09-10 17:46:48.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"329e9d50d39c5b7ecdbf940254e17ad4bf3688c5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-10 19:38:17.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/57/747157/9/check/DellEMC_Unity_Manila_dhss_true/0bd1215/DellEMC_Unity_Manila_dhss_true/394 : FAILURE in 28m 31s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/57/747157/9/check/DellEMC_Unity_Manila_dhss_false/b86591f/DellEMC_Unity_Manila_dhss_false/377 : FAILURE in 30m 13s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"b966f487c6eeefd986bb1bb66de9bbb2a01d95d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 20:28:07.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7412da3890e248819ef0a665104510fe : SUCCESS in 3m 53s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/327723b822a84f03a3bca459e57b5b09 : SUCCESS in 36m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/07066ac3a092431fb9ab65be5927d945 : SUCCESS in 36m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1816631650b9478c8f7cd979a3541f51 : SUCCESS in 1h 02m 23s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/851a5ae435fe4cc1a60c83fb7b2d5c6f : FAILURE in 1h 37m 33s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/620ceee7f8ad4142a317922f00ad5056 : SUCCESS in 59m 54s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e35401359ae3470cae6064fa4089d5d2 : FAILURE in 1h 50m 17s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b7117caede444ff08ad87a48b3a32c4c : SUCCESS in 41m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9fc52e8340bd4ef295161178bc9c5ff3 : FAILURE in 2h 36m 53s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9a29f57eac664e918b30c4245ea2c3dc : FAILURE in 28m 48s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"da040ca736f0a87c64a2afe48a17bdf64784a718","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-09-10 20:38:01.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/57/747157/9/check/DellEMC_VNX_Manila/f26e42a/DellEMC_VNX_Manila/214 : FAILURE in 9m 35s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":9},{"id":"737bf9a601aa190bf313751f9a8457fba7b6f43f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-10 22:24:56.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"625993b82544522f06a25bc267fe97f820d25bfe","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-10 23:39:14.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/57/747157/9/check/DellEMC_Unity_Manila_dhss_true/402a095/DellEMC_Unity_Manila_dhss_true/399 : FAILURE in 29m 25s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/57/747157/9/check/DellEMC_Unity_Manila_dhss_false/bab0687/DellEMC_Unity_Manila_dhss_false/382 : FAILURE in 29m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"e20fe15ed52094135238389b64ccdbaf6a81a984","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-09-11 13:04:03.000000000","message":"Patch Set 9: Code-Review+2\n\nLGTM, nice work!","accounts_in_message":[],"_revision_number":9},{"id":"91ea7679593819a601143134e154f5cb5e6444e9","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-09-11 14:07:45.000000000","message":"Patch Set 9: Code-Review+2\n\nCode looks good to me, errors on the zfsonlinux and generic tempest gates are caused for other tests, this are passing ok. Thanks Andre!","accounts_in_message":[],"_revision_number":9},{"id":"610435f25f09180d2a8ce7bd0305c9e78bb963f9","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-09-11 17:42:38.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"28c6c2198727bcba10cf15d71e418cd65450e01a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 17:43:16.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"58edc4965af6ecdb524eb0f42190d405ef20ea86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 18:54:12.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc9eaf210b2240bb8de09cef68ed4321 : SUCCESS in 3m 07s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cc266c7d628b4123a8888fd30ee4a3d5 : SUCCESS in 27m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/768a81506107459b970d5fe3f2c008f1 : SUCCESS in 34m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/42e843ccd7814ab69a874ba831dfd4a8 : SUCCESS in 1h 09m 14s","accounts_in_message":[],"_revision_number":9},{"id":"3f3180380ce20a04f524c4367402ddef9fae0c14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 18:54:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"d9e13137e03bd5d1e9da9f6772aa089a61f5d723","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-11 21:41:21.000000000","message":"Patch Set 9:\n\n(32 comments)\n\nHi, sorry these are late breaking comments - please address them in a separate patch, by RC. We don\u0027t need a bug to track this","accounts_in_message":[],"_revision_number":9},{"id":"92051c5e9aef523360950ae6a915c405e3a38949","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-09-11 22:19:03.000000000","message":"Patch Set 9:\n\nHi Goutham, thank you very much for the review. ASAP I will provide the fix for your comments.","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"0d523bb8d4cd62186616103b59bba7d933d2cee4","revisions":{"47ae791f78dd17e4cddb054db58695248f98224a":{"kind":"REWORK","_number":1,"created":"2020-08-20 13:08:15.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/57/747157/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/57/747157/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/1"}}},"commit":{"parents":[{"commit":"1369e54a2d8c4874f2b90f98c77e115349efb22a","subject":"Fix skip condition for \"test_default_share_group_type_applied\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1369e54a2d8c4874f2b90f98c77e115349efb22a"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-08-20 12:48:49.000000000","tz":0},"committer":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-08-20 13:08:14.000000000","tz":0},"subject":"Add tests for share server migration","message":"Add tests for share server migration\n\nThis patch adds new positive and negative tests for\nshare server migration.\n\nPartially-implements: bp share-server-migration\nCo-Authored-By: Daniel Tapia \u003cdanielarthurt@gmail.com\u003e\n\nChange-Id: Ice973bd2ad15bfb432e51da595b53c0480adba14\nDepends-On: Ic0751027d2c3f1ef7ab0f7836baff3070a230cfd\nDepends-On: I46a0cee0a4625d950f02fa7a5bf612de926451b5\nDepends-On: I813801fd64d4d605c0a4fe06c7a130ad00865b6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/47ae791f78dd17e4cddb054db58695248f98224a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/47ae791f78dd17e4cddb054db58695248f98224a"}]},"branch":"refs/heads/master"},"60991ee82df5e9576baefaa9e94c860e6667e406":{"kind":"REWORK","_number":2,"created":"2020-08-20 17:56:19.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/57/747157/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/57/747157/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/2"}}},"commit":{"parents":[{"commit":"1369e54a2d8c4874f2b90f98c77e115349efb22a","subject":"Fix skip condition for \"test_default_share_group_type_applied\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1369e54a2d8c4874f2b90f98c77e115349efb22a"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-08-20 12:48:49.000000000","tz":0},"committer":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-08-20 17:55:54.000000000","tz":0},"subject":"Add tests for share server migration","message":"Add tests for share server migration\n\nThis patch adds new positive and negative tests for\nshare server migration.\n\nPartially-implements: bp share-server-migration\nCo-Authored-By: Daniel Tapia \u003cdanielarthurt@gmail.com\u003e\n\nChange-Id: Ice973bd2ad15bfb432e51da595b53c0480adba14\nDepends-On: Ic0751027d2c3f1ef7ab0f7836baff3070a230cfd\nDepends-On: I46a0cee0a4625d950f02fa7a5bf612de926451b5\nDepends-On: I813801fd64d4d605c0a4fe06c7a130ad00865b6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/60991ee82df5e9576baefaa9e94c860e6667e406"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/60991ee82df5e9576baefaa9e94c860e6667e406"}]},"branch":"refs/heads/master"},"5600858dca38a4f5e28cd22aaa8a11f0eaf85f39":{"kind":"REWORK","_number":3,"created":"2020-08-20 19:22:53.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/57/747157/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/57/747157/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/3"}}},"commit":{"parents":[{"commit":"1369e54a2d8c4874f2b90f98c77e115349efb22a","subject":"Fix skip condition for \"test_default_share_group_type_applied\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1369e54a2d8c4874f2b90f98c77e115349efb22a"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-08-20 12:48:49.000000000","tz":0},"committer":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-08-20 19:22:14.000000000","tz":0},"subject":"Add tests for share server migration","message":"Add tests for share server migration\n\nThis patch adds new positive and negative tests for\nshare server migration.\n\nPartially-implements: bp share-server-migration\nCo-Authored-By: Daniel Tapia \u003cdanielarthurt@gmail.com\u003e\n\nChange-Id: Ice973bd2ad15bfb432e51da595b53c0480adba14\nDepends-On: Ic0751027d2c3f1ef7ab0f7836baff3070a230cfd\nDepends-On: I46a0cee0a4625d950f02fa7a5bf612de926451b5\nDepends-On: I813801fd64d4d605c0a4fe06c7a130ad00865b6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5600858dca38a4f5e28cd22aaa8a11f0eaf85f39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5600858dca38a4f5e28cd22aaa8a11f0eaf85f39"}]},"branch":"refs/heads/master"},"976da5b638a377224f359c896670402b34ef41d3":{"kind":"REWORK","_number":4,"created":"2020-08-21 21:39:20.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/57/747157/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/57/747157/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/4"}}},"commit":{"parents":[{"commit":"1369e54a2d8c4874f2b90f98c77e115349efb22a","subject":"Fix skip condition for \"test_default_share_group_type_applied\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1369e54a2d8c4874f2b90f98c77e115349efb22a"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-08-20 12:48:49.000000000","tz":0},"committer":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-08-21 21:38:40.000000000","tz":0},"subject":"Add tests for share server migration","message":"Add tests for share server migration\n\nThis patch adds new positive and negative tests for\nshare server migration.\n\nPartially-implements: bp share-server-migration\nCo-Authored-By: Daniel Tapia \u003cdanielarthurt@gmail.com\u003e\n\nChange-Id: Ice973bd2ad15bfb432e51da595b53c0480adba14\nDepends-On: Ic0751027d2c3f1ef7ab0f7836baff3070a230cfd\nDepends-On: I46a0cee0a4625d950f02fa7a5bf612de926451b5\nDepends-On: I813801fd64d4d605c0a4fe06c7a130ad00865b6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/976da5b638a377224f359c896670402b34ef41d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/976da5b638a377224f359c896670402b34ef41d3"}]},"branch":"refs/heads/master"},"248a13a869a0c4ff7f264aa554ba38a9730eba8b":{"kind":"REWORK","_number":5,"created":"2020-08-27 13:44:00.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/57/747157/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/57/747157/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/5"}}},"commit":{"parents":[{"commit":"1ef4c6af99498dacfe8017cbec19df07b80f0b44","subject":"Merge \"Ensure share server is created when driver handles share servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1ef4c6af99498dacfe8017cbec19df07b80f0b44"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-08-20 12:48:49.000000000","tz":0},"committer":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-08-27 13:43:27.000000000","tz":0},"subject":"Add tests for share server migration","message":"Add tests for share server migration\n\nThis patch adds new positive and negative tests for\nshare server migration.\n\nPartially-implements: bp share-server-migration\nCo-Authored-By: Daniel Tapia \u003cdanielarthurt@gmail.com\u003e\n\nChange-Id: Ice973bd2ad15bfb432e51da595b53c0480adba14\nDepends-On: Ic0751027d2c3f1ef7ab0f7836baff3070a230cfd\nDepends-On: I46a0cee0a4625d950f02fa7a5bf612de926451b5\nDepends-On: I813801fd64d4d605c0a4fe06c7a130ad00865b6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/248a13a869a0c4ff7f264aa554ba38a9730eba8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/248a13a869a0c4ff7f264aa554ba38a9730eba8b"}]},"branch":"refs/heads/master"},"8f52a3b59abfc06a6d10bdaa80005e39b58d320f":{"kind":"REWORK","_number":6,"created":"2020-08-27 23:15:07.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/57/747157/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/57/747157/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/6"}}},"commit":{"parents":[{"commit":"1ef4c6af99498dacfe8017cbec19df07b80f0b44","subject":"Merge \"Ensure share server is created when driver handles share servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1ef4c6af99498dacfe8017cbec19df07b80f0b44"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-08-20 12:48:49.000000000","tz":0},"committer":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-08-27 23:12:07.000000000","tz":0},"subject":"Add tests for share server migration","message":"Add tests for share server migration\n\nThis patch adds new positive and negative tests for\nshare server migration.\n\nPartially-implements: bp share-server-migration\nCo-Authored-By: Daniel Tapia \u003cdanielarthurt@gmail.com\u003e\n\nChange-Id: Ice973bd2ad15bfb432e51da595b53c0480adba14\nDepends-On: Ic0751027d2c3f1ef7ab0f7836baff3070a230cfd\nDepends-On: I46a0cee0a4625d950f02fa7a5bf612de926451b5\nDepends-On: I813801fd64d4d605c0a4fe06c7a130ad00865b6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8f52a3b59abfc06a6d10bdaa80005e39b58d320f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8f52a3b59abfc06a6d10bdaa80005e39b58d320f"}]},"branch":"refs/heads/master"},"6bc778a97ba8b693c47fe55640c72dc4031a4d65":{"kind":"REWORK","_number":7,"created":"2020-09-01 16:08:51.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/57/747157/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/57/747157/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/7"}}},"commit":{"parents":[{"commit":"f904fd620453eb367430aa647f9def80caf64653","subject":"Merge \"Add flake8-import-order extension to tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f904fd620453eb367430aa647f9def80caf64653"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-08-20 12:48:49.000000000","tz":0},"committer":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-09-01 16:08:35.000000000","tz":0},"subject":"Add tests for share server migration","message":"Add tests for share server migration\n\nThis patch adds new positive and negative tests for\nshare server migration.\n\nPartially-implements: bp share-server-migration\nCo-Authored-By: Daniel Tapia \u003cdanielarthurt@gmail.com\u003e\n\nChange-Id: Ice973bd2ad15bfb432e51da595b53c0480adba14\nDepends-On: Ic0751027d2c3f1ef7ab0f7836baff3070a230cfd\nDepends-On: I46a0cee0a4625d950f02fa7a5bf612de926451b5\nDepends-On: I813801fd64d4d605c0a4fe06c7a130ad00865b6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6bc778a97ba8b693c47fe55640c72dc4031a4d65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6bc778a97ba8b693c47fe55640c72dc4031a4d65"}]},"branch":"refs/heads/master"},"97d698c28f37949b42d6844d0e4aceda01e8244c":{"kind":"REWORK","_number":8,"created":"2020-09-10 16:57:05.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/57/747157/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/57/747157/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/8"}}},"commit":{"parents":[{"commit":"2e0e0814142b970dffe1f7675026a325f3cd73ad","subject":"Merge \"Remove experimental flag from share replication feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e0e0814142b970dffe1f7675026a325f3cd73ad"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-08-20 12:48:49.000000000","tz":0},"committer":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-09-10 16:57:03.000000000","tz":0},"subject":"Add tests for share server migration","message":"Add tests for share server migration\n\nThis patch adds new positive and negative tests for\nshare server migration.\n\nPartially-implements: bp share-server-migration\nCo-Authored-By: Daniel Tapia \u003cdanielarthurt@gmail.com\u003e\n\nChange-Id: Ice973bd2ad15bfb432e51da595b53c0480adba14\nDepends-On: Ic0751027d2c3f1ef7ab0f7836baff3070a230cfd\nDepends-On: I46a0cee0a4625d950f02fa7a5bf612de926451b5\nDepends-On: I813801fd64d4d605c0a4fe06c7a130ad00865b6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/97d698c28f37949b42d6844d0e4aceda01e8244c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/97d698c28f37949b42d6844d0e4aceda01e8244c"}]},"branch":"refs/heads/master"},"0d523bb8d4cd62186616103b59bba7d933d2cee4":{"kind":"REWORK","_number":9,"created":"2020-09-10 17:46:48.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/57/747157/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/57/747157/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/747157/9"}}},"commit":{"parents":[{"commit":"2e0e0814142b970dffe1f7675026a325f3cd73ad","subject":"Merge \"Remove experimental flag from share replication feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e0e0814142b970dffe1f7675026a325f3cd73ad"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-08-20 12:48:49.000000000","tz":0},"committer":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-09-10 17:46:19.000000000","tz":0},"subject":"Add tests for share server migration","message":"Add tests for share server migration\n\nThis patch adds new positive and negative tests for\nshare server migration.\n\nPartially-implements: bp share-server-migration\nCo-Authored-By: Daniel Tapia \u003cdanielarthurt@gmail.com\u003e\n\nChange-Id: Ice973bd2ad15bfb432e51da595b53c0480adba14\nDepends-On: Ic0751027d2c3f1ef7ab0f7836baff3070a230cfd\nDepends-On: I46a0cee0a4625d950f02fa7a5bf612de926451b5\nDepends-On: I813801fd64d4d605c0a4fe06c7a130ad00865b6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0d523bb8d4cd62186616103b59bba7d933d2cee4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0d523bb8d4cd62186616103b59bba7d933d2cee4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
