)]}'
{"id":"openstack%2Fmanila-tempest-plugin~747276","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I221b4a48b2f66327c83a78514b471cae51e20ea8","project":"openstack/manila-tempest-plugin","branch":"master","topic":"tracker.ceph.com/issues/47236","hashtags":[],"change_id":"I221b4a48b2f66327c83a78514b471cae51e20ea8","subject":"Revert CephFS test image to Bionic Beaver","status":"MERGED","created":"2020-08-20 18:51:00.000000000","updated":"2020-09-02 14:14:39.000000000","submitted":"2020-09-02 14:14:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747276-1599056079796-941d6ed0","meta_rev_id":"8a2f3b65d6f0616d150b8863237ae4f70612873b","_number":747276,"virtual_id_number":747276,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-02 14:14:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-01 19:48:04.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-01 17:08:55.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":2,"date":"2020-09-02 12:59:29.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":1,"date":"2020-09-02 12:59:29.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-31 23:55:19.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-01 17:08:55.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-09-01 19:48:04.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-09-02 12:59:29.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-09-02 14:14:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"662084b2ebb69f6cbe5572606aaaf230467bf97e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-20 18:51:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8258480aea790a651286d1aab1e584c5c1a01977","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 20:05:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0abceedd39ff4f2f8c5f8614a525f0a8 : FAILURE in 1h 10m 34s","accounts_in_message":[],"_revision_number":1},{"id":"6f61af7e1a2648f5a867202c435b62f639e313e5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-31 22:34:56.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"fc5157d27a845db0f4335cbfbe7637c7ff94e44b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-31 23:29:13.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/76/747276/2/check/DellEMC_Unity_Manila_dhss_true/ba6d550/DellEMC_Unity_Manila_dhss_true/335 : FAILURE in 53m 48s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/76/747276/2/check/DellEMC_Unity_Manila_dhss_false/f905015/DellEMC_Unity_Manila_dhss_false/317 : SUCCESS in 54m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"0757038d0fbe3a46d95d5f274dd1df16d81aa5c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 23:43:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b222319372404ecfb24a2ef1b53e71d6 : SUCCESS in 1h 02m 07s","accounts_in_message":[],"_revision_number":2},{"id":"f2afe4e558199048e517dfe316455cf7526bb6a3","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-08-31 23:55:19.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/76/747276/2/check/DellEMC_VNX_Manila/393c086/DellEMC_VNX_Manila/190 : FAILURE in 11m 43s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"f57b41b1743e625c5b60009b02f3a981ac739d52","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-01 01:26:36.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/76/747276/2/check/DellEMC_Unity_Manila_dhss_true/e04b03a/DellEMC_Unity_Manila_dhss_true/337 : FAILURE in 49m 36s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/76/747276/2/check/DellEMC_Unity_Manila_dhss_false/486ebcc/DellEMC_Unity_Manila_dhss_false/319 : SUCCESS in 48m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"3ff3baf7430e38077c81b7fa1ed6ca2a37887bca","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-01 16:36:45.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"7347c20b97e5079fc9d99722b20d0c0f9f8b2aec","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-01 16:36:58.000000000","message":"Topic changed from cephfs-debug to tracker.ceph.com/issues/47236","accounts_in_message":[],"_revision_number":3},{"id":"af3ff6cb2afe3cc7b7cdfa9fbbaeb45e60032d77","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-09-01 17:08:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"eeffa484c3a6666b669f3246f1dc1c71a2bbe77d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-01 17:42:06.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/76/747276/3/check/DellEMC_Unity_Manila_dhss_true/bffeed5/DellEMC_Unity_Manila_dhss_true/345 : FAILURE in 1h 04m 21s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/76/747276/3/check/DellEMC_Unity_Manila_dhss_false/f88c724/DellEMC_Unity_Manila_dhss_false/327 : SUCCESS in 56m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"4adecb55e6947a5365f8d83945add06ce4d8a24e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 18:38:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e7c70b0d0954d748d59d33e6195525c : SUCCESS in 3m 55s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e4d3f2ee23e34ed3a6d7f54a55d0324b : SUCCESS in 29m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/60c8de0bb3cc46a78b9facbe48837b33 : SUCCESS in 33m 23s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/725d8fd0bb924cf689a649880a2f930c : SUCCESS in 1h 18m 45s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/90c1422269b1497fa248c4884d97c076 : SUCCESS in 1h 17m 24s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/18681dca2c7343da9ca9ad67c77affcb : SUCCESS in 56m 36s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a122d92f0955426b97b13e18a83edb9e : FAILURE in 58m 48s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4a159a92ad414f37984330d225144022 : SUCCESS in 46m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e6b73ffee752459e88c56f74f46c6cee : FAILURE in 1h 55m 57s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b3404622c7464968ab6bebb718f37e77 : SUCCESS in 18m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cb94d796a3f85794091688386f2d456229b09d6c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-01 19:48:04.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/76/747276/3/check/DellEMC_Unity_Manila_dhss_true/98f9ebd/DellEMC_Unity_Manila_dhss_true/347 : FAILURE in 1h 00m 59s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/76/747276/3/check/DellEMC_Unity_Manila_dhss_false/bc1ee36/DellEMC_Unity_Manila_dhss_false/329 : SUCCESS in 48m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"d6362fa6cfdd58a4e7ddf14afafc918caf795deb","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-09-02 12:59:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThanks Goutham. I gave this some thought and seems to be the best course of action. Blacklisting seems better at first so we could continue testing with Focal Fossa but, TBH, not sure when this bug will be fixed and it\u0027s also a bad idea to reduce the test coverage for manila for a bug that is in other component that doesn\u0027t depend on us. Hence, my +2.","accounts_in_message":[],"_revision_number":3},{"id":"5ca27570149fc0fb413549affb033fb519645a4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 12:59:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7ac4d7c231274654d13e2ca14cc3a35a7caa1f8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 14:14:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efed52f6cc8b4a83b841a4443110dab3 : SUCCESS in 3m 45s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/70435417894e44fcb66fbe2831fb3d04 : SUCCESS in 37m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/58df187ffafa4dc39f2e056a0a232d19 : SUCCESS in 33m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/33d2ffecf74c48f691d5b5de9035b75a : SUCCESS in 1h 11m 34s","accounts_in_message":[],"_revision_number":3},{"id":"2f740459759b9020faa0c88642f9997f46a474b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 14:14:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"75f0d60731d91e6d1c6512d792c9a6cd3862187f","revisions":{"e055a2e13a0e88f7d48d1501bebf8a2ec545a8e8":{"kind":"REWORK","_number":1,"created":"2020-08-20 18:51:00.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/76/747276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/76/747276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/747276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/747276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/747276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/747276/1"}}},"commit":{"parents":[{"commit":"1369e54a2d8c4874f2b90f98c77e115349efb22a","subject":"Fix skip condition for \"test_default_share_group_type_applied\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1369e54a2d8c4874f2b90f98c77e115349efb22a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-20 18:50:48.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-20 18:50:48.000000000","tz":-420},"subject":"DNM: Debug cephfs scenario tests","message":"DNM: Debug cephfs scenario tests\n\nChange-Id: I221b4a48b2f66327c83a78514b471cae51e20ea8\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e055a2e13a0e88f7d48d1501bebf8a2ec545a8e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e055a2e13a0e88f7d48d1501bebf8a2ec545a8e8"}]},"branch":"refs/heads/master"},"bbbbcf710dadf753dec9142b589ad93f64ad7015":{"kind":"REWORK","_number":2,"created":"2020-08-31 22:34:56.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/76/747276/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/76/747276/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/747276/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/747276/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/747276/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/747276/2"}}},"commit":{"parents":[{"commit":"1369e54a2d8c4874f2b90f98c77e115349efb22a","subject":"Fix skip condition for \"test_default_share_group_type_applied\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1369e54a2d8c4874f2b90f98c77e115349efb22a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-20 18:50:48.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-31 22:34:54.000000000","tz":0},"subject":"DNM: Debug cephfs scenario tests","message":"DNM: Debug cephfs scenario tests\n\nChange-Id: I221b4a48b2f66327c83a78514b471cae51e20ea8\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bbbbcf710dadf753dec9142b589ad93f64ad7015"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bbbbcf710dadf753dec9142b589ad93f64ad7015"}]},"branch":"refs/heads/master"},"75f0d60731d91e6d1c6512d792c9a6cd3862187f":{"kind":"REWORK","_number":3,"created":"2020-09-01 16:36:45.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/76/747276/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/76/747276/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/747276/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/747276/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/747276/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/747276/3"}}},"commit":{"parents":[{"commit":"1369e54a2d8c4874f2b90f98c77e115349efb22a","subject":"Fix skip condition for \"test_default_share_group_type_applied\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1369e54a2d8c4874f2b90f98c77e115349efb22a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-20 18:50:48.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-09-01 16:36:41.000000000","tz":0},"subject":"Revert CephFS test image to Bionic Beaver","message":"Revert CephFS test image to Bionic Beaver\n\nThere\u0027s an issue with CephFS in Focal Fossa (Ubuntu\n20.04 LTS) [1]. While this is being investigated,\nlets revert to using Ubuntu Bionic Beaver (18.04 LTS)\nfor the test VMs.\n\n[1] https://tracker.ceph.com/issues/47236\nChange-Id: I221b4a48b2f66327c83a78514b471cae51e20ea8\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/75f0d60731d91e6d1c6512d792c9a6cd3862187f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/75f0d60731d91e6d1c6512d792c9a6cd3862187f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
