)]}'
{"id":"openstack%2Fmanila-tempest-plugin~747339","triplet_id":"openstack%2Fmanila-tempest-plugin~master~If1d581730c42b060e74600ef2a05298921b1a3c6","project":"openstack/manila-tempest-plugin","branch":"master","topic":"fix-ssh-with-storage-network","hashtags":[],"change_id":"If1d581730c42b060e74600ef2a05298921b1a3c6","subject":"SSH over IPv4 when using a storage network","status":"MERGED","created":"2020-08-21 06:47:54.000000000","updated":"2020-09-07 14:26:07.000000000","submitted":"2020-09-07 14:26:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747339-1599488767686-d5824c7c","meta_rev_id":"f4287b440c51baa7447e45b0ae6e59cf81cbc2cb","_number":747339,"virtual_id_number":747339,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-01 01:26:58.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-31 22:40:45.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-01 02:19:31.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-07 14:26:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"recommended":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2020-09-01 11:02:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":2,"date":"2020-09-07 13:23:03.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-01 05:37:23.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":1,"date":"2020-09-07 13:23:03.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-21 11:58:41.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-01 01:26:58.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-01 02:19:31.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-09-01 05:37:23.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2020-09-01 11:02:52.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2020-09-07 13:23:03.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-09-07 14:26:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5f4e4aad535b3d774b36d21e5ed12cf8303f0617","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-21 06:47:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"376bd381d444d224069f8e0e1bd4d23f6cf694c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 09:55:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d447cde780184e06837bcdc0ab0c9616 : SUCCESS in 5m 33s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e9df09f672fc43b1b44afd07b1a8c2f5 : SUCCESS in 28m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7e4ed7507dfd479f86b70947f905a39a : SUCCESS in 40m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7d6dfbc2e29b42169fcd4b996c9b3cb3 : SUCCESS in 59m 43s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7c4131c840ff4ea5b97b98d7daa96bdf : FAILURE in 1h 18m 48s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/636676a243e6429285e95d109182e7e9 : FAILURE in 1h 30m 40s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f7bfd3d358e746b48302cfb3e39f22c5 : FAILURE in 47m 02s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/09ce9a5a057f42dea8fc05bf4e69c6be : SUCCESS in 46m 44s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a36d07f97e9649679eed21498c287832 : FAILURE in 2h 53m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/bf8aaed567d845878834d9b2cd2c766f : SUCCESS in 25m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d926ad973a2da6be2f56c28fe04de1f3f87c7fb3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-21 11:14:01.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/39/747339/1/check/DellEMC_Unity_Manila_dhss_true/203d110/DellEMC_Unity_Manila_dhss_true/297 : FAILURE in 1h 18m 37s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/39/747339/1/check/DellEMC_Unity_Manila_dhss_false/40f2f09/DellEMC_Unity_Manila_dhss_false/280 : SUCCESS in 1h 00m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"d05b85451a2e5a38d4c48f64fdae7fad1612d67b","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-08-21 11:15:52.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/39/747339/1/check/DellEMC_VNX_Manila/7f74ff2/DellEMC_VNX_Manila/174 : SUCCESS in 1h 20m 26s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"0d7910615e677e1a63773b33aec291b729661c71","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-21 11:58:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/39/747339/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/bae280c/ : SUCCESS in 2h 03m 11s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/39/747339/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/12a66b1/ : SUCCESS in 1h 23m 20s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"512cd309db89e210b12bb56996744ac1df8bebdf","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-23 15:10:49.000000000","message":"Patch Set 1:\n\nThis fix does work.\nBut I have some problems with it:\n1. I think it\u0027s confusing because enabling \"run_ipv6_tests\" means for me the private network is ipv6.\n   As I understood \"run_ipv6_tests\" refers to the private share and not to the storage network.\n2. In case we have an environment which supports both versions for private and storage networks and \n   for some reason I want to test ipv4 private network with ipv6 storage network, with this I can not \n   do it. Adding a condition option for the storage network version gives you the flexibility to test \n   any versions combination.","accounts_in_message":[],"_revision_number":1},{"id":"33b6fe142f1c8dae6948caf1aa438b0a065e943d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-24 22:20:39.000000000","message":"Patch Set 1:\n\n\u003e This fix does work.\n\u003e But I have some problems with it:\n\u003e 1. I think it\u0027s confusing because enabling \"run_ipv6_tests\" means for me the private network is ipv6.\n\u003e   As I understood \"run_ipv6_tests\" refers to the private share and not to the storage network.\n\u003e 2. In case we have an environment which supports both versions for private and storage networks and \n\u003e   for some reason I want to test ipv4 private network with ipv6 storage network, with this I can not \n\u003e   do it. Adding a condition option for the storage network version gives you the flexibility to test \n\u003e   any versions combination.\n\nThanks Liron, \n\nI can improve the help text/description for the \"run_ipv6_tests\" configuration option. The intent of these IP version specific NFS tests is two fold:\n\n1) Test whether IPv6 export locations work (in turn, some backends are tested entirely over IPv6 management path)\n2) Test whether IPv6 access control works\n\nHistorically, without the \"storage_network\" configuration option, we wanted to keep things simple;\nso for example, on a devstack machine, we\u0027re able to setup dual stack IPv4 and IPv6 setup where\nour private network will suffice as the network with with we will:\n\n- SSH to the machine\n- Perform access control, because that\u0027s also the network which we will use to\n- Mount shares over/with\n\nHowever, when using a \"storage_network\", the network you would SSH with isn\u0027t the same as the\nnetwork with which you would perform access control or mount shares with. That\u0027s why with this fix, \nI\u0027m ignoring the whole IP version setup on the network with which we\u0027d perform SSH (tenant private network).\n\nThe SSH network has never been the object under test for manila tempest tests - so I believe this is a good workaround - with this fix, \nwe\u0027d even be able to test dual-stack networking when using the storage network - i.e., for example, if a share has two export paths\non the storage nfs network, one IPv4 and one IPv6 - the tests will automatically use the IPv4 address in case of IPv4 tests and\nIPv6 address in case of IPv6 tests.","accounts_in_message":[],"_revision_number":1},{"id":"55f21f985d32713da819d1989619216a950cc2f2","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-30 05:03:22.000000000","message":"Patch Set 1:\n\nSorry for the late respond.\nI\u0027m fine with this fix.\nLet\u0027s just improve the description like you suggested.\nOnce this patch is merged I will update https://review.opendev.org/#/c/744851/","accounts_in_message":[],"_revision_number":1},{"id":"cead0af89c3249b794a79b3c2cd97736b142df4c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-31 17:51:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"756df0afa473c7d544c2c369479c034bdcee7543","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-31 18:59:08.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/39/747339/2/check/DellEMC_Unity_Manila_dhss_true/b7533d2/DellEMC_Unity_Manila_dhss_true/334 : FAILURE in 1h 07m 38s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/39/747339/2/check/DellEMC_Unity_Manila_dhss_false/583e18c/DellEMC_Unity_Manila_dhss_false/316 : SUCCESS in 50m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"ea793fcc4baad7a7bfe3801ee7e3172034d54fe1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 20:07:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06e366f8b0e64a829f1277ca8dabae41 : SUCCESS in 3m 56s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d20776b224da4760a07306ee3e17479a : SUCCESS in 25m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4a86d57809d64331880060b5e8797d18 : SUCCESS in 31m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a6766c55a63a4bf1a656c98abc9b451b : FAILURE in 1h 21m 55s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a69aa2f346cb4915a99e89104da1af26 : FAILURE in 1h 18m 36s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/2119c1c0809143239afa79bfce2aabe0 : FAILURE in 1h 28m 04s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2404dc5ca45e4618926ad21b66afe50f : FAILURE in 1h 13m 31s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/64877fe3e7404949a5cea995c48044ad : SUCCESS in 1h 00m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e7346ee614a94e58be1f4aa0f805ae4a : FAILURE in 2h 14m 28s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/aa6749cbff1049e0bc4c521ca1effa78 : SUCCESS in 20m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dadfed56aeb2827f5d4696bdcd002ea8c67af7c0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-31 22:40:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7415f38ea174a75c8f0da34238a60c4e7b3be860","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 01:18:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab423df8acf54f26a494e11e830c9b24 : SUCCESS in 4m 10s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9c1154fdd8a84e7ea2b35cff3fd6232a : SUCCESS in 25m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/498e87eee3fc43159e692d9ec5994954 : SUCCESS in 32m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bacc7bd349dc428aaed2047af8e8b0b8 : SUCCESS in 1h 13m 07s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4b876387d4de4e64965c5d0749478d05 : SUCCESS in 42m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/07133a25d18347d7b7a505501845cf03 : FAILURE in 1h 37m 12s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/873c00e1dfb44a4f9b820ed41896d3d7 : SUCCESS in 1h 02m 32s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/64bc012fb2814dd5b182973c477815d9 : SUCCESS in 30m 38s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2c92532a60d940f0b473ac5d33d47ef5 : FAILURE in 2h 32m 00s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1726d5672f1c4d62b9b93a6cbb719741 : SUCCESS in 25m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5b1728bb8d35f04a8a49e8669c59632d870529eb","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-09-01 01:26:58.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/39/747339/2/check/DellEMC_VNX_Manila/3fb8e1c/DellEMC_VNX_Manila/191 : FAILURE in 8m 08s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"080be795146a41a7e2251b48a92a5b15bde5469f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-01 02:19:31.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/39/747339/2/check/DellEMC_Unity_Manila_dhss_true/7cf0583/DellEMC_Unity_Manila_dhss_true/338 : FAILURE in 52m 49s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/39/747339/2/check/DellEMC_Unity_Manila_dhss_false/057cd30/DellEMC_Unity_Manila_dhss_false/320 : SUCCESS in 52m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"cd9805ff834c55664b4ca0dee0cebbcd9ed7914a","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-09-01 05:37:23.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks","accounts_in_message":[],"_revision_number":2},{"id":"1921c561a58fbdd2b32d1926e52f7c2ce167bbd4","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2020-09-01 11:02:52.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM, thanks Goutham.","accounts_in_message":[],"_revision_number":2},{"id":"82db574a2772177b9916fec38d47464a47d27c55","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-09-07 13:23:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ab9174b24ef27a0c47aa4acd6d44bd43d950332e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 13:23:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2b70b04dd6d6b8f9e8cc2ca1ffb01253455264b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 14:26:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ebe4e13105c41c7b3e7c8accd3565c5 : SUCCESS in 3m 59s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1f0ff3a88460407bb8bb84902b132abe : SUCCESS in 28m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f2efcbd015064449870e619cfeea0eb6 : SUCCESS in 28m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0cb84618d1e54102bbb1cd033532a59f : SUCCESS in 1h 01m 16s","accounts_in_message":[],"_revision_number":2},{"id":"d049752991cb5b30dd0cf15addc6727e79dfd283","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 14:26:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2b308742004a6b345cebc77bb8319955310cfb35","revisions":{"d11339d90ae711a3b1f896c13cac23fc285efd79":{"kind":"REWORK","_number":1,"created":"2020-08-21 06:47:54.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/39/747339/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/39/747339/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/747339/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/747339/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/747339/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/747339/1"}}},"commit":{"parents":[{"commit":"1369e54a2d8c4874f2b90f98c77e115349efb22a","subject":"Fix skip condition for \"test_default_share_group_type_applied\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1369e54a2d8c4874f2b90f98c77e115349efb22a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-21 06:47:11.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-21 06:47:11.000000000","tz":-420},"subject":"SSH over IPv4 when using a storage network","message":"SSH over IPv4 when using a storage network\n\nA storage network can be configured to connect\nVMs in scenario tests to their shares. When\nusing a storage network, we don\u0027t need to force\nthe private network that we use to SSH to the\nVM to be on a particular IP version.\n\nWe need the private network subnet to be of\na particular IP version only when there is\nno storage network, and we\u0027re connecting to\nshares over the same network as SSH.\n\nChange-Id: If1d581730c42b060e74600ef2a05298921b1a3c6\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d11339d90ae711a3b1f896c13cac23fc285efd79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d11339d90ae711a3b1f896c13cac23fc285efd79"}]},"branch":"refs/heads/master"},"2b308742004a6b345cebc77bb8319955310cfb35":{"kind":"REWORK","_number":2,"created":"2020-08-31 17:51:21.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/39/747339/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/39/747339/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/747339/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/747339/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/747339/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/747339/2"}}},"commit":{"parents":[{"commit":"1369e54a2d8c4874f2b90f98c77e115349efb22a","subject":"Fix skip condition for \"test_default_share_group_type_applied\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1369e54a2d8c4874f2b90f98c77e115349efb22a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-21 06:47:11.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-31 17:51:10.000000000","tz":-480},"subject":"SSH over IPv4 when using a storage network","message":"SSH over IPv4 when using a storage network\n\nA storage network can be configured to connect\nVMs in scenario tests to their shares. When\nusing a storage network, we don\u0027t need to force\nthe private network that we use to SSH to the\nVM to be on a particular IP version.\n\nWe need the private network subnet to be of\na particular IP version only when there is\nno storage network, and we\u0027re connecting to\nshares over the same network as SSH.\n\nChange-Id: If1d581730c42b060e74600ef2a05298921b1a3c6\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2b308742004a6b345cebc77bb8319955310cfb35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2b308742004a6b345cebc77bb8319955310cfb35"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
