)]}'
{"id":"openstack%2Fmanila-tempest-plugin~747593","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I998154886a1c242359a31d1406ffed8dd6f42d52","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bp/graduate-share-replication-feature","hashtags":[],"change_id":"I998154886a1c242359a31d1406ffed8dd6f42d52","subject":"Remove experimental flag from share replication feature","status":"MERGED","created":"2020-08-23 18:52:49.000000000","updated":"2020-09-09 15:22:10.000000000","submitted":"2020-09-09 15:22:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747593-1599664930727-c6eda660","meta_rev_id":"977f66f5699f6c1494014619b4c4312944726eef","_number":747593,"virtual_id_number":747593,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"date":"2020-09-08 21:52:30.000000000","_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-08-25 00:06:43.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"date":"2020-09-09 02:45:42.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-09 15:22:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2020-09-08 23:31:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-09 13:43:41.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-09 13:43:41.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-25 00:06:43.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-08 21:52:30.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-09-08 23:31:28.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-09-09 02:45:42.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-09-09 13:43:41.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-09-09 15:22:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2cf7b05f064756b3ee2ed4dd843dbc50dad075be","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-08-23 18:52:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"edd129e809ab58fc794063142a68376af272a895","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-23 19:20:55.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/93/747593/1/check/DellEMC_Unity_Manila_dhss_true/8fb8503/DellEMC_Unity_Manila_dhss_true/315 : FAILURE in 27m 58s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/93/747593/1/check/DellEMC_Unity_Manila_dhss_false/12f3f55/DellEMC_Unity_Manila_dhss_false/298 : FAILURE in 26m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"c45fb27da68ce6a80bd5c98498480176a1de0845","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-08-23 21:06:06.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"84642781a4060c90206a439840cb66ddfac85c12","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-23 21:34:45.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/93/747593/2/check/DellEMC_Unity_Manila_dhss_true/ff9e863/DellEMC_Unity_Manila_dhss_true/316 : FAILURE in 27m 06s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/93/747593/2/check/DellEMC_Unity_Manila_dhss_false/e58f1e7/DellEMC_Unity_Manila_dhss_false/299 : FAILURE in 28m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"93fd1ff058133647422781526668dc16a7cf759b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 23:38:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58c1a972048c4cc5b286331cc4ecdf92 : SUCCESS in 2m 52s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e2c9dbb03628402688b2dfd621bf0bdb : SUCCESS in 35m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/683c7c9e63da48b6b00cb36600d7060e : SUCCESS in 28m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f4b638f39a594384951cc3be8e684352 : FAILURE in 1h 32m 09s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/04a118d9e4da46d6b67ebac7e15401c9 : FAILURE in 1h 22m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6cb10c59fe1b47329bf7602d4f0dc4b1 : FAILURE in 1h 56m 56s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/166ed73c14c848519658488ea91629cf : FAILURE in 1h 24m 14s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9e19e7d287e54545bc6e5226fe14eeac : SUCCESS in 1h 07m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3a21ed27e45b41708dac321e2ef7af6b : FAILURE in 2h 30m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9a90fba0298a474cab592ebd036e153e : FAILURE in 37m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b9b7af2bb314f3968ed35e37495c537df3786126","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-08-24 11:42:08.000000000","message":"Patch Set 2:\n\nrecheck\nFailed test in the LVM job is a known intermittent issue.","accounts_in_message":[],"_revision_number":2},{"id":"c3b6a4c7e54f2e1da96f314cae447f450bcb5b03","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-08-24 13:22:57.000000000","message":"Topic set to graduate-share-replication-feature","accounts_in_message":[],"_revision_number":2},{"id":"fc7bb690d0606f731294494ab9c7db31adb98243","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-08-24 13:23:51.000000000","message":"Topic changed from graduate-share-replication-feature to bp/graduate-share-replication-feature","accounts_in_message":[],"_revision_number":2},{"id":"1cb71076889aa23a793d8151ea311a064cad59ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 13:39:23.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b47cf99d6bac45ac88b51c0926296006 : SUCCESS in 3m 42s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bc0c0ab6c9a9414284d3a481afb2bf73 : SUCCESS in 27m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0106bf3242e046f68647c67ac3c61339 : SUCCESS in 40m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6c23f2de452047beaec8a8850fa3f55f : FAILURE in 1h 13m 34s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1c1c23eaef8b4ed982eb956cce74471a : FAILURE in 1h 25m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d79d84d7a55c4097bd27d4339d32ddf4 : FAILURE in 1h 41m 56s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e8c46a27ad9a457e82997060f48c940d : FAILURE in 1h 06m 06s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e5233c54c4674273b0c54969af999f29 : SUCCESS in 46m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/08ebc5a5befb468381ab75e40116dbd0 : FAILURE in 1h 50m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f9041c22b9264e2fbdb884798b54ea44 : SUCCESS in 25m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"77c692a58895fc413e513a97b21d3d7aa3d5651a","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-08-24 19:11:31.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"1762beb2d13c01b3bad5bd4e2831253d040d26aa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-24 19:41:21.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/93/747593/3/check/DellEMC_Unity_Manila_dhss_true/16ddc17/DellEMC_Unity_Manila_dhss_true/322 : FAILURE in 29m 40s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/93/747593/3/check/DellEMC_Unity_Manila_dhss_false/3e863a6/DellEMC_Unity_Manila_dhss_false/304 : FAILURE in 27m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"50273a7b70cc995fb5f1e82571202b7a89731501","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 21:58:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5df56ea005ad406ebdc28b0ff6a6ce1a : SUCCESS in 3m 16s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/037d2ba50ee54f00bc19534648356b8e : SUCCESS in 29m 47s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a8f26f7001c24d0daa8acb8a7d03cda5 : SUCCESS in 32m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b9d3358bc4074c52bbfb8b999d7441fa : SUCCESS in 1h 30m 57s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/dcdb0bcc9eed4a1aa7811bf6f767f7cf : FAILURE in 1h 11m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/cd595b34231f44868922e4569b8915db : FAILURE in 1h 30m 33s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f464a33910bf429494df52ca00da6d2f : FAILURE in 1h 00m 25s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bcd328ae31b54930a3a958d60553846e : SUCCESS in 39m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2dc8b155a8a84e4f8060c1afbbefbc95 : FAILURE in 2h 33m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/57f693a2f4b04c00b9b3ff8e9639e1b5 : SUCCESS in 24m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"622d032f3d5baed35f0c6f61aeb44d821cc6fc65","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-24 22:28:02.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/93/747593/3/check/DellEMC_Unity_Manila_dhss_true/9397239/DellEMC_Unity_Manila_dhss_true/323 : FAILURE in 29m 17s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/93/747593/3/check/DellEMC_Unity_Manila_dhss_false/d3d3ac9/DellEMC_Unity_Manila_dhss_false/305 : FAILURE in 28m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"a61ab405ec5dae80b26f5311596b0c633bddd458","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-25 00:06:43.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/93/747593/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/deb7521/ : SUCCESS in 2h 07m 37s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/93/747593/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/e22bd09/ : FAILURE in 1h 26m 47s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"90ac5d48459467afb7e64e35abc3949f8d45626b","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-09-08 21:52:30.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"deb9fd39ddf7295bdafb437096887df4cb2ca7ee","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-08 23:31:28.000000000","message":"Patch Set 3: Code-Review+2\n\nrecheck\n\nWould like to see the zfsonlinux job passing..","accounts_in_message":[],"_revision_number":3},{"id":"c81e004bf62c49e544292ba51d41d6796232ba53","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-08 23:31:42.000000000","message":"Patch Set 3:\n\nThanks for the changes, Carlos!","accounts_in_message":[],"_revision_number":3},{"id":"710f41d01ff13bfe43222d4efc6adc759ccac272","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 23:55:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4831aabfafbf4b5db5c8d0ac43231984 : SUCCESS in 3m 53s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/200a63d45ae8463f99ccd1bd57764dcc : SUCCESS in 24m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/bcf14854211a4717933cac02db1059f8 : SUCCESS in 29m 08s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/446ded2054ab4b69983269ce1ff0d9ce : SUCCESS in 1h 03m 44s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3aa09570b70b42659ce9a43089828eb8 : FAILURE in 58m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/53f51b4ea0e648f6abdc4b1044690f24 : SUCCESS in 57m 00s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6e533d91a5244d4b974c74a26873700b : FAILURE in 1h 04m 31s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d5397e80b6da483f8c121d6e04f0e644 : SUCCESS in 39m 14s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/91f402ea9e004a118b6e3aa3cfe1c575 : FAILURE in 1h 48m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6f79e513f7b54988ab13f7c570b2c371 : FAILURE in 32m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"047fcc638b4caaecb7012235a5883f1b5c01237f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-09 00:01:44.000000000","message":"Patch Set 3:\n\nrecheck\n\nDifferent failures on the zfs job, none related to this patch","accounts_in_message":[],"_revision_number":3},{"id":"8c041e61e3564c3f86ee37296a111c1d4fb55efe","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-09 00:36:23.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/93/747593/3/check/DellEMC_Unity_Manila_dhss_true/51e684a/DellEMC_Unity_Manila_dhss_true/367 : FAILURE in 25m 45s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/93/747593/3/check/DellEMC_Unity_Manila_dhss_false/566a6b7/DellEMC_Unity_Manila_dhss_false/349 : FAILURE in 27m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"32420ae131a1d1947c608cc2fcb8ed23447582cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 01:50:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac913b3e07a34cc48f5afc7730159102 : SUCCESS in 5m 12s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a95e5a25183648dc8b386de16d1aa6bd : SUCCESS in 28m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1ab099f39ae64ebf89823a20b494cb35 : SUCCESS in 31m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0145879f8dd94d4ba5ed14737b186423 : SUCCESS in 1h 10m 16s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/78527ba0a8804c54a497db45897ea33c : SUCCESS in 1h 06m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/2461647ee1c4480d844c977929c7e7a5 : SUCCESS in 58m 06s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/772743375c72426caeac20833b9bbf8c : FAILURE in 1h 07m 15s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/769a8247934643cd8e692bf328a45a8c : SUCCESS in 44m 05s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/33214e5851494e4f83e87b2f12c45e7d : FAILURE in 1h 45m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6fdf0a82d5ab4e0f8380cf3403bdea41 : SUCCESS in 24m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d22fa0738198ee24c55179efa6aecff92f760ba8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-09 02:45:42.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/93/747593/3/check/DellEMC_Unity_Manila_dhss_true/562e625/DellEMC_Unity_Manila_dhss_true/370 : FAILURE in 26m 23s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/93/747593/3/check/DellEMC_Unity_Manila_dhss_false/f772b15/DellEMC_Unity_Manila_dhss_false/352 : FAILURE in 26m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"f0feb239ae619f3d119d576736ae45a453f01456","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-09-09 13:43:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThanks, Carlos!","accounts_in_message":[],"_revision_number":3},{"id":"724e459e3ca9044bb196226a8a2af64aa873865e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 13:43:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1b33f941a0b39026ddd4442b50c1f6a5c2bf790e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 15:22:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc8a7771efc14c00ba34b8c2efbd057e : SUCCESS in 5m 15s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f4427659711a40a4b6ef78c17d361832 : SUCCESS in 32m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7063fe17419c46058fd1b7b6782bc43a : SUCCESS in 36m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0c76bbd4ea3e45e08b03ea48045b947a : SUCCESS in 1h 32m 29s","accounts_in_message":[],"_revision_number":3},{"id":"977f66f5699f6c1494014619b4c4312944726eef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 15:22:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d354d672f8e69d3b3ef44110cf1cc6eb958c8e57","revisions":{"48e8ea15a384c5bb38a9bcf74d3ba195e3be3a71":{"kind":"REWORK","_number":1,"created":"2020-08-23 18:52:49.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/93/747593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/93/747593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/93/747593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/93/747593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/93/747593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/93/747593/1"}}},"commit":{"parents":[{"commit":"1ef4c6af99498dacfe8017cbec19df07b80f0b44","subject":"Merge \"Ensure share server is created when driver handles share servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1ef4c6af99498dacfe8017cbec19df07b80f0b44"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2020-08-23 18:49:52.000000000","tz":0},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2020-08-23 18:49:52.000000000","tz":0},"subject":"[WIP] Remove experimental flag from share replication feature","message":"[WIP] Remove experimental flag from share replication feature\n\nThis patch removes the experimental flag from the tests related\nto the share replication feature.\n\nChange-Id: I998154886a1c242359a31d1406ffed8dd6f42d52\nDepends-On: Idf8af1c96df373fbcbb4024db490cb4dab42faf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/48e8ea15a384c5bb38a9bcf74d3ba195e3be3a71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/48e8ea15a384c5bb38a9bcf74d3ba195e3be3a71"}]},"branch":"refs/heads/master"},"5e27ffa586cafbfc253d129062011b04221ea305":{"kind":"REWORK","_number":2,"created":"2020-08-23 21:06:06.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/93/747593/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/93/747593/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/93/747593/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/93/747593/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/93/747593/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/93/747593/2"}}},"commit":{"parents":[{"commit":"1ef4c6af99498dacfe8017cbec19df07b80f0b44","subject":"Merge \"Ensure share server is created when driver handles share servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1ef4c6af99498dacfe8017cbec19df07b80f0b44"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2020-08-23 18:49:52.000000000","tz":0},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2020-08-23 21:06:00.000000000","tz":0},"subject":"[WIP] Remove experimental flag from share replication feature","message":"[WIP] Remove experimental flag from share replication feature\n\nThis patch removes the experimental flag from the tests related\nto the share replication feature.\n\nChange-Id: I998154886a1c242359a31d1406ffed8dd6f42d52\nDepends-On: Idf8af1c96df373fbcbb4024db490cb4dab42faf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5e27ffa586cafbfc253d129062011b04221ea305"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5e27ffa586cafbfc253d129062011b04221ea305"}]},"branch":"refs/heads/master"},"d354d672f8e69d3b3ef44110cf1cc6eb958c8e57":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-08-24 19:11:31.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/93/747593/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/93/747593/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/93/747593/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/93/747593/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/93/747593/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/93/747593/3"}}},"commit":{"parents":[{"commit":"1ef4c6af99498dacfe8017cbec19df07b80f0b44","subject":"Merge \"Ensure share server is created when driver handles share servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1ef4c6af99498dacfe8017cbec19df07b80f0b44"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2020-08-23 18:49:52.000000000","tz":0},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2020-08-24 19:11:23.000000000","tz":0},"subject":"Remove experimental flag from share replication feature","message":"Remove experimental flag from share replication feature\n\nThis patch removes the experimental flag from the tests related\nto the share replication feature.\n\nPartially-implements: bp graduate-share-replication-feature\nChange-Id: I998154886a1c242359a31d1406ffed8dd6f42d52\nDepends-On: Idf8af1c96df373fbcbb4024db490cb4dab42faf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d354d672f8e69d3b3ef44110cf1cc6eb958c8e57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d354d672f8e69d3b3ef44110cf1cc6eb958c8e57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
