)]}'
{"id":"openstack%2Fmanila-tempest-plugin~752312","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I4dcd677d53d56aad544d9a7e88b8160b704c241d","project":"openstack/manila-tempest-plugin","branch":"master","hashtags":[],"change_id":"I4dcd677d53d56aad544d9a7e88b8160b704c241d","subject":"Fix share server migration tests","status":"MERGED","created":"2020-09-16 20:40:29.000000000","updated":"2020-09-30 20:28:13.000000000","submitted":"2020-09-30 20:28:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":35,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752312-1601497693106-9817b790","meta_rev_id":"aba7783c318d5ee75b64a352e99d707eda1e9d60","_number":752312,"virtual_id_number":752312,"owner":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-22 19:59:59.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},{"value":0,"date":"2020-09-22 18:33:16.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-22 20:34:50.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-30 20:28:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-29 21:48:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2020-09-22 17:00:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-29 21:48:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-22 17:00:33.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-09-22 18:33:16.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-22 19:59:59.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-22 20:34:50.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-09-29 21:48:35.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-09-30 20:28:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e016677c347af881166f31d173bc26b6767f5980","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-09-16 20:40:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50b03116bceaa2af9b1a94b778621034ad86cf35","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-16 21:35:31.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/12/752312/1/check/DellEMC_Unity_Manila_dhss_true/3eb9ee4/DellEMC_Unity_Manila_dhss_true/424 : FAILURE in 54m 54s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/12/752312/1/check/DellEMC_Unity_Manila_dhss_false/699d3d1/DellEMC_Unity_Manila_dhss_false/405 : SUCCESS in 52m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"b07755b11c91d65afd6e769109b9a2161f62956d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 23:06:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a84dc5414bed44c7bab26e4252be146f : SUCCESS in 3m 07s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6c8b1b6cd71740fa99b26c2b1e886616 : SUCCESS in 34m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/980607cb379f44f99def876654404be1 : SUCCESS in 41m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d684f3a56e3e419fb926b8b5ae3e85b2 : SUCCESS in 1h 33m 16s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/669786cdb2bc4cc5b538ddfc76b078e7 : SUCCESS in 1h 11m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e39b62ea9adb461e9fb981f7ad2ebe22 : SUCCESS in 1h 02m 32s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4d6612a400504bd9a64dbab3645d17de : FAILURE in 1h 08m 20s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6285585975464bc2989bb2526e181712 : SUCCESS in 51m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/206f1566531e4a05a8b3f57f9b0e3d68 : FAILURE in 2h 15m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2602fd06f997449aa94f7994fb192c64 : SUCCESS in 26m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2510ba547d7daff124e4dde630e1d266f991a44a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-17 00:05:28.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/12/752312/1/check/DellEMC_Unity_Manila_dhss_true/8bb8cd7/DellEMC_Unity_Manila_dhss_true/425 : FAILURE in 58m 30s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/12/752312/1/check/DellEMC_Unity_Manila_dhss_false/9141ee5/DellEMC_Unity_Manila_dhss_false/406 : SUCCESS in 55m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"53423b4be2efc26d79c27986fc30b0d2ba90cf8d","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-09-17 00:25:46.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/12/752312/1/check/DellEMC_VNX_Manila/42e7db8/DellEMC_VNX_Manila/233 : SUCCESS in 1h 18m 43s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"7fe82ca2fcbcd84c170b18e14c1ecd2ea1cc9863","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-17 01:32:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/12/752312/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/be99d96/ : SUCCESS in 2h 25m 02s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/12/752312/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/f10a31d/ : SUCCESS in 1h 39m 23s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"da10eb5812db11978a97b471b09be4508b4e328f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-17 17:48:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(12 comments)\n\nThanks for the update Andre, some questions and several comments inline","accounts_in_message":[],"_revision_number":1},{"id":"3d487306c02d7fccfd83baec6d92fef7634cd9df","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-09-17 20:07:43.000000000","message":"Patch Set 1:\n\n(12 comments)\n\nHi Goutham, thanks for the comments. Some questions inline.","accounts_in_message":[],"_revision_number":1},{"id":"843aa148ed92d925dcc9ff9812b4aecc4e011b3e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-17 22:16:53.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"128e8ac726b970ddf1792c2c668445db848538a9","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-09-18 12:47:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e930e5d4c987b9d6529b01f6a3be27b6412293b4","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-09-18 12:49:20.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nHi Goutham, thanks for the comments. I removed the share network creation in the shares client v2. It\u0027s not necessary, thanks for the clarification.","accounts_in_message":[],"_revision_number":1},{"id":"eaf02bafb7a4e961b9d98b5820abb13bd532f504","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-18 13:01:10.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/12/752312/2/check/DellEMC_Unity_Manila_dhss_true/53bcade/DellEMC_Unity_Manila_dhss_true/438 : FAILURE in 12m 33s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/12/752312/2/check/DellEMC_Unity_Manila_dhss_false/27414e9/DellEMC_Unity_Manila_dhss_false/419 : FAILURE in 13m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"39a40e722fdbf77bd644e6a860222ff0d3010815","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 14:55:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/589154f3216e46f98f462d5cff8e59c4 : SUCCESS in 3m 50s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f5d197ec99834885be74f24143821d77 : SUCCESS in 28m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/94b55ceffa5c460784ca00e5349e8f64 : SUCCESS in 32m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5374853460f6414bbd119435263e0f1a : SUCCESS in 1h 04m 19s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5bed3f16e47d4205adc176c51fe29fff : SUCCESS in 1h 11m 43s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6862759b202947948a404d4cb81c8e33 : SUCCESS in 1h 01m 24s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/89a572c07eb941d1a058102af0a59f4c : FAILURE in 1h 28m 20s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/94c9e0d01af54e2393d8bc6cb525ce6c : SUCCESS in 50m 26s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5587e210318e4bbba2ed3f2bef190d05 : FAILURE in 2h 00m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3b265e17403e46c8b8e13cbdd00d406e : SUCCESS in 26m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"23b341e8785c5ad53048c562da68fcc53caebc9c","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-09-18 15:09:28.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/12/752312/2/check/DellEMC_VNX_Manila/93ccfdc/DellEMC_VNX_Manila/243 : FAILURE in 13m 57s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"a3055a1e0e483f6b592d0aead383eac262dd2342","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-18 15:09:36.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/12/752312/2/check/DellEMC_Unity_Manila_dhss_true/d58689f/DellEMC_Unity_Manila_dhss_true/439 : FAILURE in 14m 10s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/12/752312/2/check/DellEMC_Unity_Manila_dhss_false/b19dace/DellEMC_Unity_Manila_dhss_false/420 : FAILURE in 13m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"f8ad8e7237e75f79052b2150c53d17ab1e0ea1c5","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-18 17:09:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/12/752312/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/bfc8a76/ : SUCCESS in 2h 14m 17s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/12/752312/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/58ad89f/ : SUCCESS in 1h 26m 39s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"4e9e652db4d824db7833e2d89480ac0106da62e1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-18 19:12:16.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nLooks great now, one more issue inline to avoid potential failures..","accounts_in_message":[],"_revision_number":2},{"id":"32adce87296a91490f22d713b26bb39cec38029f","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-09-19 13:15:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6ad4542d60146bca10038721fba087b6f213cc0a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-19 14:09:06.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/12/752312/3/check/DellEMC_Unity_Manila_dhss_true/27637ea/DellEMC_Unity_Manila_dhss_true/445 : FAILURE in 53m 48s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/12/752312/3/check/DellEMC_Unity_Manila_dhss_false/10e4c8b/DellEMC_Unity_Manila_dhss_false/426 : SUCCESS in 44m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"6fba59aadae619c7c521ad31a6d32e9c8a1f1800","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-19 15:10:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28daa2d06581449bb33482ec07ec03be : SUCCESS in 3m 25s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/49ad06e5ca5049ad8e0b7138dad750a8 : SUCCESS in 23m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5809f3fc6c884acb8a835b6b6263373b : SUCCESS in 37m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2366f15fa65d473a992c9a01305a6ddc : SUCCESS in 1h 22m 42s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/99defb060a354d7ba07757e659fa8d3a : FAILURE in 1h 23m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/86c205bcb47b428ab71e32199c82dbbb : SUCCESS in 1h 28m 44s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f3370105a9dc476681d3ca52c6ed20c6 : FAILURE in 1h 14m 57s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d78ceaada92a434fb8a78de0c3a10f55 : SUCCESS in 38m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c881b2b55eac488f9addb5e92a11066f : FAILURE in 1h 54m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ca2518c0fa3644c9a5c2e5314ca03edf : SUCCESS in 26m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f96c5131d1a4006a570891a9596598c47071c943","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-19 16:02:13.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/12/752312/3/check/DellEMC_Unity_Manila_dhss_true/175880e/DellEMC_Unity_Manila_dhss_true/446 : FAILURE in 51m 09s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/12/752312/3/check/DellEMC_Unity_Manila_dhss_false/a5823d2/DellEMC_Unity_Manila_dhss_false/427 : SUCCESS in 47m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"1f6f48c3e79d84b8b4df04e92f77fe6d02d15354","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-19 17:16:51.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/12/752312/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/6c2b682/ : SUCCESS in 2h 05m 37s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/12/752312/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/2939525/ : SUCCESS in 1h 30m 12s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"0afd6ad4e48cea693bf46258b194942fe04b333b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-21 21:32:57.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2b5d3e8c98d17e3d82ebb169bfdb8cd0b0b876b1","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-09-22 12:53:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fee8b817c2ef3675121226619dd9ede3af8e4761","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-09-22 12:53:54.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nHi Goutham, thanks for this reminder! :D","accounts_in_message":[],"_revision_number":3},{"id":"359776781c8c7f21180be6bbc726ace9aa483846","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-22 13:57:52.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/12/752312/4/check/DellEMC_Unity_Manila_dhss_true/fb3fd92/DellEMC_Unity_Manila_dhss_true/470 : FAILURE in 58m 48s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/12/752312/4/check/DellEMC_Unity_Manila_dhss_false/c8e92b5/DellEMC_Unity_Manila_dhss_false/453 : SUCCESS in 1h 04m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"38ca6a833724e8e306c80c93f6436a76283b387d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 14:39:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/482a813d345b44959683e5fc0e9713cf : SUCCESS in 4m 18s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/84458f0739ef4b71a650b722612d93d7 : SUCCESS in 27m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7755cdf576ee41008d751f7e02e16e1f : SUCCESS in 35m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6a3a66553d6f46e99ccc49741cb5401b : SUCCESS in 59m 56s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a6033b3ffdd4477da6c6f8fe66d8af40 : SUCCESS in 1h 13m 28s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/2ac325030fdf460eb4bbe070eb339641 : SUCCESS in 1h 01m 02s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a4799f616a8345eab520d39c9bde12a3 : FAILURE in 51m 49s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/58a47b99baf349c1b3a042d96202fdb4 : SUCCESS in 42m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d84dfaecfa7f42dab7f965f9aefd7397 : FAILURE in 1h 37m 02s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7c3ec15bb08949baa2cacad40d636b15 : SUCCESS in 20m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f38c619198e9944060abb608e0e9e147c17d4ff4","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-09-22 14:58:26.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"9b65a9f7ad12abda65440d85034f03e5ba368156","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-22 16:52:29.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/12/752312/5/check/DellEMC_Unity_Manila_dhss_true/f9f6ceb/DellEMC_Unity_Manila_dhss_true/473 : FAILURE in 58m 59s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/12/752312/5/check/DellEMC_Unity_Manila_dhss_false/bbcacf3/DellEMC_Unity_Manila_dhss_false/456 : SUCCESS in 57m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"e6720189b35bfc7a37c9e47fbf1a60e31df5d8ce","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-22 17:00:33.000000000","message":"Patch Set 5: Code-Review+2\n\nThanks, Andre!","accounts_in_message":[],"_revision_number":5},{"id":"8cd3c1d155a7c5b14ffe7ac9bd545c6668823f08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 17:03:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2849e8ba00384894b2e26a8a68977978 : SUCCESS in 4m 21s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0f0f0351c50345099227aeb47e60a9e7 : SUCCESS in 32m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/59c59d53451b4c20a0947b3ef6b7e1ea : SUCCESS in 37m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7c10a9fa2b2b4408a43bd34a31d4064b : SUCCESS in 1h 06m 06s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/06bf8cda916a4154a361dc23e8923e40 : SUCCESS in 1h 04m 35s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1bdd848eaaba4988ba8dd9ffa10d8713 : SUCCESS in 1h 09m 27s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5c6583bb87394511a32130cbc5da0ba0 : FAILURE in 1h 07m 23s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/85d749601b004cc29e4e2322598c15dc : SUCCESS in 40m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/72f7bb25cdca49eeb066ed7e7b4c36e6 : FAILURE in 1h 53m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9380bfe2e2cb4d899851db5dde2bec70 : SUCCESS in 23m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"122cc19fc056898b6e8fa91d94818c5a563fff0b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-22 18:33:16.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/12/752312/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/1aeb927/ : POST_FAILURE in 4m 17s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/12/752312/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/511b71c/ : SUCCESS in 1h 29m 33s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"6920d74d845ea32d0749291a2db5c3b520c535fd","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-09-22 19:59:59.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/12/752312/5/check/DellEMC_VNX_Manila/049c56f/DellEMC_VNX_Manila/263 : SUCCESS in 1h 16m 45s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"e932892df00ec4e395ed735fdc39f600e94640d5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-22 20:34:50.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/12/752312/5/check/DellEMC_Unity_Manila_dhss_true/d9d3359/DellEMC_Unity_Manila_dhss_true/477 : FAILURE in 51m 02s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/12/752312/5/check/DellEMC_Unity_Manila_dhss_false/0c4bc02/DellEMC_Unity_Manila_dhss_false/460 : SUCCESS in 53m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"835c4c5408e48f4f01f778b964ea02599d10e882","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-09-29 21:48:35.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nThanks Andre, looks great!","accounts_in_message":[],"_revision_number":5},{"id":"2b68ceb1bbe1f2274dd3d9f9345dad4f68e83cb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 21:48:54.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"bba5e0f3ab3dbe0012f9d603a5c3e727b6f9c446","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 22:57:03.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36c8aafa9e6042f5b3741d2a08841555 : SUCCESS in 2m 52s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/af7fe670e8354da48fb668c10206c873 : SUCCESS in 32m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b5421907375d4142aa34fab19b936e09 : SUCCESS in 37m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b9a1150a17f641bf9502adc629e36e80 : FAILURE in 1h 07m 23s","accounts_in_message":[],"_revision_number":5},{"id":"3b0543631ad6f0cbce4f6faa59bf3798a8faaf7d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-29 23:32:45.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"66b68cd7dcb9259dbbd8828e37ac30b42263c994","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-30 15:58:50.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"c3c60cb065bb7b7fb3e0ce22dffa0787b1606f1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 18:59:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96054c5d11fb42e3b95aebe6496fcae1 : SUCCESS in 6m 23s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f0d7b5973b134d119404ab30dfd144dc : SUCCESS in 32m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9f1a4572c02b464cbe4392e39351fa00 : SUCCESS in 31m 23s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8a982f49379a47898733443cfe08b6e8 : SUCCESS in 1h 05m 17s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2402d14468dd4e5d92a7ad97d9f85929 : SUCCESS in 1h 20m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1824735dd76345a28edb4c89af252024 : SUCCESS in 1h 00m 16s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b1b5bfc7997f4ce4a4a005d850b9e39f : FAILURE in 47m 37s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/69b46c32e746498985e78f95dd9d85e9 : SUCCESS in 40m 08s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f5670caf98544097b53a89de87d99133 : FAILURE in 2h 56m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1cee70d048504702bb83fc19d8d9f6f1 : SUCCESS in 29m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"61c814372c536df9ea90ddc3ab86381c96a21fe5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 18:59:23.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4deec113acbb5eb6710c5098dea2e9edacdd90d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 20:28:12.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af04650979f64957bcb55eb2e2d7bdfa : SUCCESS in 3m 41s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/322f8f03a47f4e9db780b4af15a80a65 : SUCCESS in 28m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/89c84d03a3994f8bbef524f0c3ea833d : SUCCESS in 32m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bf06593e70c141f986e13ba53f83b067 : SUCCESS in 1h 24m 51s","accounts_in_message":[],"_revision_number":5},{"id":"aba7783c318d5ee75b64a352e99d707eda1e9d60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 20:28:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b3089a7ee3da99cc249e4d5bf3e4bad353116877","revisions":{"c121efd612892c2a5169715d4026ae4057e458ce":{"kind":"REWORK","_number":1,"created":"2020-09-16 20:40:29.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/12/752312/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/12/752312/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/752312/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/752312/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/752312/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/752312/1"}}},"commit":{"parents":[{"commit":"2e12123d97d04113fdc0f23458fa27ee18978da0","subject":"Merge \"Add tests for share server migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e12123d97d04113fdc0f23458fa27ee18978da0"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-09-15 21:06:04.000000000","tz":0},"committer":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-09-16 20:38:54.000000000","tz":0},"subject":"Fix share server migration tests","message":"Fix share server migration tests\n\nThis patch provides small fixes to the share server migration tests.\n\nChange-Id: I4dcd677d53d56aad544d9a7e88b8160b704c241d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c121efd612892c2a5169715d4026ae4057e458ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c121efd612892c2a5169715d4026ae4057e458ce"}]},"branch":"refs/heads/master"},"2a79eb495f7099bc82734f38a6c07502a6f0f315":{"kind":"REWORK","_number":2,"created":"2020-09-18 12:47:38.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/12/752312/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/12/752312/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/752312/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/752312/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/752312/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/752312/2"}}},"commit":{"parents":[{"commit":"2e12123d97d04113fdc0f23458fa27ee18978da0","subject":"Merge \"Add tests for share server migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e12123d97d04113fdc0f23458fa27ee18978da0"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-09-15 21:06:04.000000000","tz":0},"committer":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-09-18 12:47:15.000000000","tz":0},"subject":"Fix share server migration tests","message":"Fix share server migration tests\n\nThis patch provides small fixes to the share server migration tests.\n\nChange-Id: I4dcd677d53d56aad544d9a7e88b8160b704c241d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2a79eb495f7099bc82734f38a6c07502a6f0f315"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2a79eb495f7099bc82734f38a6c07502a6f0f315"}]},"branch":"refs/heads/master"},"2ca8fd78891e690e0d1f1a9ef97574cd26da5688":{"kind":"REWORK","_number":3,"created":"2020-09-19 13:15:10.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/12/752312/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/12/752312/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/752312/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/752312/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/752312/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/752312/3"}}},"commit":{"parents":[{"commit":"2e12123d97d04113fdc0f23458fa27ee18978da0","subject":"Merge \"Add tests for share server migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e12123d97d04113fdc0f23458fa27ee18978da0"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-09-15 21:06:04.000000000","tz":0},"committer":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-09-19 13:13:32.000000000","tz":0},"subject":"Fix share server migration tests","message":"Fix share server migration tests\n\nThis patch provides small fixes to the share server migration tests.\n\nChange-Id: I4dcd677d53d56aad544d9a7e88b8160b704c241d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2ca8fd78891e690e0d1f1a9ef97574cd26da5688"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2ca8fd78891e690e0d1f1a9ef97574cd26da5688"}]},"branch":"refs/heads/master"},"ca4dd84dcca81bb7dfc02825eccc01e4afcbb37d":{"kind":"REWORK","_number":4,"created":"2020-09-22 12:53:22.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/12/752312/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/12/752312/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/752312/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/752312/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/752312/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/752312/4"}}},"commit":{"parents":[{"commit":"2e12123d97d04113fdc0f23458fa27ee18978da0","subject":"Merge \"Add tests for share server migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e12123d97d04113fdc0f23458fa27ee18978da0"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-09-15 21:06:04.000000000","tz":0},"committer":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-09-22 12:52:56.000000000","tz":0},"subject":"Fix share server migration tests","message":"Fix share server migration tests\n\nThis patch provides small fixes to the share server migration tests.\n\nChange-Id: I4dcd677d53d56aad544d9a7e88b8160b704c241d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ca4dd84dcca81bb7dfc02825eccc01e4afcbb37d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ca4dd84dcca81bb7dfc02825eccc01e4afcbb37d"}]},"branch":"refs/heads/master"},"b3089a7ee3da99cc249e4d5bf3e4bad353116877":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-09-22 14:58:26.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/12/752312/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/12/752312/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/752312/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/752312/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/752312/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/752312/5"}}},"commit":{"parents":[{"commit":"3b68c8174d7860905a9a30521c011fa29b158b84","subject":"Merge \"Don\u0027t ignore invalid export location ip version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3b68c8174d7860905a9a30521c011fa29b158b84"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-09-15 21:06:04.000000000","tz":0},"committer":{"name":"Andre Beltrami","email":"debeltrami@gmail.com","date":"2020-09-22 14:58:26.000000000","tz":0},"subject":"Fix share server migration tests","message":"Fix share server migration tests\n\nThis patch provides small fixes to the share server migration tests.\n\nChange-Id: I4dcd677d53d56aad544d9a7e88b8160b704c241d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b3089a7ee3da99cc249e4d5bf3e4bad353116877"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b3089a7ee3da99cc249e4d5bf3e4bad353116877"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
