)]}'
{"id":"openstack%2Fmanila-tempest-plugin~755183","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I724eba28d1a88317655f9cb7da94716772cd01ab","project":"openstack/manila-tempest-plugin","branch":"master","topic":"manila_remove_six","attention_set":{},"removed_from_attention_set":{"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2021-12-12 18:50:14.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2021-12-12 18:50:14.000000000","reason":"Change was submitted"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2021-12-12 17:26:18.000000000","reason":"Liron Kuchlani replied on the change"}},"hashtags":[],"change_id":"I724eba28d1a88317655f9cb7da94716772cd01ab","subject":"Remove all usage of six library","status":"MERGED","created":"2020-09-30 05:51:33.000000000","updated":"2021-12-12 18:50:14.000000000","submitted":"2021-12-12 18:50:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755183-manila_remove_six","meta_rev_id":"05dcfe05cbcba6c4faa8081cedc435be728c4ae7","_number":755183,"virtual_id_number":755183,"owner":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-12 18:50:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2021-12-09 19:39:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":2,"date":"2021-12-12 17:26:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":1,"date":"2021-12-12 17:26:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-30 06:35:32.000000000","updated_by":{"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},"reviewer":{"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},"state":"REVIEWER"},{"updated":"2020-09-30 11:58:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-30 13:43:18.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-09 03:05:41.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-02-05 14:01:52.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-05 16:15:21.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-02-12 14:13:54.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2021-12-07 21:53:03.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-12-07 21:53:03.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"}],"messages":[{"id":"dc4822059ba05b3226b7bf17462c621a0f87217e","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-09-30 05:51:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07fb8043a8183667b356a67e71f738ec5495afc5","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-09-30 05:52:19.000000000","message":"Topic set to remove-six","accounts_in_message":[],"_revision_number":1},{"id":"3ab7cb3e23f0b58b18d638504976c465be0202c7","author":{"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},"date":"2020-09-30 06:35:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf805fd0886f06492ca99a7bd204944fcb253379","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 07:47:42.000000000","message":"Patch Set 1: Verified-1\n\n(22 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d7866b921bb48e99d0af4d55160009c : FAILURE in 2m 58s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cd60eb72a1064e779025e2bbac598c21 : SUCCESS in 35m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/57ee80c308ca4479abbf068836b270c4 : SUCCESS in 37m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4d1eb6e0e2454e6aa3d3e1c67faa19d9 : SUCCESS in 1h 11m 21s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/92cb363f471649ac920b2ba3756983b0 : SUCCESS in 1h 06m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d7dafddb1d924d0b8b15ce9eb33bdcc9 : SUCCESS in 1h 07m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/344385ccf8f5448884f0ba21d2f332c6 : FAILURE in 1h 00m 12s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c4611f9d3b0a4824bff26d45331d7ce8 : SUCCESS in 45m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4a1b0cf973ed4f6aab14087517086a3b : FAILURE in 1h 50m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0aedbf05f493403da60fc87703c07653 : SUCCESS in 29m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8e8c84fd11137541f4c6d3243f5f937e6c54cd5","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-09-30 09:12:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"286a066c9cf39a0c963fba7762cbd7c008687e00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 11:58:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4484a81c4024946a8b9cbef2153ee91 : SUCCESS in 3m 07s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9514e53923fc4da69249f1b0efc17800 : SUCCESS in 30m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/751cb64e91b944ffaef70f9ae09ddc08 : SUCCESS in 30m 24s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/09614774d8a3405a9cda9bd36f17a369 : SUCCESS in 1h 11m 41s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a1796e96324f421da0a097d3226187e2 : SUCCESS in 1h 02m 49s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f1acb4efbcc141d2adabc617f80c5835 : FAILURE in 1h 08m 55s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/be4a4adc4b8e44b6a56bd06219eff38b : FAILURE in 48m 17s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a2297c1916e24bdd99e4df948d18d95e : SUCCESS in 1h 02m 08s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b32b9f140d054c4b8cf447ff8e964c13 : FAILURE in 2h 37m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5b208f73f2774a0db1bf5ab41aa57e14 : SUCCESS in 26m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bfb6f96cd8583d0c601aae9a2a7966345ecfea62","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-30 13:43:18.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/83/755183/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/ecf1adf/ : FAILURE in 23m 13s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/83/755183/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/049e914/ : SUCCESS in 1h 44m 25s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"18d48b49cafd986dcdb08bb46ee20fadf0870f6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-02-05 08:57:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b4bf7f017e2dce8f5c61191537cf77cd3f88b8ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 10:56:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccf47d2267e64cd3a3ff2c36e0f554af : SUCCESS in 3m 05s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/479a7c55e6114ad59a06274b06fd803d : SUCCESS in 45m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5eaf9942e7244685b84746944ceafa65 : SUCCESS in 48m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ae61ecbcf0f541bab42de493a7c46a24 : SUCCESS in 1h 19m 34s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4d3d315be1124fcb985b93a56dbbf70c : SUCCESS in 1h 05m 38s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6deda62b453842e785fd14539ea3f564 : SUCCESS in 1h 04m 09s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c82f9a9cf4764b92978fb626e7adc178 : FAILURE in 1h 29m 41s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2a73a7b8334f45ada8496b088dbbf3f8 : SUCCESS in 47m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/742684a0499d40ad920eda58edc3ebda : FAILURE in 1h 50m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/32e86e26f2434d7c9ff44f14d1931f68 : FAILURE in 28m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"72655d6bf26a7835483b8af9c0b2fe2595b3a786","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-05 13:53:19.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/83/755183/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/b4d47f0/ : SUCCESS in 2h 31m 31s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/83/755183/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/9d32eea/ : SUCCESS in 1h 51m 39s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"669a0222c3f51e03ff5575b74d49b946f629af93","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-05 14:01:52.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/83/755183/3/check/DellEMC_VNX_Manila/9ca4477/DellEMC_VNX_Manila/158 : SUCCESS in 1h 15m 00s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"89e4e3b474f6e6fa8584e8e04b8017da94db204e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-05 16:15:21.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/83/755183/3/check/DellEMC_Unity_Manila_dhss_true/162cc07/DellEMC_Unity_Manila_dhss_true/265 : FAILURE in 1h 42m 02s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/83/755183/3/check/DellEMC_Unity_Manila_dhss_false/e6bc0e7/DellEMC_Unity_Manila_dhss_false/248 : SUCCESS in 50m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"bac45efaa0f9ef9a4fac2035cd73b3dc46ea3b78","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-02-12 14:13:54.000000000","message":"Patch Set 3: Code-Review-1\n\nGreat job Haixin, thanks! Can you also remove the six usage in /manila_tempest_tests/common/waiters.py?","accounts_in_message":[],"_revision_number":3},{"id":"991c79f5f1edc1ab74ab5cad28b24ca8e0c2234b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-02-18 01:23:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"456ea219b882571a20143a4632f0ad70267cce55","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-02-18 01:29:46.000000000","message":"Patch Set 4:\n\nhi,Victoria Martinez de la Cruz.\ni have fix it, thanks for your review.","accounts_in_message":[],"_revision_number":4},{"id":"4e2d8c0f0cba0533940726695af536ed60166fae","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-18 02:29:45.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/83/755183/4/check/DellEMC_Unity_Manila_dhss_true/9547e34/DellEMC_Unity_Manila_dhss_true/348 : FAILURE in 1h 05m 52s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/83/755183/4/check/DellEMC_Unity_Manila_dhss_false/87c28ff/DellEMC_Unity_Manila_dhss_false/332 : SUCCESS in 56m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"569cab71a3551fce315dd8a0ca7ac8d20cbe0d9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 03:08:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58ec379adfe642418e74ecfd8e4208d4 : SUCCESS in 3m 27s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4389150029f2490282d505f91ed7dc2f : FAILURE in 33m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8911768f072d4e39a0096454b7d095cc : SUCCESS in 37m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5045bc1a1ed4446ca403fdabc8c9d933 : SUCCESS in 1h 16m 16s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/58a47a3069f5456c909943c0e4ca2b1d : SUCCESS in 1h 19m 05s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/da0de562b2aa42e4a83e15f67ffcdfd8 : SUCCESS in 1h 17m 30s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/85a033f3d5ac4657ab5cb1ae3aa10354 : SUCCESS in 1h 26m 24s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/48c89f6d658c41358f67639dc078457d : SUCCESS in 44m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c3eb169a67f642c8820dcc5b1f75382a : FAILURE in 1h 35m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/711e7723220c4fdd8b950abb34c6ab59 : SUCCESS in 26m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"75eb2afded3ff72b57273d46a87fb199a1f62698","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-02-18 08:53:55.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"352c8aef90c65e10766ce6ac4583a529341d3e6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 10:39:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d377bf74df8346e1aae8106fe46c0b2a : SUCCESS in 3m 22s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6b58d68b8f0446d69684f34b8df88ab9 : SUCCESS in 29m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ed7558d44796466297254c367f5c78b2 : SUCCESS in 38m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/91d1093082fd4f49ade439a2c52c3b00 : SUCCESS in 54m 52s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ea3d8f93622745548e397e7f53491bba : SUCCESS in 1h 24m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/31552dae796945d9a195f21ae8e75008 : SUCCESS in 1h 22m 37s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5508e759ae9146cab63f7b4064b77639 : FAILURE in 1h 23m 04s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4ad5d8c6ece34743be083bb42d3fd3bb : SUCCESS in 46m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/eb1fcb6944aa4bc6aa9a8436608a41c9 : FAILURE in 1h 38m 52s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/83e494d92d114c699380027a686bd2fd : SUCCESS in 26m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7b7e656dcc71af88fa3d13d9b4975fe284af9e2d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-18 12:36:10.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/83/755183/4/check/DellEMC_Unity_Manila_dhss_true/9978af1/DellEMC_Unity_Manila_dhss_true/352 : FAILURE in 59m 51s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/83/755183/4/check/DellEMC_Unity_Manila_dhss_false/1aa19f7/DellEMC_Unity_Manila_dhss_false/336 : SUCCESS in 54m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"79bb8c7e3b984b1cd7e0edfb3102ef9219f947d7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-18 13:03:15.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/83/755183/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/6b1093c/ : SUCCESS in 2h 23m 27s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/83/755183/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/aeb1ee0/ : SUCCESS in 1h 47m 35s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"8cfda4f78bf6334021754d8f113fe8e3a27acf74","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-18 13:21:15.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/83/755183/4/check/DellEMC_VNX_Manila/ebaead9/DellEMC_VNX_Manila/198 : SUCCESS in 1h 23m 28s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"6a1c22731430552d9d3ae6622612d3ccfa1d4f42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-12-02 01:48:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"dc517c635ced724f047d657a69a688d8fe5d0173","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-12-02 02:56:36.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/83/755183/5/check/DellEMC_Unity_Manila_dhss_true/864f9f1/DellEMC_Unity_Manila_dhss_true/134 : FAILURE in 46m 50s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/83/755183/5/check/DellEMC_Unity_Manila_dhss_false/e2530b9/DellEMC_Unity_Manila_dhss_false/127 : FAILURE in 41m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"df9ae0e536e39e54799fa95281f40a22ed1da954","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 03:06:41.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b37837aff164b8e8b91c41f3fd46d3e : FAILURE in 3m 42s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3c7522b0e1b84bd29017680e5c7d5b13 : FAILURE in 21m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4e7e288cf96f4e22966b9f776693c94b : FAILURE in 22m 08s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6d8353c6fab244aba3447bad641a1f38 : FAILURE in 24m 28s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/99f1aea7911043d3aa15ff5191cc5ef7 : FAILURE in 1h 09m 20s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/961b51ee949344589cce4a3cbc8ccfa0 : FAILURE in 1h 08m 51s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/765ed9e50bb544e3932acff4afc1d428 : FAILURE in 1h 00m 20s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/30eab4a1ae044cbdbad0f25fd993d064 : FAILURE in 21m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/18614320b3ae4a7ea7294a4ccc68d1cc : FAILURE in 26m 40s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/855198fca42a4ce8b3e4ee977f034c59 : FAILURE in 25m 31s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cddd40584a674bd3990648f5cc918e5c : FAILURE in 30m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e2e837d0e8c2445bbc79801da6f18cab : FAILURE in 25m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0483ae9a94da453d99c04c61d0944af3 : FAILURE in 22m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ce42977bb023264aab3b26aa8d9423e57c8bbfee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-12-02 07:21:08.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7c757fccea1fb10a3546ab8ff414d867a0128e14","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-12-02 08:48:31.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/83/755183/6/check/DellEMC_Unity_Manila_dhss_true/d00eae0/DellEMC_Unity_Manila_dhss_true/137 : FAILURE in 47m 55s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/83/755183/6/check/DellEMC_Unity_Manila_dhss_false/adc71cf/DellEMC_Unity_Manila_dhss_false/130 : FAILURE in 45m 13s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"af80359d41cbd58a1919f1daf01dca7bcaba9006","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 09:16:55.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90a5ad2742e24f2fa8a8ba2c999b3781 : SUCCESS in 4m 04s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d9a74b4ff0f54a50b9504a332a41f57a : SUCCESS in 31m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4590fe7d04e6491893a9fb82dc43ccc6 : SUCCESS in 37m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e06587cb04a84b1590f4556ddd5dc713 : FAILURE in 32m 11s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/9306151c1e124992913c341da1e4ea85 : SUCCESS in 1h 17m 32s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/11be0f8016be4624bc64c14d9580e457 : SUCCESS in 1h 47m 38s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/d630e3eccf6647e5b315f1be430a9d7a : SUCCESS in 1h 15m 13s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6f49da777a97414bbbd8ad065f77e1ad : FAILURE in 28m 12s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/cf974a3d986a47d495164835368eb46b : FAILURE in 30m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4c2b6d5d98d54aadb857a5d73d1fb636 : FAILURE in 30m 50s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e7647ddc8a3e4ef9b269319bb2182cf2 : FAILURE in 37m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a2bf60678e3f4cb5b6b0ef13e24c1568 : FAILURE in 32m 06s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0db6fc8a9aff4f08b27b3b8be4262451 : SUCCESS in 33m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"958b6cf5b1e2ce131cf964ef91c38a387f74189e","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-12-03 09:02:25.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2628aec107cd5dad8e0eae28333264694660adf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 10:15:59.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9caa4981e6543859660b6083a1959c8 : SUCCESS in 3m 47s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/78db17bfb27e47d28e0428caaf758bef : SUCCESS in 30m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5dc7d9ad9f1247d094b63ec0d8b8d6fa : SUCCESS in 33m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6d51c479a0e74df1a0dc7498a2ec5620 : FAILURE in 24m 15s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/73793aab06b14c779cf873c250a7768c : SUCCESS in 1h 05m 57s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/c2b03a4af5d44937b34c9c015d9c4fef : SUCCESS in 1h 10m 50s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/ae287432036d4dbd928645989c9b3ad4 : SUCCESS in 1h 10m 14s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9f86a2c2c1f24a6b8bfda0b8f5be03db : FAILURE in 35m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f19c1116181044b981aa5f3bf0feaa55 : FAILURE in 30m 46s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6b2dd6504ad045e3b6358c0dc0845b9f : FAILURE in 32m 57s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4a13117dbb994fd6bd868b798585e9e3 : FAILURE in 30m 07s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d918df7751dc4b1c93662c954a881623 : FAILURE in 37m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/47ff6589d09a40a083ca6506699848e9 : SUCCESS in 29m 25s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"da92032a3954fb8200d389f3f719980c522192a3","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-12-06 03:12:49.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fbf6d4dba6cf13e4828a3f1eb2a97e59e8e02688","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 05:33:03.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2093b5d31b394785be02c5106fb029ef : SUCCESS in 3m 17s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/aeb6ecc030e845cf8c0e5c32b583c3db : SUCCESS in 40m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d7fc7a081a334461b029a0decd4376f7 : SUCCESS in 44m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2d966f79fb0f4797a30995bd18a85b60 : FAILURE in 1h 34m 56s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/fbff4f2f50ef41f9ab673b0673f14314 : SUCCESS in 1h 43m 16s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/30065b3df58346e19763e3532af0726c : SUCCESS in 1h 12m 28s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/e9136ec10ae84288996d68f9fbcb7f6b : FAILURE in 47m 24s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/696093f0336640c1a2667fd8f0406509 : SUCCESS in 1h 31m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/73ef3abb5a904e239f55cf4f7fd9453a : SUCCESS in 1h 31m 39s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/723b9cc15ff042a5b0136c7655bc8fbe : FAILURE in 2h 15m 00s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b7303f4baf66449cb952685b9e6ae66d : SUCCESS in 53m 12s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/cfe6f5af1bd348fda719903fe0f5e9ea : FAILURE in 2h 13m 44s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9d6cbce751a14e99910d1534c40fae0b : SUCCESS in 26m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"33ee55458425d0165c2f953c397fcc30311d2853","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-12-06 05:41:41.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"65169b3f3cf4d2e97a3986018194d6184d947485","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 08:06:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe9bb929612f40c29671b78e4214eb34 : SUCCESS in 3m 06s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/149b51dc931d426f89f3b3d8312b92f2 : SUCCESS in 24m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a2dd485d98084d408887a6e2135fd5a8 : SUCCESS in 30m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d955b61123a344d5b73b12f51360f402 : SUCCESS in 1h 09m 39s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d521b6a28c18495980431b4e451b9430 : SUCCESS in 1h 26m 24s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/d2ffc7a6d07b4a6ba6f249425afda09b : SUCCESS in 1h 28m 12s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/7755fa906ed74e6ebba3db17cde632fe : SUCCESS in 1h 14m 44s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e3e3c22a21784e078116737d5c9b9395 : FAILURE in 1h 22m 35s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/34c6cf65565446bebc396d39fd8a7f44 : SUCCESS in 1h 31m 58s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/cead2ee0169b4ae18d3dfcbf62b1d752 : FAILURE in 1h 52m 18s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9b2b142d66c84fc6baf17db3f00bc5d5 : SUCCESS in 47m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/87ad1feaf9c947b8a8b487c8e5f8a91b : FAILURE in 2h 17m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5ad3bd3abbbe4bbb914fc9d5b86ff1a5 : SUCCESS in 22m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"406072cf22bf53ef153a2090b76f482e7d57e898","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-12-06 08:07:07.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/83/755183/6/upstream-check/manila-cDOT-manila-tempest-plugin-ss/7d9bcbe/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/83/755183/6/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/a2b582a/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"c19c064a90e542193b0faf9bd14c71d5d02b1153","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-12-07 00:09:42.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"735891828369896ad0f217280db76d44f62188ca","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-12-07 00:09:47.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"dff06d6caa54d7b61b9e6c349803237b085215f2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-12-08 00:49:29.000000000","message":"Topic remove-six removed","accounts_in_message":[],"_revision_number":6},{"id":"60dfca39259944301dd2a7ae33cf472802beae93","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-12-08 00:49:33.000000000","message":"Topic set to manila_remove_six","accounts_in_message":[],"_revision_number":6},{"id":"35f21f82b3a66cc960959e8554adfc6803b9e84b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-12-09 08:51:11.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"af87d4c8ce25f02776620aadc0fde3b4e30d59c0","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-12-09 08:51:21.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"46472691e8effd4af5bbc5048aa107ccc32bcf53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 10:54:05.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd3175a7054f4ebfbd03e00b8b78cdc7 : SUCCESS in 2m 56s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6d5a65e1c652429c97ea013c82166b2d : SUCCESS in 26m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/eef7b3a0595d48bbb5e8fe4626c4bf1e : SUCCESS in 30m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f2b52dc0cae54d328778614274eb6a50 : SUCCESS in 1h 20m 12s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d7fdda71071941c890cdc17bc5c194e4 : SUCCESS in 1h 15m 41s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/e3bc018d5c04473b9088e6859fac142a : SUCCESS in 1h 04m 47s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/e7c0bdf421574ea98fa11ad51e142b37 : SUCCESS in 58m 09s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/eb7226cdb372423aad20365938d1b1cf : FAILURE in 1h 28m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/23984d79b39e428191560038c12b4c2f : SUCCESS in 1h 14m 17s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f0de099b0e7b45b59c4656b813a1d398 : FAILURE in 1h 55m 16s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/636215d9fb374aeba0651d7f6e4d1297 : SUCCESS in 36m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/90459f18a5654054b365ff75852b6a28 : FAILURE in 1h 59m 44s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/853206fb33e94ede9fafb3dc68897fa9 : SUCCESS in 20m 02s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6e84a385c0efc30307442d4c2323d4b7711ae86b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-12-09 10:55:00.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/83/755183/7/upstream-check/manila-cDOT-manila-tempest-plugin-ss/69f355b/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/83/755183/7/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/8f077e2/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"f219192190f8223afb814ea3d578e78def87684c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-12-09 19:39:20.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7756fb845ed30713158f6d73343785c0701048c8","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-12-12 17:26:18.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"01e6ed985fa1626409d2f797524c2e582bfea874","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-12 17:27:02.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"ebf834bb4eb96adb6248a498727e5d9dc4f4b120","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-12 18:50:13.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2b6bc5047b54ae9baa55e2829e2421a : SUCCESS in 3m 14s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f77ee2e38b7e4242802d526a84dd6843 : SUCCESS in 27m 12s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fb315bc88c26468b96f31e3955a8be07 : SUCCESS in 39m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6dee7be95ff14c348a8314f0265a0ef2 : SUCCESS in 1h 19m 01s","accounts_in_message":[],"_revision_number":7},{"id":"05dcfe05cbcba6c4faa8081cedc435be728c4ae7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-12 18:50:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"48895819769e94ae050a08275fa6a9606ef16b73","revisions":{"e5de94b86b63969fe9c327b2f1b487399d85224f":{"kind":"REWORK","_number":1,"created":"2020-09-30 05:51:33.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/83/755183/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/83/755183/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/1"}}},"commit":{"parents":[{"commit":"ac5126fc8543dd4460cb9f64f5706b42524754c6","subject":"Merge \"Fix export location sent to manage share operation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ac5126fc8543dd4460cb9f64f5706b42524754c6"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-09-30 05:50:37.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-09-30 05:50:45.000000000","tz":480},"subject":"Remove all usage of six library","message":"Remove all usage of six library\n\nReplace six with Python 3 style code.\n\nChange-Id: I724eba28d1a88317655f9cb7da94716772cd01ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e5de94b86b63969fe9c327b2f1b487399d85224f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e5de94b86b63969fe9c327b2f1b487399d85224f"}]},"branch":"refs/heads/master"},"0bd4eba1551a2d2d9a91827787eef3a843d3656a":{"kind":"REWORK","_number":2,"created":"2020-09-30 09:12:09.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/83/755183/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/83/755183/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/2"}}},"commit":{"parents":[{"commit":"ac5126fc8543dd4460cb9f64f5706b42524754c6","subject":"Merge \"Fix export location sent to manage share operation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ac5126fc8543dd4460cb9f64f5706b42524754c6"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-09-30 05:50:37.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-09-30 09:11:02.000000000","tz":480},"subject":"Remove all usage of six library","message":"Remove all usage of six library\n\nReplace six with Python 3 style code.\n\nChange-Id: I724eba28d1a88317655f9cb7da94716772cd01ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0bd4eba1551a2d2d9a91827787eef3a843d3656a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0bd4eba1551a2d2d9a91827787eef3a843d3656a"}]},"branch":"refs/heads/master"},"74735ea26658681403e07d1ef1b8495f26179ada":{"kind":"REWORK","_number":3,"created":"2021-02-05 08:57:52.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/83/755183/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/83/755183/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/3"}}},"commit":{"parents":[{"commit":"d2138765888fe16e68f713ed99121f41e01a3c51","subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-09-30 05:50:37.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-02-05 08:57:18.000000000","tz":480},"subject":"Remove all usage of six library","message":"Remove all usage of six library\n\nReplace six with Python 3 style code.\n\nChange-Id: I724eba28d1a88317655f9cb7da94716772cd01ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/74735ea26658681403e07d1ef1b8495f26179ada"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/74735ea26658681403e07d1ef1b8495f26179ada"}]},"branch":"refs/heads/master"},"79b6666a36293f58c2ed60e893855fd9e41d2143":{"kind":"REWORK","_number":4,"created":"2021-02-18 01:23:34.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/83/755183/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/83/755183/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/4"}}},"commit":{"parents":[{"commit":"d2138765888fe16e68f713ed99121f41e01a3c51","subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-09-30 05:50:37.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-02-18 01:22:58.000000000","tz":480},"subject":"Remove all usage of six library","message":"Remove all usage of six library\n\nReplace six with Python 3 style code.\n\nChange-Id: I724eba28d1a88317655f9cb7da94716772cd01ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/79b6666a36293f58c2ed60e893855fd9e41d2143"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/79b6666a36293f58c2ed60e893855fd9e41d2143"}]},"branch":"refs/heads/master"},"7a1a0fd8a28b6c82975fab46526f6b4aebaea662":{"kind":"REWORK","_number":5,"created":"2021-12-02 01:48:33.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/83/755183/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/83/755183/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/5"}}},"commit":{"parents":[{"commit":"8866127d8b9bc7c2811c9025be8acf2228adb0f8","subject":"Add unstable tag on some replication tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8866127d8b9bc7c2811c9025be8acf2228adb0f8"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-09-30 05:50:37.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2021-12-02 01:48:00.000000000","tz":480},"subject":"Remove all usage of six library","message":"Remove all usage of six library\n\nReplace six with Python 3 style code.\n\nChange-Id: I724eba28d1a88317655f9cb7da94716772cd01ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7a1a0fd8a28b6c82975fab46526f6b4aebaea662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7a1a0fd8a28b6c82975fab46526f6b4aebaea662"}]},"branch":"refs/heads/master"},"6878774d8df652015f5b4f5ee2a596338c64cba0":{"kind":"REWORK","_number":6,"created":"2021-12-02 07:21:08.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/83/755183/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/83/755183/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/6"}}},"commit":{"parents":[{"commit":"8866127d8b9bc7c2811c9025be8acf2228adb0f8","subject":"Add unstable tag on some replication tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8866127d8b9bc7c2811c9025be8acf2228adb0f8"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-09-30 05:50:37.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2021-12-02 07:20:54.000000000","tz":480},"subject":"Remove all usage of six library","message":"Remove all usage of six library\n\nReplace six with Python 3 style code.\n\nChange-Id: I724eba28d1a88317655f9cb7da94716772cd01ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6878774d8df652015f5b4f5ee2a596338c64cba0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6878774d8df652015f5b4f5ee2a596338c64cba0"}]},"branch":"refs/heads/master"},"48895819769e94ae050a08275fa6a9606ef16b73":{"kind":"REWORK","_number":7,"created":"2021-12-09 08:51:11.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/83/755183/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/83/755183/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/755183/7"}}},"commit":{"parents":[{"commit":"8866127d8b9bc7c2811c9025be8acf2228adb0f8","subject":"Add unstable tag on some replication tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8866127d8b9bc7c2811c9025be8acf2228adb0f8"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-09-30 05:50:37.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2021-12-09 08:50:57.000000000","tz":480},"subject":"Remove all usage of six library","message":"Remove all usage of six library\n\nReplace six with Python 3 style code.\n\nChange-Id: I724eba28d1a88317655f9cb7da94716772cd01ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/48895819769e94ae050a08275fa6a9606ef16b73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/48895819769e94ae050a08275fa6a9606ef16b73"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
