)]}'
{"id":"openstack%2Fmanila-tempest-plugin~755859","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I9cbd26016238c25487ac8104c1188cd2cf4f467e","project":"openstack/manila-tempest-plugin","branch":"master","topic":"ordered-ddt-data","hashtags":[],"change_id":"I9cbd26016238c25487ac8104c1188cd2cf4f467e","subject":"Preserve the items order when calling ddt.data()","status":"MERGED","created":"2020-10-02 18:11:49.000000000","updated":"2020-10-07 00:57:10.000000000","submitted":"2020-10-07 00:57:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755859-1602032230870-3e89e6fb","meta_rev_id":"b2c73b848e7e621c9fb02065bf90ea553d79081d","_number":755859,"virtual_id_number":755859,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-10-03 17:41:30.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-07 00:57:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2020-10-05 17:20:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-06 23:55:02.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-06 23:55:02.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-03 17:41:30.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-05 17:20:48.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-10-06 23:55:02.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-10-07 00:57:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"30cd010f687bfc4cd2972dbfe3d290bb6801eb94","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-10-02 18:11:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58036cac23963015c253b7ecefbbf5248ab5873b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 18:58:02.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4373b54afe3949919a25b0f4ec9aeeea : FAILURE in 3m 35s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cc288c0e9f534a79943133794ead12ac : FAILURE in 23m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3c6ef7cea53b44939465ba6c472a8878 : FAILURE in 16m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d78623225a1946ffba4ff4a9af34f004 : FAILURE in 33m 49s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/806fa2646acb4ee3ba07ab9f9fc9cb09 : FAILURE in 28m 10s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f7725e4ff48446468a5b37165c6c0cfd : FAILURE in 36m 44s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/312e8fad88fd4df9a8721353246f7180 : FAILURE in 27m 29s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/171359803e6147c29b4bf25bdc65e84d : FAILURE in 33m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/af1b7ce42b8e487990c7ff00a2a7a535 : FAILURE in 33m 49s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/930188a16eb54511a805fe94b8018417 : FAILURE in 22m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b4eea87078a91155d108e8bb689694eac135d109","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-10-02 19:27:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"285df1fffe458abb440b5737ee5be4454f8aa0e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 21:20:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1b9650c972b4069be30ef1d58b4fccb : SUCCESS in 3m 22s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/defeae43d75e4aa8a695272b90ff2617 : FAILURE in 34m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8166164ca3bd4dfcbd59661fa25f7dab : FAILURE in 36m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/08ff2412af64444f980dada9b91a6e6c : FAILURE in 1h 11m 55s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8db4b9dcc0d3447c9b2c47c059908269 : FAILURE in 1h 15m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/54558de25d084d4584deaa2b1e14a2d3 : FAILURE in 1h 09m 53s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/116451041eaa4d28a79003e5c0c9894a : FAILURE in 1h 00m 41s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a4ea44ace93649b6bda7a8a17cc447aa : FAILURE in 46m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e127410c9142422abf4307f988139220 : FAILURE in 1h 45m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c4711b4b77c44114bff74ddd80a950c0 : FAILURE in 24m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3cadaabc888a0664f8e8b49c90908b6612e87723","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-10-03 13:19:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e6b47b16bb7e6370b58c2484a07f9a4d1fe4a2c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 15:17:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98b412a8ed15496ca745730553600bf3 : SUCCESS in 4m 17s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9009decef4fe4242bb2888be62052728 : SUCCESS in 31m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/efecb31d943b4459af574c85dc2e22a3 : SUCCESS in 43m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a1916299868240fba7d2b40534d01144 : SUCCESS in 1h 07m 39s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/07a2599f670d4a5cb85671be82907307 : FAILURE in 1h 06m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1f825778d72042339578fb6fddd74ce8 : SUCCESS in 1h 11m 26s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e6ed6c31c81b4ca99dcbe9c3c1746e7b : FAILURE in 54m 23s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bc01c3b7591848f0a29217c82fff59a4 : SUCCESS in 52m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d55eef7c7e3b4e9891334088a7880e03 : FAILURE in 1h 52m 29s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0b309f8811c24d7fae030224a0ce58ac : SUCCESS in 25m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"39ff1f13e1c1299408790fd1b17568b03407fce7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-10-03 17:41:30.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/59/755859/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/d65bbf3/ : SUCCESS in 2h 23m 13s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/59/755859/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/4c6a5c4/ : SUCCESS in 1h 41m 01s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"523129510071f6a07117d92cc07c8a8ff3a5f45d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-10-05 17:20:48.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks, LGTM.","accounts_in_message":[],"_revision_number":3},{"id":"3fbfd4670a33721f0a95c1b5618ef71b2c57e8e8","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-10-06 23:55:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThanks, Luigi, good stuff!","accounts_in_message":[],"_revision_number":3},{"id":"16c3bf5c9479b6b766440629a8990f0fdd6651b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 23:55:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ba7f5c70359e1702a86599d426c73201f0239a29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 00:57:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3999169360ee4afe8c700c67c30fdc93 : SUCCESS in 3m 00s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/05d401ce654e44fd8f6b98e904436089 : SUCCESS in 29m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0ccfe5f1b63c47018c5ac4d7b4422593 : SUCCESS in 31m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7ba8b44ba5ff43e1b2c5f2ed92bd34f5 : SUCCESS in 1h 00m 51s","accounts_in_message":[],"_revision_number":3},{"id":"60c8e9696ab70c7f3a5ebf3ea3fab3a1f60fefcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 00:57:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d91870b8e63f2a2301af645fd8632601a67ec742","revisions":{"2945aadd17d842422fd9e63e798e14c9e2288f3c":{"kind":"REWORK","_number":1,"created":"2020-10-02 18:11:49.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/59/755859/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/59/755859/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/755859/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/755859/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/755859/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/755859/1"}}},"commit":{"parents":[{"commit":"9ff6b341370b10a80fddc6d1fd70c6375542b846","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9ff6b341370b10a80fddc6d1fd70c6375542b846"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-10-02 17:53:44.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-10-02 17:53:44.000000000","tz":120},"subject":"Preserve the items order when calling ddt.data()","message":"Preserve the items order when calling ddt.data()\n\nSeveral calls to ddt.data() change the lists of items into a set\nto de-duplicate them. This happens, for example, when the list\nconstains microversions which may come from some variable.\n\nIn order to generate a consistent and predictable list of tests,\nwhich can be compared over time (as ddt adds the index of the\ntest to the test name), replace the usage of set with\na new function which removes the duplicates but keeps the order.\n\nChange-Id: I9cbd26016238c25487ac8104c1188cd2cf4f467e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2945aadd17d842422fd9e63e798e14c9e2288f3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2945aadd17d842422fd9e63e798e14c9e2288f3c"}]},"branch":"refs/heads/master"},"c495d5bfe61ad55b8600628177602ed3abaa3a70":{"kind":"REWORK","_number":2,"created":"2020-10-02 19:27:21.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/59/755859/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/59/755859/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/755859/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/755859/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/755859/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/755859/2"}}},"commit":{"parents":[{"commit":"9ff6b341370b10a80fddc6d1fd70c6375542b846","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9ff6b341370b10a80fddc6d1fd70c6375542b846"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-10-02 19:24:53.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-10-02 19:24:53.000000000","tz":120},"subject":"Preserve the items order when calling ddt.data()","message":"Preserve the items order when calling ddt.data()\n\nSeveral calls to ddt.data() change the lists of items into a set\nto de-duplicate them. This happens, for example, when the list\nconstains microversions which may come from some variable.\n\nIn order to generate a consistent and predictable list of tests,\nwhich can be compared over time (as ddt adds the index of the\ntest to the test name), replace the usage of set with\na new function which removes the duplicates but keeps the order.\n\nChange-Id: I9cbd26016238c25487ac8104c1188cd2cf4f467e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c495d5bfe61ad55b8600628177602ed3abaa3a70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c495d5bfe61ad55b8600628177602ed3abaa3a70"}]},"branch":"refs/heads/master"},"d91870b8e63f2a2301af645fd8632601a67ec742":{"kind":"REWORK","_number":3,"created":"2020-10-03 13:19:18.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/59/755859/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/59/755859/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/755859/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/755859/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/755859/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/59/755859/3"}}},"commit":{"parents":[{"commit":"9ff6b341370b10a80fddc6d1fd70c6375542b846","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9ff6b341370b10a80fddc6d1fd70c6375542b846"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-10-03 13:17:23.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-10-03 13:17:23.000000000","tz":120},"subject":"Preserve the items order when calling ddt.data()","message":"Preserve the items order when calling ddt.data()\n\nSeveral calls to ddt.data() change the lists of items into a set\nto de-duplicate them. This happens, for example, when the list\nconstains microversions which may come from some variable.\n\nIn order to generate a consistent and predictable list of tests,\nwhich can be compared over time (as ddt adds the index of the\ntest to the test name), replace the usage of set with\na new function which removes the duplicates but keeps the order.\n\nChange-Id: I9cbd26016238c25487ac8104c1188cd2cf4f467e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d91870b8e63f2a2301af645fd8632601a67ec742"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d91870b8e63f2a2301af645fd8632601a67ec742"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
