)]}'
{"id":"openstack%2Fmanila-tempest-plugin~756570","triplet_id":"openstack%2Fmanila-tempest-plugin~master~If1cbcf01595028e7c44620e989b91fd79db50b35","project":"openstack/manila-tempest-plugin","branch":"master","topic":"756570","hashtags":[],"change_id":"If1cbcf01595028e7c44620e989b91fd79db50b35","subject":"[ci] Copy ganesha logs in the ceph-nfs job","status":"MERGED","created":"2020-10-07 18:51:32.000000000","updated":"2020-10-09 00:15:29.000000000","submitted":"2020-10-09 00:15:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756570-1602202529497-48b48ccc","meta_rev_id":"67458c6e568c52468d68c309fcc20332c8b9c67a","_number":756570,"virtual_id_number":756570,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-10-08 22:52:17.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-10-08 00:16:52.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-09 00:15:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-08 23:10:21.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-08 23:41:16.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-08 23:10:21.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-08 00:16:52.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-08 23:10:21.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-10-08 23:41:16.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-10-09 00:15:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a28aeff114b9e6e26e13674caa8b99b1e1c8ddc","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-10-07 18:51:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18954c53ee5fbaa77dcee23ccf76efe908c2e19b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-10-07 19:11:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0c10c7619bdfa5180affcaea922c8924deefd702","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 21:45:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/662657ba7a244d4fa831d32871fec8b1 : SUCCESS in 3m 17s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3c5a37b4d8e240fd8ce9b1ad35608cea : SUCCESS in 31m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/44e84495deab47638c0fac3d1bf34a9a : SUCCESS in 36m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4d00345707564ce98f0c3b21304f7837 : SUCCESS in 1h 16m 21s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8e9ff13dc5404c20bfedb59bf3719583 : FAILURE in 1h 01m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/405c983108ed4b6190847132f25d58d1 : SUCCESS in 1h 03m 51s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1fe1ba852a914fbb9f240c743cec107c : FAILURE in 56m 25s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/aa3769104ab24152aeecc6ab0d036aed : SUCCESS in 43m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/519ad494e8fa48178390d833a925fd22 : FAILURE in 2h 26m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5a27217736864aaa84a8dc53d67dcaae : SUCCESS in 27m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b02431c5502124cb31254fb33977ece7d8b189ed","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-10-08 00:16:52.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/70/756570/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/ac086de/ : FAILURE in 2h 31m 21s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/70/756570/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/770e04d/ : SUCCESS in 1h 42m 42s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"b2af97e5016794738884c44d108e6e9c55712b88","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-10-08 22:52:17.000000000","message":"Patch Set 2:\n\nTom, Victoria: lets get this merged?","accounts_in_message":[],"_revision_number":2},{"id":"1738fa1e93f6ea573ccf34f3ed0f9a4f2cafd01b","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-10-08 23:10:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThe logs are captured, let\u0027s do this.  Thanks, Goutham!","accounts_in_message":[],"_revision_number":2},{"id":"2cb9930da93a17c3217e922f503d4428a6e181d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 23:15:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"23558aba00cff6ad222698a11da6d0bf4c8f6abb","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-10-08 23:41:16.000000000","message":"Patch Set 2: Code-Review+2\n\nLate +2. Already being tested on https://review.opendev.org/#/c/756127/","accounts_in_message":[],"_revision_number":2},{"id":"8907dbe21a11195a4779973eb572121e4fe8e73d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 00:15:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3928235c9dc219cf673ac4ee73da7e7cfdf83386","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 00:15:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ebb8c7bb86748eabd6c06be5b6ab79e : SUCCESS in 3m 56s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3e8cdedd89ce4491ae2f5eaff279ff5a : SUCCESS in 30m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e6b13f836d9c4ede963202c29cb1e7f5 : SUCCESS in 36m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ad7e90607cdf4b3ba61e5809d465e969 : SUCCESS in 57m 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a4042648ee3874dd5d281561fe0647e014010e67","revisions":{"5951969b5656d0a38ec2bec37b0f2572c659ab5b":{"kind":"REWORK","_number":1,"created":"2020-10-07 18:51:32.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/70/756570/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/70/756570/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/70/756570/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/70/756570/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/70/756570/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/70/756570/1"}}},"commit":{"parents":[{"commit":"e0c51e81c30c7fbe997e391d00eb06facc792078","subject":"Keep few integration jobs running on Bionic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e0c51e81c30c7fbe997e391d00eb06facc792078"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-10-07 18:51:24.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-10-07 18:51:24.000000000","tz":-420},"subject":"[ci] Copy ganesha logs in the ceph-nfs job","message":"[ci] Copy ganesha logs in the ceph-nfs job\n\nChange-Id: If1cbcf01595028e7c44620e989b91fd79db50b35\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5951969b5656d0a38ec2bec37b0f2572c659ab5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5951969b5656d0a38ec2bec37b0f2572c659ab5b"}]},"branch":"refs/heads/master"},"a4042648ee3874dd5d281561fe0647e014010e67":{"kind":"REWORK","_number":2,"created":"2020-10-07 19:11:45.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/70/756570/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/70/756570/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/70/756570/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/70/756570/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/70/756570/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/70/756570/2"}}},"commit":{"parents":[{"commit":"e0c51e81c30c7fbe997e391d00eb06facc792078","subject":"Keep few integration jobs running on Bionic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e0c51e81c30c7fbe997e391d00eb06facc792078"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-10-07 18:51:24.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-10-07 19:11:38.000000000","tz":-420},"subject":"[ci] Copy ganesha logs in the ceph-nfs job","message":"[ci] Copy ganesha logs in the ceph-nfs job\n\nChange-Id: If1cbcf01595028e7c44620e989b91fd79db50b35\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a4042648ee3874dd5d281561fe0647e014010e67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a4042648ee3874dd5d281561fe0647e014010e67"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
