)]}'
{"id":"openstack%2Fmanila-tempest-plugin~757024","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I56acb90a3a5d863ec0ee2d8a058c57672bd29179","project":"openstack/manila-tempest-plugin","branch":"master","topic":"manila2","hashtags":[],"change_id":"I56acb90a3a5d863ec0ee2d8a058c57672bd29179","subject":"add testcase for test access rules","status":"NEW","created":"2020-10-09 08:36:06.000000000","updated":"2020-10-10 10:34:33.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"70b5d55f36691a8aa5df8ba794ec361a4ed62a2a","_number":757024,"virtual_id_number":757024,"owner":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-10-10 05:54:14.000000000","_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},{"date":"2020-10-10 10:34:33.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-10 08:13:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-10 08:13:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-10 10:34:33.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c970f91e1c235a944dd2718fd6f5e700e88da760","author":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"date":"2020-10-09 08:36:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"181041c0f4b822e87328fda32e8cbcd6cdd9387d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 11:04:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f30a47d42e004945ac4757d94282ea2c : SUCCESS in 4m 27s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cbd9f53aff5b44ce9d15c1359a06569b : FAILURE in 45m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d81d44b6d4b842b8bd9b71c18f364d5b : FAILURE in 44m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/285dc04926be4e59ae46f1d4f301a5f0 : SUCCESS in 1h 10m 53s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/69aae428a8ca4bb09ad67c81ae03ac70 : SUCCESS in 1h 04m 42s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f2068be672684a8fb2fbbbd61e989df1 : FAILURE in 1h 15m 13s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/106050f585444509a98a75bb81c7107c : FAILURE in 55m 29s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d2b9332016e747eeac27e85bce3e7408 : FAILURE in 49m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9e174839ba764a998948ee39f4f1dc37 : FAILURE in 2h 18m 47s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/dfe4c475671047deb12cb79923df07d6 : FAILURE in 24m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"25c133ace1414ce8a73dfcea4c6288dafd09fa20","author":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"date":"2020-10-10 01:42:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"72a95387873700d019ebf95526baaf86b104a5fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 03:59:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d6fe17206794e659b0b9fd6a4aaf263 : SUCCESS in 3m 26s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/53e9c3946d5149fe9696b0cc55bbf041 : SUCCESS in 29m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0e0eb56dd6cf4fa5b3a4ff4e1a09104d : SUCCESS in 35m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2591f2a24ff54950becf03f5d242864d : FAILURE in 1h 18m 09s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/108744acc85740f18bd60232aa16d134 : SUCCESS in 1h 13m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/2169de3842264367a16d780ce1bf393d : SUCCESS in 1h 01m 31s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d60296315d45469ab6bae34d5e68eee3 : FAILURE in 1h 01m 55s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/52c775e6783448479a3f6b2634f8f191 : SUCCESS in 52m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/78458fe1361a46219ad9a207c8e0388d : FAILURE in 1h 50m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f3fcb51474894d4aaa4855e41ebe41c5 : SUCCESS in 28m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4e1b302180db51d9ac4d167adccf4ae406e5d459","author":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"date":"2020-10-10 05:54:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5b94be8c60c7bc8b36addb3eb28a2749c415ea8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 08:13:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c899ed2330b42f78ba1559bd021f472 : SUCCESS in 3m 25s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9564f3e444054cfc983b35ac2381891d : SUCCESS in 30m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3ebe2b034f134d759d6e7fefa1a4c5b8 : SUCCESS in 36m 03s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e7654ec47e134a25af544c7e4b9f460b : SUCCESS in 1h 15m 37s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/51af9d211162400ca5ecd8d8d346852b : SUCCESS in 1h 00m 35s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/eef2f9295a0c4146b8a0fdc49f34a340 : SUCCESS in 1h 09m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/05ae62560b92441181576d3dbf7884fc : FAILURE in 58m 25s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2c09e31cced44a61ae797da9a0069f18 : SUCCESS in 46m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d1123f7b2af7486abae7c7a0d7e14156 : FAILURE in 2h 16m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6bd7d49b9b604bd38289c3d8dd8ff9b4 : FAILURE in 24m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"890764414061eda38aa748b5deabf1013c13b222","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-10-10 10:34:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/24/757024/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/836b2ff/ : SUCCESS in 2h 20m 25s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/24/757024/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/fd659eb/ : FAILURE in 50m 25s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bd410f8250f7a19b908dc4a4f6e923fb20a63e4b","revisions":{"59d5b1eaa4bbc14cddf94ff2fd206d71874e8d09":{"kind":"REWORK","_number":1,"created":"2020-10-09 08:36:06.000000000","uploader":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"ref":"refs/changes/24/757024/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/24/757024/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/757024/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/757024/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/757024/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/757024/1"}}},"commit":{"parents":[{"commit":"3a44bf7401d85b5dea97aa63145bf1f8b1a6e0a9","subject":"Merge \"[ci] Copy ganesha logs in the ceph-nfs job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3a44bf7401d85b5dea97aa63145bf1f8b1a6e0a9"}]}],"author":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-10-09 08:34:02.000000000","tz":480},"committer":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-10-09 08:34:09.000000000","tz":480},"subject":"add testcase for test access rules","message":"add testcase for test access rules\n\nAdd case list access rules for file test_access_rules_metadata.py\n\nChange-Id: I56acb90a3a5d863ec0ee2d8a058c57672bd29179\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/59d5b1eaa4bbc14cddf94ff2fd206d71874e8d09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/59d5b1eaa4bbc14cddf94ff2fd206d71874e8d09"}]},"branch":"refs/heads/master"},"bd410f8250f7a19b908dc4a4f6e923fb20a63e4b":{"kind":"REWORK","_number":2,"created":"2020-10-10 01:42:30.000000000","uploader":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"ref":"refs/changes/24/757024/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/24/757024/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/757024/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/757024/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/757024/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/757024/2"}}},"commit":{"parents":[{"commit":"3a44bf7401d85b5dea97aa63145bf1f8b1a6e0a9","subject":"Merge \"[ci] Copy ganesha logs in the ceph-nfs job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3a44bf7401d85b5dea97aa63145bf1f8b1a6e0a9"}]}],"author":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-10-09 08:34:02.000000000","tz":480},"committer":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-10-10 01:42:20.000000000","tz":480},"subject":"add testcase for test access rules","message":"add testcase for test access rules\n\nAdd case list access rules for file test_access_rules_metadata.py\n\nChange-Id: I56acb90a3a5d863ec0ee2d8a058c57672bd29179\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bd410f8250f7a19b908dc4a4f6e923fb20a63e4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bd410f8250f7a19b908dc4a4f6e923fb20a63e4b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
