)]}'
{"id":"openstack%2Fmanila-tempest-plugin~760245","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I142038f39c611df70aacc387a6225f28040c42c0","project":"openstack/manila-tempest-plugin","branch":"master","topic":"fix-ipv6-cephfsnfs","hashtags":[],"change_id":"I142038f39c611df70aacc387a6225f28040c42c0","subject":"Enable IPV6 in the CephFS NFS job","status":"ABANDONED","created":"2020-10-28 21:20:22.000000000","updated":"2021-01-25 10:55:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5458f2cdddc4affd64b61031c6c5aaf237c7fac3","_number":760245,"virtual_id_number":760245,"owner":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-11-16 18:56:51.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-16 17:43:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-16 20:18:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"all":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-11-17 12:45:11.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-29 04:28:03.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 17:43:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 18:56:51.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 20:18:39.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"4eb2ab19e84e9d739f5ee5af300b414dc1de16ea","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-10-28 21:20:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6cd02faba7b87c6fb11176aed0d57276fa38a0e","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-10-28 22:31:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"87fc7553a29aeaed13fa276d336786948605ea62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 00:24:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ed620729703470aa197ea6d5b5a5531 : SUCCESS in 3m 03s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/66388490b19c443ab7ab0cbbf934981f : SUCCESS in 32m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a2d01e3a03bd448991fa5342aef12d5f : SUCCESS in 42m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/28b9cb3efed246b4b5d74f7c652dd77b : SUCCESS in 1h 05m 38s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/59271ff3210a4a1d8741249dfaabf892 : FAILURE in 1h 17m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/34ac079c855c43e3905f68040a7a3eab : SUCCESS in 1h 02m 56s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e72d86d0b83a4e178ef68268ff2c402a : SUCCESS in 45m 38s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f152a9154c7041ae9657c5cf5519bf34 : SUCCESS in 40m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/037e013efa6f49eaa957de07d4e7ee00 : FAILURE in 1h 44m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/786612dd36064dec92233dc67d573e78 : SUCCESS in 24m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"75284654bfd95705efcd3adeed550cc85bf65bb5","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-10-29 01:32:03.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/45/760245/2/check/DellEMC_VNX_Manila/9429a50/DellEMC_VNX_Manila/10 : SUCCESS in 1h 07m 11s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"9ef3a659d5b90188636300522cdaa3e5c485be2d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-10-29 04:28:03.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/45/760245/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/b45d8af/ : FAILURE in 3h 50m 13s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/45/760245/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/1a67543/ : SUCCESS in 1h 45m 32s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"b70546b8b0419ec3ac2321b197a92fd37216894c","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-10-29 22:51:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a2e1b18910d78149a33115ff09795a6a9c917460","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-10-29 22:52:11.000000000","message":"Topic changed from disable-ipv6-cephfsnfs to fix-ipv6-cephfsnfs","accounts_in_message":[],"_revision_number":3},{"id":"1d7cf47733578bd7418f7806a9c3f513fc0d75a4","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-10-29 22:54:21.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"fb2641d0078955b945e0a99b0348127ef88e7f46","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-10-29 22:55:53.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"9f5fb4706788a65ce6a930f53375f22748c5730b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 02:01:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e75804a5e26445459f19a672821b8d91 : SUCCESS in 3m 53s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8f61bb3a44cc43c680499953539d4c74 : SUCCESS in 30m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b83ead33382e40588a1d0513246aac23 : SUCCESS in 35m 23s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c15704b947ec4fbe9a499ac8c2ee5b57 : FAILURE in 1h 03m 41s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6653b71db79a41dca3aae07ec8d7bc41 : SUCCESS in 1h 26m 20s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1fa2e9016ad3494cae54052cafe16e28 : SUCCESS in 1h 04m 06s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6e95cfe48a044fa08259569d573550ea : TIMED_OUT in 3h 02m 21s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/549077ba0ddc403ea079646cef8e6267 : SUCCESS in 40m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2568a5c4aabb44ac9ec3718d669827de : FAILURE in 2h 28m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/866b9cc872864e0997583517681550b4 : FAILURE in 25m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2032afa7bd29ae4dc96c3cfd5d97f9e923f65db2","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-11-16 10:41:09.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"4a3f52117376ff17a40335f86ff817f5cc6773f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 13:57:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9175f305fe0c4deb9ccdeb1c104e5f9a : SUCCESS in 3m 42s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/97c80ce8ea504cf89efede6fe462f352 : SUCCESS in 40m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/780d345d08b54ca8a7c44b46156f1225 : SUCCESS in 45m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9f12eeefc4a0463fabe9da3490642859 : SUCCESS in 1h 06m 16s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b2ab21cf3c904bc2a474bb3d387eefba : SUCCESS in 1h 23m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/122dad15b11f43fcb6dec88ecca15c4a : SUCCESS in 1h 10m 19s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8b980c9e767f4cbabd30654e0277c62a : TIMED_OUT in 3h 02m 16s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ce7392736d694522adcca1763bce14a9 : SUCCESS in 41m 14s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/30c69fd437cd47dda040c3e410354634 : FAILURE in 2h 01m 59s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/58e731af34034705b5190dab4889ff01 : FAILURE in 21m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"906bafbf402a0ade9e793fba6307ca1bd3d52ea2","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-11-16 15:09:07.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/45/760245/5/check/DellEMC_VNX_Manila/0efdb13/DellEMC_VNX_Manila/43 : SUCCESS in 1h 11m 36s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"618f871dc81a8aaa8d8d88ad9f6d7f27cbdfe81f","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-11-16 15:10:19.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"63a69182b10e7b1d24ea04d3d09fd114430e1382","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 17:43:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aac486d63aed47f0869accc91057d4dd : SUCCESS in 3m 04s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a4ef76159db245db8224b28889f5f0ea : SUCCESS in 36m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a57f1817c68e451bac75060ae089188e : SUCCESS in 33m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/64703d70b2c743ef9de2ee332568f23a : SUCCESS in 1h 01m 50s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8fe87ad135b441359e6a8c9e0f820cab : FAILURE in 55m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b6564c79d893439c877d4d639e49e776 : SUCCESS in 1h 30m 29s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/dd5f0812d5bc4a62a2f99d6e5209a887 : FAILURE in 1h 11m 51s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7241896410b24bc9a17a138c6ddc1709 : SUCCESS in 44m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/24e252ba6be049959c627f2c310054f2 : FAILURE in 2h 30m 11s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7275ba7874a24926b94d01b548824d58 : SUCCESS in 20m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b789d2196dc56d68421e3fd7e9c19ad6f0f66c81","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-11-16 18:56:51.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/45/760245/6/check/DellEMC_VNX_Manila/59162a0/DellEMC_VNX_Manila/45 : SUCCESS in 1h 13m 14s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":6},{"id":"c721452bf3c516726a438977894c0f1fdf941079","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-16 20:18:39.000000000","message":"Patch Set 6: Code-Review+2\n\nAs discussed in person, this change is fine; we don\u0027t really need to test the IPv6 control plane in all the IPv6 test jobs, but no harm doing that :)","accounts_in_message":[],"_revision_number":6},{"id":"3aa5ecd5e6d107e64bab9b845a88b019cdd10d07","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-11-17 12:45:11.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"5458f2cdddc4affd64b61031c6c5aaf237c7fac3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-01-25 10:55:19.000000000","message":"Abandoned\n\nNot actually needed change","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"69b533bb2d37dd782843ccdcfd96bec6c08805fa","revisions":{"75d4e34e3872b7a059bc840dccbaf9dd645ef5c5":{"kind":"REWORK","_number":1,"created":"2020-10-28 21:20:22.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/45/760245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/45/760245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/1"}}},"commit":{"parents":[{"commit":"0d523bb8d4cd62186616103b59bba7d933d2cee4","subject":"Add tests for share server migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0d523bb8d4cd62186616103b59bba7d933d2cee4"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-10-28 21:19:49.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-10-28 21:19:49.000000000","tz":-180},"subject":"[DNM] Testing CephFS NFS scenario without IPv6","message":"[DNM] Testing CephFS NFS scenario without IPv6\n\nChange-Id: I142038f39c611df70aacc387a6225f28040c42c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/75d4e34e3872b7a059bc840dccbaf9dd645ef5c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/75d4e34e3872b7a059bc840dccbaf9dd645ef5c5"}]},"branch":"refs/heads/master"},"90b3c50558e7b9edbffa5ed2797a49d81e68d0ab":{"kind":"REWORK","_number":2,"created":"2020-10-28 22:31:29.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/45/760245/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/45/760245/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/2"}}},"commit":{"parents":[{"commit":"0d523bb8d4cd62186616103b59bba7d933d2cee4","subject":"Add tests for share server migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0d523bb8d4cd62186616103b59bba7d933d2cee4"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-10-28 21:19:49.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-10-28 22:31:18.000000000","tz":-180},"subject":"[DNM] Testing CephFS NFS scenario without IPv6","message":"[DNM] Testing CephFS NFS scenario without IPv6\n\nChange-Id: I142038f39c611df70aacc387a6225f28040c42c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/90b3c50558e7b9edbffa5ed2797a49d81e68d0ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/90b3c50558e7b9edbffa5ed2797a49d81e68d0ab"}]},"branch":"refs/heads/master"},"e4486a11213f1444530d1e72b92ddf093ac773f6":{"kind":"REWORK","_number":3,"created":"2020-10-29 22:51:38.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/45/760245/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/45/760245/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/3"}}},"commit":{"parents":[{"commit":"3a44bf7401d85b5dea97aa63145bf1f8b1a6e0a9","subject":"Merge \"[ci] Copy ganesha logs in the ceph-nfs job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3a44bf7401d85b5dea97aa63145bf1f8b1a6e0a9"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-10-28 21:19:49.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-10-29 22:48:25.000000000","tz":-180},"subject":"Enable IPV6 in the CephFS NFS job","message":"Enable IPV6 in the CephFS NFS job\n\nThe CephFS NFS job was running ipv6 tests but we were not\nsetting Devstack properly. Precisely, we were missing the\nSERVICE_IP_VERSION\u003d\"6\" flag in the local.conf file.\n\nIn this patch set we fix this by inheriting from\nthe manila-tempest-plugin-ipv6-base job.\n\nChange-Id: I142038f39c611df70aacc387a6225f28040c42c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e4486a11213f1444530d1e72b92ddf093ac773f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e4486a11213f1444530d1e72b92ddf093ac773f6"}]},"branch":"refs/heads/master"},"489a4207474f5e4a3422e6484f3aa08a1c29ae97":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-10-29 22:54:21.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/45/760245/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/45/760245/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/4"}}},"commit":{"parents":[{"commit":"3a44bf7401d85b5dea97aa63145bf1f8b1a6e0a9","subject":"Merge \"[ci] Copy ganesha logs in the ceph-nfs job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3a44bf7401d85b5dea97aa63145bf1f8b1a6e0a9"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-10-28 21:19:49.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-10-29 22:52:56.000000000","tz":-180},"subject":"Enable IPV6 in the CephFS NFS job","message":"Enable IPV6 in the CephFS NFS job\n\nThe CephFS NFS job was running ipv6 tests but we were not\nsetting Devstack properly. Precisely, we were missing the\nSERVICE_IP_VERSION\u003d\"6\" flag in the local.conf file.\n\nIn this patch set we fix this by inheriting from\nthe manila-tempest-plugin-ipv6-base job.\n\nDepends-On: I47c1b18c29ea80e87addf06ee3ff4915d07303e8\n\nChange-Id: I142038f39c611df70aacc387a6225f28040c42c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/489a4207474f5e4a3422e6484f3aa08a1c29ae97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/489a4207474f5e4a3422e6484f3aa08a1c29ae97"}]},"branch":"refs/heads/master"},"8d0fd351d7fc7858e62a36c086089a889033b67b":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-10-29 22:55:53.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/45/760245/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/45/760245/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/5"}}},"commit":{"parents":[{"commit":"3a44bf7401d85b5dea97aa63145bf1f8b1a6e0a9","subject":"Merge \"[ci] Copy ganesha logs in the ceph-nfs job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3a44bf7401d85b5dea97aa63145bf1f8b1a6e0a9"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-10-28 21:19:49.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-10-29 22:55:32.000000000","tz":-180},"subject":"Enable IPV6 in the CephFS NFS job","message":"Enable IPV6 in the CephFS NFS job\n\nThe CephFS NFS job was running ipv6 tests but we were not\nsetting Devstack properly. Precisely, we were missing the\nSERVICE_IP_VERSION\u003d\"6\" flag in the local.conf file.\n\nIn this patch set we fix this by inheriting from\nthe manila-tempest-plugin-ipv6-base job.\n\nChange-Id: I142038f39c611df70aacc387a6225f28040c42c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8d0fd351d7fc7858e62a36c086089a889033b67b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8d0fd351d7fc7858e62a36c086089a889033b67b"}]},"branch":"refs/heads/master"},"69b533bb2d37dd782843ccdcfd96bec6c08805fa":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2020-11-16 15:10:19.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/45/760245/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/45/760245/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/760245/6"}}},"commit":{"parents":[{"commit":"3a44bf7401d85b5dea97aa63145bf1f8b1a6e0a9","subject":"Merge \"[ci] Copy ganesha logs in the ceph-nfs job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3a44bf7401d85b5dea97aa63145bf1f8b1a6e0a9"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-10-28 21:19:49.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-11-16 15:10:12.000000000","tz":0},"subject":"Enable IPV6 in the CephFS NFS job","message":"Enable IPV6 in the CephFS NFS job\n\nThe CephFS NFS job was running ipv6 tests but we were not\nsetting Devstack properly. Precisely, we were missing the\nSERVICE_IP_VERSION\u003d\"6\" flag in the local.conf file.\n\nIn this patch set we fix this by inheriting from\nthe manila-tempest-plugin-ipv6-base job.\n\nDepends-On: https://review.opendev.org/#/c/760397/\n\nChange-Id: I142038f39c611df70aacc387a6225f28040c42c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/69b533bb2d37dd782843ccdcfd96bec6c08805fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/69b533bb2d37dd782843ccdcfd96bec6c08805fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
