)]}'
{"id":"openstack%2Fmanila-tempest-plugin~762218","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ifb3cb87512c3d8f7cf3af44f7012b967d502bd8e","project":"openstack/manila-tempest-plugin","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"Ifb3cb87512c3d8f7cf3af44f7012b967d502bd8e","subject":"[ci] Switch generic driver job to focal fossa","status":"MERGED","created":"2020-11-10 23:24:56.000000000","updated":"2020-11-17 12:33:55.000000000","submitted":"2020-11-17 12:33:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762218-1605616435284-b0ade860","meta_rev_id":"66a25b60e9e17d52174016e901cd5a49b5afdb02","_number":762218,"virtual_id_number":762218,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-11-12 01:50:12.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-11-12 03:03:08.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-17 12:33:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-17 11:33:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-16 19:43:54.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-17 11:33:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-12 01:50:12.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-12 03:03:08.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 19:43:54.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-11-17 11:33:51.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-11-17 12:33:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d942d10b037d99979215b648e3012a801e5cefc","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-10 23:24:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be90e2f2c5f79e4c1869755aa5e1398dc0f0556a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-10 23:25:41.000000000","message":"Topic changed from bug/1896672 to migrate-to-focal","accounts_in_message":[],"_revision_number":1},{"id":"9e96628a9be7ea8aaccc0c1a35779ecdbd4eabb9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-10 23:27:55.000000000","message":"Patch Set 1: Workflow-1\n\nTesting","accounts_in_message":[],"_revision_number":1},{"id":"366624aca905b471218de4d2382efc6c9048cb09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 01:44:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95413ccac990470ba897f89b9914384d : SUCCESS in 3m 28s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9e6b1bcd601a4b49a01868b442a610d4 : SUCCESS in 30m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ce0e5bf2493a4dadbe1b5d754f443de6 : SUCCESS in 32m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/27c8eb0150194025bb5c3d4d7b1c7bae : SUCCESS in 1h 12m 04s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b9cb4939583e48758e6d064fc02b81ea : FAILURE in 1h 07m 53s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/16e55c92bc134899ae9b9692c77c2865 : SUCCESS in 1h 09m 14s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3ca0322260c546c1aff941c318ce484e : FAILURE in 1h 09m 05s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/664751e112e04ae4974799c89365f4b9 : SUCCESS in 42m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a1c0e7478c8e4c1dae528120054a7a6a : FAILURE in 2h 09m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8cb11806f76649f887c1c7bb093caa6f : SUCCESS in 24m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d160bfe785f27ed8d7aa14bcc496e3d8e44b968","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-11 04:26:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/18/762218/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/e70f9ed/ : SUCCESS in 2h 22m 36s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/18/762218/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/888889b/ : SUCCESS in 1h 42m 22s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"1e148ab6d8de776ef0a39bdaf5bb25185298fae0","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-11-11 04:36:27.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/18/762218/1/check/DellEMC_VNX_Manila/dfe46b9/DellEMC_VNX_Manila/30 : SUCCESS in 1h 10m 31s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"c0a9ee320ca192a6364c5720f132bc68afba8cbb","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-11 05:59:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"104a0b0ab19bc23fe3cf421e6771b3b040e79349","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 07:58:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d334b544a9904d3883228371c99dcd09 : SUCCESS in 3m 33s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e987d3c144104f6ab2a985075f58ee77 : SUCCESS in 41m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/79152aac8d8b4ac695f5f9f39d0c2295 : SUCCESS in 37m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5405a8ae72084a1f9cb5b8ddcf6b4b4e : SUCCESS in 1h 12m 31s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2d0ca178a666445a8175b35161581cda : SUCCESS in 1h 11m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/66811b77eda34849822bd0f9173a94ba : SUCCESS in 1h 34m 03s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/eaf18a43b3f14281b455e8a9962128a7 : FAILURE in 1h 07m 09s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fdb67a00bb4e4d0b8ab9586a6bb72339 : SUCCESS in 47m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1238ee4121cf4587a7c16611e811757c : FAILURE in 1h 51m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/733c0cc31bc3479ba2c0642ce21bbb44 : SUCCESS in 26m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a648f7b67f7e354541b150b16b9e34ae52b3653c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-11 08:29:26.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/18/762218/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/635bd6f/ : FAILURE in 2m 34s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/18/762218/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/5171394/ : FAILURE in 30m 47s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"654488540f0d1c6177fc38e54aacb72dbcae5fa7","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-11-11 09:04:52.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/18/762218/1/check/DellEMC_VNX_Manila/ed1e456/DellEMC_VNX_Manila/33 : SUCCESS in 1h 06m 18s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"efec89d35a470cf92a54382190b33591a71b3464","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-11 22:32:35.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"de2091564411d9a5a9ce4e471718e6c2338ffbcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 00:38:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bee1f7cc3ec74bbab473801f9b59dd1e : SUCCESS in 4m 29s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/14fa8da4545f4705ad965c627c30be20 : SUCCESS in 35m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4844862fe18a4594aab6939251e87693 : SUCCESS in 39m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e8a60c02485446249c720617a5c1491c : SUCCESS in 1h 09m 23s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3fe8f1447e6b47e7b9ba146a39eb2574 : FAILURE in 1h 03m 02s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/8b2905b2fc2641d88b0c1a0fa489df1a : SUCCESS in 58m 51s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4b0df12ec95f43aabff9bb5d00cd0a45 : FAILURE in 1h 00m 26s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/49f09adffe594719ad71b7efee5cb476 : SUCCESS in 44m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b30ce2868f99465aaa93da00f41dec6b : FAILURE in 2h 02m 49s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1171fc3ea6a44b179dd7f5b1e6180140 : SUCCESS in 21m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0b791843951112f9615bc4f108ea41334e993ec4","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-11-12 01:50:12.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/18/762218/2/check/DellEMC_VNX_Manila/d4905f8/DellEMC_VNX_Manila/35 : SUCCESS in 1h 11m 01s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"52632049b7fa210ef9bf59e832876e3ecbac6f44","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-12 03:03:08.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/18/762218/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/2a04228/ : SUCCESS in 2h 23m 53s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/18/762218/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/359a6b9/ : SUCCESS in 1h 51m 03s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"41e58cd1b0c547cc312b51176310ba8f74e5d892","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-11-16 19:43:54.000000000","message":"Patch Set 2: Code-Review+2\n\nYeah, we can switch and address the test that was already failing separately.","accounts_in_message":[],"_revision_number":2},{"id":"2400da21344ea9a052d2842ff1e92dba7670bf82","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-11-17 11:33:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nI agree, thanks!","accounts_in_message":[],"_revision_number":2},{"id":"9476f28de3a42cf16e8d3cad1d741b7044ac80ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 11:34:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8348b6a82df2cd606f19b7b2cb36e85d4fb76487","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 12:33:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95a2145109b340daa5a7dbb8e26b9d88 : SUCCESS in 2m 44s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d0de95712e124d10a5117515cde974ec : SUCCESS in 34m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b7b985190bc14c2d848acbb37fdaf97d : SUCCESS in 39m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8abeabf26a404243a168dc234c90678d : SUCCESS in 56m 36s","accounts_in_message":[],"_revision_number":2},{"id":"6505d4c2042304c499e07c951469f8301084ea99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 12:33:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5d7eaf0e57bf8e0a970f6c7b283bc88eacc672c4","revisions":{"ec2e81c252535451eda00182fb969d92dce3faa1":{"kind":"REWORK","_number":1,"created":"2020-11-10 23:24:56.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/18/762218/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/18/762218/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/762218/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/762218/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/762218/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/762218/1"}}},"commit":{"parents":[{"commit":"a79ba06596269233713c53e11d9f6d1fd86c693a","subject":"Use the same access rule for both instances","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a79ba06596269233713c53e11d9f6d1fd86c693a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-11-10 23:24:47.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-11-10 23:24:47.000000000","tz":-480},"subject":"[ci] Switch generic driver job to focal fossa","message":"[ci] Switch generic driver job to focal fossa\n\nChange-Id: Ifb3cb87512c3d8f7cf3af44f7012b967d502bd8e\nPartial-Bug: #1896672\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ec2e81c252535451eda00182fb969d92dce3faa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ec2e81c252535451eda00182fb969d92dce3faa1"}]},"branch":"refs/heads/master"},"5d7eaf0e57bf8e0a970f6c7b283bc88eacc672c4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-11 22:32:35.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/18/762218/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/18/762218/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/762218/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/762218/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/762218/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/18/762218/2"}}},"commit":{"parents":[{"commit":"a79ba06596269233713c53e11d9f6d1fd86c693a","subject":"Use the same access rule for both instances","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a79ba06596269233713c53e11d9f6d1fd86c693a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-11-10 23:24:47.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-11-11 22:32:30.000000000","tz":0},"subject":"[ci] Switch generic driver job to focal fossa","message":"[ci] Switch generic driver job to focal fossa\n\nThere is one failing test. See bug #1903922 - this failure\nis observed on older versions of Ubuntu as well, and doesn\u0027t\nseem to be an artifact of changing the underlying OS.\n\nChange-Id: Ifb3cb87512c3d8f7cf3af44f7012b967d502bd8e\nPartial-Bug: #1896672\nRelated-Bug: #1903922\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5d7eaf0e57bf8e0a970f6c7b283bc88eacc672c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5d7eaf0e57bf8e0a970f6c7b283bc88eacc672c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
