)]}'
{"id":"openstack%2Fmanila-tempest-plugin~762219","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I15fe5a95037a31d313450a5a4c1ca394f0c259b6","project":"openstack/manila-tempest-plugin","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"I15fe5a95037a31d313450a5a4c1ca394f0c259b6","subject":"[ci] Switch lvm driver job to focal fossa","status":"MERGED","created":"2020-11-10 23:26:54.000000000","updated":"2020-11-17 18:01:12.000000000","submitted":"2020-11-17 13:37:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762219-1605620245617-e490bc4f","meta_rev_id":"b7e12cfb18bee2a119c68a76741eee7cb19ccae9","_number":762219,"virtual_id_number":762219,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-11-12 21:37:36.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-11-12 23:01:59.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-17 13:37:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-17 11:39:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-16 19:42:31.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-17 11:39:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-12 21:37:36.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-12 23:01:59.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 19:42:31.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-11-17 11:39:22.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-11-17 13:37:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"06a7ada1d3023796fa925e502a4a9431b882838a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-10 23:26:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48d9815418e8a82bf67e7efc54471ab86c60f5b7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-10 23:28:14.000000000","message":"Patch Set 1: Workflow-1\n\nTesting","accounts_in_message":[],"_revision_number":1},{"id":"bc50a2e6a90b2be47b22dfee1ce80233a465d27a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 01:32:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c909195d8694b768702feb43f550e12 : SUCCESS in 3m 10s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/66ec653382ce4af496bd9e16e498f707 : SUCCESS in 26m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e0d349f7cf9a41fca7bb70f9cc136b1e : SUCCESS in 33m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/057c1a7f41b34faf9d0bde0f94f37aa5 : FAILURE in 1h 12m 59s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ca4f311e9578403db266c18b1208623e : FAILURE in 1h 17m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3f90793187a541b2a74c90079ca334f0 : FAILURE in 2h 01m 05s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b7d40e9b45864f27b2207a7f144b2dab : FAILURE in 1h 36m 11s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8d9b2c2aaaad48ebb98a81095b46f21d : SUCCESS in 45m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/96b9dcba9a8e481cbc591d9e366738b6 : FAILURE in 1h 50m 53s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/af786b7d95aa4dd5be617859c45d2a9c : SUCCESS in 25m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"196c159e4b5cbaa2452bcbab4b7a28c5186c7568","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-12 18:23:34.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4ca13c5d1fbf15262300eec48820c028edc9c94c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 20:30:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edf69f36478d40ca8e2fc9cf7b6b94a9 : SUCCESS in 3m 18s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4ea8e4579d1940d88cf1eb19417d3247 : SUCCESS in 32m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1a7ff66ba6a941c1b19b46549f256062 : SUCCESS in 34m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/12c90378fdae4be2b9febacfeba8a22a : SUCCESS in 1h 09m 01s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f88dd531778847ce87a436194350b297 : SUCCESS in 1h 11m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c24478b7ec794ee2b37a7ce818423533 : SUCCESS in 1h 00m 49s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d2ecac1434a94f038e270abd52573432 : FAILURE in 50m 29s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e5ef3ee6c28e4bfea897c896a04bf8d1 : SUCCESS in 41m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e4118d0a0426468fa11110bee6fd4984 : FAILURE in 2h 02m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2fd0c8690d6d4b4ea5313b1849f84d6d : SUCCESS in 20m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f6a70149794f9ca71f9b7b43f6494e3d32d4120a","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-11-12 21:37:36.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/19/762219/2/check/DellEMC_VNX_Manila/d248e4a/DellEMC_VNX_Manila/38 : SUCCESS in 1h 07m 05s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"98f728b83b993d27d71e23770694221c4fad5652","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-12 23:01:59.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/19/762219/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/3365f5d/ : SUCCESS in 2h 31m 23s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/19/762219/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/5a78c87/ : FAILURE in 1h 42m 30s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"f3551024a160246c506c0fad10bcab69263fe9e4","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-11-16 19:42:31.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":2},{"id":"5708355af46b29e9ddf14279e69cb8b6eed73aba","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-11-17 11:39:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks Goutham!","accounts_in_message":[],"_revision_number":2},{"id":"805147b23d0113350cac64bd5417fcf14b79cca2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 11:39:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0ae89185211216d272a5ace95d7367227b43e72a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 13:37:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9317d6199cf44e0ec421d1d280e0c04af4d29c83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 13:37:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f378c46529042238d9aac5d7f96303e : SUCCESS in 3m 54s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6c51cecb80f747adb4ddc3eeb4778cac : SUCCESS in 27m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/700ec5de352e47038a70a54404906d11 : SUCCESS in 31m 18s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e0d454c6bac2419e973ec06c78361d3d : SUCCESS in 1h 13m 49s","accounts_in_message":[],"_revision_number":2},{"id":"ee052551b41ca1a78de25ce8dd0589552aaa1564","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-17 18:01:12.000000000","message":"Patch Set 2:\n\nWe *may* need to workaround this LVM job on stable/victoria; the bugfix that this depends on needs to be backported to stable/victoria: https://review.opendev.org/#/q/topic:bug/1903773+(status:open+OR+status:merged)\n\nWe\u0027ll know in a bit; if the gate is broken on stable/victoria - i would like to temporarily make the LVM job non-voting to land the fix and revert it to voting after the fix has merged.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1214b82277f3a1e438231de7094453ce8d9a8cd9","revisions":{"4408373f053a0f87ad407a4193ebd80b76ca9981":{"kind":"REWORK","_number":1,"created":"2020-11-10 23:26:54.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/19/762219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/762219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/762219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/762219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/762219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/762219/1"}}},"commit":{"parents":[{"commit":"a79ba06596269233713c53e11d9f6d1fd86c693a","subject":"Use the same access rule for both instances","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a79ba06596269233713c53e11d9f6d1fd86c693a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-11-10 23:26:44.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-11-10 23:26:44.000000000","tz":-480},"subject":"[ci] Switch lvm driver job to focal fossa","message":"[ci] Switch lvm driver job to focal fossa\n\nChange-Id: I15fe5a95037a31d313450a5a4c1ca394f0c259b6\nPartial-Bug: #1896672\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4408373f053a0f87ad407a4193ebd80b76ca9981"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4408373f053a0f87ad407a4193ebd80b76ca9981"}]},"branch":"refs/heads/master"},"1214b82277f3a1e438231de7094453ce8d9a8cd9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-12 18:23:34.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/19/762219/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/762219/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/762219/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/762219/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/762219/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/762219/2"}}},"commit":{"parents":[{"commit":"a79ba06596269233713c53e11d9f6d1fd86c693a","subject":"Use the same access rule for both instances","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a79ba06596269233713c53e11d9f6d1fd86c693a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-11-10 23:26:44.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-11-12 18:23:32.000000000","tz":0},"subject":"[ci] Switch lvm driver job to focal fossa","message":"[ci] Switch lvm driver job to focal fossa\n\nDepends-On: https://review.opendev.org/762468/\nChange-Id: I15fe5a95037a31d313450a5a4c1ca394f0c259b6\nPartial-Bug: #1896672\nRelated-Bug: #1903773\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1214b82277f3a1e438231de7094453ce8d9a8cd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1214b82277f3a1e438231de7094453ce8d9a8cd9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
