)]}'
{"id":"openstack%2Fmanila-tempest-plugin~763009","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ib8f72404c481a343ca53f045877444060c640f3d","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1896672","hashtags":[],"change_id":"Ib8f72404c481a343ca53f045877444060c640f3d","subject":"[CI] Switch cephfs-nfs job to focal fossa","status":"MERGED","created":"2020-11-17 13:28:31.000000000","updated":"2021-01-12 22:05:09.000000000","submitted":"2021-01-12 22:05:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763009-bug/1896672","meta_rev_id":"08684c6f3ef5ba966ebfa5fb52eb0b79ce237930","_number":763009,"virtual_id_number":763009,"owner":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-12 22:05:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-12 20:37:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2021-01-12 12:56:41.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-12 20:37:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-18 23:59:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-19 01:23:30.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-19 02:18:30.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-11-19 02:36:06.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-11 22:11:07.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-01-12 20:37:33.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"1909b017c3007e307c32a09f29f983eb9293fd43","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-11-17 13:28:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5129fce14307d7269b7c2934fc0e1977c0bd70d","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-11-17 13:31:47.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"109a162890ee874f64bdf45521bd631f114a4d99","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-11-17 15:25:50.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/09/763009/2/check/DellEMC_Unity_Manila_dhss_true/4007962/DellEMC_Unity_Manila_dhss_true/33 : FAILURE in 1h 53m 49s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/09/763009/2/check/DellEMC_Unity_Manila_dhss_false/01b94fb/DellEMC_Unity_Manila_dhss_false/32 : SUCCESS in 55m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"6d77337d2c4103fc9f1a0fea5f7fb5be1fd17ee1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 15:43:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e75dc3f6ffce42eabbf72c3a30c76b36 : SUCCESS in 2m 55s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4123eb72249a467c93cd2b4ac451af25 : SUCCESS in 32m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8a997485a5e34b14b4f7278c68757a4f : SUCCESS in 39m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ec95f23445dc49df935518420715bccc : SUCCESS in 1h 02m 37s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4c23ea6030ce46618c93f95d57954c21 : FAILURE in 1h 20m 06s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0a84425496ba4390b049d3c3db3af953 : SUCCESS in 56m 58s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7eb38f09e89d40f5a448e9705d643157 : FAILURE in 9m 03s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9bb440d3614b4af58a953d46edd7e803 : SUCCESS in 48m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4a8cdebdfa2947cd88cfbc33bec4d1fe : FAILURE in 2h 06m 28s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/70b4740c2bd74747b52bc0def624d104 : SUCCESS in 25m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6734179df09e283864be5e8db87ca37b35c20bd2","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-11-17 16:17:21.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a88f1be10a11e52af0a5f3ccb98d445d212d88bd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-11-17 18:08:55.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/09/763009/3/check/DellEMC_Unity_Manila_dhss_true/22535df/DellEMC_Unity_Manila_dhss_true/35 : SUCCESS in 1h 51m 22s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/09/763009/3/check/DellEMC_Unity_Manila_dhss_false/1ddeefe/DellEMC_Unity_Manila_dhss_false/34 : SUCCESS in 50m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"d793ce0cd41b536e7b14eec4e03ec98bd954ae6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 18:25:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec9f498521364c9abc88bf8635b9123c : SUCCESS in 3m 13s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/725af027f7824169bf5b8115cf3b7b53 : SUCCESS in 33m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d5cf2c4ec9b843c3a9e1267f1ef5c13e : SUCCESS in 42m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7465a02d59f14725b568967f0d2d184b : SUCCESS in 1h 07m 33s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b316010308834802bcfca9f48acc4683 : FAILURE in 1h 27m 55s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/2e0c0242585d478887e287aa9b004b8a : SUCCESS in 1h 32m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f1dca20ae40d42d280ef566488efe316 : FAILURE in 55m 57s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2504cce1b9a54b8ba39f6cdd27381aad : SUCCESS in 51m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0feb6925a82c464e81a5111412db2e9c : FAILURE in 2h 02m 51s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/02dd7075435446c1ad22307b04e67080 : SUCCESS in 30m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e6d1784b12a4944e71c74efca191d6342a8863cd","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-11-17 19:42:22.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/09/763009/3/check/DellEMC_VNX_Manila/a3d8281/DellEMC_VNX_Manila/51 : SUCCESS in 1h 16m 20s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"8ca02a09cac4c3272a642bce29aada237611cc50","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-11-17 20:19:49.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/09/763009/3/check/DellEMC_Unity_Manila_dhss_true/7a0ac72/DellEMC_Unity_Manila_dhss_true/36 : SUCCESS in 1h 53m 53s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/09/763009/3/check/DellEMC_Unity_Manila_dhss_false/bf0628d/DellEMC_Unity_Manila_dhss_false/35 : SUCCESS in 50m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"6a99801a6e75ee59a1dfb9d4f79b64c9d36f52d9","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-17 21:01:34.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/09/763009/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/0a40d7c/ : SUCCESS in 2h 35m 23s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/09/763009/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/2e24bbc/ : SUCCESS in 1h 52m 59s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"f14d764f866d61e3e872ee6cb4c252c25915542f","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-11-18 21:37:30.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"501f0e93c9806c1ed4b619f47d6eee766448b966","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-11-18 22:38:38.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/09/763009/4/check/DellEMC_Unity_Manila_dhss_true/22db259/DellEMC_Unity_Manila_dhss_true/41 : FAILURE in 1h 01m 00s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/09/763009/4/check/DellEMC_Unity_Manila_dhss_false/5dee93a/DellEMC_Unity_Manila_dhss_false/40 : SUCCESS in 50m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"aea4f1917878061ae1050161995ad6c008e5da39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 23:59:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c8725241e8441639857d127588580c2 : SUCCESS in 3m 19s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/25b9ce6dbf4e4378a5254582ca70a6f7 : SUCCESS in 31m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/abc05e1b925a4ed599122c19ceef7515 : SUCCESS in 38m 19s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cd083ebd99e245c7a0be25470c10abb6 : SUCCESS in 1h 08m 12s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/53f7f6037d754ad5a0f6c1aa1d8c2ce9 : FAILURE in 1h 08m 35s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ecce3824d8394f7a843b799cc0c67993 : SUCCESS in 59m 29s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5a7dfda6dab84385acdade519efc7425 : FAILURE in 1h 15m 16s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/de5e45e7b0934234a7d6c1425a1d3848 : SUCCESS in 57m 17s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d64c98e51fa34222b0f6bd44cd532de1 : FAILURE in 2h 11m 44s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1587c1dbf50148599af82c43a5ffd0f8 : SUCCESS in 26m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a7c8eface6fb7179c3068e9a421ae7efa7e96aa9","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-11-19 01:23:30.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/09/763009/4/check/DellEMC_VNX_Manila/62ad6b7/DellEMC_VNX_Manila/55 : SUCCESS in 1h 24m 05s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"8493930100530347b8a130422f771c495603b127","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-11-19 02:18:30.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/09/763009/4/check/DellEMC_Unity_Manila_dhss_true/5c88fb6/DellEMC_Unity_Manila_dhss_true/44 : FAILURE in 54m 25s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/09/763009/4/check/DellEMC_Unity_Manila_dhss_false/7871bbc/DellEMC_Unity_Manila_dhss_false/43 : SUCCESS in 50m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"efc0226d3c661490286c31c6aa221edb5b16178d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-19 02:36:06.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/09/763009/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/937ee71/ : SUCCESS in 2h 26m 29s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/09/763009/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/e553d46/ : SUCCESS in 1h 42m 00s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"29a6fd669c4cfe9e28fa322278f46cc8a1e53b5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-01-11 21:25:34.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"ea50ccf95c1234eebbb67aef1704770203cda169","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-01-11 22:08:53.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"b908329a863f560840bfbd5b4619512b3a27091d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 00:54:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17e8f512008c402590faca19cc9ed9ee : SUCCESS in 4m 08s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/175aba43387d4cecb0f8a2b3fa2636c3 : SUCCESS in 32m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/32deedef96d0431eb80cbbb147d37a0f : SUCCESS in 38m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7e2625b3447f42ce80f0b7174ec4a8c3 : SUCCESS in 1h 29m 15s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e4f40bfed9f1444384d622450d204817 : FAILURE in 1h 01m 20s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f81dec4fc0be437aaff421d16f958b2a : SUCCESS in 1h 28m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e316a78e66ed4cc9b1c0be5fcc7fb2fc : FAILURE in 1h 02m 08s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/78d3dedb2ad341ba9854879848723220 : SUCCESS in 56m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0da38ad296034cad91a7613db826541b : FAILURE in 2h 42m 00s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8b50118bcc3343fb824c1d01992dd6d3 : FAILURE in 23m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2077a42d221feb37ab9f111197ea860fb9caaf96","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-12 02:05:41.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/09/763009/6/check/DellEMC_VNX_Manila/11e6494/DellEMC_VNX_Manila/125 : SUCCESS in 1h 10m 33s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":6},{"id":"20c2841ed357422357ef6cfa4fb7069cfd546d36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-12 07:26:37.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"7ffe0b7e93ffe5ed8329611abd86e710c8f70b4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 09:32:27.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1439b40996504ac7bc5d45ba50912cac : SUCCESS in 4m 08s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/67074bd130de446987950e1cd22975e8 : SUCCESS in 33m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e96045d80a47456b956147428c0aac4e : SUCCESS in 39m 18s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/244ebe026f12405e9f8d42a16e18b864 : SUCCESS in 1h 32m 52s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c8c28ad382e74d28b065c6d39b52785d : SUCCESS in 1h 12m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/65a6da23840544dea7238215c57215a3 : SUCCESS in 1h 00m 14s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/fb39e6953aed45e98952f6570fae4df5 : SUCCESS in 1h 21m 18s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7d26ac45d5944111817a91971b097c8c : SUCCESS in 45m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fc9cd302e1d54cb0b9c044286064345f : FAILURE in 1h 59m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/bb518d5db7684db1bddf6cbb83d83839 : SUCCESS in 31m 09s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ba32f01a6246db2f94378b4ac4e7bde5a1bd356c","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-12 12:56:41.000000000","message":"Patch Set 7: Code-Review+2\n\nThanks, Victoria!","accounts_in_message":[],"_revision_number":7},{"id":"eb2ac9a0ab3f83bcf731e3744e77bd313a2ef74a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-12 20:37:33.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nGreat! LGTM, glad to close https://bugs.launchpad.net/manila/+bug/1896672! Thanks Victoria","accounts_in_message":[],"_revision_number":7},{"id":"99c922e2690b6418c43a8d4ca02c29e719467d88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 20:37:59.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"01a9b502a2fd628ff4d5607f172d7bdf43d993ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 22:05:08.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b1e6ee9338647bbbb0a799720417b94 : SUCCESS in 3m 25s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2a25102cb9f24ec0b74cc06c75ac94b8 : SUCCESS in 40m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c5d85554139748928f9fe04518567c79 : SUCCESS in 35m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5080a13414844506b05ad99d512acf06 : SUCCESS in 1h 17m 28s","accounts_in_message":[],"_revision_number":7},{"id":"08684c6f3ef5ba966ebfa5fb52eb0b79ce237930","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 22:05:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"b3a3c420c7cdb72fee275aa71491e5fba8c304d7","revisions":{"86728ab0311b0033a2beee0329c29e6776802d7a":{"kind":"REWORK","_number":1,"created":"2020-11-17 13:28:31.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/09/763009/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/09/763009/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/1"}}},"commit":{"parents":[{"commit":"b498ce8e85e8e2a5a9aeb771c48b906a6b698e13","subject":"Merge \"[ci] Switch zfsonlinux job to focal fossa\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b498ce8e85e8e2a5a9aeb771c48b906a6b698e13"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-11-17 13:27:38.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-11-17 13:27:38.000000000","tz":-180},"subject":"ci] Switch cephfs-nfs job to focal fossa","message":"ci] Switch cephfs-nfs job to focal fossa\n\nChange-Id: Ib8f72404c481a343ca53f045877444060c640f3d\nPartial-Bug: #1896672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/86728ab0311b0033a2beee0329c29e6776802d7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/86728ab0311b0033a2beee0329c29e6776802d7a"}]},"branch":"refs/heads/master"},"0ca927e235a40485f08b720af1ca3bcbc4fd310e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-17 13:31:47.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/09/763009/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/09/763009/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/2"}}},"commit":{"parents":[{"commit":"b498ce8e85e8e2a5a9aeb771c48b906a6b698e13","subject":"Merge \"[ci] Switch zfsonlinux job to focal fossa\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b498ce8e85e8e2a5a9aeb771c48b906a6b698e13"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-11-17 13:27:38.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-11-17 13:31:43.000000000","tz":0},"subject":"[CI] Switch cephfs-nfs job to focal fossa","message":"[CI] Switch cephfs-nfs job to focal fossa\n\nChange-Id: Ib8f72404c481a343ca53f045877444060c640f3d\nPartial-Bug: #1896672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0ca927e235a40485f08b720af1ca3bcbc4fd310e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0ca927e235a40485f08b720af1ca3bcbc4fd310e"}]},"branch":"refs/heads/master"},"d9eea0d2592c65dfc38023a6ea5429a0a5cc4160":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-11-17 16:17:21.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/09/763009/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/09/763009/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/3"}}},"commit":{"parents":[{"commit":"b498ce8e85e8e2a5a9aeb771c48b906a6b698e13","subject":"Merge \"[ci] Switch zfsonlinux job to focal fossa\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b498ce8e85e8e2a5a9aeb771c48b906a6b698e13"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-11-17 13:27:38.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-11-17 16:17:15.000000000","tz":0},"subject":"[CI] Switch cephfs-nfs job to focal fossa","message":"[CI] Switch cephfs-nfs job to focal fossa\n\nDepends-On: https://review.opendev.org/#/c/756127/\n\nChange-Id: Ib8f72404c481a343ca53f045877444060c640f3d\nPartial-Bug: #1896672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d9eea0d2592c65dfc38023a6ea5429a0a5cc4160"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d9eea0d2592c65dfc38023a6ea5429a0a5cc4160"}]},"branch":"refs/heads/master"},"c93c7cda7cb1476d4e2cb1eaf81e1a5508e0c38b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-11-18 21:37:30.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/09/763009/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/09/763009/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/4"}}},"commit":{"parents":[{"commit":"b498ce8e85e8e2a5a9aeb771c48b906a6b698e13","subject":"Merge \"[ci] Switch zfsonlinux job to focal fossa\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b498ce8e85e8e2a5a9aeb771c48b906a6b698e13"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-11-17 13:27:38.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-11-18 21:37:26.000000000","tz":0},"subject":"[CI] Switch cephfs-nfs job to focal fossa","message":"[CI] Switch cephfs-nfs job to focal fossa\n\nDepends-On: https://review.opendev.org/#/c/756127/\nDepends-On: https://review.opendev.org/#/c/756600/\n\nChange-Id: Ib8f72404c481a343ca53f045877444060c640f3d\nPartial-Bug: #1896672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c93c7cda7cb1476d4e2cb1eaf81e1a5508e0c38b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c93c7cda7cb1476d4e2cb1eaf81e1a5508e0c38b"}]},"branch":"refs/heads/master"},"cdb4800d93157d169597cb4d5c9e39e39afc0145":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-01-11 21:25:34.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/09/763009/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/09/763009/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/5"}}},"commit":{"parents":[{"commit":"8cce66b545025d105dacaf1ee02942a84afcd79d","subject":"Add doc/requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8cce66b545025d105dacaf1ee02942a84afcd79d"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-11-17 13:27:38.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-01-11 21:25:34.000000000","tz":0},"subject":"[CI] Switch cephfs-nfs job to focal fossa","message":"[CI] Switch cephfs-nfs job to focal fossa\n\nDepends-On: https://review.opendev.org/#/c/756127/\nDepends-On: https://review.opendev.org/#/c/756600/\n\nChange-Id: Ib8f72404c481a343ca53f045877444060c640f3d\nPartial-Bug: #1896672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cdb4800d93157d169597cb4d5c9e39e39afc0145"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cdb4800d93157d169597cb4d5c9e39e39afc0145"}]},"branch":"refs/heads/master","description":"Rebase"},"a983fab90e497e312b4bf06492615fb06f51bbb7":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2021-01-11 22:08:53.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/09/763009/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/09/763009/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/6"}}},"commit":{"parents":[{"commit":"8cce66b545025d105dacaf1ee02942a84afcd79d","subject":"Add doc/requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8cce66b545025d105dacaf1ee02942a84afcd79d"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-11-17 13:27:38.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-01-11 22:08:41.000000000","tz":0},"subject":"[CI] Switch cephfs-nfs job to focal fossa","message":"[CI] Switch cephfs-nfs job to focal fossa\n\nChange-Id: Ib8f72404c481a343ca53f045877444060c640f3d\nPartial-Bug: #1896672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a983fab90e497e312b4bf06492615fb06f51bbb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a983fab90e497e312b4bf06492615fb06f51bbb7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b3a3c420c7cdb72fee275aa71491e5fba8c304d7":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-01-12 07:26:37.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/09/763009/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/09/763009/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/763009/7"}}},"commit":{"parents":[{"commit":"cee8ed17ac763b8772a41731d7b4a131b32af9a7","subject":"Tests: reduce concurrency for cephfs-nfs job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cee8ed17ac763b8772a41731d7b4a131b32af9a7"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-11-17 13:27:38.000000000","tz":-180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-01-12 07:26:37.000000000","tz":0},"subject":"[CI] Switch cephfs-nfs job to focal fossa","message":"[CI] Switch cephfs-nfs job to focal fossa\n\nChange-Id: Ib8f72404c481a343ca53f045877444060c640f3d\nPartial-Bug: #1896672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b3a3c420c7cdb72fee275aa71491e5fba8c304d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b3a3c420c7cdb72fee275aa71491e5fba8c304d7"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
