)]}'
{"id":"openstack%2Fmanila-tempest-plugin~763282","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I9f7997ad4f82877998d647c01940e37532dc16e2","project":"openstack/manila-tempest-plugin","branch":"master","topic":"stable-branches","hashtags":[],"change_id":"I9f7997ad4f82877998d647c01940e37532dc16e2","subject":"[ci] Add job variants for stable branches","status":"ABANDONED","created":"2020-11-18 23:39:07.000000000","updated":"2021-05-18 15:32:37.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"c104b7008d9fee42f9cc64e1ff35634f4f737b79","_number":763282,"virtual_id_number":763282,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-17 21:22:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-05-17 19:19:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"REVIEWER":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 01:38:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-19 02:57:47.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-19 03:07:39.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-11-19 05:10:12.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-08 15:48:24.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2021-04-13 20:21:51.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"CC"}],"messages":[{"id":"6c761995dd6d0897b7664e85a0d1070d1cd2437b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-18 23:39:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6716777d03b0a60842772b41a2249dde50021634","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-11-19 00:34:52.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/82/763282/1/check/DellEMC_Unity_Manila_dhss_true/8359b33/DellEMC_Unity_Manila_dhss_true/42 : FAILURE in 49m 37s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/82/763282/1/check/DellEMC_Unity_Manila_dhss_false/7aefa47/DellEMC_Unity_Manila_dhss_false/41 : SUCCESS in 55m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"cabd3a8d79b416da65e467387e27915531e1a2ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 01:38:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d703ba4b90fd42fb8b9f55d3240b1150 : SUCCESS in 3m 17s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7c1490ddc9fb4a0ba5d6b18c2dc9eba8 : SUCCESS in 38m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b59ccb3b6bb542fc8481fd8db633017e : SUCCESS in 39m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9db58216a6424eaa8d4a02e83951bada : SUCCESS in 1h 45m 33s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3e206ea28e0049da9993e90df0762422 : SUCCESS in 1h 12m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/44fed804f45048eab2a3d4fc299a63c9 : SUCCESS in 1h 05m 29s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7d24eb2b45094668817d388083a3a4a3 : FAILURE in 1h 04m 56s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2d4b65cfdcb44ad49f884758f083fd18 : SUCCESS in 47m 11s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/92f86a94efa1478491f2b61c3a75d4f3 : FAILURE in 1h 48m 37s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f97eee912dba4b2c944f4e41ae260eff : SUCCESS in 22m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a32cd4c5376d16d2f50aad90b1c51f6073a95a6f","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-11-19 02:57:47.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/82/763282/1/check/DellEMC_VNX_Manila/d8a8f28/DellEMC_VNX_Manila/56 : SUCCESS in 1h 19m 03s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"68c953324f1efa68b8ef490d2c8d86a3a6eb0451","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-11-19 03:07:39.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/82/763282/1/check/DellEMC_Unity_Manila_dhss_true/d28c259/DellEMC_Unity_Manila_dhss_true/45 : FAILURE in 51m 54s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/82/763282/1/check/DellEMC_Unity_Manila_dhss_false/a50aa94/DellEMC_Unity_Manila_dhss_false/44 : SUCCESS in 49m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"2a0adb8720b2214686a94e423781bc21649062ef","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-19 05:10:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/82/763282/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/f4e3bdc/ : SUCCESS in 2h 32m 50s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/82/763282/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/3773ce6/ : SUCCESS in 1h 49m 42s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"553c7e722ad2506a437a30f8f5aa9eee1d9ef4f7","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-04-13 06:01:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5b26e492af623f98e1050640c65d482d27233ee5","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-04-13 20:21:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f546077ab91cc42195fd36cec4b1cb4361a26f1","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-04-14 00:56:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c2bd08e2e77c9d8969250470414af05657790d67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-17 18:27:48.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d08850907c5c1f0a37f69d3ccefa69f720dfca2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-17 19:18:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9c8a2f7ec0ba4bc44cb19afa63ba9505b51c3e46","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-17 19:19:29.000000000","message":"Patch Set 3: Workflow-1\n\nStill testing, will merge with https://review.opendev.org/c/openstack/manila-tempest-plugin/+/725692","accounts_in_message":[],"_revision_number":3},{"id":"1f5978995da99217a443ef4d9d00f789fe5f8281","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 21:22:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52b12f97e2d84865ad3e722ab7df0e29 : SUCCESS in 4m 09s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bdaf5553694e4d94937dd3663125febd : SUCCESS in 23m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f405b9a13299407587d87e95db85c18e : SUCCESS in 30m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/97e904079e0849d280bd1cd1d8eeedbe : SUCCESS in 1h 01m 32s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e56eaa5c783443cdbe3d2e41052ffab6 : SUCCESS in 56m 11s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/2f1d94e6077240449066bb02437b7632 : FAILURE in 1h 07m 11s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/618ce752957949e8b0ed42ee162e5fbc : FAILURE in 1h 02m 48s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a379e5a4642446f68e33e14faf248fb8 : FAILURE in 38m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5202e8b8615a4fcebabd59f1e9828b4e : FAILURE in 2h 01m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6989adfe35b54119864ab5eb44194bc1 : SUCCESS in 21m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2908aab4e24ac6d6cebe07422945c981775367b2","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-05-17 22:19:49.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/82/763282/3/check/DellEMC_VNX_Manila/b40b552/DellEMC_VNX_Manila/39 : FAILURE in 46m 23s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"0e42480771771621dfed687d3d4ffe09c34ba831","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-18 00:03:25.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/82/763282/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/0052db6/ : SUCCESS in 2h 29m 10s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/82/763282/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/a1b2e58/ : SUCCESS in 1h 42m 03s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"c104b7008d9fee42f9cc64e1ff35634f4f737b79","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-18 15:32:37.000000000","message":"Abandoned\n\nMerged into https://review.opendev.org/c/openstack/manila-tempest-plugin/+/725692","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f74b7f131a81792fd375a4402c194d097912f944","revisions":{"dda57683ec2c3c46576dfa746025ff98fb5819a6":{"kind":"REWORK","_number":1,"created":"2020-11-18 23:39:07.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/82/763282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/82/763282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/763282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/763282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/763282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/763282/1"}}},"commit":{"parents":[{"commit":"6f593f486e85c117854ed0ebfa7b1d80962c2c41","subject":"Merge \"[ci] Switch lvm driver job to focal fossa\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6f593f486e85c117854ed0ebfa7b1d80962c2c41"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-11-18 23:38:46.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-11-18 23:38:46.000000000","tz":-480},"subject":"[ci] Add job variants for stable/victoria","message":"[ci] Add job variants for stable/victoria\n\nChange-Id: I9f7997ad4f82877998d647c01940e37532dc16e2\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dda57683ec2c3c46576dfa746025ff98fb5819a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dda57683ec2c3c46576dfa746025ff98fb5819a6"}]},"branch":"refs/heads/master"},"024f004256bd04d497b0de69d628f6e0f9893c13":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-05-17 18:27:48.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/82/763282/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/82/763282/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/763282/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/763282/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/763282/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/763282/2"}}},"commit":{"parents":[{"commit":"0031ba11387091fcc830047cd265d7ed9f52c129","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0031ba11387091fcc830047cd265d7ed9f52c129"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-11-18 23:38:46.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-05-17 18:27:48.000000000","tz":0},"subject":"[ci] Add job variants for stable/victoria","message":"[ci] Add job variants for stable/victoria\n\nChange-Id: I9f7997ad4f82877998d647c01940e37532dc16e2\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/024f004256bd04d497b0de69d628f6e0f9893c13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/024f004256bd04d497b0de69d628f6e0f9893c13"}]},"branch":"refs/heads/master","description":"Rebase"},"f74b7f131a81792fd375a4402c194d097912f944":{"kind":"REWORK","_number":3,"created":"2021-05-17 19:18:52.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/82/763282/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/82/763282/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/763282/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/763282/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/763282/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/763282/3"}}},"commit":{"parents":[{"commit":"0031ba11387091fcc830047cd265d7ed9f52c129","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0031ba11387091fcc830047cd265d7ed9f52c129"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-11-18 23:38:46.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-05-17 19:18:42.000000000","tz":-420},"subject":"[ci] Add job variants for stable branches","message":"[ci] Add job variants for stable branches\n\nWe need different configuration options\nfor stable branch jobs.\n\nChange-Id: I9f7997ad4f82877998d647c01940e37532dc16e2\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f74b7f131a81792fd375a4402c194d097912f944"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f74b7f131a81792fd375a4402c194d097912f944"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
