)]}'
{"id":"openstack%2Fmanila-tempest-plugin~763283","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I4ed0d1f1b5614842ce9359a411cfa0944e4ce0f8","project":"openstack/manila-tempest-plugin","branch":"master","topic":"glusterfs","hashtags":[],"change_id":"I4ed0d1f1b5614842ce9359a411cfa0944e4ce0f8","subject":"DNM - testing glusterfs-native","status":"ABANDONED","created":"2020-11-18 23:42:55.000000000","updated":"2022-07-05 15:08:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fe6d6da73d4f3c17a5d9bd210b4ca5121c8a7853","_number":763283,"virtual_id_number":763283,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-25 00:13:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 00:33:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-19 01:27:44.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"}],"messages":[{"id":"5302e4f225e81f21a66866251aa8b1cf3a41ac1e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-18 23:42:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"071352f13714fe96ce3046cf28d7182dc8a9e8f2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-19 00:03:26.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"5c87de60507a9738396d99021b41b07d0d28dc65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 00:33:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/1beb44ce49074dd8aea2b2313e9a44d7 : FAILURE in 37m 26s","accounts_in_message":[],"_revision_number":1},{"id":"582ca9da6520d850d6fe9f6dd461279d305ed8f3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-11-19 01:27:44.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/83/763283/1/check/DellEMC_Unity_Manila_dhss_true/f5014ae/DellEMC_Unity_Manila_dhss_true/43 : FAILURE in 51m 06s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/83/763283/1/check/DellEMC_Unity_Manila_dhss_false/42b5bad/DellEMC_Unity_Manila_dhss_false/42 : SUCCESS in 52m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"f8c81b28cc45d32dfadbee86d9b2a9e97359e32d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-19 18:36:53.000000000","message":"Patch Set 1:\n\nrecheck\n\nA couple of real failures in the m-shr logs; want to see if this is consistent","accounts_in_message":[],"_revision_number":1},{"id":"7df989c86cd4965160c3b7af95aac200b652bdd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 19:08:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/a9d1a09bfbf54a138877a108191e6902 : FAILURE in 29m 51s","accounts_in_message":[],"_revision_number":1},{"id":"40c7b50f6526768f309b3af99502f29399d4673e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-24 23:43:08.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b9f3f1996299dfadb4972b1215b7ff3ab030de67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 00:13:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/92f593ea83114059acaf6e7481fc66ab : FAILURE in 27m 27s","accounts_in_message":[],"_revision_number":2},{"id":"e62a27873f8c1b403b1386199609b3de3be8d99a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-25 00:32:18.000000000","message":"Patch Set 2:\n\nrecheck\n\nLets get a second run","accounts_in_message":[],"_revision_number":2},{"id":"d02e8284d815fd7782b94b29ca7fc4f98fad02c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 01:27:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/0c8cbf9d25804ed6b1b503177bdb61f2 : FAILURE in 41m 48s","accounts_in_message":[],"_revision_number":2},{"id":"fe6d6da73d4f3c17a5d9bd210b4ca5121c8a7853","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-05 15:08:51.000000000","message":"Abandoned\n\nNo longer needed","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3ff870c484b9f64585c16c0513c73700b1a66359","revisions":{"ab164d096ea76cfcf813e257e5187305257ebd78":{"kind":"REWORK","_number":1,"created":"2020-11-18 23:42:55.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/83/763283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/83/763283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/763283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/763283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/763283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/763283/1"}}},"commit":{"parents":[{"commit":"6f593f486e85c117854ed0ebfa7b1d80962c2c41","subject":"Merge \"[ci] Switch lvm driver job to focal fossa\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6f593f486e85c117854ed0ebfa7b1d80962c2c41"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-11-18 23:42:43.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-11-18 23:42:43.000000000","tz":-480},"subject":"DNM - testing glusterfs-native","message":"DNM - testing glusterfs-native\n\nChange-Id: I4ed0d1f1b5614842ce9359a411cfa0944e4ce0f8\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ab164d096ea76cfcf813e257e5187305257ebd78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ab164d096ea76cfcf813e257e5187305257ebd78"}]},"branch":"refs/heads/master"},"3ff870c484b9f64585c16c0513c73700b1a66359":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-24 23:43:08.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/83/763283/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/83/763283/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/763283/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/763283/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/763283/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/83/763283/2"}}},"commit":{"parents":[{"commit":"6f593f486e85c117854ed0ebfa7b1d80962c2c41","subject":"Merge \"[ci] Switch lvm driver job to focal fossa\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6f593f486e85c117854ed0ebfa7b1d80962c2c41"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-11-18 23:42:43.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-11-24 23:43:03.000000000","tz":0},"subject":"DNM - testing glusterfs-native","message":"DNM - testing glusterfs-native\n\nDepends-On: I96d49f84122a34701328909c929ede4d66746911\nChange-Id: I4ed0d1f1b5614842ce9359a411cfa0944e4ce0f8\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3ff870c484b9f64585c16c0513c73700b1a66359"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3ff870c484b9f64585c16c0513c73700b1a66359"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
