)]}'
{"id":"openstack%2Fmanila-tempest-plugin~764392","triplet_id":"openstack%2Fmanila-tempest-plugin~master~If3508cd17a8640e88b21110d1d61dc424b94709e","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1905796","hashtags":[],"change_id":"If3508cd17a8640e88b21110d1d61dc424b94709e","subject":"Adds get default share type function","status":"MERGED","created":"2020-11-26 23:33:31.000000000","updated":"2020-12-09 20:54:49.000000000","submitted":"2020-12-09 20:54:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764392-bug/1905796","meta_rev_id":"b48fcfcb50d6118b2ddb021ebd68effdafe3d0b7","_number":764392,"virtual_id_number":764392,"owner":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-09 20:54:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2020-12-09 19:16:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2020-12-09 19:16:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-27 01:41:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-27 02:58:45.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-27 04:07:20.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-09 19:16:44.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"393cc4dcc8986f7445cf1d113488aefa92559ea9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-11-26 23:33:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3410675ea17f35a6ba8390b18e7925f677078dbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 01:41:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18cbd08fb9b542b3a718009b9641a6bd : SUCCESS in 2m 48s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2a5e44be56aa4a928fbcd3b36b00b1e8 : SUCCESS in 26m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4ab10d774329482ab0b90a993dbc5013 : SUCCESS in 35m 37s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4dfb5a1cb1b945d9913a9f8d9a3088c0 : SUCCESS in 1h 02m 57s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b5195a4f507c449aa1c9972fb8623981 : SUCCESS in 1h 06m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0dc5d32629ba49feabcfcc3561f0d0e5 : SUCCESS in 54m 40s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f762512b2ce54a29845e2cd6065ea952 : FAILURE in 58m 07s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d91c547f90eb4a5aa12c063a1c0a9794 : SUCCESS in 45m 12s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/165272dd353442548ca0f531670d95c8 : FAILURE in 2h 05m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b533817899714bad988cfee960e3e7f8 : SUCCESS in 22m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fb8ad278803d6350baad6f42d4be96dfbfac5e5c","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-11-27 02:58:45.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/92/764392/1/check/DellEMC_VNX_Manila/ade6715/DellEMC_VNX_Manila/66 : SUCCESS in 1h 16m 41s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"5df710a90f54c7256a5eaebab709ab8b41526e8f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-27 04:07:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/92/764392/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/a157623/ : SUCCESS in 2h 23m 28s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/92/764392/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/74456d4/ : SUCCESS in 1h 42m 28s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"0ce3951f20dec59ec25d589d2be3f5547bb36104","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-12-09 19:16:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThank you, Victoria, LGTM","accounts_in_message":[],"_revision_number":1},{"id":"69313a778b2a4d1b1393423923567b0994031ea5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 19:16:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b2eee6796b62441b2679d84d8ad37eb99af40c50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 20:54:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4539c14b228e4ed0a150319d520f73df : SUCCESS in 4m 13s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/949551f0d2274702bed166b1344caa63 : SUCCESS in 42m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a8c1361aee76422b979ebe06f997d235 : SUCCESS in 49m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7100877e796d4855ad15da427884351f : SUCCESS in 1h 30m 24s","accounts_in_message":[],"_revision_number":1},{"id":"b48fcfcb50d6118b2ddb021ebd68effdafe3d0b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 20:54:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ada5b69c78fd5d7a78bc4cf38d9e47d8274e0e3e","revisions":{"ada5b69c78fd5d7a78bc4cf38d9e47d8274e0e3e":{"kind":"REWORK","_number":1,"created":"2020-11-26 23:33:31.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/92/764392/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/92/764392/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/764392/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/764392/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/764392/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/764392/1"}}},"commit":{"parents":[{"commit":"6f593f486e85c117854ed0ebfa7b1d80962c2c41","subject":"Merge \"[ci] Switch lvm driver job to focal fossa\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6f593f486e85c117854ed0ebfa7b1d80962c2c41"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-11-26 23:27:53.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-11-26 23:33:28.000000000","tz":0},"subject":"Adds get default share type function","message":"Adds get default share type function\n\nThis patch set adds the get default share type\nfunction to client used by our tempest tests.\n\nAlso leverages this new function to get the default\nshare type, instead of using the get share types\ncommand which was breaking in case the default\nshare type name was different than \"default\"\n\nCloses-Bug: #1905796\nChange-Id: If3508cd17a8640e88b21110d1d61dc424b94709e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ada5b69c78fd5d7a78bc4cf38d9e47d8274e0e3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ada5b69c78fd5d7a78bc4cf38d9e47d8274e0e3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
