)]}'
{"id":"openstack%2Fmanila-tempest-plugin~766917","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I149acb076a9e42f2b86186efc0924a489aa1fa83","project":"openstack/manila-tempest-plugin","branch":"master","hashtags":[],"change_id":"I149acb076a9e42f2b86186efc0924a489aa1fa83","subject":"Update TOX_CONSTRAINTS_FILE","status":"MERGED","created":"2020-12-14 10:00:10.000000000","updated":"2020-12-15 03:52:05.000000000","submitted":"2020-12-15 03:31:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"766917","meta_rev_id":"6807b31d83215777a9aac2eaceb5690b45da1b0f","_number":766917,"virtual_id_number":766917,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-15 03:31:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2020-12-15 03:23:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2020-12-15 03:23:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-14 10:12:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-14 11:48:23.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-14 20:04:56.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"94a7aa2ce688dd1d408ed022c12e123781ca530e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-12-14 10:00:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e95d4c4d608eeb9a2395d6b31327d5159d41d841","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 10:12:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7927063413e141cc9146c33373fa2e9f : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"b436b9ab3e45eb6c4fc1703a5cff6c3c60a69078","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-12-14 11:48:23.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/17/766917/1/check/DellEMC_VNX_Manila/288c8ac/DellEMC_VNX_Manila/81 : FAILURE in 1h 35m 26s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"5572cf38485266eb026c5112f95f2ff87840dfc2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-12-14 20:04:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nThanks for this change, please see my comment inline","accounts_in_message":[],"_revision_number":1},{"id":"7608e3f759d511f3c02ac7c608430a8ccb5a3520","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-12-15 00:56:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"38c9e7a5885bca7da1f7e593dd9af5b1d037a867","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-12-15 00:59:05.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e Thanks for this change, please see my comment inline\n\nThanks for this review , done.","accounts_in_message":[],"_revision_number":2},{"id":"f5e461b3ef75ca933223e2df58374d762c94faf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 01:04:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8fdf09d8dae45c5b823fd7f75e2d59b : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":2},{"id":"ebad87de3925ed42dde71cfecc72659ec580e7e5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-12-15 03:23:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks, LGTM","accounts_in_message":[],"_revision_number":2},{"id":"964e24abc6bcf6fe7148e3586f76dc61f536cd4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 03:23:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f1ab53241f8a6432be4866136074ee610e170dbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 03:31:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7162a12ee7d4226a3c91a70ce25fb7c : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":2},{"id":"41cc51b057a7576b59f3a2b4a01daff0a4591e57","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 03:31:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6807b31d83215777a9aac2eaceb5690b45da1b0f","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-12-15 03:52:05.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/17/766917/2/check/DellEMC_VNX_Manila/b8ede69/DellEMC_VNX_Manila/83 : SUCCESS in 1h 28m 42s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a5c2b7d467d03afba8a90dfd99fcc647e198ba0f","revisions":{"cae5bd81a7967635d11e8c4fc4463985b15c5000":{"kind":"REWORK","_number":1,"created":"2020-12-14 10:00:10.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/17/766917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/17/766917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/766917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/766917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/766917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/766917/1"}}},"commit":{"parents":[{"commit":"ada5b69c78fd5d7a78bc4cf38d9e47d8274e0e3e","subject":"Adds get default share type function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ada5b69c78fd5d7a78bc4cf38d9e47d8274e0e3e"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-12-14 09:59:28.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-12-14 10:00:05.000000000","tz":480},"subject":"Update TOX_CONSTRAINTS_FILE","message":"Update TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\n-https://zuul-ci.org/docs/zuul-jobs/python-roles.html#rolevar-tox.tox_constraints_file\nThis allows to use lower-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\nChange-Id: I149acb076a9e42f2b86186efc0924a489aa1fa83\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cae5bd81a7967635d11e8c4fc4463985b15c5000"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cae5bd81a7967635d11e8c4fc4463985b15c5000"}]},"branch":"refs/heads/master"},"a5c2b7d467d03afba8a90dfd99fcc647e198ba0f":{"kind":"REWORK","_number":2,"created":"2020-12-15 00:56:41.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/17/766917/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/17/766917/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/766917/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/766917/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/766917/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/766917/2"}}},"commit":{"parents":[{"commit":"ada5b69c78fd5d7a78bc4cf38d9e47d8274e0e3e","subject":"Adds get default share type function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ada5b69c78fd5d7a78bc4cf38d9e47d8274e0e3e"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-12-14 09:59:28.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-12-15 00:56:28.000000000","tz":480},"subject":"Update TOX_CONSTRAINTS_FILE","message":"Update TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\n-https://zuul-ci.org/docs/zuul-jobs/python-roles.html#rolevar-tox.tox_constraints_file\nThis allows to use lower-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\nChange-Id: I149acb076a9e42f2b86186efc0924a489aa1fa83\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a5c2b7d467d03afba8a90dfd99fcc647e198ba0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a5c2b7d467d03afba8a90dfd99fcc647e198ba0f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
