)]}'
{"id":"openstack%2Fmanila-tempest-plugin~769978","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I460a0cc3c92168644ceac34440549d6f01892430","project":"openstack/manila-tempest-plugin","branch":"master","topic":"T","hashtags":[],"change_id":"I460a0cc3c92168644ceac34440549d6f01892430","subject":"DNM -  testing the tests ...","status":"ABANDONED","created":"2021-01-08 18:24:46.000000000","updated":"2021-02-11 13:14:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a797d3386ff9abbff24ba0fce03e3e43bf21daf7","_number":769978,"virtual_id_number":769978,"owner":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-01-14 20:36:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-08 18:25:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-08 22:21:18.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-11 20:05:27.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"c7eeb5a1fb7b42e32ac55ee0adfcbab2ec5f0e06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-08 18:24:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5440ee59460e2f574b2150c89a6ff912dc556a78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 18:25:53.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/devstack-plugin-ceph on branch master.  The error was:\n\n  Job manila-tempest-plugin-cephfs-native not defined\n\nThe error appears in the following job stanza:\n\n  job:\n      name: devstack-plugin-ceph-cephfs-native\n      description: |\n        Runs manila tempest plugin tests with Native CephFS as a manila back\n        end (DHSS\u003dFalse)\n      parent: manila-tempest-plugin-cephfs-native\n\n  in \"openstack/devstack-plugin-ceph/.zuul.yaml@master\", line 152, column 3","accounts_in_message":[],"_revision_number":1},{"id":"3087875107df9100004032f0c335a0b1ed36b7c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-08 18:40:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"491830b1b31cadaea9c4d817745611bfe9e07695","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 18:40:56.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/devstack-plugin-ceph on branch master.  The error was:\n\n  Job manila-tempest-plugin-cephfs-native not defined\n\nThe error appears in the following job stanza:\n\n  job:\n      name: devstack-plugin-ceph-cephfs-native\n      description: |\n        Runs manila tempest plugin tests with Native CephFS as a manila back\n        end (DHSS\u003dFalse)\n      parent: manila-tempest-plugin-cephfs-native\n\n  in \"openstack/devstack-plugin-ceph/.zuul.yaml@master\", line 152, column 3","accounts_in_message":[],"_revision_number":2},{"id":"7ef4e58569fe3ede83492977e0608a74f357ff9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-08 19:07:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6432dff7f1a8ea63ec1e48097a4bf56131b5a607","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 19:08:18.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-image-elements on branch master.  The error was:\n\n  Job manila-tempest-plugin-container not defined\n\nThe error appears in the following job stanza:\n\n  job:\n      name: manila-image-elements-container-scenario\n      parent: manila-tempest-plugin-container\n      pre-run: playbooks/manila-buildimage-docker/run.yaml\n      required-projects:\n        - openstack/manila-image-elements\n        - openstack/manila\n        - openstack/manila-tempest-plugin\n        - openstack/python-manilaclient\n        - openstack/tempest\n      vars:\n        # NOTE(gouthamr): Not running scenario tests yet, need to test them\n        # against the container driver first\n        tempest_test_regex: \u0027(^manila_tempest_tests.tests)(?\u003d.*\\[.*\\bbackend\\b.*\\])\u0027\n        devstack_localrc:\n          # \"docker build\" uploads the image we need to the local registry\n          # manila\u0027s devstack plugin just needs the name of the image to look up\n          MANILA_DOCKER_IMAGE_URL: \"file:///home/zuul/src/opendev.org/openstack/manila-image-elements/manila-docker-container.tar.gz\"\n\n  in \"openstack/manila-image-elements/zuul.d/manila-image-elements-jobs.yaml@master\", line 34, column 3","accounts_in_message":[],"_revision_number":3},{"id":"480ae02d0e7666c2aa1793ea456113b82b9c65d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-08 19:26:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"52a32bd7b91099e4403824bb477ba0474f8390fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 21:11:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8ce5046259944a2aa33451ddf3806cd0 : SUCCESS in 1h 38m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f84b7514acbb71341f44c5f7ffa7694850d2a3d7","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-08 21:31:07.000000000","message":"Patch Set 4:\n\nrecheck\n\npasses 1/1 with scenario tests only at concurrency 1.  If this is consistent, we\u0027ll titrate the concurrency upwards.","accounts_in_message":[],"_revision_number":4},{"id":"86189f75ce15f11fe2cdd97877ea32ef7e807874","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-08 22:21:18.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/78/769978/4/check/DellEMC_VNX_Manila/4e03294/DellEMC_VNX_Manila/110 : SUCCESS in 1h 09m 25s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"8ffbb90bc6476281249c5cc049d0c1fe3d0b9ecc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 23:06:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/39f084ee8c2b444bb7a477d7bbd8f389 : SUCCESS in 1h 31m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b52a1ad43bc741f7c3b36c3a852f4baeef28692c","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-09 00:09:57.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/78/769978/4/check/DellEMC_VNX_Manila/a3d0934/DellEMC_VNX_Manila/111 : SUCCESS in 1h 03m 39s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"178987eeeae307cbe50addc706d7bdf5fbb4b221","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-09 12:43:10.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"227da7c066afb218ba085933e0f2d3835a311be7","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-09 12:44:50.000000000","message":"Patch Set 5:\n\n2 of 2 Ceph-NFS passed at concurrency 1.  I think the issue is not with the back end but\nsomething about the test environment (resource overload?)\n\nTrying with concurrency\u003d4.","accounts_in_message":[],"_revision_number":5},{"id":"227c3cf2017a5bf465587073ef697b8bb9c72450","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-09 13:51:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a143f572b31b471b9c6fae9ba5101813 : FAILURE in 1h 07m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"426b39aac578799b898811766b5199d61391f34c","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-09 14:56:44.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/78/769978/5/check/DellEMC_VNX_Manila/e19d979/DellEMC_VNX_Manila/112 : SUCCESS in 1h 04m 47s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"d0071ad45f1e7158ad5c8638f4ad6d1ad9747482","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-09 15:33:44.000000000","message":"Patch Set 5:\n\nPassed two times with concurrency\u003d1, failed right away with concurrency\u003d4.\n\nTry with concurrency\u003d2.\n\nAim here is not necessarily to \"fix\" the issue by reducing concurrency, but first to get a minimum reproducer -- smallest concurrency value needed to reproduce, then the smallest set of tests within that level of concurrency.","accounts_in_message":[],"_revision_number":5},{"id":"d0f5b6c4a51de93f88a85e6b4af8302c81d0f01a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-09 15:33:52.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"3b97625c64cf36d480244cbcf23b1db842d5fd0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-09 16:48:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0d8dd6260a0e4b02a55c2a4b699e09e3 : SUCCESS in 1h 13m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"906bf56765ce8c9a4111629c2415a2801196750c","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-09 16:59:37.000000000","message":"Patch Set 6:\n\n1/1 passed with concurrency\u003d2","accounts_in_message":[],"_revision_number":6},{"id":"11a3c88510165ede0528c542a7882a52bc766129","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-09 16:59:48.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"3691d4a430a5e64957e7952d62462645247e38b2","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-09 17:54:21.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/78/769978/6/check/DellEMC_VNX_Manila/a5f9510/DellEMC_VNX_Manila/113 : SUCCESS in 1h 05m 19s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":6},{"id":"d7c66a39e4d317844293ce2d373b2df5cf0727c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-09 18:15:21.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3781d5dbbb6441a9aee3e1b4ad133518 : SUCCESS in 1h 13m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"db419e32f0aced07454a60e8e163bd8929bd89e5","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-09 18:28:49.000000000","message":"Patch Set 6:\n\nSeems to work at concurrency\u003d2, failed at concurrency\u003d4 -- bumping to 3.","accounts_in_message":[],"_revision_number":6},{"id":"7a431970811d7b50480ae9379af4edcff4e03019","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-09 18:28:56.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"a6b4ca4675d36dc9f5c9260d014002ca9723003c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-09 20:05:41.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3a4af27fc6bd40408b38bf4362dc4984 : FAILURE in 1h 07m 15s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f740a9fe915d455cc810cf8c3a19f9184ce6358f","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-09 20:28:25.000000000","message":"Patch Set 7:\n\nrecheck\n\n2 test cases failed at concurrency\u003d3.  Let\u0027s see how consistent this is.","accounts_in_message":[],"_revision_number":7},{"id":"97b62e627cdbd45b794bdb099c6a3eb811f8ee68","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-09 21:10:07.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/78/769978/7/check/DellEMC_VNX_Manila/e1270cd/DellEMC_VNX_Manila/115 : SUCCESS in 1h 04m 11s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"f54a968afe908e298ac7bf4d1efacdf9c4cbba40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-09 21:46:22.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/95adbe9dff77445286fcb876ef98f58b : FAILURE in 1h 17m 04s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b04cb8d839370561f375efb730b16b5665ec7da3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-09 21:56:46.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"d85c0290b8659cea5529516ebdab31e000e81e5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-09 22:44:44.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/27e7784650444949b126354cb89bde44 : SUCCESS in 42m 49s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a10d27a223b9220cc369ee408614236f8e6dfcc7","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-09 23:52:08.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/78/769978/8/check/DellEMC_VNX_Manila/1b1a935/DellEMC_VNX_Manila/117 : SUCCESS in 1h 07m 08s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":8},{"id":"c0cf52a9213eecdf278da1a583301bdf4ea404fa","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-10 01:31:43.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"e0538ec80876c3aadb4b990361d77850347fd08d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-10 02:12:55.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6afc1b3e1d3545a3a9060298d2582005 : SUCCESS in 40m 34s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"10a0af734f73c177760023f667ef3f47c0715af5","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-10 03:20:04.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/78/769978/8/check/DellEMC_VNX_Manila/d704d88/DellEMC_VNX_Manila/118 : SUCCESS in 1h 06m 52s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":8},{"id":"c2ba510cdc31c6d452c244faea40647210771734","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-11 16:04:20.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"59497e1f92e41eda4fb82b99515bf1c3c3a2496d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 17:25:30.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/bbcb386772f34ed5ad75c8252cab3e71 : SUCCESS in 1h 13m 30s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"719c5c201916a80524cdefaa27b38b3b86aa8f65","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-11 18:40:42.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/78/769978/9/check/DellEMC_VNX_Manila/d94df86/DellEMC_VNX_Manila/123 : SUCCESS in 1h 14m 56s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":9},{"id":"c822fda3b931f4d3b6c4d9d8102811c9572426c6","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-11 20:05:27.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/78/769978/9/upstream-check/manila-cDOT-manila-tempest-plugin-ss/f622af4/ : SUCCESS in 2h 39m 25s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/78/769978/9/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/92bb8b7/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"1453b5e10aafa055dee304ac4692b82f957506c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-14 17:17:47.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"537a7573880baf57a00cccfc3f32995fc83d18bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 17:18:10.000000000","message":"Patch Set 10: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  The key \"tempest_concurrency\" appears more than once; duplicate keys\n  are not permitted.\n\nThe error appears in the following stanza:\n\n  tempest_concurrency: 2\n        devstack_plugins:\n          devstack-plugin-ceph: https://opendev.org/openstack/devstack-plugin-ceph\n          neutron-dynamic-routing: https://opendev.org/openstack/neutron-dynamic-routing\n        devstack_localrc:\n          VOLUME_BACKING_FILE_SIZE: 60GB\n          SHARE_DRIVER: manila.share.drivers.cephfs.driver.CephFSDriver\n          MANILA_ENABLED_BACKENDS: cephfsnfs\n          ENABLED_SHARE_PROTOCOLS: NFS\n          MANILA_DEFAULT_SHARE_TYPE_EXTRA_SPECS: \u0027snapshot_support\u003dTrue\u0027\n          MANILA_SERVICE_IMAGE_ENABLED: true\n          MANILA_OPTGROUP_cephfsnfs_share_driver: manila.share.drivers.cephfs.driver.CephFSDriver\n          MANILA_OPTGROUP_cephfsnfs_driver_handles_share_servers: false\n          MANILA_OPTGROUP_cephfsnfs_cephfs_protocol_helper_type: NFS\n          MANILA_OPTGROUP_cephfsnfs_cephfs_conf_path: /etc/ceph/ceph.conf\n          MANILA_OPTGROUP_cephfsnfs_cephfs_auth_id: manila\n          MANILA_CEPH_DRIVER: cephfsnfs\n          MANILA_SETUP_IPV6: true\n          NEUTRON_CREATE_INITIAL_NETWORKS: false\n          MANILA_RESTORE_IPV6_DEFAULT_ROUTE: false\n        zuul_copy_output:\n          \u0027/var/log/ganesha\u0027: logs\n          \u0027/etc/ganesha\u0027: logs\n        devstack_local_conf:\n          test-config:\n            $TEMPEST_CONFIG:\n              share:\n                default_share_type_name: default\n                run_ipv6_tests: true\n                run_share_group_tests: false\n                multitenancy_enabled: false\n                suppress_errors_in_cleanup: true\n                backend_names: CEPHFSNFS\n                enable_protocols: nfs\n                capability_storage_protocol: NFS\n                capability_create_share_from_snapshot_support: false\n                image_password: manila\n        #tempest_test_regex: (manila_tempest_tests\\.tests\\.scenario\\.(test_share_extend|test_share_shrink)).*$\n        tempest_concurrency: 1\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 336, column 7","accounts_in_message":[],"_revision_number":10},{"id":"560cfd17a46a09b6119598a3cf6b7e1904e18b72","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-14 17:19:37.000000000","message":"Patch Set 10:\n\nJust logging that we\u0027ve reduced concurrency on the ceph-nfs job to 2.  Seems to improve\nstability but we still see some failures.  So I\u0027m going to run with concurrency\u003d1 here\nin hopes of being able to focus on the failures when they happen.  I am expecting they\nwill be rare when the tests are single threaded.","accounts_in_message":[],"_revision_number":10},{"id":"69bb435428e5cc8db35fd4fe49690e9572b34a3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-14 17:21:48.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"6461f085d68f5500967bc241c3fcc4fc58d85151","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 17:22:21.000000000","message":"Patch Set 11: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":11},{"id":"87f8da5075f0c152751c7be17bda618a7067b34f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-14 17:46:18.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"244edd008c5668ebac551591b8b16320aea70271","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 20:36:45.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/77fdae3cfb6f4b3285d654a7b4d8cf93 : SUCCESS in 2h 03m 29s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"262b46b0006e70f92f8bd65b342befd82ab4e38f","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-14 20:48:58.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"839ef959e7ca40f3fa07e79ff3ae24483bbeac62","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-14 23:00:43.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/78/769978/12/upstream-check/manila-cDOT-manila-tempest-plugin-ss/7eab683/ : SUCCESS in 2h 23m 41s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/78/769978/12/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/4f53094/ : FAILURE in 1h 48m 00s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"3fae317bf569a43a63076c599793c608872e5ef1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 23:30:19.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/370393115df442e9b2d2040cd27774b3 : SUCCESS in 2h 38m 37s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"ab3371ee2ef747584172887984216529cefc998e","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-15 00:01:35.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"22760a3fca6e79620cdeadb354b09e3786bab0e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 02:07:42.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f8ce96cc13d5428e9456f7605d281114 : FAILURE in 2h 01m 47s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"a522b7345fce813c368c78852c7dcd29f6fb80f7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-15 03:23:12.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/78/769978/12/upstream-check/manila-cDOT-manila-tempest-plugin-ss/cc23ed9/ : FAILURE in 3h 01m 15s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/78/769978/12/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/945d279/ : FAILURE in 1h 57m 27s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"a797d3386ff9abbff24ba0fce03e3e43bf21daf7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-02-11 13:14:19.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"676db0aeae40ca0492e5b2bfc0739b918c079f31","revisions":{"83b8bc19efc4e3325e18c747385e1ffac805f53f":{"kind":"REWORK","_number":1,"created":"2021-01-08 18:24:46.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/78/769978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/769978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/1"}}},"commit":{"parents":[{"commit":"a5c2b7d467d03afba8a90dfd99fcc647e198ba0f","subject":"Update TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a5c2b7d467d03afba8a90dfd99fcc647e198ba0f"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-08 18:23:19.000000000","tz":-300},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-08 18:24:25.000000000","tz":-300},"subject":"DNM -  testing the tests ...","message":"DNM -  testing the tests ...\n\nChange-Id: I460a0cc3c92168644ceac34440549d6f01892430\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/83b8bc19efc4e3325e18c747385e1ffac805f53f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/83b8bc19efc4e3325e18c747385e1ffac805f53f"}]},"branch":"refs/heads/master"},"cf5db8e2c54eff2aa03cf814bcde4ba9d8a30905":{"kind":"REWORK","_number":2,"created":"2021-01-08 18:40:30.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/78/769978/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/769978/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/2"}}},"commit":{"parents":[{"commit":"a5c2b7d467d03afba8a90dfd99fcc647e198ba0f","subject":"Update TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a5c2b7d467d03afba8a90dfd99fcc647e198ba0f"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-08 18:23:19.000000000","tz":-300},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-08 18:40:14.000000000","tz":-300},"subject":"DNM -  testing the tests ...","message":"DNM -  testing the tests ...\n\nChange-Id: I460a0cc3c92168644ceac34440549d6f01892430\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cf5db8e2c54eff2aa03cf814bcde4ba9d8a30905"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cf5db8e2c54eff2aa03cf814bcde4ba9d8a30905"}]},"branch":"refs/heads/master"},"139518c16f9fbbfc37ea17135b58fcbc9cbcb38e":{"kind":"REWORK","_number":3,"created":"2021-01-08 19:07:58.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/78/769978/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/769978/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/3"}}},"commit":{"parents":[{"commit":"a5c2b7d467d03afba8a90dfd99fcc647e198ba0f","subject":"Update TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a5c2b7d467d03afba8a90dfd99fcc647e198ba0f"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-08 18:23:19.000000000","tz":-300},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-08 19:05:48.000000000","tz":-300},"subject":"DNM -  testing the tests ...","message":"DNM -  testing the tests ...\n\nChange-Id: I460a0cc3c92168644ceac34440549d6f01892430\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/139518c16f9fbbfc37ea17135b58fcbc9cbcb38e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/139518c16f9fbbfc37ea17135b58fcbc9cbcb38e"}]},"branch":"refs/heads/master"},"45b38029da028d8cd5d488b633055083d202bd35":{"kind":"REWORK","_number":4,"created":"2021-01-08 19:26:12.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/78/769978/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/769978/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/4"}}},"commit":{"parents":[{"commit":"a5c2b7d467d03afba8a90dfd99fcc647e198ba0f","subject":"Update TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a5c2b7d467d03afba8a90dfd99fcc647e198ba0f"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-08 18:23:19.000000000","tz":-300},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-08 19:25:47.000000000","tz":-300},"subject":"DNM -  testing the tests ...","message":"DNM -  testing the tests ...\n\nChange-Id: I460a0cc3c92168644ceac34440549d6f01892430\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/45b38029da028d8cd5d488b633055083d202bd35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/45b38029da028d8cd5d488b633055083d202bd35"}]},"branch":"refs/heads/master"},"bd718c74a0b5bf5b47ca5a641d507411892189d8":{"kind":"REWORK","_number":5,"created":"2021-01-09 12:43:10.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/78/769978/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/769978/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/5"}}},"commit":{"parents":[{"commit":"a5c2b7d467d03afba8a90dfd99fcc647e198ba0f","subject":"Update TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a5c2b7d467d03afba8a90dfd99fcc647e198ba0f"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-08 18:23:19.000000000","tz":-300},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-09 12:42:58.000000000","tz":0},"subject":"DNM -  testing the tests ...","message":"DNM -  testing the tests ...\n\nChange-Id: I460a0cc3c92168644ceac34440549d6f01892430\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bd718c74a0b5bf5b47ca5a641d507411892189d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bd718c74a0b5bf5b47ca5a641d507411892189d8"}]},"branch":"refs/heads/master"},"5eb7b125241980c227bb6098d934c898d29f5e7a":{"kind":"REWORK","_number":6,"created":"2021-01-09 15:33:52.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/78/769978/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/769978/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/6"}}},"commit":{"parents":[{"commit":"a5c2b7d467d03afba8a90dfd99fcc647e198ba0f","subject":"Update TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a5c2b7d467d03afba8a90dfd99fcc647e198ba0f"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-08 18:23:19.000000000","tz":-300},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-09 15:31:24.000000000","tz":0},"subject":"DNM -  testing the tests ...","message":"DNM -  testing the tests ...\n\nChange-Id: I460a0cc3c92168644ceac34440549d6f01892430\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5eb7b125241980c227bb6098d934c898d29f5e7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5eb7b125241980c227bb6098d934c898d29f5e7a"}]},"branch":"refs/heads/master"},"5363115d910ac6e4dcea4ed1d9dcb7086e6bc72b":{"kind":"REWORK","_number":7,"created":"2021-01-09 18:28:56.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/78/769978/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/769978/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/7"}}},"commit":{"parents":[{"commit":"a5c2b7d467d03afba8a90dfd99fcc647e198ba0f","subject":"Update TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a5c2b7d467d03afba8a90dfd99fcc647e198ba0f"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-08 18:23:19.000000000","tz":-300},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-09 18:27:57.000000000","tz":0},"subject":"DNM -  testing the tests ...","message":"DNM -  testing the tests ...\n\nChange-Id: I460a0cc3c92168644ceac34440549d6f01892430\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5363115d910ac6e4dcea4ed1d9dcb7086e6bc72b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5363115d910ac6e4dcea4ed1d9dcb7086e6bc72b"}]},"branch":"refs/heads/master"},"93bd003df666d5120076bebfed84d773c8454dab":{"kind":"REWORK","_number":8,"created":"2021-01-09 21:56:46.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/78/769978/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/769978/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/8"}}},"commit":{"parents":[{"commit":"a5c2b7d467d03afba8a90dfd99fcc647e198ba0f","subject":"Update TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a5c2b7d467d03afba8a90dfd99fcc647e198ba0f"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-08 18:23:19.000000000","tz":-300},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-09 21:56:41.000000000","tz":0},"subject":"DNM -  testing the tests ...","message":"DNM -  testing the tests ...\n\nChange-Id: I460a0cc3c92168644ceac34440549d6f01892430\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/93bd003df666d5120076bebfed84d773c8454dab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/93bd003df666d5120076bebfed84d773c8454dab"}]},"branch":"refs/heads/master"},"5d09deec12dfa996bec38921f103dc3b0fa71c4c":{"kind":"REWORK","_number":9,"created":"2021-01-11 16:04:20.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/78/769978/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/769978/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/9"}}},"commit":{"parents":[{"commit":"a5c2b7d467d03afba8a90dfd99fcc647e198ba0f","subject":"Update TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a5c2b7d467d03afba8a90dfd99fcc647e198ba0f"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-08 18:23:19.000000000","tz":-300},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-11 16:04:07.000000000","tz":0},"subject":"DNM -  testing the tests ...","message":"DNM -  testing the tests ...\n\nChange-Id: I460a0cc3c92168644ceac34440549d6f01892430\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5d09deec12dfa996bec38921f103dc3b0fa71c4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5d09deec12dfa996bec38921f103dc3b0fa71c4c"}]},"branch":"refs/heads/master"},"4a3aeccbeba36b1b286bca73e832fa5eac390b8e":{"kind":"REWORK","_number":10,"created":"2021-01-14 17:17:47.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/78/769978/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/769978/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/10"}}},"commit":{"parents":[{"commit":"a5c2b7d467d03afba8a90dfd99fcc647e198ba0f","subject":"Update TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a5c2b7d467d03afba8a90dfd99fcc647e198ba0f"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-08 18:23:19.000000000","tz":-300},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-14 17:17:34.000000000","tz":0},"subject":"DNM -  testing the tests ...","message":"DNM -  testing the tests ...\n\nChange-Id: I460a0cc3c92168644ceac34440549d6f01892430\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4a3aeccbeba36b1b286bca73e832fa5eac390b8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4a3aeccbeba36b1b286bca73e832fa5eac390b8e"}]},"branch":"refs/heads/master"},"08835aada4350f73e24b665d6ebe161c2ce7438c":{"kind":"REWORK","_number":11,"created":"2021-01-14 17:21:48.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/78/769978/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/769978/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/11"}}},"commit":{"parents":[{"commit":"a5c2b7d467d03afba8a90dfd99fcc647e198ba0f","subject":"Update TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a5c2b7d467d03afba8a90dfd99fcc647e198ba0f"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-08 18:23:19.000000000","tz":-300},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-14 17:21:42.000000000","tz":0},"subject":"DNM -  testing the tests ...","message":"DNM -  testing the tests ...\n\nChange-Id: I460a0cc3c92168644ceac34440549d6f01892430\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/08835aada4350f73e24b665d6ebe161c2ce7438c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/08835aada4350f73e24b665d6ebe161c2ce7438c"}]},"branch":"refs/heads/master"},"676db0aeae40ca0492e5b2bfc0739b918c079f31":{"kind":"REWORK","_number":12,"created":"2021-01-14 17:46:18.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/78/769978/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/769978/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/769978/12"}}},"commit":{"parents":[{"commit":"b3a3c420c7cdb72fee275aa71491e5fba8c304d7","subject":"[CI] Switch cephfs-nfs job to focal fossa","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b3a3c420c7cdb72fee275aa71491e5fba8c304d7"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-08 18:23:19.000000000","tz":-300},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-01-14 17:46:01.000000000","tz":-300},"subject":"DNM -  testing the tests ...","message":"DNM -  testing the tests ...\n\nChange-Id: I460a0cc3c92168644ceac34440549d6f01892430\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/676db0aeae40ca0492e5b2bfc0739b918c079f31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/676db0aeae40ca0492e5b2bfc0739b918c079f31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
