)]}'
{"id":"openstack%2Fmanila-tempest-plugin~770954","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I15bfbf6e1ca604f2bf2fcf12dcd4e67bc728fe5e","project":"openstack/manila-tempest-plugin","branch":"master","topic":"fix-more-ddtdata-randomness","hashtags":[],"change_id":"I15bfbf6e1ca604f2bf2fcf12dcd4e67bc728fe5e","subject":"Remove further randomness when calling ddt.data()","status":"MERGED","created":"2021-01-15 15:51:50.000000000","updated":"2021-01-22 21:28:49.000000000","submitted":"2021-01-22 21:15:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"770954-fix-more-ddtdata-randomness","meta_rev_id":"1ffdcdb8ae9cfc808318b54abdb61c70c1bf75a2","_number":770954,"virtual_id_number":770954,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-22 21:15:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2021-01-22 19:53:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2021-01-22 20:01:34.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":2,"date":"2021-01-22 19:08:27.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2021-01-22 20:02:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-15 17:25:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-15 17:47:47.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-01-16 19:48:34.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-16 20:58:05.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-21 17:49:43.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-01-22 15:35:50.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"CC"},{"updated":"2021-01-22 15:37:07.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-01-22 19:08:27.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"}],"messages":[{"id":"734362ee0bdfec7fad3f1360f971da9ee654870b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-01-15 15:51:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30805a9b0e519f92e07bd37f574309b482ba9030","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 17:25:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/939573d1df934c22919fb02ffb32dfa7 : SUCCESS in 3m 23s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bf2abd3abbc04ba88b480d406f076192 : SUCCESS in 33m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/daa215d8c1e048f2bb13863f4a4d8c7e : SUCCESS in 30m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ee31090675284061bb87f98c174c5837 : FAILURE in 1h 17m 05s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ad87363b0d0d46dba9e2f7d261234b68 : FAILURE in 1h 06m 48s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0a9a2e9bedcb4303878ead4d5eeeffc9 : SUCCESS in 1h 00m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/98252d007ecf45088a40e0401bb22361 : SUCCESS in 1h 24m 13s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0529d4a376fa48dfb6aa29a78a69e651 : SUCCESS in 51m 11s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fc752ef10b144360820b20b0ae89225f : POST_FAILURE in 1h 20m 16s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8f034d04e4054383bc0765055e784fcd : SUCCESS in 22m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"474a8333eef55b52fc5ac1abf2a4d173c7ac4ce8","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-01-15 17:45:26.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nrecheck\n\nThe failures don\u0027t seem related to the patch; when the affected jobs are executed, they seem to be generated in the proper order.","accounts_in_message":[],"_revision_number":1},{"id":"91842c781016f3bbb070f960be99e9038473c585","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-15 17:47:47.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/54/770954/1/check/DellEMC_Unity_Manila_dhss_true/856bef8/DellEMC_Unity_Manila_dhss_true/168 : SUCCESS in 1h 55m 39s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/54/770954/1/check/DellEMC_Unity_Manila_dhss_false/d48f08f/DellEMC_Unity_Manila_dhss_false/153 : SUCCESS in 50m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"1d02bde4464a857e30d33c01f518e87c685a887b","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-01-16 16:31:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"faff38e2e5a1cea6d6dc2d9585e2133509471368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 18:33:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32c0ccf04f3d47f5825b5c67b41344d1 : SUCCESS in 9m 45s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/24118e5b7bfc40a5927f17d864e455b3 : SUCCESS in 26m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f5cc4b3f08be465490bfcf25b944f20f : SUCCESS in 35m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cad267a727864cc1937f3718d5f62a23 : SUCCESS in 1h 30m 42s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/06be917a51cb4a898e97161b458e737d : FAILURE in 1h 25m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d02ca43fe339406fae42f3ae5980adbc : SUCCESS in 1h 02m 07s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e9123e29db4a4645bc254272ab71e2ad : SUCCESS in 1h 35m 58s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7897c9cebfab4a378b7f4dd9da37a43c : SUCCESS in 51m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5048d95e9fcd43799d0f5e543c631873 : FAILURE in 1h 57m 07s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c93fec63c2834cbaa4a85642af4ab17e : SUCCESS in 27m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b798bfd6ab41ac4a2ac31bf3b98941156d456c62","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-16 19:48:34.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/54/770954/1/check/DellEMC_VNX_Manila/6f1cca9/DellEMC_VNX_Manila/134 : SUCCESS in 1h 14m 40s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"16850d0c31993cfd2f48ddadf8cea8f9f1c5093b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-16 20:25:16.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/54/770954/1/check/DellEMC_Unity_Manila_dhss_true/3b4b813/DellEMC_Unity_Manila_dhss_true/171 : SUCCESS in 1h 51m 23s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/54/770954/1/check/DellEMC_Unity_Manila_dhss_false/09bcb24/DellEMC_Unity_Manila_dhss_false/156 : SUCCESS in 52m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"1f05f50bbe9fc1408cb13ffc1d88dbf8ec56983d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-16 20:58:05.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/54/770954/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/d70541a/ : SUCCESS in 2h 24m 10s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/54/770954/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/49c5fa1/ : FAILURE in 30m 22s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"71c513b7d45625e63677591f97864a879212d238","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-21 17:49:43.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nThanks, LGTM","accounts_in_message":[],"_revision_number":1},{"id":"23af35fb00abae29b1486a6328a916c8ada33341","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-01-22 15:35:50.000000000","message":"Patch Set 1:\n\nLGTM.  Noticed a minor typo in commit message, de-deplicate","accounts_in_message":[],"_revision_number":1},{"id":"d9ce8a217d05c752f5424f18e31dc3ea2ea6b68f","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-01-22 15:37:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5ac847d0208b6ef4438729552f6dfa0a138317ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-01-22 15:42:18.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"10b7ff9a9489dd0a62516fc93d8539260675513d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-22 16:38:40.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/54/770954/2/check/DellEMC_Unity_Manila_dhss_true/e099346/DellEMC_Unity_Manila_dhss_true/202 : FAILURE in 55m 53s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/54/770954/2/check/DellEMC_Unity_Manila_dhss_false/492048d/DellEMC_Unity_Manila_dhss_false/187 : SUCCESS in 52m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"e65aae222f4ab2f9c8004b7122e30ab49fb8ee06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 18:57:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69e0c5a1c2da47c095ff84c8e1cd3a5b : SUCCESS in 4m 03s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3bb9c8e90e5047e2844faa740bfc0992 : SUCCESS in 33m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4c2cd8ae411c4c4c8ab09a2a939fc2d9 : SUCCESS in 42m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/75f1a48c559c4fa997f21abaf8ab42a1 : SUCCESS in 1h 10m 22s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/97dad10ed0a2455cb3b5307f8510e7f3 : SUCCESS in 1h 16m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/5db92180e7cb49be86659da26be5b122 : SUCCESS in 1h 00m 24s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/24acf24b11a647ca8b56a7b2137451ab : SUCCESS in 1h 18m 54s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d3d2401629ea4a87814c5db274dbf83e : SUCCESS in 45m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b2cfa750b3e64e57ab2ebfdd02f16f8d : TIMED_OUT in 3h 05m 09s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1105f751e4f04086b9f4916817fa3f86 : SUCCESS in 25m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b66cb661ed26ac27a2a9910d8a8a46aa6d2619d2","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-22 19:08:27.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks, Luigi!","accounts_in_message":[],"_revision_number":2},{"id":"d2fb3f6a8e1f95726f5910b9221b31198f2245e4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-22 19:53:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"321f61d22d3248eba6f5f51dc106466f2c55226c","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-01-22 20:01:34.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c621befb8868bb394f816837509156f97bcb80b1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-22 20:02:18.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6cef7d261ca430b8559c32f338cf886ec4ba6f36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 20:02:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"af5015f3df045348ea85008b41217013cde1cc3f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-22 20:05:06.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/54/770954/2/check/DellEMC_Unity_Manila_dhss_true/c463f19/DellEMC_Unity_Manila_dhss_true/205 : FAILURE in 52m 15s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/54/770954/2/check/DellEMC_Unity_Manila_dhss_false/bc5a737/DellEMC_Unity_Manila_dhss_false/190 : SUCCESS in 1h 02m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"5550429134df98fd52e53b02bba909134b3c721f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 21:15:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f652933b982645a582c4d252d900da1a : SUCCESS in 3m 35s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1670496f815e4892a5b18d7c5b9978ed : SUCCESS in 37m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2afdbf80cb4b4b77af71eea9c5799a10 : SUCCESS in 37m 24s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/169eea6f18e84c7c96e9e60714eec8b9 : SUCCESS in 1h 06m 54s","accounts_in_message":[],"_revision_number":2},{"id":"f0983b779833206f444660e639422974650ab198","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 21:15:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1ffdcdb8ae9cfc808318b54abdb61c70c1bf75a2","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-22 21:28:49.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/54/770954/2/check/DellEMC_VNX_Manila/2d9b716/DellEMC_VNX_Manila/144 : FAILURE in 1h 23m 11s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fe6e603e4dbcea2e9fd632ce983c7d69cc18dcd4","revisions":{"bfd218b45d09e18bf36ae4ef6ba33040a6dfd544":{"kind":"REWORK","_number":1,"created":"2021-01-15 15:51:50.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/54/770954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/54/770954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/770954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/770954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/770954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/770954/1"}}},"commit":{"parents":[{"commit":"b3a3c420c7cdb72fee275aa71491e5fba8c304d7","subject":"[CI] Switch cephfs-nfs job to focal fossa","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b3a3c420c7cdb72fee275aa71491e5fba8c304d7"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2021-01-15 15:50:58.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2021-01-15 15:50:58.000000000","tz":60},"subject":"Remove further randomness when calling ddt.data()","message":"Remove further randomness when calling ddt.data()\n\nAvoid the usage of set to de-deplicate the lists and\ninstead use the specific utils.deduplicate function,\nwhich preserves the original order.\n\nThis is a follow-up of: I9cbd26016238c25487ac8104c1188cd2cf4f467e\n\nChange-Id: I15bfbf6e1ca604f2bf2fcf12dcd4e67bc728fe5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bfd218b45d09e18bf36ae4ef6ba33040a6dfd544"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bfd218b45d09e18bf36ae4ef6ba33040a6dfd544"}]},"branch":"refs/heads/master"},"fe6e603e4dbcea2e9fd632ce983c7d69cc18dcd4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-22 15:42:18.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/54/770954/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/54/770954/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/770954/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/770954/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/770954/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/770954/2"}}},"commit":{"parents":[{"commit":"b3a3c420c7cdb72fee275aa71491e5fba8c304d7","subject":"[CI] Switch cephfs-nfs job to focal fossa","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b3a3c420c7cdb72fee275aa71491e5fba8c304d7"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2021-01-15 15:50:58.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2021-01-22 15:42:09.000000000","tz":0},"subject":"Remove further randomness when calling ddt.data()","message":"Remove further randomness when calling ddt.data()\n\nAvoid the usage of set to dedeplicate the lists and\ninstead use the specific utils.deduplicate function,\nwhich preserves the original order.\n\nThis is a follow-up of: I9cbd26016238c25487ac8104c1188cd2cf4f467e\n\nChange-Id: I15bfbf6e1ca604f2bf2fcf12dcd4e67bc728fe5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fe6e603e4dbcea2e9fd632ce983c7d69cc18dcd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fe6e603e4dbcea2e9fd632ce983c7d69cc18dcd4"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
