)]}'
{"id":"openstack%2Fmanila-tempest-plugin~771382","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I477d64cbda446c53f749d9f36acfbd2d198a5272","project":"openstack/manila-tempest-plugin","branch":"master","topic":"skip_checks","hashtags":[],"change_id":"I477d64cbda446c53f749d9f36acfbd2d198a5272","subject":"Moving skip conditions under skip_check classmethod","status":"MERGED","created":"2021-01-19 08:44:20.000000000","updated":"2021-01-19 22:33:09.000000000","submitted":"2021-01-19 22:33:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771382-skip_checks","meta_rev_id":"9e13f921587be72b44e8d4363e3727b9f41dcce5","_number":771382,"virtual_id_number":771382,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-19 22:33:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2021-01-19 19:00:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2021-01-19 21:05:15.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2021-01-19 21:05:15.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"reviewers":{"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-19 08:48:11.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2021-01-19 08:48:11.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-01-19 08:48:11.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-01-19 08:48:11.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-01-19 10:35:28.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-01-19 10:59:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-19 15:26:24.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-19 18:32:13.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"8e73dd4bdb59f4db7c156ad35c478269d86eefdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-19 08:44:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd8e5c23d8f330785f7d9554106a6c839c92a82c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-19 10:35:28.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/82/771382/1/check/DellEMC_Unity_Manila_dhss_true/02f1000/DellEMC_Unity_Manila_dhss_true/174 : FAILURE in 1h 50m 50s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/82/771382/1/check/DellEMC_Unity_Manila_dhss_false/c7f4773/DellEMC_Unity_Manila_dhss_false/159 : FAILURE in 50m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"c3b609bd672e7582784072ba14a77c5720b772cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 10:59:34.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b28f24cb9a2e4277aee9d1558c8db361 : FAILURE in 3m 09s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/611b9d2e1c374ae3bd1318e860887f67 : FAILURE in 33m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/bcac930396c84f0b9172e76e0012be63 : FAILURE in 30m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/072af2134aa84ad6b71e46a8542bebba : FAILURE in 1h 19m 24s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f5c4ef2bb90b496db10aa34e8afdf9be : FAILURE in 1h 17m 35s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/5d6de90927f14bf4b70f89f562b12096 : FAILURE in 58m 44s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/178db39d5eba4d73af9172bbf0cc6802 : FAILURE in 1h 18m 32s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5d36a329b7e746d9ab891b2cd94b077c : FAILURE in 43m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/01d2d889990a40d1b29e13f956aa2b80 : FAILURE in 2h 08m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1974e3fa518f45dba6415e3bec60a290 : FAILURE in 24m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1aa99cb597ef0be3bb8502cfc5d573e6c3683d28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-19 11:58:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e7f64756ba94c5eedcd7ea6d03f9c554f3268783","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-19 13:05:13.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/82/771382/2/check/DellEMC_Unity_Manila_dhss_true/7536457/DellEMC_Unity_Manila_dhss_true/176 : FAILURE in 1h 06m 06s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/82/771382/2/check/DellEMC_Unity_Manila_dhss_false/3b52a39/DellEMC_Unity_Manila_dhss_false/161 : SUCCESS in 50m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"f697b6d1b20e8c66c4cbf8cd60f1010aec876773","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 14:12:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a68f5fe11de4be3972343849dde354a : SUCCESS in 3m 17s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e9a8b61b304e4c20b16721a9b5f84c14 : SUCCESS in 28m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5c7dd07a3d154ae8b827971755e39cb8 : SUCCESS in 34m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f5ff3dd70e2e4e54ba72ddba59a85e82 : SUCCESS in 1h 17m 53s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/26d964a6befe4a9cab65fd94f49e1413 : SUCCESS in 1h 04m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/276c33522de54ec599df1d3462ddaa42 : SUCCESS in 1h 16m 52s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1cda0cdec75947f891221fbf53a9b1b6 : FAILURE in 1h 12m 03s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a4a396bd6fd64d849685c6581701f874 : SUCCESS in 45m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3851c0daad594ceda8df606958152dd5 : POST_FAILURE in 1h 15m 05s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/536e17db26a7445cacad36e26f08c0fd : FAILURE in 22m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7fc29985c3bfc3d734bcc4d453d5e116575764c2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-19 15:08:29.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/82/771382/2/check/DellEMC_Unity_Manila_dhss_true/4194f79/DellEMC_Unity_Manila_dhss_true/178 : FAILURE in 55m 17s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/82/771382/2/check/DellEMC_Unity_Manila_dhss_false/5a56a0e/DellEMC_Unity_Manila_dhss_false/163 : SUCCESS in 53m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"0bfd138e47cf62dfb8529186d25f811c77e863c1","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-19 15:26:24.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/82/771382/2/check/DellEMC_VNX_Manila/f0f013c/DellEMC_VNX_Manila/136 : SUCCESS in 1h 13m 12s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"e7f1063ccf81c88bf36b7b3f86ce97eb238f9ced","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-19 18:32:13.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/82/771382/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/2793358/ : SUCCESS in 3h 38m 57s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/82/771382/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/0965fc7/ : SUCCESS in 1h 48m 22s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"cfe39df40041489242513586e63918121c564d43","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-19 19:00:02.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks, Liron - LGTM","accounts_in_message":[],"_revision_number":2},{"id":"3950206426bc1030b0ef52f9c45d4554c78f9778","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-19 21:05:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks, Liron.  I agree with this change.","accounts_in_message":[],"_revision_number":2},{"id":"d86f669d9c0810cb23a94fc4ae7e9fa8f5785653","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 21:05:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e2a4927c4ba7dc13ab8cd10ed66ebf8f80a6f7a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 22:33:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fe67bfc554c48f98a298abe790c74d6 : SUCCESS in 2m 49s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c60464e4eb394a3e901005d3907b9fdb : SUCCESS in 30m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a3db5059179d4c3c8e26b6576e63cf25 : SUCCESS in 36m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ef92a3aab742475d97bd49aa43731676 : SUCCESS in 1h 22m 20s","accounts_in_message":[],"_revision_number":2},{"id":"9e13f921587be72b44e8d4363e3727b9f41dcce5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 22:33:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7f8e0ac2a190210f69d283af937c3aed0d199d83","revisions":{"7de9694917b8cc0a99bfbb8a5156047f81c9ae5a":{"kind":"REWORK","_number":1,"created":"2021-01-19 08:44:20.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/82/771382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/82/771382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/771382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/771382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/771382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/771382/1"}}},"commit":{"parents":[{"commit":"b3a3c420c7cdb72fee275aa71491e5fba8c304d7","subject":"[CI] Switch cephfs-nfs job to focal fossa","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b3a3c420c7cdb72fee275aa71491e5fba8c304d7"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-19 08:32:17.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-19 08:32:17.000000000","tz":120},"subject":"Moving skip conditions under skip_check classmethod","message":"Moving skip conditions under skip_check classmethod\n\nSkipping conditions shouldn\u0027t be under resource_setup\nclassmethod. This method is used to create resources\nfor test methods.\n\nChange-Id: I477d64cbda446c53f749d9f36acfbd2d198a5272\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7de9694917b8cc0a99bfbb8a5156047f81c9ae5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7de9694917b8cc0a99bfbb8a5156047f81c9ae5a"}]},"branch":"refs/heads/master"},"7f8e0ac2a190210f69d283af937c3aed0d199d83":{"kind":"REWORK","_number":2,"created":"2021-01-19 11:58:52.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/82/771382/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/82/771382/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/771382/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/771382/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/771382/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/771382/2"}}},"commit":{"parents":[{"commit":"b3a3c420c7cdb72fee275aa71491e5fba8c304d7","subject":"[CI] Switch cephfs-nfs job to focal fossa","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b3a3c420c7cdb72fee275aa71491e5fba8c304d7"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-19 08:32:17.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-19 11:58:32.000000000","tz":120},"subject":"Moving skip conditions under skip_check classmethod","message":"Moving skip conditions under skip_check classmethod\n\nSkipping conditions shouldn\u0027t be under resource_setup\nclassmethod. This method is used to create resources\nfor test methods.\n\nChange-Id: I477d64cbda446c53f749d9f36acfbd2d198a5272\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7f8e0ac2a190210f69d283af937c3aed0d199d83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7f8e0ac2a190210f69d283af937c3aed0d199d83"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
