)]}'
{"id":"openstack%2Fmanila-tempest-plugin~771391","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I9c4e24c098dc510f3e9593a27c2ea22540da7a30","project":"openstack/manila-tempest-plugin","branch":"master","topic":"tag","hashtags":[],"change_id":"I9c4e24c098dc510f3e9593a27c2ea22540da7a30","subject":"Fix test tag","status":"MERGED","created":"2021-01-19 09:41:08.000000000","updated":"2021-01-19 20:06:59.000000000","submitted":"2021-01-19 20:06:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771391-tag","meta_rev_id":"e0022bceef5cba0f6fe06f9546bfe8a9dc802a7b","_number":771391,"virtual_id_number":771391,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-19 20:06:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2021-01-19 18:46:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2021-01-19 14:36:24.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":2,"date":"2021-01-19 12:16:17.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2021-01-19 18:46:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"reviewers":{"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-19 09:41:58.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-01-19 09:41:58.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-01-19 09:41:58.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-01-19 09:41:58.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2021-01-19 11:52:28.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-01-19 12:27:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-19 13:33:34.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-19 13:40:14.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"0269a93cdbeede251f84d7cb113a3d2dfe524497","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-19 09:41:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d15fefac5bafb1f2828fd3f733d16daea15f590b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-19 11:52:28.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/91/771391/1/check/DellEMC_Unity_Manila_dhss_true/2b8938c/DellEMC_Unity_Manila_dhss_true/175 : FAILURE in 1h 16m 55s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/91/771391/1/check/DellEMC_Unity_Manila_dhss_false/ba4df9c/DellEMC_Unity_Manila_dhss_false/160 : SUCCESS in 50m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"868ec89e1c567226542a54a0236185b5e837e515","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-19 12:16:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3b438f9fb4164dab61e4608a4398fb147ffbfe75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 12:27:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a5ca2fe1db44b148ef3c761248227c8 : SUCCESS in 3m 30s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f8301a67802f4dbcb2a5693d93111d7d : SUCCESS in 31m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cc03c0e9159e426eba180c3783a69cdd : SUCCESS in 38m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5494ff6644da43fe85550ec8d803246d : SUCCESS in 1h 07m 41s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9df06701cfe940c6b1f6222838130180 : FAILURE in 1h 16m 36s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/479a19c0cdbd4fe2b302060a7628110f : SUCCESS in 1h 06m 13s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/fe570d5533634d339f0418cf7b915b65 : SUCCESS in 1h 28m 50s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a74f5c7533c14cbda7dc7d770dff5932 : SUCCESS in 44m 56s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8d0a73b4bd864642aeeb7efa916e0366 : FAILURE in 2h 19m 10s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e7dbdb122e3a482b9b754480d42b9ffd : SUCCESS in 32m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31d371e05abacdf7d63199c54f7eebbce6978565","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-19 13:33:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/91/771391/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/4e76945/ : FAILURE in 1h 06m 09s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/91/771391/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/2e89148/ : FAILURE in 53s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"da6d8d26ac8053883fc6bb30752938a6b86d8cf2","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-19 13:40:14.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/91/771391/1/check/DellEMC_VNX_Manila/b623da4/DellEMC_VNX_Manila/135 : SUCCESS in 1h 12m 53s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"bbc2e71f0888c1f42c7d5181b5406668fe1c3598","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-19 14:03:12.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/91/771391/1/check/DellEMC_Unity_Manila_dhss_true/738b45a/DellEMC_Unity_Manila_dhss_true/177 : FAILURE in 57m 53s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/91/771391/1/check/DellEMC_Unity_Manila_dhss_false/a4ee3c7/DellEMC_Unity_Manila_dhss_false/162 : SUCCESS in 48m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"4fecd42b9eb88c4cd2bd2fefa660ef14f2dcf887","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-01-19 14:36:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"932332adaf0a8bc01330ee7ade56917593a7e608","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-19 18:46:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"46fe8b414a7031fabcb47a4a027a432c9915e212","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 18:47:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"45323d81826f519f958a676b989115561a394655","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 20:06:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b8901b2cbe442d29448c52702206972 : SUCCESS in 3m 05s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5a4edbc44dde4fa9863f199cc946b7d7 : SUCCESS in 36m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f6976fc79d3443f7a47a35281a7ba003 : SUCCESS in 1h 02m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/809033bb0667495d8124d756c8ea5004 : SUCCESS in 1h 12m 24s","accounts_in_message":[],"_revision_number":1},{"id":"e0022bceef5cba0f6fe06f9546bfe8a9dc802a7b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 20:06:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7cba5e98d6cbe11ab98208e21c0005579888d514","revisions":{"7cba5e98d6cbe11ab98208e21c0005579888d514":{"kind":"REWORK","_number":1,"created":"2021-01-19 09:41:08.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/91/771391/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/91/771391/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/91/771391/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/91/771391/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/91/771391/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/91/771391/1"}}},"commit":{"parents":[{"commit":"b3a3c420c7cdb72fee275aa71491e5fba8c304d7","subject":"[CI] Switch cephfs-nfs job to focal fossa","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b3a3c420c7cdb72fee275aa71491e5fba8c304d7"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-19 08:49:20.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-19 08:49:20.000000000","tz":120},"subject":"Fix test tag","message":"Fix test tag\n\nThis test requires all manila services running\nand should be tagged as TAG_API_WITH_BACKEND.\n\nChange-Id: I9c4e24c098dc510f3e9593a27c2ea22540da7a30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7cba5e98d6cbe11ab98208e21c0005579888d514"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7cba5e98d6cbe11ab98208e21c0005579888d514"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
