)]}'
{"id":"openstack%2Fmanila-tempest-plugin~771485","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I9f0d50a325139e6067e4339533d4b01a322df7a8","project":"openstack/manila-tempest-plugin","branch":"master","topic":"waiters","hashtags":[],"change_id":"I9f0d50a325139e6067e4339533d4b01a322df7a8","subject":"Moving resource waiters to a new module","status":"MERGED","created":"2021-01-19 16:09:35.000000000","updated":"2021-01-21 20:30:56.000000000","submitted":"2021-01-21 20:30:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"771485-waiters","meta_rev_id":"fdf4415a8424a46b9690769af939aadd0b9aa4c8","_number":771485,"virtual_id_number":771485,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-21 20:30:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2021-01-21 17:41:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2021-01-21 19:18:13.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2021-01-21 19:18:13.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-20 00:38:15.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-01-20 00:59:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-20 10:13:06.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2021-01-20 10:13:06.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-01-20 10:13:06.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-01-20 10:13:06.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-01-20 13:46:52.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-20 14:59:27.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"4f4c0d570fd86df6d819e114597c87c0133faad9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-19 16:09:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e32c4f66f3d3c8fa67c36bfcb33a987e274bd31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-19 19:39:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5577922b596ef97b4dcda88e0ba4164d1dc03e65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-19 19:59:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"63b806f85d3044c00b47b375b21e127ed9dbcd1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-19 20:48:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b21a72ec553e6329b624ee3593bd758934eed873","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-20 00:38:15.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/85/771485/4/check/DellEMC_Unity_Manila_dhss_true/31b723b/DellEMC_Unity_Manila_dhss_true/182 : FAILURE in 51m 47s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/85/771485/4/check/DellEMC_Unity_Manila_dhss_false/4bea225/DellEMC_Unity_Manila_dhss_false/167 : FAILURE in 3h 49m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"cdb02caa93e9cd9c4442c4fd22cda7e6929a3e1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 00:59:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62665d81b26b437485737603d772ed34 : SUCCESS in 3m 13s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6331cb3970474692b90e6ec3694d9f4e : TIMED_OUT in 1h 02m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d0f496f8a28e4ec2a87387c4cb3b9bb1 : TIMED_OUT in 1h 02m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/46d8f73491694f15bb10fbf83941d6fb : TIMED_OUT in 3h 02m 41s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b19618afc9c7468196413e87347dc47c : TIMED_OUT in 3h 02m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b26a0478e5614fddbc92dac3a43acc71 : TIMED_OUT in 3h 02m 23s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9c37cad5801f46a495afee6fc7f22478 : TIMED_OUT in 3h 02m 27s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/590dcf1fa9644b7ca9e278f627a83506 : FAILURE in 49m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d8001cddbec547de88ccdee335e5b48f : TIMED_OUT in 3h 04m 01s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b7c908e457ab439ca1a817ebb37a70d7 : FAILURE in 19m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"29d60240b070c400d99eda08e621c03d0a2dd5de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-20 05:55:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1ea826fd6d55a04f97731b2c162b83e69de944a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-20 07:46:55.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9e4c4536566bb2139bf03074d731797e22d314be","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-20 08:43:13.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/85/771485/6/check/DellEMC_Unity_Manila_dhss_true/b80eb1e/DellEMC_Unity_Manila_dhss_true/184 : FAILURE in 55m 22s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/85/771485/6/check/DellEMC_Unity_Manila_dhss_false/7303601/DellEMC_Unity_Manila_dhss_false/169 : SUCCESS in 55m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"62007b7a3364d7f530c32cb8805e6f9c4a6f1d82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 09:47:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f24884168ff4b63970ad902fd7da4f1 : SUCCESS in 2m 47s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d4bcc01573594e2996cacf07062a7d5c : SUCCESS in 43m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/de9de08bde3644389da9db98b5050ad8 : SUCCESS in 34m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b607f44034b94c39a450d90f2298a012 : SUCCESS in 1h 20m 02s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/133a927861cb434793398e1c22c93fce : SUCCESS in 1h 11m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3dcf747c86e04e778a53e032886f5261 : FAILURE in 1h 00m 28s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/bea38f72446c494a86b1fbe3fd4dcff5 : SUCCESS in 1h 20m 26s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1f559f6bbcb849ac97f10a5ff57f3c48 : SUCCESS in 40m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4c5f109871734f3b9933c02c4fc2f5be : FAILURE in 1h 48m 43s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e989553236564a89a9419dd68ea12d76 : FAILURE in 34m 11s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c1e30a47ecdffc4338cee53399ec6c58512d56ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-20 09:55:07.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"cdb31d964a6b5aac535200a6af906b81d4c07a01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-20 10:11:23.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"d895c5ca8baaee7f9b2445bcda17dd7b0bea0f36","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-20 11:10:54.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/85/771485/8/check/DellEMC_Unity_Manila_dhss_true/1f5b56d/DellEMC_Unity_Manila_dhss_true/187 : FAILURE in 59m 06s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/85/771485/8/check/DellEMC_Unity_Manila_dhss_false/e11feb9/DellEMC_Unity_Manila_dhss_false/172 : SUCCESS in 54m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"3caf8c18775a2a68f7436bfda8c6b61a8960e138","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 12:28:51.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53493c37dae8462291c50a2089e24490 : SUCCESS in 4m 07s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c27fa709057b449ba4004c8239d3b001 : SUCCESS in 30m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6eb1a534028e475bb4cf00f264dbed0f : SUCCESS in 38m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f83181809b41436aba1d96e7397780fb : SUCCESS in 1h 15m 31s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1936545c91864347864d52df7e743366 : SUCCESS in 1h 17m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c1a8b37700b44cf492c15b8e1c7e7689 : SUCCESS in 1h 02m 21s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/66f7196f38c047ffab8e2c3ade7df223 : SUCCESS in 1h 35m 27s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/79da611e2c2d49fda84773c2896e4042 : SUCCESS in 41m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b833490781234b4498350a4b798b5826 : FAILURE in 2h 09m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ea8bfa5297e3405fa82dc179261638d6 : SUCCESS in 27m 46s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b73ff0711fa9995efbc2fd309295b2832210037b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-20 13:25:24.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/85/771485/8/check/DellEMC_Unity_Manila_dhss_true/af26697/DellEMC_Unity_Manila_dhss_true/188 : FAILURE in 53m 46s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/85/771485/8/check/DellEMC_Unity_Manila_dhss_false/7b71d67/DellEMC_Unity_Manila_dhss_false/173 : SUCCESS in 56m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"dd79b221a7af6f62951679223e4ff3fbcbc1006e","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-20 13:46:52.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/85/771485/8/check/DellEMC_VNX_Manila/7fb7d37/DellEMC_VNX_Manila/138 : SUCCESS in 1h 17m 45s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":8},{"id":"8149d4839fa8ac9d302cfb3f9c61ad5fc14712c9","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-20 14:59:27.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/85/771485/8/upstream-check/manila-cDOT-manila-tempest-plugin-ss/32a128b/ : SUCCESS in 2h 30m 18s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/85/771485/8/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/f2a296a/ : SUCCESS in 1h 42m 25s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"5b5c436ee62a3953b7cf50499bc4c339451c419f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-20 22:04:23.000000000","message":"Patch Set 8:\n\n(2 comments)\n\nThanks for this refactoring - can you please take a look at the comment inline?","accounts_in_message":[],"_revision_number":8},{"id":"779fc64fd64dcd010757cbe59906f98ddbd87505","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-21 08:14:36.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f4375602cf9cb0e23b7fecf38f2705efef7a951d","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-21 08:16:09.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"d878fe736222c89411530e820d5a78780d457b99","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-21 09:12:09.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/85/771485/9/check/DellEMC_Unity_Manila_dhss_true/4246301/DellEMC_Unity_Manila_dhss_true/189 : FAILURE in 56m 30s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/85/771485/9/check/DellEMC_Unity_Manila_dhss_false/636b184/DellEMC_Unity_Manila_dhss_false/174 : SUCCESS in 51m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"6e8c646c3f87390449f47f48639438d90ea10e6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 10:16:52.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d676dc414cb420c94ca51b43be98a7a : SUCCESS in 3m 38s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/de7d3f1906c445328d2ddfc21f19bae3 : SUCCESS in 30m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3b17593e01b84c7e8b589fa35f2baebf : SUCCESS in 33m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a7c1242e18f246e2bc66dd665c7c5455 : SUCCESS in 1h 14m 20s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/69155792448141b8a4d6b95087c3b958 : SUCCESS in 1h 09m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/8a045b9373f748b18d578e0d7656d593 : SUCCESS in 1h 02m 28s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/fee2702bfaec4b2eaa768ba635bdbc66 : SUCCESS in 1h 17m 24s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/68f11aee399049cf8de178f98cb9ed61 : SUCCESS in 49m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9c25eee3c4964eb88a2aa3ba02c0e2be : FAILURE in 1h 55m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f07653b36e85491cb1aa19f7dc06c966 : SUCCESS in 28m 09s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9b8b3bf10dbbc6cd30894a72a6d743bdfff6664b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-21 11:20:47.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/85/771485/9/check/DellEMC_Unity_Manila_dhss_true/efff56f/DellEMC_Unity_Manila_dhss_true/190 : FAILURE in 55m 27s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/85/771485/9/check/DellEMC_Unity_Manila_dhss_false/e3e50de/DellEMC_Unity_Manila_dhss_false/175 : SUCCESS in 1h 03m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"ce3409a3d96bcfaeedb2488bcccb8d20ae04afbb","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-21 11:32:11.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/85/771485/9/check/DellEMC_VNX_Manila/19cc3a3/DellEMC_VNX_Manila/139 : SUCCESS in 1h 15m 01s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":9},{"id":"423acb88e2c5e552b7f460f6010324aed4147bda","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-21 12:50:37.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/85/771485/9/upstream-check/manila-cDOT-manila-tempest-plugin-ss/ccc1366/ : SUCCESS in 2h 33m 25s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/85/771485/9/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/01d2626/ : SUCCESS in 1h 43m 45s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"ec95b010752403e6a22ce9b96d7a9be7efeadda4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-21 17:41:37.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)\n\nThanks Liron, lets follow up on more consolidation - maybe a nice exercise for someone interested in contributing newly to this project","accounts_in_message":[],"_revision_number":9},{"id":"08acdb024da3be34d65673ea06bc18f4e6a4fab7","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-21 19:18:13.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\nThanks, Liron.  I like this consolidation of the waier functions!","accounts_in_message":[],"_revision_number":9},{"id":"e02c2b008cfcb18f56f235ec8cc9238bdf44b319","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 19:18:31.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"5433c294c6425fadb0d61401ce79ef032f02d382","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-21 20:06:39.000000000","message":"Patch Set 9:\n\n\u003e Patch Set 9: Code-Review+2\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e Thanks Liron, lets follow up on more consolidation - maybe a nice exercise for someone interested in contributing newly to this project\n\nDefinitely possible but I like the less because it requires adding a new parameter.\nBut as it seems this is the only option.","accounts_in_message":[],"_revision_number":9},{"id":"5fb0923f6ea55f34c39a80ee022638f7040fa963","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 20:30:54.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/889d5e972e614721bd4b9f0ebab40892 : SUCCESS in 4m 47s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c39ddabf47474549913f4ebc9920e523 : SUCCESS in 32m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1f13026f24374ec79e77b9943a9e8a77 : SUCCESS in 33m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2bd762afe8334286abfd99ce602387ec : SUCCESS in 1h 10m 08s","accounts_in_message":[],"_revision_number":9},{"id":"fdf4415a8424a46b9690769af939aadd0b9aa4c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 20:30:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"540e74ac3da07cf62a00ab7976fb9382cf2e59a8","revisions":{"16b1fec9863a488fb8c94856dea6d007c6980aa7":{"kind":"REWORK","_number":1,"created":"2021-01-19 16:09:35.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/85/771485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/771485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/1"}}},"commit":{"parents":[{"commit":"b3a3c420c7cdb72fee275aa71491e5fba8c304d7","subject":"[CI] Switch cephfs-nfs job to focal fossa","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b3a3c420c7cdb72fee275aa71491e5fba8c304d7"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-19 16:08:25.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-19 16:09:32.000000000","tz":120},"subject":"[WIP] Move waiter methods under waiters.py","message":"[WIP] Move waiter methods under waiters.py\n\nChange-Id: I9f0d50a325139e6067e4339533d4b01a322df7a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/16b1fec9863a488fb8c94856dea6d007c6980aa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/16b1fec9863a488fb8c94856dea6d007c6980aa7"}]},"branch":"refs/heads/master"},"864d36c557db10f010ff2583cafeb4b9d8dcf3e9":{"kind":"REWORK","_number":2,"created":"2021-01-19 19:39:59.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/85/771485/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/771485/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/2"}}},"commit":{"parents":[{"commit":"b3a3c420c7cdb72fee275aa71491e5fba8c304d7","subject":"[CI] Switch cephfs-nfs job to focal fossa","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b3a3c420c7cdb72fee275aa71491e5fba8c304d7"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-19 16:08:25.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-19 19:39:42.000000000","tz":120},"subject":"[WIP] Move waiter methods under waiters.py","message":"[WIP] Move waiter methods under waiters.py\n\nChange-Id: I9f0d50a325139e6067e4339533d4b01a322df7a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/864d36c557db10f010ff2583cafeb4b9d8dcf3e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/864d36c557db10f010ff2583cafeb4b9d8dcf3e9"}]},"branch":"refs/heads/master"},"d07c7457df574c8ba57eb98c151ec1b9ed75161f":{"kind":"REWORK","_number":3,"created":"2021-01-19 19:59:24.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/85/771485/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/771485/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/3"}}},"commit":{"parents":[{"commit":"b3a3c420c7cdb72fee275aa71491e5fba8c304d7","subject":"[CI] Switch cephfs-nfs job to focal fossa","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b3a3c420c7cdb72fee275aa71491e5fba8c304d7"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-19 16:08:25.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-19 19:59:02.000000000","tz":120},"subject":"[WIP] Move waiter methods under waiters.py","message":"[WIP] Move waiter methods under waiters.py\n\nChange-Id: I9f0d50a325139e6067e4339533d4b01a322df7a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d07c7457df574c8ba57eb98c151ec1b9ed75161f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d07c7457df574c8ba57eb98c151ec1b9ed75161f"}]},"branch":"refs/heads/master"},"b950aac04a41171ef06f40bbc08a65d13278803f":{"kind":"REWORK","_number":4,"created":"2021-01-19 20:48:49.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/85/771485/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/771485/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/4"}}},"commit":{"parents":[{"commit":"b3a3c420c7cdb72fee275aa71491e5fba8c304d7","subject":"[CI] Switch cephfs-nfs job to focal fossa","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b3a3c420c7cdb72fee275aa71491e5fba8c304d7"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-19 16:08:25.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-19 20:48:26.000000000","tz":120},"subject":"[WIP] Move waiter methods under waiters.py","message":"[WIP] Move waiter methods under waiters.py\n\nChange-Id: I9f0d50a325139e6067e4339533d4b01a322df7a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b950aac04a41171ef06f40bbc08a65d13278803f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b950aac04a41171ef06f40bbc08a65d13278803f"}]},"branch":"refs/heads/master"},"e179a42e66dcf0ed6ec0f2b525365d6afe0498ec":{"kind":"REWORK","_number":5,"created":"2021-01-20 05:55:10.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/85/771485/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/771485/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/5"}}},"commit":{"parents":[{"commit":"b3a3c420c7cdb72fee275aa71491e5fba8c304d7","subject":"[CI] Switch cephfs-nfs job to focal fossa","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b3a3c420c7cdb72fee275aa71491e5fba8c304d7"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-19 16:08:25.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-20 05:54:52.000000000","tz":120},"subject":"[WIP] Move waiter methods under waiters.py","message":"[WIP] Move waiter methods under waiters.py\n\nChange-Id: I9f0d50a325139e6067e4339533d4b01a322df7a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e179a42e66dcf0ed6ec0f2b525365d6afe0498ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e179a42e66dcf0ed6ec0f2b525365d6afe0498ec"}]},"branch":"refs/heads/master"},"b9c56ed6372986d22eb91869200b8ea384ec9ce3":{"kind":"REWORK","_number":6,"created":"2021-01-20 07:46:55.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/85/771485/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/771485/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/6"}}},"commit":{"parents":[{"commit":"b3a3c420c7cdb72fee275aa71491e5fba8c304d7","subject":"[CI] Switch cephfs-nfs job to focal fossa","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b3a3c420c7cdb72fee275aa71491e5fba8c304d7"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-19 16:08:25.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-20 07:46:39.000000000","tz":120},"subject":"[WIP] Move waiter methods under waiters.py","message":"[WIP] Move waiter methods under waiters.py\n\nChange-Id: I9f0d50a325139e6067e4339533d4b01a322df7a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b9c56ed6372986d22eb91869200b8ea384ec9ce3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b9c56ed6372986d22eb91869200b8ea384ec9ce3"}]},"branch":"refs/heads/master"},"fcf1b4a976cf9609e7d923ee9e0ff31773e6c904":{"kind":"REWORK","_number":7,"created":"2021-01-20 09:55:07.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/85/771485/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/771485/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/7"}}},"commit":{"parents":[{"commit":"b3a3c420c7cdb72fee275aa71491e5fba8c304d7","subject":"[CI] Switch cephfs-nfs job to focal fossa","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b3a3c420c7cdb72fee275aa71491e5fba8c304d7"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-19 16:08:25.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-20 09:51:25.000000000","tz":120},"subject":"Moving resource waiters to a new module","message":"Moving resource waiters to a new module\n\nCurrently, resource waiters methods are placed under\nservice clients modules. There are three duplicated\nwaiters in both v1 and v2 modules:\n- wait_for_share_status\n- wait_for_snapshot_status\n- wait_for_access_rule_status\n\nThis patch suggests to separate the waiter functions\nfrom client modules and collect them in one place.\n\nChange-Id: I9f0d50a325139e6067e4339533d4b01a322df7a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fcf1b4a976cf9609e7d923ee9e0ff31773e6c904"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fcf1b4a976cf9609e7d923ee9e0ff31773e6c904"}]},"branch":"refs/heads/master"},"8787764137be43c76864546f89845f543ebbf762":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-01-20 10:11:23.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/85/771485/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/771485/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/8"}}},"commit":{"parents":[{"commit":"f5beab4862ac3d148ec4179b82bbd7190485f8f6","subject":"Merge \"Moving skip conditions under skip_check classmethod\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f5beab4862ac3d148ec4179b82bbd7190485f8f6"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-19 16:08:25.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-20 10:10:35.000000000","tz":120},"subject":"Moving resource waiters to a new module","message":"Moving resource waiters to a new module\n\nCurrently, resource waiters methods are placed under\nservice clients modules. There are three duplicated\nwaiters in both v1 and v2 modules:\n- wait_for_share_status\n- wait_for_snapshot_status\n- wait_for_access_rule_status\n\nThis patch suggests to separate the waiter functions\nfrom client modules and collect them in one place.\n\nChange-Id: I9f0d50a325139e6067e4339533d4b01a322df7a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8787764137be43c76864546f89845f543ebbf762"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8787764137be43c76864546f89845f543ebbf762"}]},"branch":"refs/heads/master"},"540e74ac3da07cf62a00ab7976fb9382cf2e59a8":{"kind":"REWORK","_number":9,"created":"2021-01-21 08:14:36.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/85/771485/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/771485/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/771485/9"}}},"commit":{"parents":[{"commit":"f5beab4862ac3d148ec4179b82bbd7190485f8f6","subject":"Merge \"Moving skip conditions under skip_check classmethod\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f5beab4862ac3d148ec4179b82bbd7190485f8f6"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-19 16:08:25.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-21 08:14:19.000000000","tz":120},"subject":"Moving resource waiters to a new module","message":"Moving resource waiters to a new module\n\nCurrently, resource waiters methods are placed under\nservice clients modules. There are three duplicated\nwaiters in both v1 and v2 modules:\n- wait_for_share_status\n- wait_for_snapshot_status\n- wait_for_access_rule_status\n\nThis patch suggests to separate the waiter functions\nfrom client modules and collect them in one place.\n\nChange-Id: I9f0d50a325139e6067e4339533d4b01a322df7a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/540e74ac3da07cf62a00ab7976fb9382cf2e59a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/540e74ac3da07cf62a00ab7976fb9382cf2e59a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
