)]}'
{"id":"openstack%2Fmanila-tempest-plugin~771941","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ibdeb62ddd3d1ae813b90989ae27497874ee0dbe9","project":"openstack/manila-tempest-plugin","branch":"master","topic":"DNM","hashtags":[],"change_id":"Ibdeb62ddd3d1ae813b90989ae27497874ee0dbe9","subject":"DNM","status":"ABANDONED","created":"2021-01-22 08:07:12.000000000","updated":"2021-01-25 05:45:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7bb1d1b40afaec07287269276ec254bed2c19500","_number":771941,"virtual_id_number":771941,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-22 10:51:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-22 09:19:08.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-01-22 10:51:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c3ef01d52adf7c2f264325d6c117a7fba0aa4f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-22 08:07:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1fffa2d060971f92b7c7423974f462deb83b5005","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-22 09:19:08.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/41/771941/1/check/DellEMC_Unity_Manila_dhss_true/29cc9c7/DellEMC_Unity_Manila_dhss_true/201 : FAILURE in 58m 10s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/41/771941/1/check/DellEMC_Unity_Manila_dhss_false/34ef2be/DellEMC_Unity_Manila_dhss_false/186 : SUCCESS in 56m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"07d47d4bb9f8a3d173c77beb649ce8b6f09d0c89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 10:51:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d337fea9dc1c401cbe3ea2504022f8ce : SUCCESS in 3m 54s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f0ae61a17c014ccfb4882c8a23b768ae : SUCCESS in 33m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9ec69d0aca8648519a56b1f0507d4847 : FAILURE in 35m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/68a2f46e872a45eab353362da000a4eb : SUCCESS in 1h 04m 39s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/dcabbd894a094bf7b4d86f2e7c9396de : FAILURE in 1h 17m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7127004cfacb4f21911e7f249e21b4c6 : SUCCESS in 1h 04m 40s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0bdcf3aa0fe448c5b0d6df0caa9754e3 : SUCCESS in 1h 34m 00s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/31c90fc39dc54d989d17215caea4126b : SUCCESS in 52m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c92119ef79244d71a5b477922202fa95 : FAILURE in 1h 52m 03s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/889e46e114af45b8a57a7e4a2f0e9d62 : SUCCESS in 32m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7bb1d1b40afaec07287269276ec254bed2c19500","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-25 05:45:14.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fcd8d207f37bc42c7cecd0272cb7218735af9f64","revisions":{"fcd8d207f37bc42c7cecd0272cb7218735af9f64":{"kind":"REWORK","_number":1,"created":"2021-01-22 08:07:12.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/41/771941/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/41/771941/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/41/771941/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/41/771941/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/41/771941/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/41/771941/1"}}},"commit":{"parents":[{"commit":"540e74ac3da07cf62a00ab7976fb9382cf2e59a8","subject":"Moving resource waiters to a new module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/540e74ac3da07cf62a00ab7976fb9382cf2e59a8"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-22 08:06:56.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-22 08:06:56.000000000","tz":120},"subject":"DNM","message":"DNM\n\nChange-Id: Ibdeb62ddd3d1ae813b90989ae27497874ee0dbe9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fcd8d207f37bc42c7cecd0272cb7218735af9f64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fcd8d207f37bc42c7cecd0272cb7218735af9f64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
