)]}'
{"id":"openstack%2Fmanila-tempest-plugin~772514","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I92f4628051731a9f7e05404b48a96d2ed27a855b","project":"openstack/manila-tempest-plugin","branch":"master","topic":"waiters_consolidation","hashtags":[],"change_id":"I92f4628051731a9f7e05404b48a96d2ed27a855b","subject":"Consolidate waiters methods","status":"MERGED","created":"2021-01-26 12:54:59.000000000","updated":"2021-02-25 02:09:16.000000000","submitted":"2021-02-25 01:04:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772514-waiters_consolidation","meta_rev_id":"b40a7faf5b2ae501f2b3adf2fbdd191bc872bfc2","_number":772514,"virtual_id_number":772514,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-25 01:04:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2021-02-23 22:11:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2021-02-23 20:07:30.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-23 19:20:40.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2021-02-23 22:11:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"}],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-26 16:30:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-26 20:54:42.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-02-01 14:56:06.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-02-02 16:36:55.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2021-02-02 16:36:55.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-02-02 16:36:55.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-02-02 17:58:01.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-19 11:28:14.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"9f8c9d866317e05edeb1e30a681b3914879acf31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-26 12:54:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb0b0c4c841186b0b24971b26d4e56bcaaacabf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-26 13:35:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7389e5e4229dac6cc5d65a07c38e7e7c823bed3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 16:30:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67521a977d5d4027922a29b92a6b0590 : SUCCESS in 3m 02s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e7a06a6dac584d1990a427c30976726d : FAILURE in 27m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/04ae975848db467aaf1bc6dab9789288 : FAILURE in 26m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/243f2035d27948648e5cccb0e4613e93 : FAILURE in 1h 10m 41s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/798de3b53019409aab650276c30c348c : FAILURE in 58m 20s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/818699c21cfc4b6886573c3568a4db4e : FAILURE in 2h 48m 49s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/217ab32abf1647968bc3536a1a3af499 : FAILURE in 1h 29m 20s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0f4ba3ab40d24883b3fe1d26d0c13d76 : FAILURE in 44m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bac70a56554d4de1bb12dcb503866824 : FAILURE in 2h 46m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/88606d04b4cc47e4bec908a71e124c6b : FAILURE in 19m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"35000a4bb61bd71f282907d8a9d4a1f81c3cfec5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-26 19:26:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"181133969225a43c1e9ba9bc8a9f8f56d9a9352c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 21:05:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cfd9ae5c1b74a4290b0de873af88244 : SUCCESS in 5m 18s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e571887678f24223a46af73d5d022859 : FAILURE in 34m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f3bed30421ce4dfe9b69bb6da1a1caa1 : FAILURE in 38m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/55201fe6b57440b0acc025537fe319ac : FAILURE in 1h 14m 49s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0d59d545f4ba446c9d31f378362255fd : FAILURE in 1h 12m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/fe62cfec427249908b0c3be9758c865e : SUCCESS in 1h 03m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1e80a520a7d44844a87b4eee2f28dc84 : SUCCESS in 1h 13m 38s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a5a5a9c4291243568e6a44908cd9a8bd : FAILURE in 42m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2c15d7ef0926493faf82c179ea3dbbf2 : FAILURE in 1h 32m 42s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/776f00690bb546699cd9d19d95e90688 : SUCCESS in 25m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8a94015503b106a9ae642aeba47983b49395e277","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-27 05:53:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"14651d0c7954e79ebce691652a850fd6988fe095","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-27 07:28:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"61258b8656602d97a1018bbdc2e69599a6597c95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-27 08:52:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"357fcbd54cb733281169f639d7473ea3c9dffa78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-27 09:47:52.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"785c8a0856afbd2128c0f0c14efb1209a1c7d33b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 11:58:34.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4bada3ba3c946918416901b16533a52 : SUCCESS in 3m 01s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d06ce23617774f7eabd63284b20f09f8 : SUCCESS in 36m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6a00134873714fd3bec2fef73e11248a : FAILURE in 33m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2cf8690e8c584a0480b4189151e49d8a : FAILURE in 1h 10m 23s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/385e19bd5ae24a9199accbc629ae0715 : SUCCESS in 1h 29m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a2877676cd2646969c35968e998bbe52 : SUCCESS in 1h 01m 31s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c301dded7daf4fa6b06295613557fbbd : SUCCESS in 1h 31m 37s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8785ea567fdf4b49acc0a2fe74bc05c3 : FAILURE in 45m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0a6febd849674585bc7353f460aded3a : FAILURE in 2h 04m 29s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a7b6ad4c54004b3085813351c3863895 : SUCCESS in 23m 59s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6f77b2fc8924543acea5ab80a90e90c1679d3f44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-27 12:55:15.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"08c1cdecef0ba4a76bb733bd3c1f4198cd60b670","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 16:07:23.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e3584ea4637491a8eb5198b52f95858 : SUCCESS in 3m 38s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2e6b99c4daa94554abc3d4db0a62f07b : SUCCESS in 34m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/27a980137c944b8086fe89ca57b330ef : FAILURE in 30m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b07d45b861074c2d868239f720d9882c : SUCCESS in 1h 07m 12s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6a68db94f56d4c8ca4c948135434b313 : SUCCESS in 1h 03m 36s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9f293d5afa0949ca8d38063653e4a29e : SUCCESS in 1h 00m 47s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f255781943734e18a5994c40a8de06dd : FAILURE in 1h 19m 30s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c003370cf02841ecbbfefcfecd91b1cb : FAILURE in 44m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5d3abf4702bb406d92f36f4ec5db6601 : FAILURE in 3h 08m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/61bb582c9aee4526af8b1a46a3be3081 : FAILURE in 36m 55s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"cf1db23a62c906797a0ddedcf83f26f5138b8787","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-01-28 08:16:33.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"cd6bd3f4216fd03f1f86c31701d90a8a2300799b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 10:28:37.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bc59671fce9491d9c54256c63c1bd3b : SUCCESS in 4m 31s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8673ffaa56cf479f860e1f004e3ad2cf : FAILURE in 27m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/59d2d1ee316540b68ae1af7bd362cd9e : FAILURE in 31m 03s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/99abafb30dbf474fa6e805bb869a6352 : SUCCESS in 1h 13m 37s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/aac31173411842d6befcf6f46ad6a8a1 : FAILURE in 1h 12m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/82d6476148f54ac8acc4afc32a5921de : SUCCESS in 59m 52s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6a456d3e7baf4cfaa0c297f809cd6800 : SUCCESS in 1h 28m 01s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/68a72867ae5245b9bb73ad8b31f942f9 : POST_FAILURE in 46m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9504cdd3080e484aa73ca8ae75e34715 : FAILURE in 2h 07m 40s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/76c7e89b25b9407990aa7cb19a639147 : SUCCESS in 25m 03s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"038e412140b74e4d56aafd888a9201ee34a91c41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-02-01 14:47:34.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"a07ef00659b8499dd509558afa79ebd090e7a780","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-01 14:56:06.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/14/772514/10/check/DellEMC_Unity_Manila_dhss_true/2d3b645/DellEMC_Unity_Manila_dhss_true/226 : FAILURE in 28s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/14/772514/10/check/DellEMC_Unity_Manila_dhss_false/8670b0a/DellEMC_Unity_Manila_dhss_false/211 : FAILURE in 8m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"95bb17e101102630265a8cabba7be6dbfd15d5fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 16:34:00.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f7e045bb99e4508bb10e9e457c2c3de : SUCCESS in 2m 48s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b695fa7b1f904c1885641d028fec2140 : FAILURE in 32m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a6666acceec14e2bab8e18066498dfad : FAILURE in 31m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c44252cab9de4e0ba81279023030215e : SUCCESS in 1h 13m 42s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c087991996154de290fb1702c2f2342b : FAILURE in 1h 03m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ac781e3350164dfd88b9c6d9cef8f717 : FAILURE in 57m 44s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d73839db58a2411e93f3a320141c8663 : FAILURE in 1h 23m 21s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5271dd4f07fa427281ca5df9acb18987 : FAILURE in 40m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/afeecabc152e4b6cb8354ff146acb32f : FAILURE in 1h 44m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/58ca4367e7834884bdb41c814d37ab39 : FAILURE in 21m 37s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b3a2125ebccaa046f4f57ea927cc07ab1b40e51d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-02-02 09:52:35.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"ffc61909626aa7673f3772e8305c27c4207bb26e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-02 10:01:14.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/14/772514/11/check/DellEMC_Unity_Manila_dhss_true/548f436/DellEMC_Unity_Manila_dhss_true/233 : FAILURE in 28s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/14/772514/11/check/DellEMC_Unity_Manila_dhss_false/33e5acd/DellEMC_Unity_Manila_dhss_false/218 : FAILURE in 8m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"05fb8dd00c0464e05e21eaa328d0f152cb8353c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 12:48:44.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/963e3cf1f3e24440bc1da08ca07eddd8 : SUCCESS in 3m 29s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3cf526d9c6ae48dbbcb93dd340104a81 : FAILURE in 28m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5a0c28b6f1db45aaa5743395a94c0b56 : SUCCESS in 33m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/54c3bf2d1419442d923f831344a5d6e1 : SUCCESS in 1h 27m 24s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cebd454d93784adb86aa94a8e432bc9f : SUCCESS in 52m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/acf0b7a145aa4ac0862af9c429703378 : SUCCESS in 55m 30s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/18779555f87148bd8868d00c83bda3f3 : SUCCESS in 1h 52m 32s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3616eb98b1f74ec59f1ae218dae7e60d : SUCCESS in 1h 08m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/cbb21f5949a048f686d79f3e4b7fe2c7 : FAILURE in 2h 51m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/76ed6b58eef74820aaf8b0b9e0225aef : SUCCESS in 20m 55s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"93efa696645253d8121f8b4c798c5335ea4e6b0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-02-02 13:02:17.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"811800740ca48384b10038b8aa19f53c9ec663e7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-02 13:12:25.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/14/772514/12/check/DellEMC_Unity_Manila_dhss_true/139bba2/DellEMC_Unity_Manila_dhss_true/235 : FAILURE in 27s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/14/772514/12/check/DellEMC_Unity_Manila_dhss_false/e90d16b/DellEMC_Unity_Manila_dhss_false/220 : FAILURE in 8m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"7c043d4102ef02feaeaec4f94f240fe3963d3396","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-02-02 13:55:45.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"7243a4de7029f36453c8e30f80f2f197415f8a10","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-02 14:04:37.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/14/772514/13/check/DellEMC_Unity_Manila_dhss_true/7214af6/DellEMC_Unity_Manila_dhss_true/236 : FAILURE in 28s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/14/772514/13/check/DellEMC_Unity_Manila_dhss_false/698389a/DellEMC_Unity_Manila_dhss_false/221 : FAILURE in 8m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"97073802b12f6a211c386a45377e14c705856e4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 16:06:44.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a93e5bb70b404198897a1739523b67bb : SUCCESS in 4m 33s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5845f25b6fa84b5f8daa007969adf576 : SUCCESS in 29m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e488f36f47b04937a18c91b2d80e5ea8 : SUCCESS in 31m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1f059c952ee84092930d5601f4b51fad : SUCCESS in 1h 08m 56s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/017c54f412504e0c864004bbc86ae426 : FAILURE in 1h 22m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/8c4fbee65a304e679c223b94d83f668f : SUCCESS in 54m 42s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/78aad439994d4c2db225f0f716f8c09c : SUCCESS in 1h 42m 19s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d4c6a70e632145dd8f5351404f04d58f : SUCCESS in 43m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/78029753d9b3427e8d0befaa20bbbeae : FAILURE in 2h 04m 10s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f419a3528bab425eaccc45e34dba0bdf : SUCCESS in 23m 55s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"a0c28630152884276f3a1734297080df771f5eda","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-02 16:15:19.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/14/772514/13/check/DellEMC_Unity_Manila_dhss_true/eb7e28a/DellEMC_Unity_Manila_dhss_true/240 : FAILURE in 30s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/14/772514/13/check/DellEMC_Unity_Manila_dhss_false/9fed98d/DellEMC_Unity_Manila_dhss_false/225 : FAILURE in 8m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"066665b7d41c20f5e277748bb07ebd963e87e388","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-02 17:58:01.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/14/772514/13/upstream-check/manila-cDOT-manila-tempest-plugin-ss/c503562/ : FAILURE in 5m 03s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/14/772514/13/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/98dfa53/ : SUCCESS in 1h 51m 00s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"b5dc354e6d469acd6c27c94307b7be4bad3bb9f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-02-03 05:34:28.000000000","message":"Patch Set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"c67aeff01d5038bd0bc442cb70c896540a6e0839","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-03 05:43:23.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/14/772514/14/check/DellEMC_Unity_Manila_dhss_true/8ff80e6/DellEMC_Unity_Manila_dhss_true/249 : FAILURE in 30s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/14/772514/14/check/DellEMC_Unity_Manila_dhss_false/c6132f9/DellEMC_Unity_Manila_dhss_false/234 : FAILURE in 8m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"603e4ee8860bcdb324e1dbec9722369cc2b09c4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 07:37:18.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06882a2a472948ccbfe744b05614ca4f : SUCCESS in 2m 55s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3444270c9333449b8aa1bbc47654e31b : SUCCESS in 33m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d22fa48ab7854606b706c0f087364d7f : SUCCESS in 41m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dad36d2a6cbd4616bab13a4bcf1d98fd : SUCCESS in 1h 02m 16s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/19224ba666cb4a3fae46e0d5ebafbc70 : FAILURE in 1h 10m 11s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f6cae12d50b1483cb187a9c987df53bc : SUCCESS in 56m 09s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ba1713da2a5f43198c58732958fc1f7a : SUCCESS in 1h 17m 51s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1f45fb1faac84015ad719a3143ec2633 : SUCCESS in 49m 26s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1aaca3c4b30e4626ad555aeaefe4c32a : FAILURE in 1h 57m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/65ee7ad90fbe478d8718d0ea4f3981d7 : SUCCESS in 23m 08s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"e1743f24f2d9f5ec2b12b013d2ed1c40b70eedf4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-03 07:47:41.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/14/772514/14/check/DellEMC_Unity_Manila_dhss_true/59d8bc3/DellEMC_Unity_Manila_dhss_true/252 : FAILURE in 31s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/14/772514/14/check/DellEMC_Unity_Manila_dhss_false/ee3b2a6/DellEMC_Unity_Manila_dhss_false/237 : FAILURE in 8m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"f90be9a41c7c086b469d00b8fa6678e458d76303","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-03 09:57:31.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/14/772514/14/upstream-check/manila-cDOT-manila-tempest-plugin-ss/e1f6c46/ : SUCCESS in 2h 19m 54s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/14/772514/14/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/e8bbdf9/ : FAILURE in 30m 35s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"0a2300f8d90f7190efe64d1ed7a67ad914b20cf4","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-02-18 12:04:41.000000000","message":"Patch Set 14: Code-Review+2\n\nThanks Liron, nice refactor! The code looks good and the failures in the jobs are not related to this change, so I think it\u0027s good to go.","accounts_in_message":[],"_revision_number":14},{"id":"6fb05e73fc3b656e59eed86fcd2b4e43f20128ca","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-02-18 19:25:51.000000000","message":"Patch Set 14: Code-Review+1\n\nLGTM, thanks Liron","accounts_in_message":[],"_revision_number":14},{"id":"15a9ffc44b89d5f597a139d837c914780f8b58c1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-02-18 19:28:06.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"62098a99719020c41f22d415f75b7ae80c7fd581","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-02-18 19:40:58.000000000","message":"Patch Set 14: Code-Review+1\n\nStill reviewing this patch - will get back to it shortly; however if any of the other core reviewers is looking at this, i\u0027m okay with them merging it if it looks alright!","accounts_in_message":[],"_revision_number":14},{"id":"4c86abcb1f0342571058f131ca5e73989feb7a1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-02-19 05:24:58.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"2c66e3e7613c40bf4dc3549143a9c5c87b86fb6d","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-02-19 05:25:17.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"a9fc93aded859b847d875dd338a12a4c25d8484a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-19 06:35:24.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/14/772514/15/check/DellEMC_Unity_Manila_dhss_true/5894523/DellEMC_Unity_Manila_dhss_true/355 : FAILURE in 1h 10m 06s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/14/772514/15/check/DellEMC_Unity_Manila_dhss_false/49fdeca/DellEMC_Unity_Manila_dhss_false/339 : SUCCESS in 1h 00m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"99e373d6736bd6b877c8d9c64969fb37093a3cb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 07:42:17.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11bb11480b8742fd8d8b2d14e3287860 : SUCCESS in 3m 20s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/58d61853046b41ee9ed7edb99de8b347 : SUCCESS in 33m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e3658a6738ca4c55a2e744f37b1a0707 : FAILURE in 37m 23s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2dc073f51a8c4568bf7358296321eed3 : SUCCESS in 1h 10m 58s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a6a38d5261f8460bb3ff050a50e0bc74 : FAILURE in 1h 19m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b998dab7967046e495078cf8944eecef : SUCCESS in 1h 20m 44s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4ce38b1758924b77adbdeb6d857d2628 : SUCCESS in 1h 28m 56s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/10231f80e0d14721a525564e2a9c5782 : SUCCESS in 46m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6bdd6965761b4907a85a4d364d275925 : FAILURE in 2h 11m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/cc8bbfe7ebf64b37ba511abc32c9f3ad : SUCCESS in 24m 49s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"c0aeb0ae6dff52ca6983c79e1c2c71e5072cb3fd","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-02-19 07:49:01.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"ca4e4574ae0595819c74f53fa479eb05e8e3341e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 10:11:13.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/386d4c9339ba4752bdda09be8be68824 : SUCCESS in 3m 50s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/053353f350704dff9a83505c9b386aa3 : SUCCESS in 31m 39s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ea5ef21a2ed44595af661256e8c8c962 : SUCCESS in 39m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/314b44dfee244755a02d3257bd7539f6 : SUCCESS in 1h 14m 14s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/13a79700354c466da1c53941643a1816 : SUCCESS in 1h 17m 49s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/cfd7573f1edc4962b8164abaf42f2708 : SUCCESS in 1h 15m 44s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/fe09dbd850824f96a7ea04601b854a6d : SUCCESS in 1h 44m 13s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bd1f400ff03e46bfa557f5d5bc59671b : SUCCESS in 51m 34s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6d189a4dd5ec49539233d4fd3ca76e5f : FAILURE in 2h 14m 44s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f7cd59958bed40198c6c77742e71cb2b : SUCCESS in 26m 02s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"5ba2e6eb32c761ef5a32ef811d66add284a9a729","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-19 11:17:49.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/14/772514/15/check/DellEMC_Unity_Manila_dhss_true/9ea789c/DellEMC_Unity_Manila_dhss_true/356 : FAILURE in 1h 06m 17s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/14/772514/15/check/DellEMC_Unity_Manila_dhss_false/73e3b1f/DellEMC_Unity_Manila_dhss_false/340 : FAILURE in 59m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"94f2b1f4d8cedcdea84abe1a69b1c1b787c4b179","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-19 11:28:14.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/14/772514/15/check/DellEMC_VNX_Manila/2d6b093/DellEMC_VNX_Manila/200 : SUCCESS in 1h 16m 42s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":15},{"id":"6bc5289d3ee3b6f7fa667d38b9bae5b76b101751","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-02-19 13:17:40.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"4e3cd2efccb2f539883a9dad3c00bb63730372ce","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-02-19 13:19:46.000000000","message":"Patch Set 16:\n\nSorry, I forgot to pass method_kwargs.","accounts_in_message":[],"_revision_number":16},{"id":"6aa5ec253f27445a543aabf0b1985011eb687d42","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-19 14:19:32.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/14/772514/16/check/DellEMC_Unity_Manila_dhss_true/f3ceb8c/DellEMC_Unity_Manila_dhss_true/357 : FAILURE in 1h 01m 32s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/14/772514/16/check/DellEMC_Unity_Manila_dhss_false/844a610/DellEMC_Unity_Manila_dhss_false/341 : FAILURE in 35m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":16},{"id":"d2838bf67dfb20d8aab14892ab5d6861029bc834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 14:54:16.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1b85b99c6dc451c9cbb3faab7287178 : SUCCESS in 3m 38s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b2f9e11ee9004dafbaa75e103602520d : SUCCESS in 31m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/51186e9569d14cc4b951550582b91ea6 : SUCCESS in 42m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f7d71c2f6e7b4e1a81f41038e2819252 : SUCCESS in 1h 23m 13s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f3cf792ef4e3411da7ac304c0f4b1b96 : FAILURE in 56m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6a9127638fcc4e9496f86f6415da303f : SUCCESS in 1h 21m 55s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e9f4064fa2e342ff9e7d0970077ff288 : FAILURE in 1h 15m 15s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/302a9b3bb2a94700bf18473f2af12840 : SUCCESS in 46m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/785ea0b058ea4ded88b85368d9f4ad71 : POST_FAILURE in 1h 32m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7d6bf608dce5467fac0131bd6c65a717 : FAILURE in 25m 53s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"e1c961fbb6c36d5fd70af144933c7858858d2490","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-19 15:31:53.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/14/772514/16/check/DellEMC_Unity_Manila_dhss_true/06b4cc8/DellEMC_Unity_Manila_dhss_true/358 : FAILURE in 25m 36s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/14/772514/16/check/DellEMC_Unity_Manila_dhss_false/4f69a8e/DellEMC_Unity_Manila_dhss_false/342 : FAILURE in 36m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":16},{"id":"102824558cf7c117d62b514f02a162c7231e86bd","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-19 16:16:28.000000000","message":"Patch Set 16:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/14/772514/16/check/DellEMC_VNX_Manila/9a6171d/DellEMC_VNX_Manila/201 : SUCCESS in 1h 21m 21s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":16},{"id":"5137242cd55e40cca96c66a37fb45c5fb6e8ac04","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-19 17:20:06.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/14/772514/16/upstream-check/manila-cDOT-manila-tempest-plugin-ss/f32e2d0/ : SUCCESS in 2h 25m 10s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/14/772514/16/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/16abbba/ : SUCCESS in 1h 43m 16s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":16},{"id":"e13abc52bc3acbc76ba55f5887b4fda8c905f0d3","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-02-23 19:20:40.000000000","message":"Patch Set 16: Code-Review+2\n\nLGTM, thanks Liron!","accounts_in_message":[],"_revision_number":16},{"id":"1f0ee3f80f92ae9d8942e9ac8f173cfa446ce363","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-02-23 20:07:30.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"f10836c80d869f0647824db89a2113e758be63f6","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-02-23 22:11:44.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1\n\nThanks, LGTM","accounts_in_message":[],"_revision_number":16},{"id":"77ce575797943fdf3471e3630ce003fde29ec004","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 22:11:57.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"99c78e938d5cf305aa018d110b1b857bda2a9129","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 23:39:19.000000000","message":"Patch Set 16: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c532d0e3305d4e7cb6c801b113c54561 : SUCCESS in 4m 37s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a245684cd9a94e0a8e2fd72eaef7eb5a : FAILURE in 29m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8b7cc3af01e443068ea6ce963a221f77 : FAILURE in 38m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8e315ec012aa4f6f993349291c233730 : SUCCESS in 1h 18m 55s","accounts_in_message":[],"_revision_number":16},{"id":"e6dc4aac8451a6c384697065e030503bf4f0a9e8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-02-24 00:10:35.000000000","message":"Patch Set 16:\n\nHmm, don\u0027t re check this change. Gate\u0027s broken: https://review.opendev.org/c/openstack/manila-tempest-plugin/+/777266","accounts_in_message":[],"_revision_number":16},{"id":"48714c74b89ab39869f288b14b534def94c59c20","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-02-24 00:30:22.000000000","message":"Patch Set 16:\n\ngrr.. can\u0027t get Zulu to ignore this change unless I misspell \"rchek\" or something - just don\u0027t do it yet :D","accounts_in_message":[],"_revision_number":16},{"id":"47973b9e12a2eaed987ffc6272cb5ac90830cd5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 02:15:49.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04ee0c8abf4242c2adf1b467d846c506 : SUCCESS in 3m 50s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/040e898eadbc4bada9c5a760b9f28399 : FAILURE in 33m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c5840d16b00249b99bd2a13ddf873c95 : FAILURE in 34m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fdb2867acd71470f888ed17efa4e054d : SUCCESS in 1h 26m 59s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2d0a899bc6eb4989b573ff7f2d8f99e3 : SUCCESS in 1h 19m 33s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/21955a03fda24316be247bd1268d5007 : FAILURE in 48m 27s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d7455225da564610830613ffca3309b0 : FAILURE in 1h 31m 10s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/40a97ef7fca74c39ac488d7b2c7ac7a1 : FAILURE in 45m 26s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f9ee7b0a39714f62b6eacee6e46437e3 : FAILURE in 1h 59m 57s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b1dcad18b9b94ce284c3d2fadf7659f8 : FAILURE in 24m 48s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"d0de39ca74526d38c165b3293f891577a11df711","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-02-24 05:56:32.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"3880ce2eb873479d0010be891e7b77c723573f44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 07:25:36.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ce9c3dce11d4f49b5b1005caeeefeee : SUCCESS in 3m 19s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3ef60b4b53d140d8ae6625132fd746ae : FAILURE in 30m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d3f34de07692403c8fa174f48e28077f : FAILURE in 46m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/544c6c77e9fb4983ab599c5edd175ac4 : FAILURE in 1h 26m 14s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9fccd3d9bb4a437db9745cf6360a3ef1 : SUCCESS in 55m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/5bb9cb0db9d249f4b73dc86d18878de7 : FAILURE in 1h 02m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4e5b22eee7aa414f8995d0bfd988c58b : FAILURE in 1h 10m 35s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6cb25dea9d93434ebae0c4758693b2a0 : FAILURE in 38m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f45a848a4f2d4bf0b79ca98730f95bee : FAILURE in 1h 27m 42s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/33f1e1cfc21a4e02b0d482a66fb91730 : FAILURE in 24m 12s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"4f0d120277572c8d73157f735664d2b86beca465","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-02-24 21:31:22.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"8f3ddddc5b83d64ea0131aea8c537ee1464e983a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 23:29:32.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b652605200a4b3a8dab832371ede461 : SUCCESS in 3m 38s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6048d8c6dbf443d7ad27baced8116af9 : SUCCESS in 33m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c390eb32f3014a3d983038fdd42dc499 : SUCCESS in 43m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/246ee9ae62a041b08c1bc77aeea48c12 : SUCCESS in 1h 04m 35s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a8dfbb8c4e1e4ee1af65dfc567ad11d3 : SUCCESS in 1h 07m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/42195f653ba94a8fb93a13322f8dbe82 : SUCCESS in 55m 04s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/42f410aac9114b1e863114785fcd2a02 : SUCCESS in 1h 51m 39s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/662b37215d1f45dd813a3280b6a2bee8 : SUCCESS in 53m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c49c46c33a8c45899fe6d447b54d5c1c : FAILURE in 1h 33m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/36078e63ca6e492495ace9059befbf44 : FAILURE in 22m 09s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"49f61e433c15a6d4cd985ad2af9257f6969fd9a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 23:29:47.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"cfc9d3f02d4db103ceab6ffdb2ec96e25201391b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-25 00:37:28.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/14/772514/16/check/DellEMC_Unity_Manila_dhss_true/00fc0dd/DellEMC_Unity_Manila_dhss_true/389 : FAILURE in 57m 37s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/14/772514/16/check/DellEMC_Unity_Manila_dhss_false/895d670/DellEMC_Unity_Manila_dhss_false/373 : SUCCESS in 1h 05m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":16},{"id":"a49b11cb3483fd37b14f704b411420bc70c170f8","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-25 00:38:54.000000000","message":"Patch Set 16:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/14/772514/16/check/DellEMC_VNX_Manila/7e26087/DellEMC_VNX_Manila/216 : SUCCESS in 1h 09m 05s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":16},{"id":"93f6dd7c36736b360e7b4f5ebd6ada3b323c51f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 01:04:51.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fe2bcdbdfa642c2bbc1c7e499e7bbf6 : SUCCESS in 4m 36s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b4e41c78241f4bd8b0a52ba5f260a493 : SUCCESS in 28m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1157514f34e7454da95572a69b31e22f : SUCCESS in 34m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/40e695f61dc24208983cd10721e922e2 : SUCCESS in 1h 32m 28s","accounts_in_message":[],"_revision_number":16},{"id":"0ee0cc5efe4e20a78fcbf8b9d8b62cc8131f4508","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 01:04:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":16},{"id":"b40a7faf5b2ae501f2b3adf2fbdd191bc872bfc2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-25 02:09:16.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/14/772514/16/upstream-check/manila-cDOT-manila-tempest-plugin-ss/b7676de/ : SUCCESS in 2h 36m 56s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/14/772514/16/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/d209062/ : SUCCESS in 1h 48m 14s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"f7fc5b6494b0cd6aaeface59fe1c378ba72b5677","revisions":{"583937f7a28d7fd4bd1c5b4dd91eba01c55d5127":{"kind":"REWORK","_number":1,"created":"2021-01-26 12:54:59.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/14/772514/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/14/772514/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/1"}}},"commit":{"parents":[{"commit":"f2340dbfd2d7289e85acd8172255cde49eade9c7","subject":"Merge \"Remove further randomness when calling ddt.data()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f2340dbfd2d7289e85acd8172255cde49eade9c7"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-26 12:53:43.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-26 12:54:54.000000000","tz":120},"subject":"[WIP] Consolidate waiters methods","message":"[WIP] Consolidate waiters methods\n\nChange-Id: I92f4628051731a9f7e05404b48a96d2ed27a855b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/583937f7a28d7fd4bd1c5b4dd91eba01c55d5127"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/583937f7a28d7fd4bd1c5b4dd91eba01c55d5127"}]},"branch":"refs/heads/master"},"1cc4a75d20e77d3410321aa1e6f52f2b56dd5cf6":{"kind":"REWORK","_number":2,"created":"2021-01-26 13:35:59.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/14/772514/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/14/772514/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/2"}}},"commit":{"parents":[{"commit":"f2340dbfd2d7289e85acd8172255cde49eade9c7","subject":"Merge \"Remove further randomness when calling ddt.data()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f2340dbfd2d7289e85acd8172255cde49eade9c7"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-26 12:53:43.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-26 13:35:42.000000000","tz":120},"subject":"[WIP] Consolidate waiters methods","message":"[WIP] Consolidate waiters methods\n\nChange-Id: I92f4628051731a9f7e05404b48a96d2ed27a855b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1cc4a75d20e77d3410321aa1e6f52f2b56dd5cf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1cc4a75d20e77d3410321aa1e6f52f2b56dd5cf6"}]},"branch":"refs/heads/master"},"f15dbeed5c4a6a5a10ea11b3029e85701113ca93":{"kind":"REWORK","_number":3,"created":"2021-01-26 19:26:19.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/14/772514/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/14/772514/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/3"}}},"commit":{"parents":[{"commit":"f2340dbfd2d7289e85acd8172255cde49eade9c7","subject":"Merge \"Remove further randomness when calling ddt.data()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f2340dbfd2d7289e85acd8172255cde49eade9c7"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-26 12:53:43.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-26 19:26:01.000000000","tz":120},"subject":"[WIP] Consolidate waiters methods","message":"[WIP] Consolidate waiters methods\n\nChange-Id: I92f4628051731a9f7e05404b48a96d2ed27a855b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f15dbeed5c4a6a5a10ea11b3029e85701113ca93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f15dbeed5c4a6a5a10ea11b3029e85701113ca93"}]},"branch":"refs/heads/master"},"90f77b5c7ccfdd5b7eed2473ea979dbdc9c1be55":{"kind":"REWORK","_number":4,"created":"2021-01-27 05:53:14.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/14/772514/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/14/772514/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/4"}}},"commit":{"parents":[{"commit":"f2340dbfd2d7289e85acd8172255cde49eade9c7","subject":"Merge \"Remove further randomness when calling ddt.data()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f2340dbfd2d7289e85acd8172255cde49eade9c7"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-26 12:53:43.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-27 05:52:56.000000000","tz":120},"subject":"[WIP] Consolidate waiters methods","message":"[WIP] Consolidate waiters methods\n\nChange-Id: I92f4628051731a9f7e05404b48a96d2ed27a855b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/90f77b5c7ccfdd5b7eed2473ea979dbdc9c1be55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/90f77b5c7ccfdd5b7eed2473ea979dbdc9c1be55"}]},"branch":"refs/heads/master"},"a59fed9b3525b3cf8a470420780df890fc06d229":{"kind":"REWORK","_number":5,"created":"2021-01-27 07:28:29.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/14/772514/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/14/772514/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/5"}}},"commit":{"parents":[{"commit":"f2340dbfd2d7289e85acd8172255cde49eade9c7","subject":"Merge \"Remove further randomness when calling ddt.data()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f2340dbfd2d7289e85acd8172255cde49eade9c7"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-26 12:53:43.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-27 07:28:14.000000000","tz":120},"subject":"[WIP] Consolidate waiters methods","message":"[WIP] Consolidate waiters methods\n\nChange-Id: I92f4628051731a9f7e05404b48a96d2ed27a855b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a59fed9b3525b3cf8a470420780df890fc06d229"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a59fed9b3525b3cf8a470420780df890fc06d229"}]},"branch":"refs/heads/master"},"b184d613796b487152717db3f098617005943627":{"kind":"REWORK","_number":6,"created":"2021-01-27 08:52:00.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/14/772514/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/14/772514/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/6"}}},"commit":{"parents":[{"commit":"f2340dbfd2d7289e85acd8172255cde49eade9c7","subject":"Merge \"Remove further randomness when calling ddt.data()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f2340dbfd2d7289e85acd8172255cde49eade9c7"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-26 12:53:43.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-27 08:51:42.000000000","tz":120},"subject":"[WIP] Consolidate waiters methods","message":"[WIP] Consolidate waiters methods\n\nChange-Id: I92f4628051731a9f7e05404b48a96d2ed27a855b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b184d613796b487152717db3f098617005943627"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b184d613796b487152717db3f098617005943627"}]},"branch":"refs/heads/master"},"a3bbe8cbc2ea93af2b6188ed04c8ece544c6d9b0":{"kind":"REWORK","_number":7,"created":"2021-01-27 09:47:52.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/14/772514/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/14/772514/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/7"}}},"commit":{"parents":[{"commit":"f2340dbfd2d7289e85acd8172255cde49eade9c7","subject":"Merge \"Remove further randomness when calling ddt.data()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f2340dbfd2d7289e85acd8172255cde49eade9c7"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-26 12:53:43.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-27 09:47:34.000000000","tz":120},"subject":"[WIP] Consolidate waiters methods","message":"[WIP] Consolidate waiters methods\n\nChange-Id: I92f4628051731a9f7e05404b48a96d2ed27a855b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a3bbe8cbc2ea93af2b6188ed04c8ece544c6d9b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a3bbe8cbc2ea93af2b6188ed04c8ece544c6d9b0"}]},"branch":"refs/heads/master"},"bbba5c923d26ebf38b696da270de6f5af001cae7":{"kind":"REWORK","_number":8,"created":"2021-01-27 12:55:15.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/14/772514/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/14/772514/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/8"}}},"commit":{"parents":[{"commit":"f2340dbfd2d7289e85acd8172255cde49eade9c7","subject":"Merge \"Remove further randomness when calling ddt.data()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f2340dbfd2d7289e85acd8172255cde49eade9c7"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-26 12:53:43.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-27 12:54:57.000000000","tz":120},"subject":"[WIP] Consolidate waiters methods","message":"[WIP] Consolidate waiters methods\n\nChange-Id: I92f4628051731a9f7e05404b48a96d2ed27a855b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bbba5c923d26ebf38b696da270de6f5af001cae7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bbba5c923d26ebf38b696da270de6f5af001cae7"}]},"branch":"refs/heads/master"},"05401c583884063e9a64212aa7ff9ea71b6b9eff":{"kind":"REWORK","_number":9,"created":"2021-01-28 08:16:33.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/14/772514/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/14/772514/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/9"}}},"commit":{"parents":[{"commit":"f2340dbfd2d7289e85acd8172255cde49eade9c7","subject":"Merge \"Remove further randomness when calling ddt.data()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f2340dbfd2d7289e85acd8172255cde49eade9c7"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-26 12:53:43.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-28 08:16:16.000000000","tz":120},"subject":"[WIP] Consolidate waiters methods","message":"[WIP] Consolidate waiters methods\n\nChange-Id: I92f4628051731a9f7e05404b48a96d2ed27a855b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/05401c583884063e9a64212aa7ff9ea71b6b9eff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/05401c583884063e9a64212aa7ff9ea71b6b9eff"}]},"branch":"refs/heads/master"},"e37bb8ac4f4d5cd774d98ce4e6f5dbf5df9b53f1":{"kind":"REWORK","_number":10,"created":"2021-02-01 14:47:34.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/14/772514/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/14/772514/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/10"}}},"commit":{"parents":[{"commit":"d2138765888fe16e68f713ed99121f41e01a3c51","subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-26 12:53:43.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-01 14:45:00.000000000","tz":120},"subject":"[WIP] Consolidate waiters methods","message":"[WIP] Consolidate waiters methods\n\nChange-Id: I92f4628051731a9f7e05404b48a96d2ed27a855b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e37bb8ac4f4d5cd774d98ce4e6f5dbf5df9b53f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e37bb8ac4f4d5cd774d98ce4e6f5dbf5df9b53f1"}]},"branch":"refs/heads/master"},"d5a41f210c6b4b0b40385df79e188aa59a069ed6":{"kind":"REWORK","_number":11,"created":"2021-02-02 09:52:35.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/14/772514/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/14/772514/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/11"}}},"commit":{"parents":[{"commit":"d2138765888fe16e68f713ed99121f41e01a3c51","subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-26 12:53:43.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-02 09:52:07.000000000","tz":120},"subject":"[WIP] Consolidate waiters methods","message":"[WIP] Consolidate waiters methods\n\nChange-Id: I92f4628051731a9f7e05404b48a96d2ed27a855b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d5a41f210c6b4b0b40385df79e188aa59a069ed6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d5a41f210c6b4b0b40385df79e188aa59a069ed6"}]},"branch":"refs/heads/master"},"c5ba8abe051d81ae4db3237aa0e88b516bf44cfb":{"kind":"REWORK","_number":12,"created":"2021-02-02 13:02:17.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/14/772514/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/14/772514/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/12"}}},"commit":{"parents":[{"commit":"d2138765888fe16e68f713ed99121f41e01a3c51","subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-26 12:53:43.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-02 12:44:45.000000000","tz":120},"subject":"Consolidate waiters methods","message":"Consolidate waiters methods\n\nCurrently, for each resource there is a waiter method.\nThe code is almost the same in all waiter methods.\nThis patch suggests proposes to consolidate these methods.\n\nChange-Id: I92f4628051731a9f7e05404b48a96d2ed27a855b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c5ba8abe051d81ae4db3237aa0e88b516bf44cfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c5ba8abe051d81ae4db3237aa0e88b516bf44cfb"}]},"branch":"refs/heads/master"},"5963d4076b55d936196dbef864e58dc4edffb32b":{"kind":"REWORK","_number":13,"created":"2021-02-02 13:55:45.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/14/772514/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/14/772514/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/13"}}},"commit":{"parents":[{"commit":"d2138765888fe16e68f713ed99121f41e01a3c51","subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-26 12:53:43.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-02 13:55:27.000000000","tz":120},"subject":"Consolidate waiters methods","message":"Consolidate waiters methods\n\nCurrently, for each resource there is a waiter method.\nThe code is almost the same in all waiter methods.\nThis patch suggests proposes to consolidate these methods.\n\nChange-Id: I92f4628051731a9f7e05404b48a96d2ed27a855b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5963d4076b55d936196dbef864e58dc4edffb32b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5963d4076b55d936196dbef864e58dc4edffb32b"}]},"branch":"refs/heads/master"},"2f89f0bb082cbf458ebf6717f26eb8268fa40da6":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2021-02-03 05:34:28.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/14/772514/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/14/772514/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/14"}}},"commit":{"parents":[{"commit":"d2138765888fe16e68f713ed99121f41e01a3c51","subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-26 12:53:43.000000000","tz":120},"committer":{"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","date":"2021-02-03 05:33:22.000000000","tz":0},"subject":"Consolidate waiters methods","message":"Consolidate waiters methods\n\nCurrently, for each resource there is a waiter method.\nThe code is almost the same in all waiter methods.\nThis patch proposes to consolidate these methods.\n\nChange-Id: I92f4628051731a9f7e05404b48a96d2ed27a855b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2f89f0bb082cbf458ebf6717f26eb8268fa40da6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2f89f0bb082cbf458ebf6717f26eb8268fa40da6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6ed30d354a0e404b208b6b7cdf4308725a95ac1e":{"kind":"REWORK","_number":15,"created":"2021-02-19 05:24:58.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/14/772514/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/14/772514/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/15"}}},"commit":{"parents":[{"commit":"8868f18b7c469ccdaf477d2247186be2884b5d17","subject":"Merge \"tox miniversion update to 3.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8868f18b7c469ccdaf477d2247186be2884b5d17"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-26 12:53:43.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-19 05:24:35.000000000","tz":120},"subject":"Consolidate waiters methods","message":"Consolidate waiters methods\n\nCurrently, for each resource there is a waiter method.\nThe code is almost the same in all waiter methods.\nThis patch suggests proposes to consolidate these methods.\n\nChange-Id: I92f4628051731a9f7e05404b48a96d2ed27a855b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6ed30d354a0e404b208b6b7cdf4308725a95ac1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6ed30d354a0e404b208b6b7cdf4308725a95ac1e"}]},"branch":"refs/heads/master"},"f7fc5b6494b0cd6aaeface59fe1c378ba72b5677":{"kind":"REWORK","_number":16,"created":"2021-02-19 13:17:40.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/14/772514/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/14/772514/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/772514/16"}}},"commit":{"parents":[{"commit":"8868f18b7c469ccdaf477d2247186be2884b5d17","subject":"Merge \"tox miniversion update to 3.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8868f18b7c469ccdaf477d2247186be2884b5d17"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-01-26 12:53:43.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-19 13:17:22.000000000","tz":120},"subject":"Consolidate waiters methods","message":"Consolidate waiters methods\n\nCurrently, for each resource there is a waiter method.\nThe code is almost the same in all waiter methods.\nThis patch suggests proposes to consolidate these methods.\n\nChange-Id: I92f4628051731a9f7e05404b48a96d2ed27a855b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f7fc5b6494b0cd6aaeface59fe1c378ba72b5677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f7fc5b6494b0cd6aaeface59fe1c378ba72b5677"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
