)]}'
{"id":"openstack%2Fmanila-tempest-plugin~772942","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I4585fa8349a98f0ac6ec7ec23fb7b26ab61d65af","project":"openstack/manila-tempest-plugin","branch":"master","topic":"fix_conflicting_sec_services","hashtags":[],"change_id":"I4585fa8349a98f0ac6ec7ec23fb7b26ab61d65af","subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","status":"MERGED","created":"2021-01-28 21:06:52.000000000","updated":"2021-01-29 23:37:15.000000000","submitted":"2021-01-29 23:37:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772942-fix_conflicting_sec_services","meta_rev_id":"35f125e8f74b9b7499d292d94458e9acb65740cb","_number":772942,"virtual_id_number":772942,"owner":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-29 23:37:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2021-01-29 21:15:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2021-01-29 20:57:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2021-01-29 22:27:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-28 21:16:16.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-01-28 23:00:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-29 01:36:47.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-29 14:37:21.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-29 15:06:34.000000000","updated_by":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"reviewer":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"state":"REVIEWER"},{"updated":"2021-01-29 18:18:26.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2021-01-29 20:57:04.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-01-29 20:57:36.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"d13c2031f54a143e22a98ea28163cfc4d83a17f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-01-28 21:06:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfeebb74eb94a80885d63c09f599b66f95fd8d64","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-28 21:16:16.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/42/772942/1/check/DellEMC_Unity_Manila_dhss_true/c90b484/DellEMC_Unity_Manila_dhss_true/211 : FAILURE in 39s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/42/772942/1/check/DellEMC_Unity_Manila_dhss_false/b18c024/DellEMC_Unity_Manila_dhss_false/196 : FAILURE in 8m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"ac756f58f9fbc3e06193e315d9c81624854433c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 23:00:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/081fc17a2fcc4995bd885793a973eb89 : SUCCESS in 2m 56s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d79d176ee5e2441389f824c01adf6680 : SUCCESS in 37m 53s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0715a4752f024b69af489ada02c364e4 : SUCCESS in 37m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/90e0d35a175d4bbd9b088a19e3b2b7b5 : SUCCESS in 1h 03m 48s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/676feebca8cc48f799eac3c88b72b997 : FAILURE in 1h 10m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e9a0a372b9e044378f38d9ab6dc02f32 : FAILURE in 52m 34s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3b46c2567e60479890e673b14f2c5f4a : SUCCESS in 1h 28m 23s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/969a7bac2258428da67aa9b085b57244 : SUCCESS in 44m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f78bfe60bf404408ae5e88dff98c3282 : FAILURE in 1h 45m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0f2da2a83c3843729f5bd514a0edb0d6 : SUCCESS in 22m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b2435d774a3f4c4eb34e6d86d422c1d672215d8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-28 23:09:21.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/42/772942/1/check/DellEMC_Unity_Manila_dhss_true/2c3b1c1/DellEMC_Unity_Manila_dhss_true/212 : FAILURE in 27s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/42/772942/1/check/DellEMC_Unity_Manila_dhss_false/0ac7e27/DellEMC_Unity_Manila_dhss_false/197 : FAILURE in 8m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"b3d328053b8e3d5a97e88a73da173840289205dd","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-29 01:36:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/42/772942/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/f9e46dc/ : SUCCESS in 2h 36m 03s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/42/772942/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/f5f30f1/ : SUCCESS in 1h 56m 44s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"3196691cbb7f711bfc622baac1e78405eea10e4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-01-29 11:52:46.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ddecaf1c5e9d5a7fa1534e267df4676e07e8ccd5","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-01-29 11:57:25.000000000","message":"Patch Set 2:\n\nAll tests that were changed are passing: \n\nhttps://logs.openstack.netapp.com/logs/42/772942/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/f9e46dc/","accounts_in_message":[],"_revision_number":2},{"id":"4da26c54434fbd2c8f6ce9cbaccf5939fb883575","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-29 12:02:11.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/42/772942/2/check/DellEMC_Unity_Manila_dhss_true/027f749/DellEMC_Unity_Manila_dhss_true/213 : FAILURE in 30s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/42/772942/2/check/DellEMC_Unity_Manila_dhss_false/cfdac31/DellEMC_Unity_Manila_dhss_false/198 : FAILURE in 9m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"420681f85c78329c1460c3843151461dfb14a558","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 14:27:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95f6eccdbada48a0bea9037132c63f8c : SUCCESS in 3m 42s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/204bd6eaa60f4aa29378e0215750b21c : SUCCESS in 27m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4fa0ebdffba443df9e8253c6bc728894 : SUCCESS in 34m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ebc6842b9b3448938ee189d4da4d130c : SUCCESS in 1h 01m 28s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a924540da04c4a46ae9975373cf2d0a8 : SUCCESS in 1h 17m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f76b4622f8f6414abcfb2f8fd8cbf2e6 : SUCCESS in 1h 02m 14s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/327c0a82e00f4e1a9965d27e069bdcc7 : SUCCESS in 1h 16m 47s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/90d3535d93554f8ea80d87e17cde350f : SUCCESS in 45m 14s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5d3d071cd790421b962edfaaed5a9cd3 : FAILURE in 2h 33m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2ca036716599414191c986485075318b : SUCCESS in 21m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a27f900b7d8e0f8ffdd3ce6c31396d245b0fbb8d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-29 14:36:26.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/42/772942/2/check/DellEMC_Unity_Manila_dhss_true/75b2f34/DellEMC_Unity_Manila_dhss_true/214 : FAILURE in 26s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/42/772942/2/check/DellEMC_Unity_Manila_dhss_false/9d24ff8/DellEMC_Unity_Manila_dhss_false/199 : FAILURE in 8m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"b800a4e92a0ca42bef99326d97213a73f66bb57e","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-29 14:37:21.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/42/772942/2/check/DellEMC_VNX_Manila/3c3a734/DellEMC_VNX_Manila/147 : FAILURE in 9m 22s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"42ad4f5a0154b311f6e19b3d6c9e6f97c32edf77","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2021-01-29 15:06:34.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM Douglas, good job.","accounts_in_message":[],"_revision_number":2},{"id":"040ff0310b45036b2b4878293e863315a3452e0d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-29 16:56:30.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/42/772942/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/ab07ad9/ : SUCCESS in 2h 28m 30s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/42/772942/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/1d553fa/ : SUCCESS in 1h 40m 47s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"a86a503cc3e0b0cfc05a0d12fc635faf3b5a019d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-29 18:18:26.000000000","message":"Patch Set 2:\n\nThanks Douglas, \n\ndoesn\u0027t this test need a fix as well? https://github.com/openstack/manila-tempest-plugin/blob/d91870b8e63f2a2301af645fd8632601a67ec742/manila_tempest_tests/tests/api/test_security_services.py#L222-L232","accounts_in_message":[],"_revision_number":2},{"id":"3fd49005d47aebdd6f573a5893a4442e07a1832b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-29 18:20:51.000000000","message":"Patch Set 2:\n\nAnd this one as well: https://github.com/openstack/manila-tempest-plugin/blob/88452b3d310c82c7cd54e0b76a23a1dd58169977/manila_tempest_tests/tests/api/test_security_services_negative.py#L102-L107","accounts_in_message":[],"_revision_number":2},{"id":"26bff51891a5049ba28405fd6418eb04eca597a2","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-01-29 18:29:55.000000000","message":"Patch Set 2:\n\n@Goutham,\n\nYes, i missed those two.","accounts_in_message":[],"_revision_number":2},{"id":"90e7b5b1e82ceca226a5a89250d63cf0ad3aee06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-01-29 18:30:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a823110397e90fbef7aa1de92300b65b3f7251df","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-01-29 18:33:58.000000000","message":"Patch Set 3:\n\nThanks! Did a search again, don\u0027t seems that I\u0027m missing any other.","accounts_in_message":[],"_revision_number":3},{"id":"4b9be89f80b6be1066e1be08c5a708fbc76282d5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-29 18:39:43.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/42/772942/3/check/DellEMC_Unity_Manila_dhss_true/6c90cfe/DellEMC_Unity_Manila_dhss_true/215 : FAILURE in 29s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/42/772942/3/check/DellEMC_Unity_Manila_dhss_false/cb59bc6/DellEMC_Unity_Manila_dhss_false/200 : FAILURE in 8m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"867e8bec35da71c969f06ff8601adced79163d76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 20:46:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94f8eca22c0943039bea165a95ae0dd0 : SUCCESS in 3m 46s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f964fef348d74014b6c1bcaf415aa5a9 : SUCCESS in 32m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8436d2eb6d16405781afdcf12d255890 : SUCCESS in 44m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c6e88c0752e5452884a75eef323acc4a : SUCCESS in 1h 07m 41s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/db8446dac4714adf91b31be895851e87 : SUCCESS in 1h 09m 06s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/5f5c1518324f42a1805a56d17f881dac : SUCCESS in 1h 15m 45s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d5c9643a12b34763957091a25858ea50 : SUCCESS in 2h 09m 38s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/92d4118cfeeb411492e388e1868fd486 : SUCCESS in 46m 44s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fd9149e4bb55452187d18f3604424e5a : FAILURE in 1h 54m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/bffc2ae4a09e4d0ebb34fc63951bd778 : SUCCESS in 28m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7105e82fd0f5ec4a46186884e04ea3b00167a16c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-29 20:55:39.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/42/772942/3/check/DellEMC_Unity_Manila_dhss_true/8b4c02a/DellEMC_Unity_Manila_dhss_true/216 : FAILURE in 30s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/42/772942/3/check/DellEMC_Unity_Manila_dhss_false/7991738/DellEMC_Unity_Manila_dhss_false/201 : FAILURE in 8m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"3f51635f05d062ee099b17e209821b56c7210801","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-29 20:56:08.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/42/772942/3/check/DellEMC_VNX_Manila/efa4c61/DellEMC_VNX_Manila/148 : FAILURE in 9m 08s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"dd09ee70bb97970b4e33cda94ebb9ab8b3b77a90","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-29 20:57:04.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks, LGTM - the scenario test failures on the generic driver are unrelated; the Dell/EMC CIs are failing much earlier in the devstack stage - this change doesn\u0027t seem to be the culprit there.","accounts_in_message":[],"_revision_number":3},{"id":"27ebeef4bf31b5b98a39ed47e05b62d40f8086e0","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-01-29 21:15:29.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM, thanks for working on this!","accounts_in_message":[],"_revision_number":3},{"id":"b766b7b653676a0efb7e24a29800c4b1e1db4987","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-01-29 22:27:48.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"319533b4d7a1f30664d6f896da50893debb84a14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 22:28:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d89874881d7ba538ce2de8c2c3c20d0429c92a28","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-29 23:21:01.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/42/772942/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/efd0baf/ : SUCCESS in 2h 34m 00s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/42/772942/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/3edcd86/ : FAILURE in 1h 41m 15s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"0be5133f5e42bc8567d0294337304780524cce38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 23:37:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f88525c30add4584be451e9af05f950c : SUCCESS in 3m 44s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a61bf319e7a647c59c97474f8c499c8e : SUCCESS in 38m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3a84444a66624523b358cc1a839d86c8 : SUCCESS in 32m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/29efeb8bbf184dd18628e60ffe9e7fd8 : SUCCESS in 1h 06m 01s","accounts_in_message":[],"_revision_number":3},{"id":"35f125e8f74b9b7499d292d94458e9acb65740cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 23:37:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d2138765888fe16e68f713ed99121f41e01a3c51","revisions":{"9ca094658a4b95c82dba3bb96b1a427825dce03a":{"kind":"REWORK","_number":1,"created":"2021-01-28 21:06:52.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/42/772942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/42/772942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/42/772942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/42/772942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/42/772942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/42/772942/1"}}},"commit":{"parents":[{"commit":"0720fc354922cffbcb91663d85fe0cc0393ab0ba","subject":"Merge \"Enable running multi-backends tests with different share protocols\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0720fc354922cffbcb91663d85fe0cc0393ab0ba"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2021-01-28 20:38:30.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2021-01-28 20:38:30.000000000","tz":-180},"subject":"WIP - Fix \u0027ss-type\u0027 conflicts in security service tests","message":"WIP - Fix \u0027ss-type\u0027 conflicts in security service tests\n\nThis patch fixes security service conflict errors when security\nservice tests are executed. In these tests, no security services\nshould be automatically created during share network creation,\nor the test may fail due conflicting types.\nThis issue can be seen when \u0027security_service\u0027 config option\nis set to use a \u0027kerberos\u0027 or \u0027ldap\u0027 security service.\n\nChange-Id: I4585fa8349a98f0ac6ec7ec23fb7b26ab61d65af\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9ca094658a4b95c82dba3bb96b1a427825dce03a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9ca094658a4b95c82dba3bb96b1a427825dce03a"}]},"branch":"refs/heads/master"},"24a6afec955bbd80cc28b8c56ded5d97838aef32":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-29 11:52:46.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/42/772942/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/42/772942/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/42/772942/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/42/772942/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/42/772942/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/42/772942/2"}}},"commit":{"parents":[{"commit":"0720fc354922cffbcb91663d85fe0cc0393ab0ba","subject":"Merge \"Enable running multi-backends tests with different share protocols\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0720fc354922cffbcb91663d85fe0cc0393ab0ba"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2021-01-28 20:38:30.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2021-01-29 11:51:27.000000000","tz":0},"subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","message":"Fix \u0027ss-type\u0027 conflicts in security service tests\n\nThis patch fixes security service conflict errors when security\nservice tests are executed. In these tests, no security services\nshould be automatically created during share network creation,\nor the test may fail due conflicting types.\nThis issue can be seen when \u0027security_service\u0027 config option\nis set to use a \u0027kerberos\u0027 or \u0027ldap\u0027 security service.\n\nChange-Id: I4585fa8349a98f0ac6ec7ec23fb7b26ab61d65af\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/24a6afec955bbd80cc28b8c56ded5d97838aef32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/24a6afec955bbd80cc28b8c56ded5d97838aef32"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d2138765888fe16e68f713ed99121f41e01a3c51":{"kind":"REWORK","_number":3,"created":"2021-01-29 18:30:54.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/42/772942/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/42/772942/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/42/772942/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/42/772942/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/42/772942/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/42/772942/3"}}},"commit":{"parents":[{"commit":"0720fc354922cffbcb91663d85fe0cc0393ab0ba","subject":"Merge \"Enable running multi-backends tests with different share protocols\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0720fc354922cffbcb91663d85fe0cc0393ab0ba"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2021-01-28 20:38:30.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2021-01-29 18:30:32.000000000","tz":-180},"subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","message":"Fix \u0027ss-type\u0027 conflicts in security service tests\n\nThis patch fixes security service conflict errors when security\nservice tests are executed. In these tests, no security services\nshould be automatically created during share network creation,\nor the test may fail due conflicting types.\nThis issue can be seen when \u0027security_service\u0027 config option\nis set to use a \u0027kerberos\u0027 or \u0027ldap\u0027 security service.\n\nChange-Id: I4585fa8349a98f0ac6ec7ec23fb7b26ab61d65af\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
