)]}'
{"id":"openstack%2Fmanila-tempest-plugin~773812","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I70e81a712f5703b7b345de223260d91984c4d6e9","project":"openstack/manila-tempest-plugin","branch":"master","topic":"waiters","hashtags":[],"change_id":"I70e81a712f5703b7b345de223260d91984c4d6e9","subject":"Fixed a sporadic failure in test_delete_share_in_manage_error","status":"ABANDONED","created":"2021-02-02 22:26:12.000000000","updated":"2021-02-02 22:41:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"50f5f9eab32de7080e1144149dccc319cfb5fe96","_number":773812,"virtual_id_number":773812,"owner":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"REVIEWER":[{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-02 22:26:12.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2021-02-02 22:34:57.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"4547e89ccd317233cbb02a223e84f8e02f89a6de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-02-02 22:26:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d721bc6f9df25149103a097a91c9023d3c83b25","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-02 22:34:57.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/12/773812/1/check/DellEMC_Unity_Manila_dhss_true/e6b35ae/DellEMC_Unity_Manila_dhss_true/244 : FAILURE in 28s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/12/773812/1/check/DellEMC_Unity_Manila_dhss_false/1c01d1a/DellEMC_Unity_Manila_dhss_false/229 : FAILURE in 8m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"50f5f9eab32de7080e1144149dccc319cfb5fe96","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-02-02 22:41:38.000000000","message":"Abandoned\n\nBad rebase","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f289cd8aeda13eb31a63b15b0f32ec59e0e0643a","revisions":{"f289cd8aeda13eb31a63b15b0f32ec59e0e0643a":{"kind":"REWORK","_number":1,"created":"2021-02-02 22:26:12.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/12/773812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/12/773812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/773812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/773812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/773812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/773812/1"}}},"commit":{"parents":[{"commit":"0720fc354922cffbcb91663d85fe0cc0393ab0ba","subject":"Merge \"Enable running multi-backends tests with different share protocols\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0720fc354922cffbcb91663d85fe0cc0393ab0ba"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2021-01-28 20:38:30.000000000","tz":-180},"committer":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2021-02-02 22:20:39.000000000","tz":-300},"subject":"Fixed a sporadic failure in test_delete_share_in_manage_error","message":"Fixed a sporadic failure in test_delete_share_in_manage_error\n\nAdded the prefix, \u0027-share-protocol\u0027 to \u0027export_path\u0027 resulting in\na unique export path value per share protocol.\n\nChange-Id: I70e81a712f5703b7b345de223260d91984c4d6e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f289cd8aeda13eb31a63b15b0f32ec59e0e0643a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f289cd8aeda13eb31a63b15b0f32ec59e0e0643a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
