)]}'
{"id":"openstack%2Fmanila-tempest-plugin~773841","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I82eba46b72ff656f00c9ecc94d47589d8292bf5f","project":"openstack/manila-tempest-plugin","branch":"master","topic":"get_share","hashtags":[],"change_id":"I82eba46b72ff656f00c9ecc94d47589d8292bf5f","subject":"[Negative] Get share from other project","status":"MERGED","created":"2021-02-03 07:30:16.000000000","updated":"2021-02-12 15:19:20.000000000","submitted":"2021-02-12 15:19:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773841-get_share","meta_rev_id":"a21b0e62d2c85e4ed9f6094919fc31f7f6ec0c34","_number":773841,"virtual_id_number":773841,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-12 15:19:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"recommended":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":1,"date":"2021-02-03 21:21:51.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":2,"date":"2021-02-03 12:01:58.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-12 13:31:42.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"all":[{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-12 13:31:42.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-03 07:31:20.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2021-02-03 07:31:20.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-02-03 07:31:20.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-02-03 07:31:20.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-02-03 07:39:00.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-02-03 09:58:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-03 14:55:34.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"79270a9dc720c501d781af4b2dc285b3647c0204","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-02-03 07:30:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e80f7c62ada0f572a80ead813bfce2e4dc8eb78","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-03 07:39:00.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/41/773841/1/check/DellEMC_Unity_Manila_dhss_true/5baba26/DellEMC_Unity_Manila_dhss_true/251 : FAILURE in 27s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/41/773841/1/check/DellEMC_Unity_Manila_dhss_false/d4def39/DellEMC_Unity_Manila_dhss_false/236 : FAILURE in 8m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"8927b5ed8bc19cfae242db72b67ee7955ab07faf","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-02-03 07:43:08.000000000","message":"Patch Set 1:\n\nI have uploaded the test even though the fix there is no fix and it expected that the test will fail.\nIn the next thread I will add a skip decorator because a bug.\nI prefer the code to be inside and once the fix merge I will remove the decorator.","accounts_in_message":[],"_revision_number":1},{"id":"082cf8c88ca9ab390849fa3e150f7e6542ebac2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 09:58:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12bca8cbe06b4b82981009860a496882 : SUCCESS in 2m 53s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bbdcb026385347cdb1476d14e4e50dc9 : FAILURE in 37m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/39ff276b659d425eae0909cbb7fe7216 : FAILURE in 40m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/aaad25faf90640e0aff29aa45913a75e : SUCCESS in 1h 21m 57s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/bd0cdc647ec14244b7ca6c4a3adebccf : SUCCESS in 57m 15s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/4b5e1642695a42e3b644721e0b93c1cc : FAILURE in 34m 13s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c6cc0786230b48ba8b8470bfa432c20d : FAILURE in 1h 10m 36s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/dfa2787685b34e52a83f84214d55bd50 : FAILURE in 51m 07s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5341f207b05542a19da18266debba36e : FAILURE in 2h 22m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2a353c34a0044abcae68958d0eeda359 : FAILURE in 22m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf891fb9ed0995d02ec8e4f259ffce3fec28c1ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-02-03 10:17:26.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7ac7b4eda4a3c597d40279c55d2aa236bfcce47f","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-02-03 12:01:58.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM, thanks!","accounts_in_message":[],"_revision_number":2},{"id":"ccedfe005a25d1328d29bd0a3b51f9719006e9bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 12:21:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/820310f937024c7cb00c6667cd0ff154 : SUCCESS in 3m 06s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e8f0f1f05be9454e9422828cfa4656d6 : SUCCESS in 35m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f346843e2c73417195cca36b558efffe : SUCCESS in 40m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bcf813cca0574645a2f9bcbb8f00d789 : SUCCESS in 1h 04m 17s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/55c3bb894db84e1ca40f8184b48cfab6 : SUCCESS in 1h 13m 06s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/dab94e52e9b54064bf2f5b46edeb3e55 : SUCCESS in 1h 01m 08s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/29807cdacd68470d8c649ac838f65e42 : SUCCESS in 1h 20m 29s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/21635a9a7dcc4107b2487a1bfe9f8772 : SUCCESS in 44m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6b8d694b44854305bcce6b5fb9f30737 : FAILURE in 1h 57m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7ca72fd2293942769c4b71854c5da084 : SUCCESS in 21m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0ad1d883c104261bd5e099d1885ec2a2e86fafd2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-03 14:55:34.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/41/773841/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/7706076/ : SUCCESS in 2h 34m 12s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/41/773841/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/1684038/ : SUCCESS in 1h 43m 10s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"1671769797e0b7d1eb1d9d823e01f62163125e6f","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-02-03 21:21:51.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks for covering this Liron.","accounts_in_message":[],"_revision_number":2},{"id":"157ec0a132ac7b1d001afafaf90be31c16bb2c0c","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-02-12 13:31:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nGreat job, thanks Liron!","accounts_in_message":[],"_revision_number":2},{"id":"149dac6a79c77b4d790911295313e3d12cd3c542","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 13:32:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"38eab35204039a7b240e8943399b2a58e9104c94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 15:19:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b3daeb7a51c4944a9dbcb98354405ea : SUCCESS in 3m 47s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fac71dba38844f0b882adde16af8f5b9 : SUCCESS in 29m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/871b75575b1e43e7b342bcbd66abd125 : SUCCESS in 43m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/da354103af4044c1b45b6ad5e5e76122 : SUCCESS in 1h 06m 32s","accounts_in_message":[],"_revision_number":2},{"id":"a21b0e62d2c85e4ed9f6094919fc31f7f6ec0c34","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 15:19:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cb3123c50c28da6681780c098ea9c7510610c8a9","revisions":{"f68f87367ed4dcd2f089217de11937ec271cd2ce":{"kind":"REWORK","_number":1,"created":"2021-02-03 07:30:16.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/41/773841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/41/773841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/41/773841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/41/773841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/41/773841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/41/773841/1"}}},"commit":{"parents":[{"commit":"d2138765888fe16e68f713ed99121f41e01a3c51","subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-03 07:23:50.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-03 07:23:50.000000000","tz":120},"subject":"[Negative] Get share from other project","message":"[Negative] Get share from other project\n\nThe test verifies that user cannot find out if a share\nexists in another project.\n\nRelated-Bug: #1914363\nChange-Id: I82eba46b72ff656f00c9ecc94d47589d8292bf5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f68f87367ed4dcd2f089217de11937ec271cd2ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f68f87367ed4dcd2f089217de11937ec271cd2ce"}]},"branch":"refs/heads/master"},"cb3123c50c28da6681780c098ea9c7510610c8a9":{"kind":"REWORK","_number":2,"created":"2021-02-03 10:17:26.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/41/773841/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/41/773841/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/41/773841/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/41/773841/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/41/773841/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/41/773841/2"}}},"commit":{"parents":[{"commit":"d2138765888fe16e68f713ed99121f41e01a3c51","subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-03 07:23:50.000000000","tz":120},"committer":{"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","date":"2021-02-03 10:16:51.000000000","tz":0},"subject":"[Negative] Get share from other project","message":"[Negative] Get share from other project\n\nThe test verifies that user cannot find out if a share\nexists in another project.\n\nRelated-Bug: #1914363\nChange-Id: I82eba46b72ff656f00c9ecc94d47589d8292bf5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cb3123c50c28da6681780c098ea9c7510610c8a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cb3123c50c28da6681780c098ea9c7510610c8a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
