)]}'
{"id":"openstack%2Fmanila-tempest-plugin~775034","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bp/add-security-service-in-use-share-networks","attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-12-09 13:47:33.000000000","reason":"A robot voted negatively on a label"},"30998":{"account":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"last_update":"2022-03-10 15:03:14.000000000","reason":"A robot voted negatively on a label"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2022-12-22 13:37:41.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46","subject":"Add functional tests for add/update security service feature","status":"NEW","created":"2021-02-10 22:03:36.000000000","updated":"2023-03-29 12:11:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":69,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"9cf1c195cec53ffdecb9e566296962eb1e9f9d21","_number":775034,"virtual_id_number":775034,"owner":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-03-29 10:04:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-11 00:19:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-11 00:33:12.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-02-12 02:12:53.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-12 02:23:07.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-15 14:41:52.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2021-02-16 11:15:32.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"CC"},{"updated":"2021-03-10 23:12:23.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2021-04-13 14:04:06.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2021-04-13 14:04:06.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2023-01-18 18:51:44.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-03-16 22:46:14.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"}],"messages":[{"id":"1170bc6ba6d990fab0266916322d27e07165841b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2021-02-10 22:03:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1d84b1847af5c86f5fdcab86b241a4880e602fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 00:19:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b14584c4877459497939e91091f69f9 : SUCCESS in 2m 59s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2eb32c16d2aa4e56927f2eedd3fc7c8c : FAILURE in 34m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0e063c717f0d4fffa912a38ce546e59a : FAILURE in 40m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f820b304d7a34f1b98e042a00b518be4 : SUCCESS in 1h 04m 03s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2212eef93d01491a99e8dc6c5e528468 : FAILURE in 1h 13m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/db68b489c12045c294b02129e88baabf : FAILURE in 1h 09m 39s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a045736bbecd45d897fd6dfbcd0766a9 : FAILURE in 1h 30m 58s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/52f6c14172af4637b4ebfe99a279d87e : FAILURE in 49m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2e75dd3a86b44d13bc8b1a2f5866f1f8 : FAILURE in 2h 10m 05s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/37a483f5dc194cd79f00c2b5e2fb07b2 : FAILURE in 25m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e3cda615d5f99285ffcc648070cca6bfecd97ccd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-11 00:33:12.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/34/775034/1/check/DellEMC_Unity_Manila_dhss_true/b42a543/DellEMC_Unity_Manila_dhss_true/283 : FAILURE in 1h 52m 22s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/34/775034/1/check/DellEMC_Unity_Manila_dhss_false/5a451d7/DellEMC_Unity_Manila_dhss_false/267 : FAILURE in 55m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"f16e6516cead1b998ffe005a932aa2f078c95b3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2021-02-11 11:35:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"827b444ab6c9ae066ec4dd3ad2dddc1edeb5dca0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2021-02-11 12:56:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9ecbc12f31adaeb1f07194488c229f8886bad6e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 12:57:06.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  The key \"security_service\" appears more than once; duplicate keys are\n  not permitted.\n\nThe error appears in the following stanza:\n\n  backend_names: ALPHA,BETA\n                backend_replication_type: readable\n                build_timeout: 180\n                capability_sg_consistent_snapshot_support: pool\n                create_networks_when_multitenancy_enabled: false\n                enable_ip_rules_for_protocols: nfs\n                enable_ro_access_level_for_protocols: nfs,cifs\n                enable_user_rules_for_protocols: cifs\n                multi_backend: true\n                multitenancy_enabled: true\n                run_driver_assisted_migration_tests: true\n                run_manage_unmanage_snapshot_tests: true\n                run_manage_unmanage_tests: true\n                run_migration_with_preserve_snapshots_tests: true\n                run_mount_snapshot_tests: true\n                run_replication_tests: true\n                run_revert_to_snapshot_tests: true\n                run_share_server_migration_tests: true\n                security_service: \"ss_type:ldap,ss_dns_ip:10.10.10.10,ss_domain:fake.ldap1.openstack.com,ss_user:Administrator,ss_password:pwd\"\n                security_service: \"ss_type:ldap,ss_dns_ip:11.11.11.11,ss_domain:fake.ldap2.openstack.com,ss_user:Administrator,ss_password:pwd\"\n                security_service: \"ss_type:active_directory,ss_dns_ip:12.12.12.12,ss_domain:fake.ad.openstack.com,ss_user:Administrator,ss_password:pwd\"\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 475, column 15","accounts_in_message":[],"_revision_number":3},{"id":"eaa06c94e7f8cb2d4c705823caa38fd5080b4ee3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-11 14:40:36.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/34/775034/3/check/DellEMC_Unity_Manila_dhss_true/39692b8/DellEMC_Unity_Manila_dhss_true/292 : FAILURE in 7m 31s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/34/775034/3/check/DellEMC_Unity_Manila_dhss_false/cf198ac/DellEMC_Unity_Manila_dhss_false/276 : SUCCESS in 56m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"e645ef54626bf56b68018cccdc8eefc93fcea512","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2021-02-11 21:15:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bdb141ec553b91a4fcf90f183479d06a1fc3975e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 23:33:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96b1fbce9c7a4fa89a5da883caefe015 : SUCCESS in 3m 33s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/aab62c2310d744de9f022297d08ddbfe : SUCCESS in 29m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/60efa13669cd4ab686be7573621d6025 : SUCCESS in 39m 33s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/10520c4b105b4afcbfa6938ab3f12050 : SUCCESS in 1h 06m 21s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/759fab3e1b2648dbaf8cbae49d504b83 : SUCCESS in 1h 10m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d596861c405f483cb768ddb0f765bbf8 : SUCCESS in 1h 16m 35s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6a858ad796744afe868c9bc4ae1810e3 : SUCCESS in 1h 31m 01s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6f683428a82646468110b409a0d9704c : FAILURE in 49m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/002cf8f94adb464bb07358fb75ba8ba7 : FAILURE in 2h 12m 41s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/dc79bc63e151413d9d8e56f9e7ff294d : SUCCESS in 21m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f2a91103bf1e38b0b4942c7f2f29f1082b8a79a8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-11 23:41:33.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/34/775034/4/check/DellEMC_Unity_Manila_dhss_true/7fed2d6/DellEMC_Unity_Manila_dhss_true/302 : FAILURE in 1h 04m 36s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/34/775034/4/check/DellEMC_Unity_Manila_dhss_false/c95190a/DellEMC_Unity_Manila_dhss_false/286 : SUCCESS in 54m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"4683918129b493de70e3b12ff5cceb7c772b37a2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-12 02:12:53.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/34/775034/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/b775ae4/ : FAILURE in 2h 33m 07s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/34/775034/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/e70a0cd/ : SUCCESS in 2h 05m 19s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"6883b2227a026591e1a3c3ba1482abd1322e6ec1","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-12 02:23:07.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/34/775034/4/check/DellEMC_VNX_Manila/658d4f0/DellEMC_VNX_Manila/178 : FAILURE in 1h 24m 46s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"8bc9d0e7649eefbc77be8d7f22118a2a4630ebc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2021-02-12 15:32:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"12d9432b3381156e360cf322df6a5d555132684c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-12 16:33:16.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/34/775034/5/check/DellEMC_Unity_Manila_dhss_true/cb73340/DellEMC_Unity_Manila_dhss_true/311 : FAILURE in 1h 00m 19s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/34/775034/5/check/DellEMC_Unity_Manila_dhss_false/1bba2ae/DellEMC_Unity_Manila_dhss_false/295 : SUCCESS in 55m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"8fa2f9cf82dbeafbe856a07e450914d69075c2e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 17:41:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77a3f40d3ed541f7b2acafe042ccc6d6 : SUCCESS in 4m 08s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e2db7b9054c841e8ba1fda26066d18c5 : SUCCESS in 31m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9b83775e1be54489b4313546e2016fd4 : SUCCESS in 38m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e734f0d840154f088987ecfc067fd41a : SUCCESS in 1h 18m 19s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/27614575cd8b4714a13334f54de250bc : SUCCESS in 1h 02m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/436562fd1c94408fbda2b3d12c894f6e : SUCCESS in 1h 09m 09s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ca2371617c4342d6a859f6c9deecb1a8 : SUCCESS in 1h 28m 17s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b8a80c753e714a8db10c239cadf536c4 : FAILURE in 51m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0c02f37cc2514dd0853a9de55c74bf6e : FAILURE in 2h 03m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d9b1277471484b5e9a7d2e8cbac7b4d3 : SUCCESS in 26m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bb1715d843703885fb57f82e615e2653f2b14c75","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-12 19:10:28.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/34/775034/5/check/DellEMC_VNX_Manila/c7788d5/DellEMC_VNX_Manila/182 : FAILURE in 1h 28m 49s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"fed62622df5e867c6588956a6d575dbd01e01e23","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-12 19:38:12.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/34/775034/5/check/DellEMC_Unity_Manila_dhss_true/4a3111f/DellEMC_Unity_Manila_dhss_true/314 : FAILURE in 56m 04s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/34/775034/5/check/DellEMC_Unity_Manila_dhss_false/6ead1b2/DellEMC_Unity_Manila_dhss_false/298 : SUCCESS in 1h 02m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"e5d35a25be12ae699ed5ff042d1113d184c42b94","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-12 20:16:05.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/34/775034/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/d39065c/ : FAILURE in 2h 33m 44s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/34/775034/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/9e72741/ : FAILURE in 1h 49m 17s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"853a0c51391b6063dfad46593227d739d49eb2c2","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-02-15 14:41:52.000000000","message":"Patch Set 5: Code-Review-1\n\n(12 comments)\n\nThanks for working on this change, André!\nI have addressed few comments for you. Please check them inline :)","accounts_in_message":[],"_revision_number":5},{"id":"4dfa0b0cf8ecf46cfa323719303153cdfab2a19c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2021-02-15 19:43:26.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"49b695ea471059d5c11f8393115051cc7dbe279d","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2021-02-15 19:46:25.000000000","message":"Patch Set 6:\n\n(10 comments)\n\nHi carloss thank you very much for the comments!","accounts_in_message":[],"_revision_number":6},{"id":"08fdd375b79343371897e125030ee843de29ef8f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-15 20:51:01.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/34/775034/6/check/DellEMC_Unity_Manila_dhss_true/38f73c7/DellEMC_Unity_Manila_dhss_true/327 : FAILURE in 1h 05m 14s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/34/775034/6/check/DellEMC_Unity_Manila_dhss_false/0f77627/DellEMC_Unity_Manila_dhss_false/311 : SUCCESS in 1h 03m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"33f3d694538a7bb6e33ed570c5104662f21fec93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 22:09:46.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76c0e03492af44d5bab644291c4ed864 : SUCCESS in 4m 16s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/588e77ba2e364b65bc60c217a91cae75 : FAILURE in 35m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3c200083ecd7457493f7d83b3f6adc07 : FAILURE in 40m 08s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8bbb150c2849486b9223e4de5e857a0d : SUCCESS in 1h 15m 44s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/df23bceb921e413faec5e9e52ae0b55c : FAILURE in 1h 10m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3197be59eca643529d94ccb0e3bcc034 : SUCCESS in 1h 00m 05s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/28c1d6cb9ef147edb667e509cb83970f : SUCCESS in 1h 38m 29s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ec16e3ca474044db8be955298685f9fb : SUCCESS in 1h 01m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4dd22021a3b64bbc8eaff3b85cafaf2c : FAILURE in 1h 59m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/064a7d733f5744199c169f9582b74b62 : SUCCESS in 30m 36s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2edf041d23d8dded48f09d538105503004723817","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-02-16 11:15:32.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"2bb574b73c4e0cfca9f6947a03653b4355bd2ea9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 13:34:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46d462fe695341dba5d9f7deb12ae6b8 : SUCCESS in 4m 03s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/11e205ba344747a6bf9fbdbd0f833061 : SUCCESS in 29m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6677ff36986d4796a0c4251104febae5 : SUCCESS in 32m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bb0cd8f3f9db4fb2840cae7072329bc0 : SUCCESS in 1h 03m 41s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8c2a807c75c74e27a79abafaf340de93 : FAILURE in 1h 23m 57s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e841b0efe1174888a4b75de1b28e7e3b : SUCCESS in 47m 32s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0f762f447e79475895ac1387c20dcc7b : FAILURE in 2h 10m 11s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/83b76006adad4c43a66720cca4338e6f : SUCCESS in 40m 58s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4dbd3db2c02446b5b683c91e08e0f1c0 : FAILURE in 2h 09m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c0b25b125aad4f9da33db61b015e9704 : SUCCESS in 20m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d6ea087e431ab495bd5f4d0341bf7b03606dacba","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-16 14:40:54.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/34/775034/6/check/DellEMC_Unity_Manila_dhss_true/76fca2a/DellEMC_Unity_Manila_dhss_true/337 : FAILURE in 1h 03m 32s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/34/775034/6/check/DellEMC_Unity_Manila_dhss_false/9b6e494/DellEMC_Unity_Manila_dhss_false/321 : SUCCESS in 1h 06m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"4c9f1f75f1821c40d939018e79e4846c12854ef6","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-16 15:05:53.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/34/775034/6/check/DellEMC_VNX_Manila/ab9148b/DellEMC_VNX_Manila/192 : SUCCESS in 1h 31m 05s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":6},{"id":"282720201a5502d6c06f1bf254cb464b773b5e51","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-16 16:05:08.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/34/775034/6/upstream-check/manila-cDOT-manila-tempest-plugin-ss/5fd354d/ : FAILURE in 2h 30m 08s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/34/775034/6/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/95674c4/ : SUCCESS in 1h 51m 35s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"900bf17856a761247869933876767e9c731f0bc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-03-10 23:12:23.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"eb8dd0c80b5f65c827a1d20e567efaafb5691b2c","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-03-10 23:16:15.000000000","message":"Patch Set 7: Workflow-1\n\nStill working in some updates for this tests. The main change has changed a little bit, and now a \u0027check\u0027 operation must be triggered before adding or updating a security service for in use share networks.","accounts_in_message":[],"_revision_number":7},{"id":"69b56aaff8107fc6f37e5459be024df0a1d8ed4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-03-11 00:32:29.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"fab946515896fec2bf081e0d581a817c5936c538","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 02:36:17.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c09561fcafd3407185825ca5d088cc9a : SUCCESS in 6m 03s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0a8962f4ae654f0dbb7547a7811cbf49 : SUCCESS in 48m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8c036d4fe7624e89b46bf45127a42f18 : SUCCESS in 47m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fbda75a5f81e4cd78ba1f74f9a8e6be0 : SUCCESS in 1h 12m 37s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4dd22578d9a14dd7839379d1feb40562 : SUCCESS in 1h 51m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a921441c530945c2b0f7b3a60a13aef7 : SUCCESS in 1h 10m 42s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7a07f6d727be4ab683d6133feb6e869d : SUCCESS in 1h 26m 34s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e9b8df8f8ce74c68b25bf7b1f8d25201 : SUCCESS in 56m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/77d56271c1744203947f58cd9f2fd72c : FAILURE in 1h 58m 29s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5f74fe1e8464462794294bb0af8bc0ed : FAILURE in 29m 20s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a395a83751a1a5a69d22f176025fc9d94ebf0063","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-11 04:24:09.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/34/775034/8/check/DellEMC_VNX_Manila/064d5c0/DellEMC_VNX_Manila/278 : SUCCESS in 1h 46m 57s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":8},{"id":"1617f84e384a11f10e5c0721cfd8f3aecbeb2725","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-11 05:15:03.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/34/775034/8/upstream-check/manila-cDOT-manila-tempest-plugin-ss/f764ba9/ : SUCCESS in 2h 38m 10s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/34/775034/8/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/8f2a896/ : SUCCESS in 1h 45m 30s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"8c550f27ac55afdfaf5e8bf0cd811fdf85603ca4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-03-11 13:37:31.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"94537ea400c3b430f130968e4ece498b1c803255","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 17:21:05.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/736b30bed1744d68a4efdf1ae31d1554 : SUCCESS in 4m 41s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f96a934ebf824662be81aa289ec791c0 : FAILURE in 37m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/50e15a8cf7a647d98cd7d37dd5698a22 : SUCCESS in 38m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/604a834b7a15430083cdc8b4d1ce2970 : SUCCESS in 1h 05m 47s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/59e582c8977e43889d8a0a1ccfc60871 : SUCCESS in 1h 36m 53s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/402bbc9b08e145f1b13e049a25813420 : SUCCESS in 1h 16m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9f7f4372f8194ee89fc715ad682f0b64 : SUCCESS in 1h 32m 53s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/88a3537ca3344e5b8ef45b80642f5e6b : SUCCESS in 49m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e117b280384447eda68a5bccecf41dd2 : FAILURE in 1h 43m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/10dcd1a7076143e2a333ab02bf4814ee : SUCCESS in 26m 04s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"366f05d513377bc13be70f53a565fb84fb36795e","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-03-11 17:22:58.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"f9f053803a48c97450aef8877d8f2e9bfc9c7867","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 19:35:25.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6bbcbbbceb64e32a94280d31b0233d1 : SUCCESS in 4m 34s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/147a7a6e230f4d1a8343ba2b2e097e64 : SUCCESS in 36m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c886cc48b36246cdb7fe3be6a9f4a778 : SUCCESS in 45m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/567b669538e44626a4e6e0432225b21d : SUCCESS in 1h 07m 08s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/861a2d6c0de6424faf69f9b879f9dbd9 : FAILURE in 1h 25m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/bb05ad18470840c2936f454c685ed0a1 : SUCCESS in 1h 04m 51s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/bb36a0617f4940b1bd17a27ccfb0b691 : FAILURE in 1h 32m 45s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/60b894c3179d4fac94b258dab58f1b4d : SUCCESS in 55m 03s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5178107d89f649088731fe17ab2f22ec : FAILURE in 2h 02m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/33cd0faed2be4361809863d2166387d4 : SUCCESS in 31m 57s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c227fb5c8c62d13df55b4e515a6de7ed2c36f243","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-11 22:07:49.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/34/775034/9/upstream-check/manila-cDOT-manila-tempest-plugin-ss/84f4865/ : SUCCESS in 2h 31m 54s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/34/775034/9/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/429a2a6/ : SUCCESS in 1h 50m 49s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"0ab30cd7ef451233e7399599b5e01ecd99233685","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-11 23:03:10.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/34/775034/9/check/DellEMC_VNX_Manila/cffd35c/DellEMC_VNX_Manila/283 : SUCCESS in 1h 35m 17s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":9},{"id":"8662f3c7736e534d70ab1fd5abe2f4f0590d840e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-03-12 18:33:09.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"26b77d0aa6e0d1cc3a712861e2a2f632a6d7f506","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 20:22:55.000000000","message":"Patch Set 10: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"b53220d397b42478da65a2a01deca4ea776aba3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-03-13 02:32:06.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"bd940449c7a7f0e810eed52b7c1e9657846acfc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 04:39:13.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6278042822644810b5a0e66bdd076588 : SUCCESS in 4m 21s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3abdb989bbad48aeb3246dfa2173d708 : SUCCESS in 30m 53s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c8a9b90887824be4afb40bd67e4b9e0a : SUCCESS in 35m 19s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d16640e358a5442c91747ebfe1f39bf0 : SUCCESS in 1h 27m 34s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b61c62d16df74f18a806434cfbb95cb2 : FAILURE in 1h 20m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/33e9a672ad904be29d653dac51131e2f : SUCCESS in 56m 10s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e4c113e287ad4f14986b52790b743036 : SUCCESS in 1h 21m 52s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/aa295662493c467195cb2cc545876009 : SUCCESS in 51m 12s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e38e8f9d7aa54fbb84b162d51254de20 : FAILURE in 2h 04m 52s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a3498040e98b4517b1d5bd73a0f20566 : FAILURE in 34m 48s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"6b8cd7c4eecac80a034bb2048cce4ee1824a5f40","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-13 08:44:06.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/34/775034/11/check/DellEMC_VNX_Manila/58710ad/DellEMC_VNX_Manila/302 : FAILURE in 31m 31s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":11},{"id":"35e643a32034b33ea782856c35b1c0bf28041aa7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-03-30 17:06:53.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"146deeb9242c0645b4b74cc60490dc0be035d552","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-03-30 19:11:49.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"94f25adc7f3d32ae27f6c9682ba26a2259f641db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 23:23:05.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdf17bb6e88245f6b3c65da8da0e9930 : SUCCESS in 4m 02s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/61266f5fa6fe4a4ab3309cd85f390e46 : SUCCESS in 25m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f2856f0c799a49e9b83f9ab6eb9e417c : SUCCESS in 42m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1f87689b421b42f0942bbcfb12fd74ba : SUCCESS in 1h 43m 06s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ae2df5451fd74fb4b0c8bedb040d0d67 : FAILURE in 1h 17m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6de6a49803884591be7543f632062651 : FAILURE in 1h 06m 56s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/56938e9af9894181bd29548398815725 : SUCCESS in 1h 58m 39s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6a0a43d5ccff47c1a1dd9d6832b984ea : FAILURE in 51m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5633493b70b34959b736c775f30e3853 : FAILURE in 2h 05m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/609697c9c07947f2957f8c4b8bf2c797 : SUCCESS in 26m 28s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"02a4cfea93c5d6d8e200f165694014184206d643","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-31 01:35:40.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/34/775034/13/check/DellEMC_VNX_Manila/8dd15e5/DellEMC_VNX_Manila/377 : FAILURE in 1h 41m 22s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":13},{"id":"c98a8a9910bf8f61e013f518d083dac4f569b5eb","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-31 01:54:34.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/34/775034/13/upstream-check/manila-cDOT-manila-tempest-plugin-ss/1263d1c/ : SUCCESS in 2h 31m 08s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/34/775034/13/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/2721403/ : SUCCESS in 1h 51m 27s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"ea943970012e7b421653db75aed5b83e244eddc5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-31 02:06:15.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/34/775034/13/check/DellEMC_Unity_Manila_dhss_true/ea593f7/DellEMC_Unity_Manila_dhss_true/564 : FAILURE in 2h 14m 14s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/34/775034/13/check/DellEMC_Unity_Manila_dhss_false/fccaa12/DellEMC_Unity_Manila_dhss_false/567 : SUCCESS in 56m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"8ee1c8d6dd8ad13983bc9092335df399e5d8bc91","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-04-06 21:04:49.000000000","message":"Patch Set 13:\n\nrun experimental-NetApp CI","accounts_in_message":[],"_revision_number":13},{"id":"675bdf88432c915e02a6589da79169c6299623be","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-04-06 21:40:04.000000000","message":"Patch Set 13:\n\nrun-experimental NetApp CI","accounts_in_message":[],"_revision_number":13},{"id":"8729aacaea817f1bb8c0dfd37e7dcce74fa0111d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-06 21:40:37.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- experimental-manila-cDOT-security-service-ss https://logs.openstack.netapp.com/logs/34/775034/13/upstream-experimental-check/experimental-manila-cDOT-security-service-ss/9fbcf58/ : NOT_REGISTERED\n\nComment \u0027run-experimental NetApp CI\u0027 to rerun these tests.","accounts_in_message":[],"_revision_number":13},{"id":"ed24ce2b9a96fbe005c366c02fe0506c81bfbc0f","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-05-06 09:20:56.000000000","message":"Patch Set 13:\n\n(19 comments)","accounts_in_message":[],"_revision_number":13},{"id":"8d81a63881d6c0115910a088212e906af143a410","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-09-02 18:23:40.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"d2aa2cb130128d4ab6e1a99c11f1d06beab2c668","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-09-02 18:27:57.000000000","message":"Patch Set 14:\n\n(22 comments)\n\nSorry for take so long to solves this issue. Thanks Liron","accounts_in_message":[],"_revision_number":14},{"id":"467068b28189168af045b9cf755ce462a7976dad","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-09-02 18:33:17.000000000","message":"Topic bp/add-security-service-in-use-share-network removed","accounts_in_message":[],"_revision_number":14},{"id":"c9536bc999b4301bc54019d8856cc853db1304ca","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-09-02 18:33:26.000000000","message":"Topic set to bp/add-security-service-in-use-share-networks","accounts_in_message":[],"_revision_number":14},{"id":"e79db622bcb26c674060a39ecee7c94754aaff6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 20:06:11.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed622b2e0b984b138cab74dd5dbe7f07 : SUCCESS in 4m 02s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2dd6f676580e4e4588206c5d9e425345 : SUCCESS in 27m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d50a84e507284a4db115484211f93146 : FAILURE in 31m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f7f06e738ad9483e8f66e8790f5dec57 : SUCCESS in 1h 19m 29s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/72c6d37e47f245ddb8496b449b8b2cc6 : SUCCESS in 59m 01s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/71c8aba5fe5145b9872b7fd275511fa7 : SUCCESS in 59m 22s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/d76cbc0c5abd4d20830694296db5026d : SUCCESS in 1h 07m 16s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/306977e3ebd14dee8eeabdd03310cd1c : SUCCESS in 1h 09m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/abf16fda84164283b9fce9d7d4f5605a : FAILURE in 1h 07m 35s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5a96593cf6a746e286ce48e5394f95e2 : FAILURE in 1h 29m 18s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4c3eac8077e2423aab7a4ebba96e9e1e : FAILURE in 49m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/616be71b5e0b4b18ba1445e4e7bbe425 : FAILURE in 1h 35m 43s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ecbfbf0939e543d0b0ed53edc3c07b31 : FAILURE in 29m 50s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"9221d7a02f5f834dd0c4d17bf3a4013b3d373992","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2021-11-22 19:28:52.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"d13c65b8f92d8be7b606b837d5ccfe14b4a208ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 21:34:51.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53edc6e9dc514b769b08db627184e179 : SUCCESS in 3m 30s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bec47b30fe914135aab6c023c3dff654 : SUCCESS in 34m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/bf27aa130ab24985a383142b90e6a53e : FAILURE in 37m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f9b38fe5785f4ba1a52b0e7bc4fe936a : SUCCESS in 1h 43m 33s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/298faeb9dada4cb3802882933acddf93 : SUCCESS in 1h 30m 01s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/53d3c5478b5145d8b333e314adf18b1b : SUCCESS in 1h 12m 06s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/09a98f82f88f40e383677a4031bef357 : FAILURE in 1h 21m 57s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/adaf295635e24dd8962b9a71b087a85f : FAILURE in 1h 35m 24s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/04f9a8cd4786416b8021bb1cbdd5f0ee : FAILURE in 1h 59m 42s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/216842f26f274cf2a77d7dcff77f1a82 : FAILURE in 1h 54m 10s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/11ff24cd447d4f7cbb370ae540e2b328 : FAILURE in 51m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/59e21c0f96c745db92d8dfa236ff47ba : FAILURE in 1h 55m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4a19e6f03d1a4a899aea305c80686fe9 : SUCCESS in 26m 31s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"5cc94f64fb22026f265b66faa9a935ab6c0c22dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2022-03-10 12:25:28.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"dfdd402bb65bee3fc9432fb95644a88f9244e434","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2022-03-10 12:27:37.000000000","message":"Patch Set 16:\n\n(4 comments)","accounts_in_message":[],"_revision_number":16},{"id":"037a702a10bc72fc9ece5509f6074fc20fc21f33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 15:03:14.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e478a336d154801ae023147650ad280 : SUCCESS in 5m 01s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f8fab011f54b44ef9890e830aed4047f : SUCCESS in 35m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/41818168dcd94de8aecfba8e4804ba77 : FAILURE in 38m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9c75547d9d044e8e8f2da9b2fbc475c8 : SUCCESS in 1h 36m 17s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/c58a360025ec4f6baf3e9f772ffb3d7b : SUCCESS in 1h 08m 06s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/9ab83c69b9814b54a6633242786ae36b : SUCCESS in 1h 26m 34s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/e6092e4e2f9a490ab9a8cfdcfc81b1e4 : SUCCESS in 1h 11m 44s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b40438a6b83740ba904197fb878d4f37 : SUCCESS in 1h 24m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3bb2bf7d649647518a542c56dadc1579 : FAILURE in 1h 52m 26s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7d0ab67e67dd4f18bdebb269e62d17ff : FAILURE in 2h 17m 20s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/56a5bc464b7346b3a2b8961f09085935 : FAILURE in 1h 21m 03s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d13cee4959024e119bdcc09647666cc2 : FAILURE in 2h 04m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f888e5a94fbb46dba08706a040987c48 : SUCCESS in 30m 22s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"f54ffcaed8ee4b44691b722280d82edb6f7f12aa","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-03-15 19:14:19.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"9a33d196c07b8313d8a45d334c8558fef4554684","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 21:24:54.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6670d0183a5f474d8c32ce63aa0b7587 : SUCCESS in 5m 44s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/04cc92ac8f414b83bac0153077834947 : SUCCESS in 30m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/027050e7029e4466b22c47645e2c623a : FAILURE in 33m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/990da4a2899d4813ab43f7780c81fecd : SUCCESS in 1h 11m 24s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/dbbb765ab7914450b4c2ac8fb61a53ee : SUCCESS in 1h 01m 44s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/7501824426294c908586038905694bb5 : SUCCESS in 1h 33m 55s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/3fb6847cdd1149728c49b80623fea456 : SUCCESS in 1h 23m 14s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fa307a80f3264905bb23d31e3db24358 : FAILURE in 1h 29m 36s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/cece3ca429c8429db351a68f663001fe : SUCCESS in 1h 31m 45s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4b19633215cc42058cbbda721cfc287d : FAILURE in 2h 00m 54s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/85e6dc6514e349019a0792dabc949a4f : FAILURE in 1h 14m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/aca5be7314444eaebb4c9c967a193495 : SUCCESS in 1h 54m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e12a9401a6094c8598e3f6e1a0cf427e : SUCCESS in 30m 56s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"15546fb61fe2f4d678fd5e541017fcf0629c1332","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-12-09 12:06:25.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"680c735c1900973ab65686c0f202c28361a1cf6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 13:47:33.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/13cde3cbfd714f0da318431372e543ae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ad2891d6260427fbbac0ccac5c53386 : SUCCESS in 3m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e81ffb42fe854054a2bb210e777d1cf7 : FAILURE in 31m 40s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/723a0568592244b38ef25eb6fa8ee239 : SUCCESS in 1h 39m 17s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4c1d595757c24a5e834e03ccd997d9dd : SUCCESS in 54m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1b6dd9c7313d40d8a4d97a3ab4eb36a4 : FAILURE in 29m 57s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/56efde3419c24a5ab71e27f7a2795a11 : FAILURE in 30m 20s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0e4bd87891a242498b8a39ddaea92851 : SUCCESS in 56m 54s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a4659244f36d46dab586eaca393f89e3 : SUCCESS in 29m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/779b509246a34de4905b70b1e6a3a395 : FAILURE in 32m 24s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/4e7fdc039bba4980bb588a2e66a88593 : FAILURE in 27m 34s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/886d88fa47444e4ea8acdfaf30357b6f : SUCCESS in 55m 30s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/a911ad956d9a49328ce38ec025fde738 : SUCCESS in 1h 02m 01s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/be2842ded06c4f62a99e7b985a59f638 : SUCCESS in 23m 28s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c9dfc3a2884b4836b5282078b09a9ff1 : FAILURE in 46m 07s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ed94fb025f814aa5ac84a538a86c9107 : FAILURE in 16m 27s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"ff6d1d990a8e12072c1abb4b9641ed605ab420f4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-12-09 15:15:23.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/34/775034/17/check/DellEMC_Unity_Manila_dhss_true/47d20c5/DellEMC_Unity_Manila_dhss_true/1071 : FAILURE in 3h 08m 41s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/34/775034/17/check/DellEMC_Unity_Manila_dhss_false/c6b456f/DellEMC_Unity_Manila_dhss_false/1143 : SUCCESS in 1h 08m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":17},{"id":"05cd2de89dedde1745587ec66daf2e68f9ae0ace","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-22 09:41:07.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"7897eff4f2535cdbfeb2ae9616d567a8e1d790b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-12-22 10:27:14.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"fad4e9819919f9c9e940888112c2623c3c0b5b5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 13:37:41.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbf70ff7b88f4d649257d6bcb92ec086\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f607c4da8484aa9998df8e55f3e8c19 : SUCCESS in 2m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cd5bf2dbf103402c954c83211ad856b8 : SUCCESS in 1h 20m 45s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/666ac731cd5b442c99ab699b901b9a25 : FAILURE in 2h 31m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/05ec4dc5809746b5a643d7def16fdf1b : TIMED_OUT in 3h 02m 28s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0666694fbc0a465daa0963634e54c7de : SUCCESS in 1h 50m 40s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/24c1c8fe9f0243d3a568626c5d91a7c3 : FAILURE in 1h 57m 37s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/aee2e5cee7b54d37b01b87b8d674c7d8 : SUCCESS in 1h 14m 34s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4a85fe3552e34d3fb80b9f36247653c2 : SUCCESS in 38m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2b461f2c337f4449b653c012b872b971 : FAILURE in 46m 31s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/0de1c602513a4e95a7cf35597cb6ec99 : SUCCESS in 1h 16m 52s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/b2cd9770468f4b7f939958e10cf2d63a : SUCCESS in 58m 07s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/86190a210d53495894e869d9a126197b : SUCCESS in 1h 06m 02s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/4c1ceba724dc42bc925a475e924a11f9 : SUCCESS in 24m 16s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3d11c690b98543229faab71826727d27 : FAILURE in 37m 17s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a5ae7d0548734be3aa043bcc243747d5 : SUCCESS in 28m 30s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"2a264eec9debe210d7ff5eb9dbe5bc99f2f40941","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-18 10:51:43.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"414fd5cf6289c8bbe55c4ae9eb6f934b2505d8a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-18 11:56:18.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"cf3217329601e51cd17813ad198439d29871482c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-18 13:41:07.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"ddbfc10fa8f79b214950df603c0c613996d05d62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-18 15:01:38.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"3d26ad8784574caf0c65118ae21a7d93ea275a3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 17:08:58.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c46e773cc06c42228140c8ca73561b66\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6aec3c991144ff28f5244ad4e2a69ca : SUCCESS in 2m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/20862519923a4960a8f7296015c0b863 : SUCCESS in 1h 42m 11s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1f7e13b8e15c49ea8c8850e77f04c0ba : SUCCESS in 2h 01m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0f386a44af9a41f180c9ae77037ca9c9 : SUCCESS in 59m 53s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/90bb60bbe09e485eaf11a4d60c3975db : SUCCESS in 1h 22m 49s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b3b2f31f9dcf48f88954c491233dbf8f : FAILURE in 1h 39m 05s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d2431beafea44fc19e9e6ab2327fb51f : SUCCESS in 1h 17m 26s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/976630656970453fb32c1ec535636d48 : SUCCESS in 36m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2f85d0c3237c4740bd93512f52232b9c : SUCCESS in 42m 18s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/c8824cde53954dedb01a4fa9944b40e6 : SUCCESS in 1h 17m 10s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/ef07cd7773fd4f00880f238d3d362b49 : SUCCESS in 1h 09m 21s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/2a94000c96164ae9a0134e8b402c3f05 : SUCCESS in 1h 12m 09s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/64331bbbc33f4b62b6f7a7ed694f692d : SUCCESS in 32m 16s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9a7fb24cfc4b4107a5e37dd582da801e : SUCCESS in 48m 10s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/07b6a174e617478e8b71527ea1d1ca5e : SUCCESS in 25m 17s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"f6543023f6555280a6982ef8eabad59a8f984deb","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-18 18:51:44.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/775034/23/upstream-check/manila-tempest-plugin-ontap-dhss/7625209/ : SUCCESS in 1h 41m 18s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/775034/23/upstream-check/manila-tempest-plugin-ontap-no-dhss/0d9dc42/ : SUCCESS in 1h 10m 53s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":23},{"id":"b57bdfc769e739ea6e28a96df950b7f8b42040dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-18 20:35:18.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"b9a94d1027c0cc72421eec88f2a2b8428aac2350","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-01-18 20:44:01.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"6443fe186213c039040821cc30fc02fc86a0b0cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 23:13:10.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d8a2e8fd42a4948a517b695963af20d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1841f95b22ed420c87a7236167c29209 : SUCCESS in 2m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/50ab2afbd5614a27b28fbd195d35f0ec : SUCCESS in 1h 17m 01s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c81004f8ebdb455a93cdc7fe103c6107 : FAILURE in 2h 23m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/aadac4af82aa4525b2bece052b1fd929 : SUCCESS in 1h 26m 16s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/acde86c451d2435fabc14e79fdddffd8 : SUCCESS in 1h 26m 25s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a8e7d63537aa4d89b0130efc4b1a6df7 : FAILURE in 2h 04m 29s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f9e4777f3e56466a99c7a49debde5e80 : FAILURE in 1h 13m 30s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1c06fca0021c4d02be408403fdc0185f : SUCCESS in 37m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/11245af989bf4abdbf250b3d0c1566c3 : SUCCESS in 44m 29s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/fca361f9e2554b4dbc2c2fd52f57e267 : SUCCESS in 1h 32m 14s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/c34c19cd997f488cb116e88db6744b1e : SUCCESS in 1h 20m 12s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d80c55f136504e84bed63af9af57d342 : SUCCESS in 1h 14m 43s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/91d8487e0a314eb392a0c6ce41d1488a : SUCCESS in 38m 12s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ecb05ed44be042dfac594364cdcccf4a : SUCCESS in 54m 15s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3ac0ffddf7734aed94ffa81054161021 : SUCCESS in 30m 19s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"11dc7d63d2d0ba70ab2172e74e6de9b238ff44d8","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-19 00:57:54.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/775034/25/upstream-check/manila-tempest-plugin-ontap-dhss/eec58d8/ : FAILURE in 1h 44m 04s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/775034/25/upstream-check/manila-tempest-plugin-ontap-no-dhss/6b20d97/ : SUCCESS in 1h 19m 05s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":25},{"id":"f9195a591b1964836b07fe7324bc869b7ee5d07e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-15 09:37:41.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"79c21e4e40d0c1b0b5252e6eb75865fb0160fde5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-15 10:29:08.000000000","message":"Patch Set 26:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/34/775034/26/check/DellEMC_Unity_Manila_dhss_true/ba8d5d0/DellEMC_Unity_Manila_dhss_true/1371 : FAILURE in 51m 06s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/34/775034/26/check/DellEMC_Unity_Manila_dhss_false/0c8e3d3/DellEMC_Unity_Manila_dhss_false/1472 : SUCCESS in 51m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":26},{"id":"4b0faed1b9b529f564376d29eb408e1360d25674","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 11:19:45.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8731dc2e7c7f4ef598173f40691f29ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5916f58646a3449daa1911cbf19e6dfb : SUCCESS in 2m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6f173820bc824126a9eb2e05d7e20ede : SUCCESS in 1h 15m 53s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/6532192e2a39427b88395c237dc454b1 : FAILURE in 1h 38m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/99d3ebbd69714325917e1e58eb71f7a9 : SUCCESS in 1h 23m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/eb210eb4a9aa43a589c7fdd134fd37f4 : SUCCESS in 1h 03m 10s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/029b8418bb0d4e35a140836f7dd0f5ca : SUCCESS in 1h 37m 48s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/93243e1852ba47e69cb6802173d86d45 : SUCCESS in 53m 02s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e952bc3a64ce466389e0f26b68ad2f0b : SUCCESS in 38m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5a8236a75b3f4ec4858f9c90a592dfae : FAILURE in 41m 33s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/394fb77f603c46f992a2e6c23fcec85f : SUCCESS in 1h 11m 02s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/2b3bdb2610764982833f36bea0dea493 : SUCCESS in 1h 19m 57s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/4e4df311b2ab43a29fa90bbe505debff : SUCCESS in 23m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/193e52186312458b9503a256a428ee54 : FAILURE in 58m 24s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/50f593556da9448fa79f44602c0dec93 : SUCCESS in 17m 24s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"d58ad4b782525063ca335df28139fa9041980295","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-15 12:42:25.000000000","message":"Topic carlos removed","accounts_in_message":[],"_revision_number":26},{"id":"0496bd23ae8b9e7f5b4029cc0a2793ff2dc2a6a0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-15 12:42:33.000000000","message":"Topic set to bp/add-security-service-in-use-share-networks","accounts_in_message":[],"_revision_number":26},{"id":"2f862f0040bb830c70ada4f3935cfc35fc93be67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-15 13:01:22.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"9414ac0a85695b00a0e286c00082b7ec711d3f99","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-15 13:54:50.000000000","message":"Patch Set 27:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/34/775034/27/check/DellEMC_Unity_Manila_dhss_true/6d25fc2/DellEMC_Unity_Manila_dhss_true/1374 : FAILURE in 50m 09s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/34/775034/27/check/DellEMC_Unity_Manila_dhss_false/6993abf/DellEMC_Unity_Manila_dhss_false/1475 : SUCCESS in 53m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":27},{"id":"966c89da72632bbd9dcaeff40362da01f241b42a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 15:05:30.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5e1f525595e4252a3ab9b3a8db8c68d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2809f6ceaf374cec89f4692472d56474 : SUCCESS in 3m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/62afe8648656425399b0e1d63520f1ef : SUCCESS in 1h 20m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9d9ac5ebafea40939ec4174b18a79eb1 : SUCCESS in 2h 02m 07s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a62174b532fc4f8fab306b2d8a9300f6 : SUCCESS in 1h 32m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/166d812cc41f4cdaa920d1646af5197f : FAILURE in 36m 34s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b78821b6373f4108a34d70b11f044799 : SUCCESS in 1h 31m 50s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9c2f38edc7894243bb9f968039a8ed32 : FAILURE in 1h 17m 36s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/beb36d3b4a9544b8aa1071271637998f : SUCCESS in 26m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cda2a630222145ee8dc4f24d6fe4411b : SUCCESS in 30m 47s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/c4e78bc3ad1c44939ae89a9b6dcf3d85 : SUCCESS in 1h 26m 48s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/04b8e0de963f41aa85b4d8307fc80be6 : SUCCESS in 1h 29m 08s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/4848d613b58f4184a329d9cc03c5428e : SUCCESS in 29m 11s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d3ae4a6aefa74d42a976e1814fd3c98e : SUCCESS in 37m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/fec4a0d6890345b5967d0ef0d2a78a9f : SUCCESS in 32m 33s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"bc98cb516d061867a42961197e4eb5964422dcc2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-15 16:17:46.000000000","message":"Patch Set 27:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/34/775034/27/check/DellEMC_Unity_Manila_dhss_true/fa5e3d0/DellEMC_Unity_Manila_dhss_true/1375 : FAILURE in 50m 56s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/34/775034/27/check/DellEMC_Unity_Manila_dhss_false/196a779/DellEMC_Unity_Manila_dhss_false/1476 : SUCCESS in 1h 11m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":27},{"id":"4db1219848eaee4062ce4fc5967da8a76aa24e22","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-03-16 22:46:14.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"53a8061961b44dd64f92c4167ad70d39063da1e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-22 14:28:11.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"64a48fb52f6640b14a97201a6bb1d9615c1c8b6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-22 15:52:33.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"34004b923636b54c6692c3d04615942eddef0704","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-22 18:14:43.000000000","message":"Patch Set 29:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/34/775034/29/check/DellEMC_Unity_Manila_dhss_true/e96eded/DellEMC_Unity_Manila_dhss_true/1405 : FAILURE in 1h 12m 05s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/34/775034/29/check/DellEMC_Unity_Manila_dhss_false/319dd59/DellEMC_Unity_Manila_dhss_false/1506 : SUCCESS in 1h 03m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":29},{"id":"f79cf603bf77b67172b1a6090ee68c212b0cc124","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 18:35:16.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bb02bcbf2334607ba58cc7e227ff8da\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d143c8615bf44ccb5a7bf040aa55ae2 : SUCCESS in 2m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3ff8e052e8754a2183c9c61c2c34bde5 : SUCCESS in 1h 26m 22s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1ae4817fd8a14a27afa3d2255dd32834 : FAILURE in 2h 36m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/987f6db00cb74d7a9aed7346552af8f1 : SUCCESS in 1h 13m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c911b6f1d3dd4109b448c71038c332ff : SUCCESS in 1h 28m 19s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6e74e8a8e04f4dbaa4bf1b428fc056be : FAILURE in 1h 17m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8a7c8d7d092f4f98bca5ab9b05336ee3 : FAILURE in 1h 09m 49s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8725b43f1e924c4896bc7a88c40f2423 : SUCCESS in 39m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5662a305b47d4cafbf079545fc5bfb81 : SUCCESS in 28m 17s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/b203eace13424424b1dd7b4056fac0e4 : SUCCESS in 1h 20m 53s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/77739e9d9e974ceeba276f4d602d03cb : SUCCESS in 1h 03m 55s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f253a96a285d4e34ade319bde2b8fd51 : SUCCESS in 37m 40s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/66b2bb2afc274271ae79d753aee8bf4f : SUCCESS in 50m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2e3f62e00abf450a90e5d6ac76d86c99 : SUCCESS in 31m 47s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"2fcd44ce5b2ee5e8224fcb998ddad9b9676f3a65","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-22 21:00:30.000000000","message":"Patch Set 29:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/34/775034/29/check/DellEMC_Unity_Manila_dhss_true/76bd89f/DellEMC_Unity_Manila_dhss_true/1407 : FAILURE in 1h 07m 57s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/34/775034/29/check/DellEMC_Unity_Manila_dhss_false/6246636/DellEMC_Unity_Manila_dhss_false/1508 : SUCCESS in 1h 16m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":29},{"id":"577fa88d419b5b39f2631b7dc7ad4987a5018994","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-23 11:54:22.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"e055421d7c958b4bc7cdba87bffc12818f698fa4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-23 13:04:48.000000000","message":"Patch Set 30:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/34/775034/30/check/DellEMC_Unity_Manila_dhss_true/3aa9cfc/DellEMC_Unity_Manila_dhss_true/1410 : FAILURE in 1h 08m 57s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/34/775034/30/check/DellEMC_Unity_Manila_dhss_false/f81394c/DellEMC_Unity_Manila_dhss_false/1511 : SUCCESS in 1h 10m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":30},{"id":"2fc4a46afbae2dd06e36881f2eb2c4fe72553d0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-23 13:26:57.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"301239c461d3ede2d713b3e35e09b23f9eeb0818","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-23 14:12:17.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"a0b47db420f5aff780b25261cfebb1a37ecbfb5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 16:24:47.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6fb03542c1c46b19f4e0e4cc603480d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f6bd6c2eb7c4c4a99f71ca7f3c85aa0 : SUCCESS in 3m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d1c8b80c58494757b7372117916152e0 : SUCCESS in 2h 03m 35s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/224221fbb73447f3942e2ebf4ea98c2e : SUCCESS in 1h 56m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4fb40ccafd3e4b1390988a1fcfb56ef2 : SUCCESS in 1h 05m 43s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/2c74ae96dfb142218d9b538d422b7090 : SUCCESS in 1h 17m 21s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/dbaa6fe8fdd84cc79c3fd9398c89ef4e : FAILURE in 1h 26m 28s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0dae9bc8c1e74d1daddee1e4aa272d6d : SUCCESS in 1h 14m 07s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1038bee336674172969879352c033299 : SUCCESS in 31m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9e8c8aac20da4cd0bf21568ff1d2e833 : SUCCESS in 44m 43s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/2c326c11cc7c4b0bb9dbb7a7b3fd229b : SUCCESS in 1h 35m 17s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/047279bbaf6a431d82477f2960b467a6 : SUCCESS in 1h 11m 35s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/b7dc15cf457d48feb1e6a8bbd9671242 : SUCCESS in 36m 08s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/38de92179e5a4d54b86b92318a1bdcaf : SUCCESS in 55m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9b87b6cdc96c470a8cd3d713a01b64f2 : SUCCESS in 29m 03s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"23ddf6dbbf53676c98cd8684669a30d53026dbf7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-23 17:07:15.000000000","message":"Patch Set 32:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/34/775034/32/check/DellEMC_Unity_Manila_dhss_true/43fe8c1/DellEMC_Unity_Manila_dhss_true/1414 : FAILURE in 54m 24s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/34/775034/32/check/DellEMC_Unity_Manila_dhss_false/3720406/DellEMC_Unity_Manila_dhss_false/1515 : SUCCESS in 59m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":32},{"id":"cf509d9db97b7bf7568ec3e5acbc876375fd3677","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-26 13:16:28.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"92e20e7afc785113bf8b6c0d55b61b6ee246e14e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-26 14:21:59.000000000","message":"Patch Set 33:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/34/775034/33/check/DellEMC_Unity_Manila_dhss_true/e2d3a0f/DellEMC_Unity_Manila_dhss_true/1423 : FAILURE in 52m 22s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/34/775034/33/check/DellEMC_Unity_Manila_dhss_false/ba2f562/DellEMC_Unity_Manila_dhss_false/1524 : SUCCESS in 1h 05m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":33},{"id":"1cee6bfd74ed2d50793ff8e74609b7064b040487","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-26 15:44:13.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8c665af05a14890b930e1f87c9a0c4a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3175a9dd0f124e5784ad6a576e408ffe : SUCCESS in 2m 33s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/567b4574e597443e97ece84167416860 : SUCCESS in 1h 20m 35s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/14e8e0a325484d13b8b813a7088c31da : SUCCESS in 2h 21m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8d0eedec00b4457ebfa77c3efb7430a8 : SUCCESS in 1h 23m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/85cddef8bdc1468ca247ece53ccbfd26 : SUCCESS in 1h 32m 28s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b89f400bec9e4a5b9945bbe136011d6b : FAILURE in 1h 52m 13s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b1d12455034c4aa7a0957d1e6303d176 : SUCCESS in 1h 17m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a3814bbbfa8a48568196e3d735fe3e6b : SUCCESS in 35m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c520cc07628548788ae3ea668f540ef3 : SUCCESS in 43m 19s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/38496049e4b245489ee59d2e6b9bce9c : SUCCESS in 2h 02m 17s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f48b8ed50b1c41b3895e4d5241522b2b : SUCCESS in 1h 13m 47s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/81e149863eff4721bfb0894a40930b0f : SUCCESS in 33m 53s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b3d864e62e934237bb57698e06be9ba6 : SUCCESS in 51m 44s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/984c3aa2b9e44fec96b592605137ba7a : SUCCESS in 29m 19s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"81ff7fe154c21f3bf4f89c665cdf69e80bd3cafa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-26 16:49:05.000000000","message":"Patch Set 33:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/34/775034/33/check/DellEMC_Unity_Manila_dhss_true/b6cf8cc/DellEMC_Unity_Manila_dhss_true/1424 : FAILURE in 55m 11s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/34/775034/33/check/DellEMC_Unity_Manila_dhss_false/097260e/DellEMC_Unity_Manila_dhss_false/1525 : SUCCESS in 1h 04m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":33},{"id":"03878e2e7d3e952e168696d65e45a9b7a1ce1f9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-28 11:39:17.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"75af0ae02809c34aa1b23804f04a89afb2a71c7e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-28 12:50:15.000000000","message":"Patch Set 34:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/34/775034/34/check/DellEMC_Unity_Manila_dhss_true/9c5d50c/DellEMC_Unity_Manila_dhss_true/1428 : FAILURE in 1h 10m 39s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/34/775034/34/check/DellEMC_Unity_Manila_dhss_false/5c7fa00/DellEMC_Unity_Manila_dhss_false/1529 : SUCCESS in 1h 05m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":34},{"id":"0871a17acd172b32978814f1c82b23290797a318","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-28 12:52:24.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"a9360284fbbf639a06b3edfc89b75d3f0ec599bb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-28 14:00:12.000000000","message":"Patch Set 35:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/34/775034/35/check/DellEMC_Unity_Manila_dhss_true/90e98eb/DellEMC_Unity_Manila_dhss_true/1429 : FAILURE in 1h 07m 33s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/34/775034/35/check/DellEMC_Unity_Manila_dhss_false/6b375a9/DellEMC_Unity_Manila_dhss_false/1530 : SUCCESS in 56m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":35},{"id":"c38213210aa677a659405eb2a9a62ac51297eb0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 15:01:35.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ead46fa45c5f4030aa4a58eb9e736291\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ba211fba75f4063bc221dc942ce0328 : SUCCESS in 2m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7748e2e53a074411a95a6e9b2045409f : SUCCESS in 1h 12m 48s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/331176d0f675433aae4d5649605cb392 : SUCCESS in 2h 07m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2ce7a4c1aaf74dc4b7dfdcd5adfc7c8d : SUCCESS in 1h 16m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/792b37eb46e14b059a662b57a6f36e5e : SUCCESS in 1h 23m 18s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8a1b084755a147cbb50195c32e468b15 : FAILURE in 2h 02m 15s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e3def95ab41047ccb149fbdf93a42c86 : SUCCESS in 59m 30s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6275f3e2a6bb40a2bcd299ad9f4093ad : SUCCESS in 35m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/dbddf97e522e4224b52ee6e94cf44dda : SUCCESS in 46m 58s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/39ca07e1af534d84bfb0086fe7189735 : SUCCESS in 1h 19m 45s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/e1800c7cb8a5472b86148ed0e19c848d : SUCCESS in 1h 23m 38s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ceec8deedaa2423c98ff92ce40969721 : SUCCESS in 24m 46s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1ac7e3b6df9c41b6b3d6607b620be589 : SUCCESS in 44m 07s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/af6dca8895ef49cdbb0393b1e8d6734f : SUCCESS in 30m 29s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"e77ee15eceddcccb5d35396aa74eb75762af3e88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-28 15:17:14.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"d16f811459af4c7e0a2417ced987c9ec8cde651c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-28 16:24:18.000000000","message":"Patch Set 36:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/34/775034/36/check/DellEMC_Unity_Manila_dhss_true/610fea9/DellEMC_Unity_Manila_dhss_true/1431 : FAILURE in 1h 06m 43s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/34/775034/36/check/DellEMC_Unity_Manila_dhss_false/782aafe/DellEMC_Unity_Manila_dhss_false/1532 : SUCCESS in 1h 03m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":36},{"id":"da2348c4126809cd9703bb45877852ec2fac6da4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 17:22:59.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e3f97e9bd424896bfdc96b841a92509\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eca654c14b314ec6b4858bf64e8e3e66 : SUCCESS in 2m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/810ea7511b2a400680f04f56f050d44b : SUCCESS in 56m 30s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/278e8ff008474a048d8c89a5e30d74ee : SUCCESS in 1h 37m 17s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4531344bff2a477baeec1316aff05da3 : SUCCESS in 1h 18m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/48d863fd4c0444e9ade4120d8cbc11cb : SUCCESS in 1h 34m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a82164ab95ab4935b145dc2afd75efe6 : FAILURE in 1h 57m 25s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d557edd83816484bb716b46748317e82 : SUCCESS in 1h 22m 29s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5ec7031dbb6040c4b1a0cc7dd606d3a3 : SUCCESS in 40m 42s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8da1631505e745a9a9abe55f29d92ece : SUCCESS in 41m 29s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/3eec0712b6214678b9eeed2e260eab52 : SUCCESS in 1h 51m 49s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/3c59cd5d7db74dfab804ee8a7be15831 : SUCCESS in 1h 13m 34s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/24211443cbeb4e7aaae7768f0479c11b : SUCCESS in 39m 36s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/83617ca9263f4fcdbdf0aa6344f0dba2 : SUCCESS in 1h 00m 47s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/517cb18bfcee432583f1284eec10206f : SUCCESS in 27m 10s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"91655f69bcfbd876ee618f10b28d294d690e3b0c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-28 18:18:10.000000000","message":"Patch Set 36:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/34/775034/36/check/DellEMC_Unity_Manila_dhss_true/43cd7f8/DellEMC_Unity_Manila_dhss_true/1432 : FAILURE in 52m 26s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/34/775034/36/check/DellEMC_Unity_Manila_dhss_false/98bc11a/DellEMC_Unity_Manila_dhss_false/1533 : SUCCESS in 54m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":36},{"id":"811ca3593f2ab46b9651e581d2a18e022efd3ecf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-28 18:59:35.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"336a8407a49c5d19f524130c9eddcb4356db69f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-28 20:00:44.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"949eebce6f7f9d8a2f5e9199d35a15ed3af1e3b2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-28 21:10:35.000000000","message":"Patch Set 38:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/34/775034/38/check/DellEMC_Unity_Manila_dhss_true/c5f27dc/DellEMC_Unity_Manila_dhss_true/1434 : FAILURE in 51m 49s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/34/775034/38/check/DellEMC_Unity_Manila_dhss_false/c0b2e70/DellEMC_Unity_Manila_dhss_false/1535 : SUCCESS in 1h 09m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":38},{"id":"4b828c6b2f7394dfd104639dc9d01dd9e65446a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 23:11:07.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/713d74ad2cd0441090dd53d845d19deb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f0e99a7f570421aabb4909c6ec586d8 : SUCCESS in 2m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d768eec623ae41cf94f18a6d243aaf40 : SUCCESS in 1h 25m 51s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e9bfe4df5f654620946e495748148f60 : TIMED_OUT in 3h 02m 42s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c60b9f86c5844ae9926b1feb5ffc66b4 : SUCCESS in 1h 24m 05s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9418eabb413e4f0d946eaa92d59a1671 : SUCCESS in 1h 26m 25s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d4b1943435394c72ae7cfd2c93a0845d : FAILURE in 2h 04m 20s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3476b696c6414623b97346d122534db0 : SUCCESS in 1h 25m 34s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b82e763e596d4dd2ac81d0c11e1d3bd7 : SUCCESS in 34m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/bed734c6218f4bf78f2674e6c250d759 : FAILURE in 41m 23s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/7bafb46ae9d64f139751454fa3586881 : SUCCESS in 1h 15m 46s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/ab38eb39e972451cb532696fb61dd23a : SUCCESS in 1h 08m 50s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d2453418089043229aa97249fbe1f114 : SUCCESS in 36m 12s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cf0117fce97f47609fd2e2be36c95eed : SUCCESS in 56m 42s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/aa30735ebf6547299b3a89c361538964 : SUCCESS in 18m 49s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"ebbdcc159b30fff81cfbb8845538aa54fe6d32f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-29 05:22:20.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"101d223dd46cf3ce0f177d4ead3d0efa16e46a8c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-29 06:14:46.000000000","message":"Patch Set 39:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/34/775034/39/check/DellEMC_Unity_Manila_dhss_true/89c66fd/DellEMC_Unity_Manila_dhss_true/1435 : FAILURE in 52m 10s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/34/775034/39/check/DellEMC_Unity_Manila_dhss_false/fbc6c2d/DellEMC_Unity_Manila_dhss_false/1536 : FAILURE in 12m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":39},{"id":"843af7dd01da3f6f8e18d5f39304882a3d0d15b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-03-29 06:54:03.000000000","message":"Uploaded patch set 40.","accounts_in_message":[],"_revision_number":40},{"id":"2e60a0ea11dd030ecd95149e1b54eda6c58515ce","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-29 07:58:32.000000000","message":"Patch Set 40:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/34/775034/40/check/DellEMC_Unity_Manila_dhss_true/572ee20/DellEMC_Unity_Manila_dhss_true/1436 : FAILURE in 51m 18s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/34/775034/40/check/DellEMC_Unity_Manila_dhss_false/d80e5e4/DellEMC_Unity_Manila_dhss_false/1537 : SUCCESS in 1h 04m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":40},{"id":"fa1fa4627a1d78a4dd95ed531b19977f8c9594e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 10:04:27.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caa9756414ac40e8b126aca6fb9ebc41\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3eb784ddae844c6b8b55018e03facd6 : SUCCESS in 2m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/faf5136e9d5340f4bffa786db732251e : SUCCESS in 1h 19m 55s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d096faf09de94da9b07100f967a73fb1 : TIMED_OUT in 3h 02m 42s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0866051063434be9aeafb9a8be4e6b50 : SUCCESS in 1h 07m 36s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/15b66d4c280a488388710b106ca4c175 : SUCCESS in 1h 30m 16s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d3b05ac81aad42719ecfda8d7f426225 : FAILURE in 1h 35m 15s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/be9d22e61bba46f7861a9d24d2e05c4a : SUCCESS in 1h 25m 34s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/693510dd5e78469390a3c3840c0421a4 : SUCCESS in 38m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2f9879ebfa814e3f9b565ef6065533c9 : SUCCESS in 42m 02s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/9e62e74e7cb1480284420d70bb489c1d : SUCCESS in 1h 41m 47s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f22133c9ac03490cb51ff947251f329d : SUCCESS in 1h 08m 15s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/b79eafd042dc428c8dacf68a7eff30d0 : SUCCESS in 33m 29s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e49c2715fbf94644a0fa68e889f345e9 : SUCCESS in 42m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/59e9fbfef5bb4419828224ba104aed54 : SUCCESS in 28m 34s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"9cf1c195cec53ffdecb9e566296962eb1e9f9d21","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-29 12:11:42.000000000","message":"Patch Set 40:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/34/775034/40/check/DellEMC_Unity_Manila_dhss_true/13f501b/DellEMC_Unity_Manila_dhss_true/1441 : FAILURE in 51m 30s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/34/775034/40/check/DellEMC_Unity_Manila_dhss_false/0e3b85e/DellEMC_Unity_Manila_dhss_false/1542 : SUCCESS in 1h 10m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":40}],"current_revision_number":40,"current_revision":"4c6fa46c9c2ffb01ad085505106e389c62aa5e0e","revisions":{"8c1383f9a8e3840f496d89ee7f67cf0b3096038c":{"kind":"REWORK","_number":1,"created":"2021-02-10 22:03:36.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/34/775034/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/1"}}},"commit":{"parents":[{"commit":"d2138765888fe16e68f713ed99121f41e01a3c51","subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-10 21:59:16.000000000","tz":0},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nDepends-On: I129a794dfd2d179fa2b9a2fed050459d6f00b0de\n\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8c1383f9a8e3840f496d89ee7f67cf0b3096038c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8c1383f9a8e3840f496d89ee7f67cf0b3096038c"}]},"branch":"refs/heads/master"},"2dc11fb3b4b0b72a64465cabd74335ceea92afbd":{"kind":"REWORK","_number":2,"created":"2021-02-11 11:35:40.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/34/775034/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/2"}}},"commit":{"parents":[{"commit":"d2138765888fe16e68f713ed99121f41e01a3c51","subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-11 11:35:23.000000000","tz":0},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nDepends-On: I129a794dfd2d179fa2b9a2fed050459d6f00b0de\n\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2dc11fb3b4b0b72a64465cabd74335ceea92afbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2dc11fb3b4b0b72a64465cabd74335ceea92afbd"}]},"branch":"refs/heads/master"},"c3558f8e56948562afaabbaa91255d07673cb070":{"kind":"REWORK","_number":3,"created":"2021-02-11 12:56:40.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/34/775034/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/3"}}},"commit":{"parents":[{"commit":"d2138765888fe16e68f713ed99121f41e01a3c51","subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-11 12:56:07.000000000","tz":0},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nDepends-On: I129a794dfd2d179fa2b9a2fed050459d6f00b0de\n\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c3558f8e56948562afaabbaa91255d07673cb070"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c3558f8e56948562afaabbaa91255d07673cb070"}]},"branch":"refs/heads/master"},"7ad2420533fe3104796bc386df60d5d4f7ef4e80":{"kind":"REWORK","_number":4,"created":"2021-02-11 21:15:33.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/34/775034/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/4"}}},"commit":{"parents":[{"commit":"d2138765888fe16e68f713ed99121f41e01a3c51","subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-11 21:14:13.000000000","tz":0},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nDepends-On: I129a794dfd2d179fa2b9a2fed050459d6f00b0de\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7ad2420533fe3104796bc386df60d5d4f7ef4e80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7ad2420533fe3104796bc386df60d5d4f7ef4e80"}]},"branch":"refs/heads/master"},"ef2c0a42a19e6cfc0002dd1d5a3b4ec94cd4fc97":{"kind":"REWORK","_number":5,"created":"2021-02-12 15:32:28.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/34/775034/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/5"}}},"commit":{"parents":[{"commit":"d2138765888fe16e68f713ed99121f41e01a3c51","subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-12 15:32:04.000000000","tz":0},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nDepends-On: I129a794dfd2d179fa2b9a2fed050459d6f00b0de\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ef2c0a42a19e6cfc0002dd1d5a3b4ec94cd4fc97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ef2c0a42a19e6cfc0002dd1d5a3b4ec94cd4fc97"}]},"branch":"refs/heads/master"},"00bedc736957ae466f10b6c6bf77606504e26d7a":{"kind":"REWORK","_number":6,"created":"2021-02-15 19:43:26.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/34/775034/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/6"}}},"commit":{"parents":[{"commit":"d2138765888fe16e68f713ed99121f41e01a3c51","subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-15 19:42:48.000000000","tz":0},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nDepends-On: I129a794dfd2d179fa2b9a2fed050459d6f00b0de\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/00bedc736957ae466f10b6c6bf77606504e26d7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/00bedc736957ae466f10b6c6bf77606504e26d7a"}]},"branch":"refs/heads/master"},"890e3de62d1b7228fd23f94601cb3585277d175a":{"kind":"REWORK","_number":7,"created":"2021-03-10 23:12:23.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/34/775034/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/7"}}},"commit":{"parents":[{"commit":"9efbc94e963613a617631d27917ba278429566f5","subject":"Merge \"Remove a duplicated skipped functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9efbc94e963613a617631d27917ba278429566f5"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2021-03-10 23:10:55.000000000","tz":-180},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nDepends-On: I129a794dfd2d179fa2b9a2fed050459d6f00b0de\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/890e3de62d1b7228fd23f94601cb3585277d175a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/890e3de62d1b7228fd23f94601cb3585277d175a"}]},"branch":"refs/heads/master"},"1093a911ba4001a7bb9806414a5b633858ac3b3b":{"kind":"REWORK","_number":8,"created":"2021-03-11 00:32:29.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/34/775034/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/8"}}},"commit":{"parents":[{"commit":"9efbc94e963613a617631d27917ba278429566f5","subject":"Merge \"Remove a duplicated skipped functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9efbc94e963613a617631d27917ba278429566f5"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2021-03-11 00:32:20.000000000","tz":-180},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nDepends-On: I129a794dfd2d179fa2b9a2fed050459d6f00b0de\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1093a911ba4001a7bb9806414a5b633858ac3b3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1093a911ba4001a7bb9806414a5b633858ac3b3b"}]},"branch":"refs/heads/master"},"53e089be5c12593ab523ded6c0181953e30e7e52":{"kind":"REWORK","_number":9,"created":"2021-03-11 13:37:31.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/34/775034/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/9"}}},"commit":{"parents":[{"commit":"9efbc94e963613a617631d27917ba278429566f5","subject":"Merge \"Remove a duplicated skipped functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9efbc94e963613a617631d27917ba278429566f5"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2021-03-11 13:36:33.000000000","tz":-180},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nDepends-On: I129a794dfd2d179fa2b9a2fed050459d6f00b0de\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/53e089be5c12593ab523ded6c0181953e30e7e52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/53e089be5c12593ab523ded6c0181953e30e7e52"}]},"branch":"refs/heads/master"},"9d7432f13b2094ccca63d967d278c4fe8b697304":{"kind":"REWORK","_number":10,"created":"2021-03-12 18:33:09.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/34/775034/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/10"}}},"commit":{"parents":[{"commit":"9efbc94e963613a617631d27917ba278429566f5","subject":"Merge \"Remove a duplicated skipped functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9efbc94e963613a617631d27917ba278429566f5"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2021-03-12 18:32:50.000000000","tz":-180},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nDepends-On: I129a794dfd2d179fa2b9a2fed050459d6f00b0de\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9d7432f13b2094ccca63d967d278c4fe8b697304"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9d7432f13b2094ccca63d967d278c4fe8b697304"}]},"branch":"refs/heads/master"},"db22b8dd52691c76ada5f230f4860fcd508cea4c":{"kind":"REWORK","_number":11,"created":"2021-03-13 02:32:06.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/34/775034/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/11"}}},"commit":{"parents":[{"commit":"9efbc94e963613a617631d27917ba278429566f5","subject":"Merge \"Remove a duplicated skipped functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9efbc94e963613a617631d27917ba278429566f5"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2021-03-13 02:31:47.000000000","tz":-180},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nDepends-On: I129a794dfd2d179fa2b9a2fed050459d6f00b0de\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/db22b8dd52691c76ada5f230f4860fcd508cea4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/db22b8dd52691c76ada5f230f4860fcd508cea4c"}]},"branch":"refs/heads/master"},"c5f6294c9b732835a327a3f45219848d176be9d4":{"kind":"REWORK","_number":12,"created":"2021-03-30 17:06:53.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/34/775034/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/12"}}},"commit":{"parents":[{"commit":"f083c71a9396a5a6f3300d3150b7fc1bee2c57c5","subject":"Add create share from snapshot in CephFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f083c71a9396a5a6f3300d3150b7fc1bee2c57c5"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2021-03-30 13:19:27.000000000","tz":-180},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nDepends-On: I129a794dfd2d179fa2b9a2fed050459d6f00b0de\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c5f6294c9b732835a327a3f45219848d176be9d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c5f6294c9b732835a327a3f45219848d176be9d4"}]},"branch":"refs/heads/master"},"608b685f09497545166eedbba67947462a903cfc":{"kind":"REWORK","_number":13,"created":"2021-03-30 19:11:49.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/34/775034/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/13"}}},"commit":{"parents":[{"commit":"f083c71a9396a5a6f3300d3150b7fc1bee2c57c5","subject":"Add create share from snapshot in CephFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f083c71a9396a5a6f3300d3150b7fc1bee2c57c5"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2021-03-30 19:07:05.000000000","tz":-180},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nDepends-On: I129a794dfd2d179fa2b9a2fed050459d6f00b0de\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/608b685f09497545166eedbba67947462a903cfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/608b685f09497545166eedbba67947462a903cfc"}]},"branch":"refs/heads/master"},"1dcea86c0aa2ccbe279418f1e2f06c1db8533254":{"kind":"REWORK","_number":14,"created":"2021-09-02 18:23:40.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/34/775034/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/14"}}},"commit":{"parents":[{"commit":"ed2b75e6ae6a1dd99dba2a20133845ba115fbc48","subject":"Merge \"Fix tests creating share without using share_size\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ed2b75e6ae6a1dd99dba2a20133845ba115fbc48"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"Douglas Viroel","email":"dviroel@redhat.com","date":"2021-09-02 18:23:06.000000000","tz":-180},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nDepends-On: I129a794dfd2d179fa2b9a2fed050459d6f00b0de\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1dcea86c0aa2ccbe279418f1e2f06c1db8533254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1dcea86c0aa2ccbe279418f1e2f06c1db8533254"}]},"branch":"refs/heads/master"},"22ae82eb15bf46fd3dd29573be86f95078b2720f":{"kind":"REWORK","_number":15,"created":"2021-11-22 19:28:52.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/34/775034/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/15"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"Andre Beltrami","email":"debeltrami@gmail.com","date":"2021-11-22 19:28:47.000000000","tz":-180},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nDepends-On: I129a794dfd2d179fa2b9a2fed050459d6f00b0de\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/22ae82eb15bf46fd3dd29573be86f95078b2720f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/22ae82eb15bf46fd3dd29573be86f95078b2720f"}]},"branch":"refs/heads/master"},"2495a845273d405cad9fb4b8aa407037f8910d10":{"kind":"REWORK","_number":16,"created":"2022-03-10 12:25:28.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/34/775034/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/16"}}},"commit":{"parents":[{"commit":"c1a3c0e54e675380ec85f8092527981efa8ce078","subject":"Add functional tests for add multiple subnets feature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c1a3c0e54e675380ec85f8092527981efa8ce078"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"andrebeltrami","email":"debeltrami@gmail.com","date":"2022-03-10 12:24:28.000000000","tz":-180},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2495a845273d405cad9fb4b8aa407037f8910d10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2495a845273d405cad9fb4b8aa407037f8910d10"}]},"branch":"refs/heads/master"},"3820a359fefe5a79b8522e6f3d2c4d614208111c":{"kind":"REWORK","_number":17,"created":"2022-12-09 12:06:25.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/34/775034/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/17"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-12-09 12:06:13.000000000","tz":-180},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3820a359fefe5a79b8522e6f3d2c4d614208111c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3820a359fefe5a79b8522e6f3d2c4d614208111c"}]},"branch":"refs/heads/master"},"a5971fc508aec001463585acb44ff6b0316b6e34":{"kind":"REWORK","_number":18,"created":"2022-12-22 09:41:07.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/34/775034/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/18"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-12-22 09:40:57.000000000","tz":120},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a5971fc508aec001463585acb44ff6b0316b6e34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a5971fc508aec001463585acb44ff6b0316b6e34"}]},"branch":"refs/heads/master"},"775104b7a7173ff47a475af7b43f80fee4d98982":{"kind":"REWORK","_number":19,"created":"2022-12-22 10:27:14.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/34/775034/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/19"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-12-22 10:27:01.000000000","tz":120},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/775104b7a7173ff47a475af7b43f80fee4d98982"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/775104b7a7173ff47a475af7b43f80fee4d98982"}]},"branch":"refs/heads/master"},"d6b55941f7bc15e69b7a8a3565cadbb8f83d4e89":{"kind":"REWORK","_number":20,"created":"2023-01-18 10:51:43.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/34/775034/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/20"}}},"commit":{"parents":[{"commit":"e42249631e502d683d0a28fcf9c120c9b4fde666","subject":"Merge \"[RBAC] Making sure export locations are viewable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e42249631e502d683d0a28fcf9c120c9b4fde666"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-18 10:50:36.000000000","tz":120},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d6b55941f7bc15e69b7a8a3565cadbb8f83d4e89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d6b55941f7bc15e69b7a8a3565cadbb8f83d4e89"}]},"branch":"refs/heads/master"},"138bc1c89f84b40cd6416f648919fd654ac70df8":{"kind":"REWORK","_number":21,"created":"2023-01-18 11:56:18.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/34/775034/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/21"}}},"commit":{"parents":[{"commit":"e42249631e502d683d0a28fcf9c120c9b4fde666","subject":"Merge \"[RBAC] Making sure export locations are viewable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e42249631e502d683d0a28fcf9c120c9b4fde666"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-18 11:56:06.000000000","tz":120},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/138bc1c89f84b40cd6416f648919fd654ac70df8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/138bc1c89f84b40cd6416f648919fd654ac70df8"}]},"branch":"refs/heads/master"},"ec1a186aebea396f2643ff3a3abf27832d88e69a":{"kind":"REWORK","_number":22,"created":"2023-01-18 13:41:07.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/34/775034/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/22"}}},"commit":{"parents":[{"commit":"e42249631e502d683d0a28fcf9c120c9b4fde666","subject":"Merge \"[RBAC] Making sure export locations are viewable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e42249631e502d683d0a28fcf9c120c9b4fde666"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-18 13:40:37.000000000","tz":120},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ec1a186aebea396f2643ff3a3abf27832d88e69a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ec1a186aebea396f2643ff3a3abf27832d88e69a"}]},"branch":"refs/heads/master"},"e09dd945aaf32ea39c1a8c875a20778ca327d2f5":{"kind":"REWORK","_number":23,"created":"2023-01-18 15:01:38.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/34/775034/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/23"}}},"commit":{"parents":[{"commit":"e42249631e502d683d0a28fcf9c120c9b4fde666","subject":"Merge \"[RBAC] Making sure export locations are viewable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e42249631e502d683d0a28fcf9c120c9b4fde666"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-18 15:01:27.000000000","tz":120},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e09dd945aaf32ea39c1a8c875a20778ca327d2f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e09dd945aaf32ea39c1a8c875a20778ca327d2f5"}]},"branch":"refs/heads/master"},"60dafff1a2c381a96e01dccbdb142eee9a9bcdd4":{"kind":"REWORK","_number":24,"created":"2023-01-18 20:35:18.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/34/775034/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/24"}}},"commit":{"parents":[{"commit":"e42249631e502d683d0a28fcf9c120c9b4fde666","subject":"Merge \"[RBAC] Making sure export locations are viewable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e42249631e502d683d0a28fcf9c120c9b4fde666"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-18 20:29:41.000000000","tz":120},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/60dafff1a2c381a96e01dccbdb142eee9a9bcdd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/60dafff1a2c381a96e01dccbdb142eee9a9bcdd4"}]},"branch":"refs/heads/master"},"4b35a13e2618c81efb104fa1dc5fd8f2dc648e1d":{"kind":"REWORK","_number":25,"created":"2023-01-18 20:44:01.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/34/775034/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/25"}}},"commit":{"parents":[{"commit":"e42249631e502d683d0a28fcf9c120c9b4fde666","subject":"Merge \"[RBAC] Making sure export locations are viewable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e42249631e502d683d0a28fcf9c120c9b4fde666"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-01-18 20:43:45.000000000","tz":120},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4b35a13e2618c81efb104fa1dc5fd8f2dc648e1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4b35a13e2618c81efb104fa1dc5fd8f2dc648e1d"}]},"branch":"refs/heads/master"},"191ae7f02b9e5905a1549669dbd7615e6dd29dc0":{"kind":"REWORK","_number":26,"created":"2023-03-15 09:37:41.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/34/775034/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/26"}}},"commit":{"parents":[{"commit":"5380c955f37e34cdae248e55114b51f1f39842c5","subject":"Merge \"Correct share group tests cleanup order\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5380c955f37e34cdae248e55114b51f1f39842c5"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-03-15 09:35:21.000000000","tz":120},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/191ae7f02b9e5905a1549669dbd7615e6dd29dc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/191ae7f02b9e5905a1549669dbd7615e6dd29dc0"}]},"branch":"refs/heads/master"},"0a13a8da6ccfafff6b2dfedf2587d594722a8c2d":{"kind":"REWORK","_number":27,"created":"2023-03-15 13:01:22.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/34/775034/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/27"}}},"commit":{"parents":[{"commit":"e42249631e502d683d0a28fcf9c120c9b4fde666","subject":"Merge \"[RBAC] Making sure export locations are viewable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e42249631e502d683d0a28fcf9c120c9b4fde666"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-03-15 13:01:13.000000000","tz":120},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0a13a8da6ccfafff6b2dfedf2587d594722a8c2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0a13a8da6ccfafff6b2dfedf2587d594722a8c2d"}]},"branch":"refs/heads/master"},"4c836af6a3c34881fdbe04836848af71770259e2":{"kind":"REWORK","_number":28,"created":"2023-03-22 14:28:11.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/34/775034/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/28"}}},"commit":{"parents":[{"commit":"5380c955f37e34cdae248e55114b51f1f39842c5","subject":"Merge \"Correct share group tests cleanup order\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5380c955f37e34cdae248e55114b51f1f39842c5"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-03-22 14:28:01.000000000","tz":120},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4c836af6a3c34881fdbe04836848af71770259e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4c836af6a3c34881fdbe04836848af71770259e2"}]},"branch":"refs/heads/master"},"d5344971622b419e79d9e55eb9548a71235ac45a":{"kind":"REWORK","_number":29,"created":"2023-03-22 15:52:33.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/34/775034/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/29"}}},"commit":{"parents":[{"commit":"5380c955f37e34cdae248e55114b51f1f39842c5","subject":"Merge \"Correct share group tests cleanup order\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5380c955f37e34cdae248e55114b51f1f39842c5"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-03-22 15:52:23.000000000","tz":120},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d5344971622b419e79d9e55eb9548a71235ac45a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d5344971622b419e79d9e55eb9548a71235ac45a"}]},"branch":"refs/heads/master"},"9eb545dd55ad79c87b75a6c0ca834f74ee18b3cb":{"kind":"REWORK","_number":30,"created":"2023-03-23 11:54:22.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/34/775034/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/30"}}},"commit":{"parents":[{"commit":"5380c955f37e34cdae248e55114b51f1f39842c5","subject":"Merge \"Correct share group tests cleanup order\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5380c955f37e34cdae248e55114b51f1f39842c5"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-03-23 11:54:11.000000000","tz":120},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9eb545dd55ad79c87b75a6c0ca834f74ee18b3cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9eb545dd55ad79c87b75a6c0ca834f74ee18b3cb"}]},"branch":"refs/heads/master"},"58a93fabdd78e0416781e74a0112a2e9e01fc8ec":{"kind":"REWORK","_number":31,"created":"2023-03-23 13:26:57.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/34/775034/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/31"}}},"commit":{"parents":[{"commit":"5380c955f37e34cdae248e55114b51f1f39842c5","subject":"Merge \"Correct share group tests cleanup order\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5380c955f37e34cdae248e55114b51f1f39842c5"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-03-23 13:26:48.000000000","tz":120},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/58a93fabdd78e0416781e74a0112a2e9e01fc8ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/58a93fabdd78e0416781e74a0112a2e9e01fc8ec"}]},"branch":"refs/heads/master"},"43e8f5293c0564a6960a3124f1740149f94e5860":{"kind":"REWORK","_number":32,"created":"2023-03-23 14:12:17.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/34/775034/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/32"}}},"commit":{"parents":[{"commit":"5380c955f37e34cdae248e55114b51f1f39842c5","subject":"Merge \"Correct share group tests cleanup order\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5380c955f37e34cdae248e55114b51f1f39842c5"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-03-23 14:12:05.000000000","tz":120},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/43e8f5293c0564a6960a3124f1740149f94e5860"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/43e8f5293c0564a6960a3124f1740149f94e5860"}]},"branch":"refs/heads/master"},"72b178bab244bb756d18fd7ab2c21580ef796318":{"kind":"REWORK","_number":33,"created":"2023-03-26 13:16:28.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/34/775034/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/33"}}},"commit":{"parents":[{"commit":"5380c955f37e34cdae248e55114b51f1f39842c5","subject":"Merge \"Correct share group tests cleanup order\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5380c955f37e34cdae248e55114b51f1f39842c5"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-03-26 13:16:18.000000000","tz":180},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/72b178bab244bb756d18fd7ab2c21580ef796318"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/72b178bab244bb756d18fd7ab2c21580ef796318"}]},"branch":"refs/heads/master"},"73c7e6a6cf2471cbd02e01bd0752732c2dbd9cb3":{"kind":"REWORK","_number":34,"created":"2023-03-28 11:39:17.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/34/775034/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/34"}}},"commit":{"parents":[{"commit":"5380c955f37e34cdae248e55114b51f1f39842c5","subject":"Merge \"Correct share group tests cleanup order\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5380c955f37e34cdae248e55114b51f1f39842c5"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-03-28 11:39:07.000000000","tz":180},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/73c7e6a6cf2471cbd02e01bd0752732c2dbd9cb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/73c7e6a6cf2471cbd02e01bd0752732c2dbd9cb3"}]},"branch":"refs/heads/master"},"97040f514574a483603751b6a4a734f6a68af871":{"kind":"REWORK","_number":35,"created":"2023-03-28 12:52:24.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/34/775034/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/35"}}},"commit":{"parents":[{"commit":"5380c955f37e34cdae248e55114b51f1f39842c5","subject":"Merge \"Correct share group tests cleanup order\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5380c955f37e34cdae248e55114b51f1f39842c5"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-03-28 12:52:15.000000000","tz":180},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/97040f514574a483603751b6a4a734f6a68af871"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/97040f514574a483603751b6a4a734f6a68af871"}]},"branch":"refs/heads/master"},"ba4ecedf8305488735dbda156a00049ebd348250":{"kind":"REWORK","_number":36,"created":"2023-03-28 15:17:14.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/34/775034/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/36"}}},"commit":{"parents":[{"commit":"5380c955f37e34cdae248e55114b51f1f39842c5","subject":"Merge \"Correct share group tests cleanup order\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5380c955f37e34cdae248e55114b51f1f39842c5"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-03-28 15:16:55.000000000","tz":180},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ba4ecedf8305488735dbda156a00049ebd348250"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ba4ecedf8305488735dbda156a00049ebd348250"}]},"branch":"refs/heads/master"},"6ebd17234f6c9f643e5395483fa43edb649775e1":{"kind":"REWORK","_number":37,"created":"2023-03-28 18:59:35.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/34/775034/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/37"}}},"commit":{"parents":[{"commit":"5380c955f37e34cdae248e55114b51f1f39842c5","subject":"Merge \"Correct share group tests cleanup order\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5380c955f37e34cdae248e55114b51f1f39842c5"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-03-28 18:59:20.000000000","tz":180},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6ebd17234f6c9f643e5395483fa43edb649775e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6ebd17234f6c9f643e5395483fa43edb649775e1"}]},"branch":"refs/heads/master"},"72e9d54e7b88a5bb1ea851aec8a7bf5043f82b60":{"kind":"REWORK","_number":38,"created":"2023-03-28 20:00:44.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/34/775034/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/38"}}},"commit":{"parents":[{"commit":"5380c955f37e34cdae248e55114b51f1f39842c5","subject":"Merge \"Correct share group tests cleanup order\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5380c955f37e34cdae248e55114b51f1f39842c5"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-03-28 20:00:31.000000000","tz":180},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/72e9d54e7b88a5bb1ea851aec8a7bf5043f82b60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/72e9d54e7b88a5bb1ea851aec8a7bf5043f82b60"}]},"branch":"refs/heads/master"},"99c21c9592daade6bf86b1b2c1bd8ba5bba23235":{"kind":"REWORK","_number":39,"created":"2023-03-29 05:22:20.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/34/775034/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/39"}}},"commit":{"parents":[{"commit":"5380c955f37e34cdae248e55114b51f1f39842c5","subject":"Merge \"Correct share group tests cleanup order\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5380c955f37e34cdae248e55114b51f1f39842c5"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-03-29 05:22:02.000000000","tz":180},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/99c21c9592daade6bf86b1b2c1bd8ba5bba23235"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/99c21c9592daade6bf86b1b2c1bd8ba5bba23235"}]},"branch":"refs/heads/master"},"4c6fa46c9c2ffb01ad085505106e389c62aa5e0e":{"kind":"REWORK","_number":40,"created":"2023-03-29 06:54:03.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/34/775034/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/775034/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/775034/40"}}},"commit":{"parents":[{"commit":"5380c955f37e34cdae248e55114b51f1f39842c5","subject":"Merge \"Correct share group tests cleanup order\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5380c955f37e34cdae248e55114b51f1f39842c5"}]}],"author":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2021-02-08 20:38:51.000000000","tz":0},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2023-03-29 06:53:52.000000000","tz":180},"subject":"Add functional tests for add/update security service feature","message":"Add functional tests for add/update security service feature\n\nThis patch adds functional tests positives and negatives\nfor the new APIs methods `manila share-network-security\n-service-update` and `manila share-network-reset-state`.\nAlso, we added and fixed tests related to\n`manila share-network-security-service-add`.\n\nPartially Implements: bp add-security-service-in-use-share-networks\nChange-Id: I8c1b882f9c4fe40a14d50a4ebb7a40ff5c94bb46\nSigned-off-by: Andre Beltrami \u003cdebeltrami@gmail.com\u003e\nCo-Authored-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4c6fa46c9c2ffb01ad085505106e389c62aa5e0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4c6fa46c9c2ffb01ad085505106e389c62aa5e0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
