)]}'
{"id":"openstack%2Fmanila-tempest-plugin~775076","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I17c64b8ffe0538309bdd823c723dbf1176a92754","project":"openstack/manila-tempest-plugin","branch":"master","topic":"share_server_dhss","hashtags":[],"change_id":"I17c64b8ffe0538309bdd823c723dbf1176a92754","subject":"Move share server creation check under admin test suite","status":"MERGED","created":"2021-02-11 07:54:11.000000000","updated":"2021-03-03 14:29:07.000000000","submitted":"2021-03-03 14:29:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775076-share_server_dhss","meta_rev_id":"623e08d2748ada84f83df164e6b6e86ab58c44e5","_number":775076,"virtual_id_number":775076,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-03 14:28:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2021-03-02 20:05:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2021-03-03 12:58:48.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":2,"date":"2021-03-03 13:11:01.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":1,"date":"2021-03-03 13:11:01.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-11 07:55:22.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2021-02-11 07:55:22.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-02-11 07:55:22.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-02-11 07:55:22.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-02-11 08:55:45.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-02-11 11:01:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-11 11:13:53.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-11 13:24:55.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"a83840112758ed72f9963971777ecd199a4168fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-02-11 07:54:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6d3c899ab122c08c722c188730e72b3f3acffe0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-11 08:55:45.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/76/775076/1/check/DellEMC_Unity_Manila_dhss_true/f934ca2/DellEMC_Unity_Manila_dhss_true/288 : FAILURE in 1h 01m 18s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/76/775076/1/check/DellEMC_Unity_Manila_dhss_false/1735741/DellEMC_Unity_Manila_dhss_false/272 : SUCCESS in 59m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"2ec58f921b7d7404f5c62576fcfa2ba8f6cda93d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 11:01:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67306ceb4bbe4fa6aded003e62c340ad : SUCCESS in 2m 59s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9797188cc6a840c898a008a396726300 : SUCCESS in 30m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/264e089f07ad46d69ccd669b4b0fd507 : SUCCESS in 35m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/87ed6e570b42472e8996ac090cf12338 : SUCCESS in 1h 05m 57s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9cc02937246d4d469d0d953049c8b846 : FAILURE in 1h 21m 20s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/cdc405fd5435472fa15c7a739009886f : SUCCESS in 59m 46s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f37443f92aca411e9dbcc4bcf1547841 : SUCCESS in 1h 16m 49s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/447034da551b44d4a9fff3a89e4676f1 : SUCCESS in 42m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ad8d22649cc44958be8c4229663d9dc2 : FAILURE in 2h 10m 24s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8b1cf3f1fd20453697cdc5b60dae4fdf : SUCCESS in 24m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9f6f18e1118c1bdeb6b4f443370bcc0e742167d9","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-11 11:13:53.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/76/775076/1/check/DellEMC_VNX_Manila/6a5234c/DellEMC_VNX_Manila/172 : FAILURE in 12m 01s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"cad359a1d2143e82dcd93c8f4f20e37cd40c30a1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-11 11:57:20.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/76/775076/1/check/DellEMC_Unity_Manila_dhss_true/9caf4e2/DellEMC_Unity_Manila_dhss_true/289 : FAILURE in 55m 13s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/76/775076/1/check/DellEMC_Unity_Manila_dhss_false/dbca146/DellEMC_Unity_Manila_dhss_false/273 : SUCCESS in 55m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"ee1810ec8196c4ba6d8b7cb5d650281eb1ff43d7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-11 13:24:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/76/775076/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/b123f9e/ : SUCCESS in 2h 23m 02s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/76/775076/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/4ab6fab/ : SUCCESS in 1h 49m 02s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"390ce32f2d6e91945d5b91ee12ba40de43a2a3b4","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-02-11 15:12:02.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM. Thanks.","accounts_in_message":[],"_revision_number":1},{"id":"a7ef43cbfb48c2b86dd81763f0fed218634861b5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-02-12 19:33:52.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks, LGTM","accounts_in_message":[],"_revision_number":1},{"id":"1b281bad389fcc1849e899a4819d2c92de05a15d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-03-02 14:28:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bff53d7cd097048a20858cd2c4361ca4f4e8e74c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 16:37:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/645fb78ae5564ac09cb05bcef7b7e919 : SUCCESS in 4m 23s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9807ee145392402f8a3538a9ebc97e9f : SUCCESS in 36m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/64ec9104b86b4cb481915e1a8540b551 : SUCCESS in 39m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/13c0df80a18b4526a9716a71b0418171 : SUCCESS in 1h 12m 24s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/58ee2549e4d24669b0571cd3d78cef98 : FAILURE in 1h 21m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ac03949bbcbd449fbc79972f16a44bac : SUCCESS in 1h 01m 29s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3c688b1ba88e40c185c282016f0aa3ad : SUCCESS in 1h 24m 33s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c995fcb09cda4fca8a1effeb8e69afb5 : SUCCESS in 42m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/eb41d84e16ac4f549342500e90092ddb : FAILURE in 2h 06m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/228675aa37f0413c8a607794b8bb3120 : SUCCESS in 23m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3dc27f4d2e856e322bd6358371bc356bd438bad4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-02 17:06:21.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/76/775076/2/check/DellEMC_Unity_Manila_dhss_true/b5f31d5/DellEMC_Unity_Manila_dhss_true/414 : FAILURE in 1h 19m 28s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/76/775076/2/check/DellEMC_Unity_Manila_dhss_false/9bd8b37/DellEMC_Unity_Manila_dhss_false/398 : FAILURE in 58m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"dd30362b4a4808ca4701b0f86dbda967e143c52c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-02 19:03:09.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/76/775076/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/71a2ee4/ : SUCCESS in 2h 25m 16s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/76/775076/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/d2cb6bd/ : SUCCESS in 1h 43m 23s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"c63ec8b40c5de5d6472e6aa0276fd3b80d68c335","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-02 19:06:11.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/76/775076/2/check/DellEMC_VNX_Manila/013536d/DellEMC_VNX_Manila/232 : SUCCESS in 1h 08m 58s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"d54c43c389d96dc8ecba4ccaef4ccc462c20cc55","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-03-02 20:05:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bce232eeff53d62c7bc8a1bf0be9a6b8ebdc3cfa","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-03-03 12:58:48.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"338168f3c2b80c8efe6b8b6ea765075e9dda4931","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-03 13:11:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0181c4c3245af0188d491c47cf29e0943cdd0a13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 13:11:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5945be0d0f53b35e9cfa84537dfc0749463f6e9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 14:28:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e778a861e73648ea8a3ec1b84c1d7880 : SUCCESS in 5m 20s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f1d604e671f14a6d806144d073243dae : SUCCESS in 39m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3aaa610f29614d1bbc127191e649a67c : SUCCESS in 35m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c4eb686839184c3dadcdda5eac31b6bd : SUCCESS in 1h 15m 35s","accounts_in_message":[],"_revision_number":2},{"id":"623e08d2748ada84f83df164e6b6e86ab58c44e5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 14:29:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ab41ebdd19534eb2dba7106b60b376325f92637d","revisions":{"1373e8b04e468225a5bbf77c938172419c625263":{"kind":"REWORK","_number":1,"created":"2021-02-11 07:54:11.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/76/775076/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/76/775076/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/775076/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/775076/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/775076/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/775076/1"}}},"commit":{"parents":[{"commit":"d2138765888fe16e68f713ed99121f41e01a3c51","subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-11 07:45:45.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-11 07:45:45.000000000","tz":120},"subject":"Move share server creation check under admin test suite","message":"Move share server creation check under admin test suite\n\nGetting \"share_server_id\" parameter requires an admin client.\nTherefore it should be placed whithin the admin test suite.\n\nChange-Id: I17c64b8ffe0538309bdd823c723dbf1176a92754\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1373e8b04e468225a5bbf77c938172419c625263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1373e8b04e468225a5bbf77c938172419c625263"}]},"branch":"refs/heads/master"},"ab41ebdd19534eb2dba7106b60b376325f92637d":{"kind":"REWORK","_number":2,"created":"2021-03-02 14:28:15.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/76/775076/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/76/775076/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/775076/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/775076/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/775076/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/775076/2"}}},"commit":{"parents":[{"commit":"d9294f9958084c2cc6eba1645910fe8e132cc44f","subject":"Merge \"Consolidate waiters methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d9294f9958084c2cc6eba1645910fe8e132cc44f"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-11 07:45:45.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-03-02 14:27:39.000000000","tz":120},"subject":"Move share server creation check under admin test suite","message":"Move share server creation check under admin test suite\n\nGetting \"share_server_id\" parameter requires an admin client.\nTherefore it should be placed whithin the admin test suite.\n\nChange-Id: I17c64b8ffe0538309bdd823c723dbf1176a92754\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ab41ebdd19534eb2dba7106b60b376325f92637d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ab41ebdd19534eb2dba7106b60b376325f92637d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
