)]}'
{"id":"openstack%2Fmanila-tempest-plugin~775148","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ie88f8f6cbbe444968026f163c99649eab623c457","project":"openstack/manila-tempest-plugin","branch":"master","topic":"default_share_type","hashtags":[],"change_id":"Ie88f8f6cbbe444968026f163c99649eab623c457","subject":"If default share type exists, use it for share creation","status":"ABANDONED","created":"2021-02-11 12:34:29.000000000","updated":"2021-08-12 05:04:11.000000000","total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"8d5d7758f00b70d27c220cfe535fe251f05634e7","_number":775148,"virtual_id_number":775148,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-26 18:08:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-11 14:03:54.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-02-11 15:00:14.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-02-11 15:26:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-14 10:22:51.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-14 11:30:51.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"3861b098c22996d556d5a5c3c0a5e6cf83206039","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-02-11 12:34:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db83ed2424f93f7f28490b3237dcb8a16a8a114f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-02-11 12:38:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b3e5d900f797f07cfb165b31690a258fba35d842","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-11 14:03:54.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/48/775148/2/check/DellEMC_Unity_Manila_dhss_true/b741528/DellEMC_Unity_Manila_dhss_true/291 : FAILURE in 1h 06m 51s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/48/775148/2/check/DellEMC_Unity_Manila_dhss_false/e453f85/DellEMC_Unity_Manila_dhss_false/275 : FAILURE in 50m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"0407bed28fc62c07f62955dab693a68ec2fbd83f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-02-11 15:00:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3336e6cfd7e18da54c04ccf6ae4bbac19da6903f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 15:26:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d7fb48fa4374e728921ed67af605f98 : SUCCESS in 4m 36s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/123d6729056f4996a84450e7c1e3a989 : FAILURE in 33m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2979292db55744449d6e266d1985edaf : FAILURE in 34m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/820142632ebb44a2a8190bdd96f48058 : FAILURE in 1h 13m 33s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2d29fbbd18d7438fa002a0c7d626b1e7 : FAILURE in 1h 20m 24s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d0435860011f41edbdc9c5faab67ec06 : SUCCESS in 59m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/26ef6fca6b9b49cfbaa2512797d12286 : SUCCESS in 1h 35m 02s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1dba1088fd1a4f288ffd1b4e7257b9e7 : FAILURE in 31m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/36eec29dd7114aa6a7f01dfecbe4cd24 : POST_FAILURE in 2h 12m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/bd30b776f3a34d2f8f75a639253c6d5c : SUCCESS in 26m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ab130b7bf04b472f37100cd6d59278f1051aeb5b","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-02-12 05:57:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9abc3106eea932f7e72ec966adda972ff7f046bf","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-02-12 06:21:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0806337bf73ad41fdb838d2de6acaca8741400d1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-02-12 19:50:31.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)\n\nCool, i understand this change better - but, it looks like some of the tests are failing as a consequence. please take a look..","accounts_in_message":[],"_revision_number":2},{"id":"795c42caba92ffa6855095c45700eb7235261484","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-02-14 06:29:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"31fc309936f65e39efd9ec2efdc73befab26ff75","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-02-14 06:30:24.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n\u003e Patch Set 2: -Code-Review\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e Cool, i understand this change better - but, it looks like some of the tests are failing as a consequence. please take a look..\n\nIt failed because I didn\u0027t cover the option for creating share from snapshot.\nWhen creating share from snapshot a new share type shouldn\u0027t be created.\nSo I fixed the condition.","accounts_in_message":[],"_revision_number":2},{"id":"a84b4c61449287aee24868b62679341fce85be19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-02-14 06:32:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"70ad3200449c311bd8ebe35af756efca45cc9e49","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-14 07:38:28.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/48/775148/4/check/DellEMC_Unity_Manila_dhss_true/ac99e80/DellEMC_Unity_Manila_dhss_true/323 : FAILURE in 54m 07s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/48/775148/4/check/DellEMC_Unity_Manila_dhss_false/2619e5d/DellEMC_Unity_Manila_dhss_false/307 : SUCCESS in 1h 05m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"ef4c3ee60e7755f0da734a7768d0e4f8537ba919","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-14 09:10:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c88046a9c9c4b86b5b54f4869bed83f : SUCCESS in 3m 26s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a676440031f340edb5fb50534b8dcafa : SUCCESS in 26m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8e840797665c4d40b638eb3f1db74160 : SUCCESS in 37m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ceda4fca069e4dbaabc60edff7f861f8 : SUCCESS in 1h 04m 53s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3e6c00ee7de9489cb37c500d1c4fef4b : SUCCESS in 1h 19m 57s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/66519b8800ee4478885c58975359bc23 : SUCCESS in 1h 18m 59s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f36b36c278804fd786270f17030ef818 : SUCCESS in 1h 07m 50s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fdb784c8fe054ff69a94f116122dca2c : SUCCESS in 49m 12s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ee17b4d051474552897670d3d43b0cd6 : FAILURE in 2h 32m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/fb49028290d34b209a6081447f1fd33b : SUCCESS in 29m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"eeb1f15383311b6cf39004aa03d0b86ec19b23e6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-14 10:14:50.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/48/775148/4/check/DellEMC_Unity_Manila_dhss_true/f792195/DellEMC_Unity_Manila_dhss_true/324 : FAILURE in 53m 11s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/48/775148/4/check/DellEMC_Unity_Manila_dhss_false/6748862/DellEMC_Unity_Manila_dhss_false/308 : SUCCESS in 1h 04m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"574184bc2f61b3ae38ef0d8a0818d4dd1570326e","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-14 10:22:51.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/48/775148/4/check/DellEMC_VNX_Manila/7d6cb03/DellEMC_VNX_Manila/187 : SUCCESS in 1h 12m 32s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"c7ad3b47f8b84377a49fc8a4aa0fc32d662e46d2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-14 11:30:51.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/48/775148/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/b54a049/ : SUCCESS in 2h 20m 29s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/48/775148/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/c2fe48f/ : FAILURE in 46m 58s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"47591e991db2a48cae718eb87b87ac01f4c7c75d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-05-23 08:50:49.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"caf43c96af097e20ca2a72aa772fb4eb6e13593c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-23 10:04:40.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/48/775148/5/check/DellEMC_Unity_Manila_dhss_true/3a26d2c/DellEMC_Unity_Manila_dhss_true/74 : FAILURE in 1h 13m 35s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/48/775148/5/check/DellEMC_Unity_Manila_dhss_false/5664260/DellEMC_Unity_Manila_dhss_false/74 : SUCCESS in 58m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"b6809ca0bd0cf7beb46f2750bd9ce4995b21ad02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-23 11:20:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1c7bc4d25d64e7785ca703e44142e49 : SUCCESS in 3m 29s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b524ffda8e8b4b04874baa9d548f6601 : SUCCESS in 26m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6bbbaecc84c245f188cc066541590b28 : SUCCESS in 28m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a5808a049b8b43a3b7d601a30204a216 : SUCCESS in 1h 06m 12s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/60ee00aad68042948f8a6e07cc4c6b47 : SUCCESS in 1h 08m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b42c06c84cb94680ac8eb13a693d2ef3 : FAILURE in 1h 05m 43s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/88a4cf9d0d3d4238be4b3fdcf83e3fa8 : FAILURE in 1h 07m 37s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b1900efda12d4c1b9796f82bb7264f70 : SUCCESS in 37m 03s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2c3889db1b784821852b0e7a62fe5542 : FAILURE in 1h 15m 15s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ab95f501014e4349b73754fe84f6c605 : SUCCESS in 18m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"92722aa0a366ffa5ee5a85418f8dd0d97421fa35","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-23 11:35:11.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/48/775148/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/a35777b/ : FAILURE in 14m 48s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/48/775148/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/9f19081/ : FAILURE in 8m 33s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"1e6deb9d55a76bc55c9661662a22bb2aaa7249c2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-23 12:19:41.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/48/775148/5/check/DellEMC_Unity_Manila_dhss_true/f8aad15/DellEMC_Unity_Manila_dhss_true/75 : FAILURE in 58m 43s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/48/775148/5/check/DellEMC_Unity_Manila_dhss_false/927c566/DellEMC_Unity_Manila_dhss_false/75 : SUCCESS in 59m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"7b0a3d617535a5e9913d8386b1fccbd30e07d1ec","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-05-23 12:24:56.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/48/775148/5/check/DellEMC_VNX_Manila/69ccbe4/DellEMC_VNX_Manila/49 : SUCCESS in 1h 04m 35s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"07767ac6af29fb65e89f2caf5af4a438bd120b76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-05-24 08:53:28.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b00834bae7e26502e83c1e0aba0b9201084112f4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-24 09:40:38.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/48/775148/6/check/DellEMC_Unity_Manila_dhss_true/4c75534/DellEMC_Unity_Manila_dhss_true/77 : FAILURE in 46m 47s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/48/775148/6/check/DellEMC_Unity_Manila_dhss_false/4404539/DellEMC_Unity_Manila_dhss_false/77 : FAILURE in 46m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"3c4122fba8d0c3ef4f6e8ea3fcd7993fb03985ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 10:34:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9daa74b2cf9e4090bb5e3aae52fa4003 : SUCCESS in 4m 22s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7637a7bec7f5433fb8fe21c22f5b4195 : FAILURE in 21m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3d86fd9426614f0893314d22eb8b364c : FAILURE in 27m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/73a44fbf75c9464cbe31c1c06e643109 : FAILURE in 59m 20s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5da81a10c7114ea28b7683f1139a4c69 : FAILURE in 1h 11m 33s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d6e73fef45b9413f9d3d9cfa59bb72e1 : FAILURE in 1h 20m 06s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5fd814e1f6eb4a0e87b4418bd9fa62e6 : FAILURE in 1h 20m 27s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/89d59ebdd644424db34b3a57e6183f03 : FAILURE in 36m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4eb2645490354efcb632e54039c1b842 : FAILURE in 1h 39m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5670ebdd90aa452685650e3d4ef7868e : FAILURE in 20m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"58581f587e369b6f56d425525f105e355c101105","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-05-24 11:46:55.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"12cc42bd90dcaaa7ece4e6ecc8725fede724f223","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-24 12:44:24.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/48/775148/7/check/DellEMC_Unity_Manila_dhss_true/bc3a1f0/DellEMC_Unity_Manila_dhss_true/78 : FAILURE in 57m 11s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/48/775148/7/check/DellEMC_Unity_Manila_dhss_false/4c1e284/DellEMC_Unity_Manila_dhss_false/78 : FAILURE in 53m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"dd113a23fa90bdd003501405f1945d5cd4e50046","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-05-24 13:39:42.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"1ab288cc383655015d48cb24f9e0facbef5a0dfa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-24 14:55:36.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/48/775148/8/check/DellEMC_Unity_Manila_dhss_true/dc3d10e/DellEMC_Unity_Manila_dhss_true/79 : FAILURE in 1h 15m 37s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/48/775148/8/check/DellEMC_Unity_Manila_dhss_false/e5ae63e/DellEMC_Unity_Manila_dhss_false/79 : FAILURE in 1h 07m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"408c442374b7e22995bec80be56765c17bdaadb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 16:01:46.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dffab2fbefa4a828e2d1749d9ce52d2 : SUCCESS in 3m 30s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/be0136da120544b0b0814891bb92724d : FAILURE in 31m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/17a6a2c09e864466ac89acb855ddf4e8 : FAILURE in 33m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/252a84e94b6c4b38af0e1771d7bf7c63 : FAILURE in 1h 09m 37s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8d08e39855294492860255bc26bad816 : FAILURE in 1h 19m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/181555c24ba74d5089b6626e20b901d3 : FAILURE in 1h 12m 20s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/90986084d881480ea855f8e052def2f5 : SUCCESS in 1h 42m 43s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/362962f867064e199118c01387ee610f : FAILURE in 47m 14s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8e743639b6a74ffd819f9e8b58ce30d5 : FAILURE in 2h 16m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4a791f7afd7041f692e2cbd5a32d40ad : SUCCESS in 24m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8f6770d1476871567c6b3a171fd7e568ca5a76ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-05-25 07:29:41.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"4fffb68e1e7c4e47719d9618b5f81997d48e6491","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-25 07:48:34.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/48/775148/9/check/DellEMC_Unity_Manila_dhss_true/da144b1/DellEMC_Unity_Manila_dhss_true/80 : FAILURE in 16m 39s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/48/775148/9/check/DellEMC_Unity_Manila_dhss_false/2ee3e03/DellEMC_Unity_Manila_dhss_false/80 : FAILURE in 18m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"4435822c7a9201ffac826259a3033e8d36439343","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 09:14:59.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f69fc13e8b3480ea6f78b84f942cf2b : SUCCESS in 3m 21s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/db4f7cf54e55475d8584a7249e296d1d : TIMED_OUT in 1h 05m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8fe7ebe68e1645388167283e3f34be34 : FAILURE in 36m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/13682ff9da0046ca8dfab62e81073542 : FAILURE in 59m 33s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/266c111f94d846c7a489fa5d5a2667bf : FAILURE in 1h 08m 15s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ae3326e0b1854669b3b406afa4948f63 : FAILURE in 46m 37s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f016de16fcf540728e03eab658fd6450 : FAILURE in 1h 28m 52s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c647498507be4588aca1b889d3c4e085 : FAILURE in 50m 12s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9335b67741904d5aae1cd1e1f92fa446 : FAILURE in 1h 43m 49s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a9606d81205247adaedeec70bf76d6a7 : FAILURE in 17m 39s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2547db5d135d28b0a3860c794c1760439cb27212","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-05-25 10:32:52.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"7b3ed87755b23be9de5c859ea56503aa2733b36d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-05-25 12:14:28.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"9b63fd3589f16a420f706f76a61e9fe83843c35f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-05-25 15:06:29.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"2bd6ebc96bffa0af3a4ab31e8847dbd660f6c52b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 17:00:41.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccff5766737849959b53ba3e4d9a43df : SUCCESS in 3m 36s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a7f648dfdcc948f1af50efdfea134c19 : FAILURE in 31m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/840226feb2a148338a0a7f0b15e01a94 : FAILURE in 40m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2446fceefa724d9382d7f377be44da7e : FAILURE in 1h 15m 12s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/70109341de41466f897005efa66c8557 : FAILURE in 1h 32m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/29730fc7ebff443a9d8583808ebd81a9 : FAILURE in 1h 22m 44s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/75f62a1b0b124bfa83ea0aead582392d : SUCCESS in 1h 38m 18s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fba8c2c1f8304a8597d0a0fdba3b4f47 : SUCCESS in 54m 19s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2acff96a15c848ab9117c4a6092df444 : FAILURE in 1h 09m 04s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5a9e84bf9d0642efb9d61f25db9257df : SUCCESS in 28m 07s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"6e2622ee54b1cf2b9991f2bb12a05422e784f603","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-05-26 05:18:41.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"5ebaaa04328b47c3960ac2e1b5696c3936c61bc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 07:00:43.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ecf87ee89b94880b8018e4b6bdbf2f7 : SUCCESS in 4m 27s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0629db3e78fb492abf7a6a28f43ff41c : FAILURE in 32m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/54b1325e52584adeabe95af66e45ac5a : FAILURE in 27m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/10e88aba3e024d94ae82c406369e1f7c : FAILURE in 1h 13m 01s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b6aea688422e47cbab205f8cc0e671e9 : FAILURE in 1h 12m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d784f0e0d8fa4cff86d2bf99a52ffb75 : SUCCESS in 1h 15m 45s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2c9ce3493ff04f94a0e1342b8db45734 : SUCCESS in 1h 37m 39s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a4e22f26070a44f2bc657bac93177509 : SUCCESS in 57m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/aaeb371838504df58084464e42c1bd95 : FAILURE in 1h 23m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1a5bc42f27f64887a59b419ab988929b : SUCCESS in 26m 24s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"fa5772672ccbcdad1f65bd7315809e47d9d48fde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-05-26 07:28:16.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"86d65307c3e1d8004b35587650fb63d6ca847ba5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-05-26 13:29:26.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"426e67f8099ad5fc5d0cc05c7e1670fdf1d18358","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 18:08:34.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32e73258a2a741f2bc398b36eed144dc : SUCCESS in 3m 58s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9687ea72a2aa49d1871c03bd7c8f35f5 : FAILURE in 30m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c0935a07d6dc465ea4927aada53c87f8 : FAILURE in 31m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4377f5b68da2416191de42c2f1a1c25a : FAILURE in 1h 30m 39s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f02da416fc094ab99df70d33ffa855c0 : FAILURE in 56m 10s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/964815ead4dc4a2d82a956efe56b5062 : FAILURE in 1h 15m 17s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/17fd7bbd6d8740b08372fa5fe5b8dc89 : SUCCESS in 1h 32m 23s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6efedd04e7664585a160ee7fee6430eb : SUCCESS in 58m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/43838895ae9f444da08f7bf7eec2a5c2 : FAILURE in 1h 27m 05s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/40952654dccb404ea4a9e46ac3b32bbb : SUCCESS in 26m 01s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"8d5d7758f00b70d27c220cfe535fe251f05634e7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-08-12 05:04:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"7ad4da25a83404c60a2a01319075af4196b4dc90","revisions":{"a86556c264355e2ec6947effc0858c1fdf0d7db4":{"kind":"REWORK","_number":1,"created":"2021-02-11 12:34:29.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/775148/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/775148/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/1"}}},"commit":{"parents":[{"commit":"d2138765888fe16e68f713ed99121f41e01a3c51","subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-11 12:04:37.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-11 12:04:37.000000000","tz":120},"subject":"If default share type exists, use it for share creation","message":"If default share type exists, use it for share creation\n\nThis patch adds \"create_share\" method under \"BaseSharesAdminTest\"\nclass. This wrapper allows us to check whether default share type\nexists, if exists use it for share creation if not create a new\none.\n\nThis patch removes duplicated code by changing the inheritances.\n\nChange-Id: Ie88f8f6cbbe444968026f163c99649eab623c457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a86556c264355e2ec6947effc0858c1fdf0d7db4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a86556c264355e2ec6947effc0858c1fdf0d7db4"}]},"branch":"refs/heads/master"},"10508af7921ebc85a74ffdb9a85754bd8738ed3b":{"kind":"REWORK","_number":2,"created":"2021-02-11 12:38:32.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/775148/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/775148/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/2"}}},"commit":{"parents":[{"commit":"d2138765888fe16e68f713ed99121f41e01a3c51","subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-11 12:04:37.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-11 12:38:09.000000000","tz":120},"subject":"If default share type exists, use it for share creation","message":"If default share type exists, use it for share creation\n\nThis patch adds \"create_share\" method under \"BaseSharesAdminTest\"\nclass. This wrapper allows us to check whether default share type\nexists, if exists use it for share creation if not create a new\none.\n\nThis patch removes duplicated code by changing the inheritances.\n\nChange-Id: Ie88f8f6cbbe444968026f163c99649eab623c457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/10508af7921ebc85a74ffdb9a85754bd8738ed3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/10508af7921ebc85a74ffdb9a85754bd8738ed3b"}]},"branch":"refs/heads/master"},"c0c93ac730fd4b25b7bfe18acb2e4718716f136b":{"kind":"REWORK","_number":3,"created":"2021-02-14 06:29:03.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/775148/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/775148/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/3"}}},"commit":{"parents":[{"commit":"d2138765888fe16e68f713ed99121f41e01a3c51","subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-11 12:04:37.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-14 06:28:43.000000000","tz":120},"subject":"If default share type exists, use it for share creation","message":"If default share type exists, use it for share creation\n\nThis patch adds \"create_share\" method under \"BaseSharesAdminTest\"\nclass. This wrapper allows us to check whether default share type\nexists, if exists use it for share creation if not create a new\none.\n\nThis patch removes duplicated code by changing the inheritances.\n\nChange-Id: Ie88f8f6cbbe444968026f163c99649eab623c457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c0c93ac730fd4b25b7bfe18acb2e4718716f136b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c0c93ac730fd4b25b7bfe18acb2e4718716f136b"}]},"branch":"refs/heads/master"},"08fdd9beef979b88c73c17b99e3eb2e3da8cae76":{"kind":"REWORK","_number":4,"created":"2021-02-14 06:32:23.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/775148/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/775148/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/4"}}},"commit":{"parents":[{"commit":"d2138765888fe16e68f713ed99121f41e01a3c51","subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-11 12:04:37.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-14 06:32:05.000000000","tz":120},"subject":"If default share type exists, use it for share creation","message":"If default share type exists, use it for share creation\n\nThis patch adds \"create_share\" method under \"BaseSharesAdminTest\"\nclass. This wrapper allows us to check whether default share type\nexists, if exists use it for share creation if not create a new\none.\n\nThis patch removes duplicated code by changing the inheritances.\n\nChange-Id: Ie88f8f6cbbe444968026f163c99649eab623c457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/08fdd9beef979b88c73c17b99e3eb2e3da8cae76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/08fdd9beef979b88c73c17b99e3eb2e3da8cae76"}]},"branch":"refs/heads/master"},"1305a981f56a70d8002f3df4ca82e1357c7c71bc":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-05-23 08:50:49.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/775148/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/775148/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/5"}}},"commit":{"parents":[{"commit":"0031ba11387091fcc830047cd265d7ed9f52c129","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0031ba11387091fcc830047cd265d7ed9f52c129"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-11 12:04:37.000000000","tz":120},"committer":{"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","date":"2021-05-23 08:50:49.000000000","tz":0},"subject":"If default share type exists, use it for share creation","message":"If default share type exists, use it for share creation\n\nThis patch adds \"create_share\" method under \"BaseSharesAdminTest\"\nclass. This wrapper allows us to check whether default share type\nexists, if exists use it for share creation if not create a new\none.\n\nThis patch removes duplicated code by changing the inheritances.\n\nChange-Id: Ie88f8f6cbbe444968026f163c99649eab623c457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1305a981f56a70d8002f3df4ca82e1357c7c71bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1305a981f56a70d8002f3df4ca82e1357c7c71bc"}]},"branch":"refs/heads/master","description":"Rebase"},"9ed758e2535f6abe70a9f40067bfc81fa1cfacdb":{"kind":"REWORK","_number":6,"created":"2021-05-24 08:53:28.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/775148/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/775148/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/6"}}},"commit":{"parents":[{"commit":"0031ba11387091fcc830047cd265d7ed9f52c129","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0031ba11387091fcc830047cd265d7ed9f52c129"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-11 12:04:37.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-05-24 08:53:06.000000000","tz":180},"subject":"If default share type exists, use it for share creation","message":"If default share type exists, use it for share creation\n\nThis patch adds \"create_share\" method under \"BaseSharesAdminTest\"\nclass. This wrapper allows us to check whether default share type\nexists, if exists use it for share creation if not create a new\none.\n\nThis patch removes duplicated code by changing the inheritances.\n\nChange-Id: Ie88f8f6cbbe444968026f163c99649eab623c457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9ed758e2535f6abe70a9f40067bfc81fa1cfacdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9ed758e2535f6abe70a9f40067bfc81fa1cfacdb"}]},"branch":"refs/heads/master"},"cd9cf057073d40d7dbb7e12d9b32f5a9c00d2df4":{"kind":"REWORK","_number":7,"created":"2021-05-24 11:46:55.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/775148/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/775148/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/7"}}},"commit":{"parents":[{"commit":"0031ba11387091fcc830047cd265d7ed9f52c129","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0031ba11387091fcc830047cd265d7ed9f52c129"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-11 12:04:37.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-05-24 11:46:32.000000000","tz":180},"subject":"If default share type exists, use it for share creation","message":"If default share type exists, use it for share creation\n\nThis patch adds \"create_share\" method under \"BaseSharesAdminTest\"\nclass. This wrapper allows us to check whether default share type\nexists, if exists use it for share creation if not create a new\none.\n\nThis patch removes duplicated code by changing the inheritances.\n\nChange-Id: Ie88f8f6cbbe444968026f163c99649eab623c457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cd9cf057073d40d7dbb7e12d9b32f5a9c00d2df4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cd9cf057073d40d7dbb7e12d9b32f5a9c00d2df4"}]},"branch":"refs/heads/master"},"c7d2733d5a24035e811a669a195c7b40789a2332":{"kind":"REWORK","_number":8,"created":"2021-05-24 13:39:42.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/775148/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/775148/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/8"}}},"commit":{"parents":[{"commit":"0031ba11387091fcc830047cd265d7ed9f52c129","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0031ba11387091fcc830047cd265d7ed9f52c129"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-11 12:04:37.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-05-24 13:39:22.000000000","tz":180},"subject":"If default share type exists, use it for share creation","message":"If default share type exists, use it for share creation\n\nThis patch adds \"create_share\" method under \"BaseSharesAdminTest\"\nclass. This wrapper allows us to check whether default share type\nexists, if exists use it for share creation if not create a new\none.\n\nThis patch removes duplicated code by changing the inheritances.\n\nChange-Id: Ie88f8f6cbbe444968026f163c99649eab623c457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c7d2733d5a24035e811a669a195c7b40789a2332"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c7d2733d5a24035e811a669a195c7b40789a2332"}]},"branch":"refs/heads/master"},"f6c65d6ed751bc79c2fcaefd8569f607b80b31c1":{"kind":"REWORK","_number":9,"created":"2021-05-25 07:29:41.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/775148/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/775148/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/9"}}},"commit":{"parents":[{"commit":"0031ba11387091fcc830047cd265d7ed9f52c129","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0031ba11387091fcc830047cd265d7ed9f52c129"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-11 12:04:37.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-05-25 07:29:18.000000000","tz":180},"subject":"If default share type exists, use it for share creation","message":"If default share type exists, use it for share creation\n\nThis patch adds \"create_share\" method under \"BaseSharesAdminTest\"\nclass. This wrapper allows us to check whether default share type\nexists, if exists use it for share creation if not create a new\none.\n\nThis patch removes duplicated code by changing the inheritances.\n\nChange-Id: Ie88f8f6cbbe444968026f163c99649eab623c457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f6c65d6ed751bc79c2fcaefd8569f607b80b31c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f6c65d6ed751bc79c2fcaefd8569f607b80b31c1"}]},"branch":"refs/heads/master"},"adc40e93b44d5a248c253f79c8f7c2418404cedf":{"kind":"REWORK","_number":10,"created":"2021-05-25 10:32:52.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/775148/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/775148/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/10"}}},"commit":{"parents":[{"commit":"0031ba11387091fcc830047cd265d7ed9f52c129","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0031ba11387091fcc830047cd265d7ed9f52c129"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-11 12:04:37.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-05-25 10:32:33.000000000","tz":180},"subject":"If default share type exists, use it for share creation","message":"If default share type exists, use it for share creation\n\nThis patch adds \"create_share\" method under \"BaseSharesAdminTest\"\nclass. This wrapper allows us to check whether default share type\nexists, if exists use it for share creation if not create a new\none.\n\nThis patch removes duplicated code by changing the inheritances.\n\nChange-Id: Ie88f8f6cbbe444968026f163c99649eab623c457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/adc40e93b44d5a248c253f79c8f7c2418404cedf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/adc40e93b44d5a248c253f79c8f7c2418404cedf"}]},"branch":"refs/heads/master"},"1b7e279d47a0f204a377c0206cf327ede85e0c9f":{"kind":"REWORK","_number":11,"created":"2021-05-25 12:14:28.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/775148/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/775148/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/11"}}},"commit":{"parents":[{"commit":"0031ba11387091fcc830047cd265d7ed9f52c129","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0031ba11387091fcc830047cd265d7ed9f52c129"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-11 12:04:37.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-05-25 12:14:07.000000000","tz":180},"subject":"If default share type exists, use it for share creation","message":"If default share type exists, use it for share creation\n\nThis patch adds \"create_share\" method under \"BaseSharesAdminTest\"\nclass. This wrapper allows us to check whether default share type\nexists, if exists use it for share creation if not create a new\none.\n\nThis patch removes duplicated code by changing the inheritances.\n\nChange-Id: Ie88f8f6cbbe444968026f163c99649eab623c457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1b7e279d47a0f204a377c0206cf327ede85e0c9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1b7e279d47a0f204a377c0206cf327ede85e0c9f"}]},"branch":"refs/heads/master"},"c7d81f7a678198e37f95723dfb033065aae3434c":{"kind":"REWORK","_number":12,"created":"2021-05-25 15:06:29.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/775148/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/775148/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/12"}}},"commit":{"parents":[{"commit":"0031ba11387091fcc830047cd265d7ed9f52c129","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0031ba11387091fcc830047cd265d7ed9f52c129"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-11 12:04:37.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-05-25 15:06:06.000000000","tz":180},"subject":"If default share type exists, use it for share creation","message":"If default share type exists, use it for share creation\n\nThis patch adds \"create_share\" method under \"BaseSharesAdminTest\"\nclass. This wrapper allows us to check whether default share type\nexists, if exists use it for share creation if not create a new\none.\n\nThis patch removes duplicated code by changing the inheritances.\n\nChange-Id: Ie88f8f6cbbe444968026f163c99649eab623c457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c7d81f7a678198e37f95723dfb033065aae3434c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c7d81f7a678198e37f95723dfb033065aae3434c"}]},"branch":"refs/heads/master"},"db4f8b02b737c73c2aab38854098b686309b40d2":{"kind":"REWORK","_number":13,"created":"2021-05-26 05:18:41.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/775148/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/775148/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/13"}}},"commit":{"parents":[{"commit":"0031ba11387091fcc830047cd265d7ed9f52c129","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0031ba11387091fcc830047cd265d7ed9f52c129"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-11 12:04:37.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-05-26 05:18:22.000000000","tz":180},"subject":"If default share type exists, use it for share creation","message":"If default share type exists, use it for share creation\n\nThis patch adds \"create_share\" method under \"BaseSharesAdminTest\"\nclass. This wrapper allows us to check whether default share type\nexists, if exists use it for share creation if not create a new\none.\n\nThis patch removes duplicated code by changing the inheritances.\n\nChange-Id: Ie88f8f6cbbe444968026f163c99649eab623c457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/db4f8b02b737c73c2aab38854098b686309b40d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/db4f8b02b737c73c2aab38854098b686309b40d2"}]},"branch":"refs/heads/master"},"8eb49758e780799638b7d1a7bc1ef462608ccb0d":{"kind":"REWORK","_number":14,"created":"2021-05-26 07:28:16.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/775148/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/775148/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/14"}}},"commit":{"parents":[{"commit":"0031ba11387091fcc830047cd265d7ed9f52c129","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0031ba11387091fcc830047cd265d7ed9f52c129"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-11 12:04:37.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-05-26 07:27:52.000000000","tz":180},"subject":"If default share type exists, use it for share creation","message":"If default share type exists, use it for share creation\n\nThis patch adds \"create_share\" method under \"BaseSharesAdminTest\"\nclass. This wrapper allows us to check whether default share type\nexists, if exists use it for share creation if not create a new\none.\n\nThis patch removes duplicated code by changing the inheritances.\n\nChange-Id: Ie88f8f6cbbe444968026f163c99649eab623c457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8eb49758e780799638b7d1a7bc1ef462608ccb0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8eb49758e780799638b7d1a7bc1ef462608ccb0d"}]},"branch":"refs/heads/master"},"7ad4da25a83404c60a2a01319075af4196b4dc90":{"kind":"REWORK","_number":15,"created":"2021-05-26 13:29:26.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/775148/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/775148/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/775148/15"}}},"commit":{"parents":[{"commit":"0031ba11387091fcc830047cd265d7ed9f52c129","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0031ba11387091fcc830047cd265d7ed9f52c129"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-02-11 12:04:37.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-05-26 13:28:26.000000000","tz":180},"subject":"If default share type exists, use it for share creation","message":"If default share type exists, use it for share creation\n\nThis patch adds \"create_share\" method under \"BaseSharesAdminTest\"\nclass. This wrapper allows us to check whether default share type\nexists, if exists use it for share creation if not create a new\none.\n\nThis patch removes duplicated code by changing the inheritances.\n\nChange-Id: Ie88f8f6cbbe444968026f163c99649eab623c457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7ad4da25a83404c60a2a01319075af4196b4dc90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7ad4da25a83404c60a2a01319075af4196b4dc90"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
