)]}'
{"id":"openstack%2Fmanila-tempest-plugin~775213","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I37ab7fd16ca3125870a1201fb3b2e86c61564f47","project":"openstack/manila-tempest-plugin","branch":"master","topic":"tox-update","hashtags":[],"change_id":"I37ab7fd16ca3125870a1201fb3b2e86c61564f47","subject":"tox miniversion update to 3.1.1","status":"MERGED","created":"2021-02-11 16:53:22.000000000","updated":"2021-02-12 21:31:52.000000000","submitted":"2021-02-12 21:31:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775213-tox-update","meta_rev_id":"33dca5a53eaac3a8882c27d5fea3683e06a45178","_number":775213,"virtual_id_number":775213,"owner":{"_account_id":32531,"name":"Dina Saparbaeva","email":"dinas27@outlook.com","username":"disap"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-12 21:31:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2021-02-12 21:21:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2021-02-12 21:21:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-11 17:03:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-11 18:14:06.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-11 18:33:40.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-02-12 21:21:15.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"f5d7a9f3b216751831e7842567b5876af960d0f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32531,"name":"Dina Saparbaeva","email":"dinas27@outlook.com","username":"disap"},"date":"2021-02-11 16:53:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7711a63a4e569b45c1253f7230f5606445ae4fcb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32531,"name":"Dina Saparbaeva","email":"dinas27@outlook.com","username":"disap"},"date":"2021-02-11 16:58:24.000000000","message":"Topic set to tox-update","accounts_in_message":[],"_revision_number":1},{"id":"f665de1b5a576c62095878ff5cc51affd661e047","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 17:03:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4da01fcd0eb4579a95ab347c267db6c : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"7f2c21cbade907ec6ffce0b9f3b9e81f29736844","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-11 18:14:06.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/13/775213/1/check/DellEMC_VNX_Manila/aa0c063/DellEMC_VNX_Manila/173 : SUCCESS in 1h 10m 30s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"131e2c049af70c0701d4064ad724c7eb409fd4f1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-11 18:33:40.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/13/775213/1/check/DellEMC_Unity_Manila_dhss_true/30d1858/DellEMC_Unity_Manila_dhss_true/296 : FAILURE in 7m 04s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/13/775213/1/check/DellEMC_Unity_Manila_dhss_false/e9dd99c/DellEMC_Unity_Manila_dhss_false/280 : SUCCESS in 55m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"5875dea12664de858ece6108ae77313cf1fe06f5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-02-12 21:21:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks, LGTM","accounts_in_message":[],"_revision_number":1},{"id":"78509db7111f5cefcab9efea24563220e299432a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 21:21:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a89963d50aa4fff940b454aae43d0e27a5601369","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 21:31:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8862ca8b93c4645ab7ba0c852862a68 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"33dca5a53eaac3a8882c27d5fea3683e06a45178","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 21:31:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"02bd40a31bedf415acd6dba08b3be32ce51bd0fb","revisions":{"02bd40a31bedf415acd6dba08b3be32ce51bd0fb":{"kind":"REWORK","_number":1,"created":"2021-02-11 16:53:22.000000000","uploader":{"_account_id":32531,"name":"Dina Saparbaeva","email":"dinas27@outlook.com","username":"disap"},"ref":"refs/changes/13/775213/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/13/775213/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/775213/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/775213/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/775213/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/775213/1"}}},"commit":{"parents":[{"commit":"0720fc354922cffbcb91663d85fe0cc0393ab0ba","subject":"Merge \"Enable running multi-backends tests with different share protocols\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0720fc354922cffbcb91663d85fe0cc0393ab0ba"}]}],"author":{"name":"Dina Saparbaeva","email":"dinas27@outlook.com","date":"2021-02-11 16:51:37.000000000","tz":360},"committer":{"name":"Dina Saparbaeva","email":"dinas27@outlook.com","date":"2021-02-11 16:53:14.000000000","tz":360},"subject":"tox miniversion update to 3.1.1","message":"tox miniversion update to 3.1.1\n\nIn 3.1.1, tox had fixed:\n\nmishandling the basepython directive [1]\nignoring PYTHONDONTWRITEBYTECODE [2]\nMore details in [3]\n\n[1] tox-dev/tox#477\n[2] tox-dev/tox@336f4f6\n[3] http://lists.openstack.org/pipermail/openstack-dev/2018-July/thread.html#132075\n\nChange-Id: I37ab7fd16ca3125870a1201fb3b2e86c61564f47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/02bd40a31bedf415acd6dba08b3be32ce51bd0fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/02bd40a31bedf415acd6dba08b3be32ce51bd0fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
