)]}'
{"id":"openstack%2Fmanila-tempest-plugin~776265","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ie554d72f5ba2991df0c7b0decd7f552d034760e0","project":"openstack/manila-tempest-plugin","branch":"master","topic":"test-cephfs-ci","hashtags":[],"change_id":"Ie554d72f5ba2991df0c7b0decd7f552d034760e0","subject":"DNM - testing cephfs-native with latest service image","status":"ABANDONED","created":"2021-02-17 20:12:04.000000000","updated":"2022-07-05 15:09:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7e252da842b5dc05e3aff54ab3f9fe31cd1eb2f8","_number":776265,"virtual_id_number":776265,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-24 18:07:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-17 21:21:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-17 21:45:25.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"60f30a0398fb8538522464e7f3f16889197f9d30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-02-17 20:12:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f430a8997d9e16db409416cdc146f63aa81e49e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 21:21:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9963fdb392f04aa2b30f45570092d269 : FAILURE in 1h 07m 10s","accounts_in_message":[],"_revision_number":1},{"id":"fc2673cd4df76c1d008d6521b6146122d89551df","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-17 21:45:25.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/65/776265/1/check/DellEMC_Unity_Manila_dhss_true/f5a5f65/DellEMC_Unity_Manila_dhss_true/346 : FAILURE in 1h 03m 39s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/65/776265/1/check/DellEMC_Unity_Manila_dhss_false/6cf892b/DellEMC_Unity_Manila_dhss_false/330 : SUCCESS in 53m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"e6cc9607bf6b64de373506e56443bff4bcdcf154","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-02-24 16:23:11.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ee4301fadec6c5cb476f8531d9a0be9101719781","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-24 17:18:51.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/65/776265/2/check/DellEMC_Unity_Manila_dhss_true/7be73ec/DellEMC_Unity_Manila_dhss_true/385 : FAILURE in 55m 21s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/65/776265/2/check/DellEMC_Unity_Manila_dhss_false/995be30/DellEMC_Unity_Manila_dhss_false/369 : FAILURE in 54m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"096cee56617b583328fb769b5d672474eb2bf2f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 18:07:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/5a3ea1762842440f8d86253e20be7992 : FAILURE in 1h 33m 05s","accounts_in_message":[],"_revision_number":2},{"id":"7e252da842b5dc05e3aff54ab3f9fe31cd1eb2f8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-05 15:09:02.000000000","message":"Abandoned\n\nno longer needed","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"53e9db238bb6ff56354de09f2cf66618635de7bd","revisions":{"a39f119a1458ad4d03c63928a0e675c6cb7432f6":{"kind":"REWORK","_number":1,"created":"2021-02-17 20:12:04.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/65/776265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/65/776265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/776265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/776265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/776265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/776265/1"}}},"commit":{"parents":[{"commit":"8868f18b7c469ccdaf477d2247186be2884b5d17","subject":"Merge \"tox miniversion update to 3.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8868f18b7c469ccdaf477d2247186be2884b5d17"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-02-17 20:11:09.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-02-17 20:11:38.000000000","tz":-480},"subject":"DNM - testing cephfs-native with latest service image","message":"DNM - testing cephfs-native with latest service image\n\nWe capped the service image to use an older\nimage due to a mismatch between clients -\nthe latest service image had an upgraded\ncephfs client that wasn\u0027t working with a\nnautilus based ceph cluster. Now that we\u0027ve\nchanged the ceph cluster infra quite a bit*,\nlets try to see if this issue still exists.\n\n* We updated the ceph package sources from\n  ceph-ci\u0027s shaman to ubuntu\n* We updated the base OS for devstack from\n  bionic beaver to focal fossa - this allows\n  us to test with ceph octopus which is\n  made available by default when we look\n  for ceph packages in the ubuntu ppa.\n\nChange-Id: Ie554d72f5ba2991df0c7b0decd7f552d034760e0\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a39f119a1458ad4d03c63928a0e675c6cb7432f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a39f119a1458ad4d03c63928a0e675c6cb7432f6"}]},"branch":"refs/heads/master"},"53e9db238bb6ff56354de09f2cf66618635de7bd":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-02-24 16:23:11.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/65/776265/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/65/776265/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/776265/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/776265/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/776265/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/776265/2"}}},"commit":{"parents":[{"commit":"8868f18b7c469ccdaf477d2247186be2884b5d17","subject":"Merge \"tox miniversion update to 3.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8868f18b7c469ccdaf477d2247186be2884b5d17"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-02-17 20:11:09.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-02-24 16:23:11.000000000","tz":0},"subject":"DNM - testing cephfs-native with latest service image","message":"DNM - testing cephfs-native with latest service image\n\nWe capped the service image to use an older\nimage due to a mismatch between clients -\nthe latest service image had an upgraded\ncephfs client that wasn\u0027t working with a\nnautilus based ceph cluster. Now that we\u0027ve\nchanged the ceph cluster infra quite a bit*,\nlets try to see if this issue still exists.\n\n* We updated the ceph package sources from\n  ceph-ci\u0027s shaman to ubuntu\n* We updated the base OS for devstack from\n  bionic beaver to focal fossa - this allows\n  us to test with ceph octopus which is\n  made available by default when we look\n  for ceph packages in the ubuntu ppa.\n\nDepends-On: I1edd5a50079f325fa143a7e0d51b3aa3bb5ed45d\nChange-Id: Ie554d72f5ba2991df0c7b0decd7f552d034760e0\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/53e9db238bb6ff56354de09f2cf66618635de7bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/53e9db238bb6ff56354de09f2cf66618635de7bd"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
