)]}'
{"id":"openstack%2Fmanila-tempest-plugin~776869","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I2a647fd5871ef6bb7d1ab45db893a44a560bed72","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1801763","hashtags":[],"change_id":"I2a647fd5871ef6bb7d1ab45db893a44a560bed72","subject":"Refactor public share test cases","status":"MERGED","created":"2021-02-22 06:04:51.000000000","updated":"2021-02-24 23:25:26.000000000","submitted":"2021-02-24 23:25:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"776869-bug/1801763","meta_rev_id":"86379bbe219548b7d7e9e162c77b0155892a7df6","_number":776869,"virtual_id_number":776869,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-24 23:25:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2021-02-24 00:27:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2021-02-23 20:05:46.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-23 21:07:34.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":1,"date":"2021-02-24 21:30:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-22 07:48:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-22 08:21:56.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-02-22 22:45:23.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2021-02-22 22:45:23.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-02-23 01:39:58.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-23 03:11:04.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-23 21:07:34.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2021-02-23 21:46:04.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"}],"messages":[{"id":"8e1b3d609070d2bbd618073f0cda286e660cda06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-02-22 06:04:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3497ba20ea688d9c6fa91b75802e943fbd2aa77f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 07:48:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5a553247233447684548f1ffde0506f : SUCCESS in 4m 48s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e5b74b77ca84428ab20cc1059773c22d : FAILURE in 36m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e8b2ebb1be85437bb0239da14497c0be : FAILURE in 34m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1b89487cadcf42528fdcd8324ecb8641 : SUCCESS in 1h 01m 20s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ddffaebdc26842469843fb984d95a72a : SUCCESS in 1h 22m 35s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/17409e899a4c4868ac04ca5f7148afd8 : FAILURE in 54m 45s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5e2ef92d66a74844af89dbd57cf7be7c : FAILURE in 1h 33m 19s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bc84a90388e64555a21afbdcc8a32c79 : FAILURE in 43m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/709cb7f86f7845fda5125380e2ce677f : FAILURE in 1h 36m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/337c59a029354d6f929d1eb402ab67ad : FAILURE in 22m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"53c2faf18825f3751509cf74045d6afc37cf2911","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-22 08:21:56.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/69/776869/1/check/DellEMC_Unity_Manila_dhss_true/c8c2dd4/DellEMC_Unity_Manila_dhss_true/363 : FAILURE in 2h 16m 47s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/69/776869/1/check/DellEMC_Unity_Manila_dhss_false/6c7e184/DellEMC_Unity_Manila_dhss_false/347 : FAILURE in 49m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"35539193151334a1e5e5d800a9359a87d0b03026","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-02-22 15:53:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"52d9f60b38e4789218d0e68edb74a8b8aa91d5bd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-22 17:47:18.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/69/776869/2/check/DellEMC_Unity_Manila_dhss_true/bb1b1c7/DellEMC_Unity_Manila_dhss_true/367 : FAILURE in 1h 53m 32s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/69/776869/2/check/DellEMC_Unity_Manila_dhss_false/1fa49a7/DellEMC_Unity_Manila_dhss_false/351 : FAILURE in 58m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"0f1633e570b2d2dadbd1848ac4c9d7fa07db635b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 18:58:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b322836396d40da82b621fc7cb2dbe5 : SUCCESS in 3m 45s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/46123aca09684c75a6869b81dcfb2cab : FAILURE in 28m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d9bbca8d07f84bdda27aabcb0b084a00 : FAILURE in 42m 23s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/29ec2d6a6a704969a05d4c1e348cc337 : SUCCESS in 1h 16m 50s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1b631c08ca6c46e3a2c17ad198c02257 : FAILURE in 1h 16m 28s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/64643a2d33fa47949173f3dc42ed3ef8 : FAILURE in 1h 03m 15s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1b67a6370d784d7485abbb10ae0904d0 : FAILURE in 1h 18m 45s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/19ba9c7b02af4e92b93e41e080a2455f : FAILURE in 55m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/df282af0e7ed4e3280eb2278276decb5 : FAILURE in 2h 59m 44s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/fb96408f24014de49501969662eea2c4 : FAILURE in 28m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b9c89bb175ee5abbb2d68af61a6f575ae2d30ea3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-02-22 18:59:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4c6c427bc18c0b8a2f7f48aa986898b6485140e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 20:52:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e66f4955730741c1aebde4ae9976fa15 : SUCCESS in 4m 23s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e906f41ce7c64bb9b0a37763f8297678 : FAILURE in 31m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/dd7df1acebd947aa96de7919f3576145 : FAILURE in 36m 08s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7754b1beab8440f9aa110d6ea268fc33 : SUCCESS in 1h 11m 02s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0787953fadc349fc953d4edf48dc56b7 : FAILURE in 1h 08m 02s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0382f5e716cf43f7b5508519400ee38a : FAILURE in 52m 17s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/97c65cf8759e4f7585412e215bea6696 : FAILURE in 1h 16m 30s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d5555e9227604eb99834b6303690a395 : FAILURE in 46m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5e95e4e119fd4856ba55bcbf9484b712 : FAILURE in 1h 48m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7215f91f03204a80836c1811a569929c : FAILURE in 21m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"700948c2e08c6a89f5549fe7078f204d8547e745","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-22 20:55:48.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/69/776869/3/check/DellEMC_Unity_Manila_dhss_true/05e16d7/DellEMC_Unity_Manila_dhss_true/368 : FAILURE in 1h 55m 31s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/69/776869/3/check/DellEMC_Unity_Manila_dhss_false/e033cd4/DellEMC_Unity_Manila_dhss_false/352 : FAILURE in 55m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"631e7d89660d6bb286bc6d92b2a272ae4443a6d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-02-22 22:35:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7d257a2b14f5b96fe5fb5904c9aefec0d7c0e516","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 00:34:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5abbd5c186504e0881081e85b7c0e90b : SUCCESS in 4m 03s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/338f10c45934483c8cd6b5e14c648dd0 : SUCCESS in 36m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5c77e8655dc843cba8cd875e49802cd2 : SUCCESS in 45m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bd6d901eafd34089b97c38ef394acfe6 : SUCCESS in 54m 42s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6bff6686bc984b3fb432dba76b0d4b4e : SUCCESS in 53m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e7ba491ef3b141e1914648a37a9b7d05 : FAILURE in 51m 24s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9c26d2f07c0b487bb390819ae583a3b5 : SUCCESS in 1h 09m 58s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9c1635257f054465b842816a7ac48bcb : SUCCESS in 32m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a8509d54ba4a4157a23d9cec647a41f1 : FAILURE in 1h 56m 17s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b0b1696601a94064a36476751e4c55a0 : SUCCESS in 23m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7aba558696bcdc9beb88654c721d9672d377f81d","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-23 01:39:58.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/69/776869/4/check/DellEMC_VNX_Manila/6550d85/DellEMC_VNX_Manila/206 : FAILURE in 1h 05m 41s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"baf264a51a3bef9306061630378db7882bde3fb7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-23 03:11:04.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/69/776869/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/e1af386/ : SUCCESS in 2h 34m 31s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/69/776869/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/4f80f5b/ : SUCCESS in 1h 42m 27s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"f5e7adec61b65d2d8c5300a3e80ab35f855c663b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-23 04:40:10.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/69/776869/4/check/DellEMC_Unity_Manila_dhss_true/220b8a6/DellEMC_Unity_Manila_dhss_true/372 : FAILURE in 3h 09m 04s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/69/776869/4/check/DellEMC_Unity_Manila_dhss_false/63f0c88/DellEMC_Unity_Manila_dhss_false/356 : FAILURE in 10m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"7271bc7f9bd6ab659e30123309023d229c649c92","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-02-23 08:06:05.000000000","message":"Patch Set 4: Code-Review+1\n\n(7 comments)\n\nThanks Goutham,\nLGTM, I added a few small comments.","accounts_in_message":[],"_revision_number":4},{"id":"a3ccf2456fc84fdf066770dc7a20fae5f5416d8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-02-23 15:13:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bf9e112995ca653a300346200c4fb96408c37062","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-02-23 15:14:42.000000000","message":"Patch Set 5:\n\n(5 comments)\n\nThanks for the comments, Liron - addressed these in a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"d62b4fe1c3c77dbe58a27f8e4a82e550ecc429ce","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-23 16:16:09.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/69/776869/5/check/DellEMC_Unity_Manila_dhss_true/a6d473e/DellEMC_Unity_Manila_dhss_true/378 : FAILURE in 1h 01m 08s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/69/776869/5/check/DellEMC_Unity_Manila_dhss_false/69abb12/DellEMC_Unity_Manila_dhss_false/362 : SUCCESS in 59m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"b9ada0fae0d143692e0b832de798d367462e5e10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 18:11:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e28879a5abba4a8db354745af12bd7e0 : SUCCESS in 3m 46s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/abd2d86b9bf342d3bf8ce1d1fa87f898 : SUCCESS in 37m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/023838f8c5e44e2dbb0e58c5478e07d3 : SUCCESS in 45m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f9137f00baad4bcf94808f7b6ecf00f3 : SUCCESS in 1h 17m 38s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2b547c651eda48cd81c2c4fa2a1244a6 : FAILURE in 1h 21m 49s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/48f176af3dcf4078a0a48e4af02142ec : SUCCESS in 1h 18m 52s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a7ec39e0fe4b4ccdb8b2397ef384d3fa : SUCCESS in 1h 11m 17s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d70da74b4dc54cc08bdca7de75c573d8 : SUCCESS in 43m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/aebf07f05281453190911179b0f4aa6a : FAILURE in 2h 50m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c2dff6ea110548aa8074e5e2056ce054 : FAILURE in 15m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0810568c48cdd6b89e0ece73da8796eeeaaba921","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-23 19:25:14.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/69/776869/5/check/DellEMC_VNX_Manila/ea6dea4/DellEMC_VNX_Manila/211 : SUCCESS in 1h 13m 43s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"437c3611659e379c1b8ae102617551979ff1eca6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-23 19:39:10.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/69/776869/5/check/DellEMC_Unity_Manila_dhss_true/d7b5bef/DellEMC_Unity_Manila_dhss_true/380 : FAILURE in 57m 15s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/69/776869/5/check/DellEMC_Unity_Manila_dhss_false/9eaba47/DellEMC_Unity_Manila_dhss_false/364 : SUCCESS in 53m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"0c829c6b60eec6c231c4b5ec0cd2ee393aa7bb07","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-02-23 20:05:46.000000000","message":"Patch Set 5: Code-Review+1\n\nLGTM. Thanks Goutham!","accounts_in_message":[],"_revision_number":5},{"id":"3e4efe9893f92af80fca0bc81514b3b1262978f3","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-23 20:32:16.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/69/776869/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/8ad4dc3/ : SUCCESS in 2h 20m 45s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/69/776869/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/f74f92b/ : FAILURE in 17m 16s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"4a87523ff8a9bfe021121e8a3d5ae7e58e975140","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-02-23 21:07:34.000000000","message":"Patch Set 5: Code-Review+2\n\nLooks good Goutham, thanks!","accounts_in_message":[],"_revision_number":5},{"id":"9dc0d4b9bd84518a261ea2083a5da6f2362232b1","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-02-24 00:27:26.000000000","message":"Patch Set 5: Code-Review+2\n\nLGTM, thanks Goutham!","accounts_in_message":[],"_revision_number":5},{"id":"651cd7725437ff5f4f0eb72500d1033a54b87e3c","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-02-24 21:30:07.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"67bbb65539dc0d53c0ae25a20be986ac4ae6a579","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 21:30:22.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7ab01f4c36951b042a631a73732fa97ff3a45d40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 23:25:13.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51f14b7829b04f7d96dfcb29d1858af7 : SUCCESS in 3m 52s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c5692e996e6147efb9af4e5d8a7553eb : SUCCESS in 31m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c495d9ab6a7141a2891a660e4c3a54cf : SUCCESS in 36m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bb84900c8f754789b076e8457d87f920 : SUCCESS in 1h 11m 42s","accounts_in_message":[],"_revision_number":5},{"id":"86379bbe219548b7d7e9e162c77b0155892a7df6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 23:25:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a3ed445fff5a0d90e64817aace08229069c37660","revisions":{"0917334235a4bdd7e14b0a15c4ccae37d7c8e701":{"kind":"REWORK","_number":1,"created":"2021-02-22 06:04:51.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/69/776869/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/69/776869/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/776869/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/776869/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/776869/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/776869/1"}}},"commit":{"parents":[{"commit":"8868f18b7c469ccdaf477d2247186be2884b5d17","subject":"Merge \"tox miniversion update to 3.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8868f18b7c469ccdaf477d2247186be2884b5d17"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-02-22 05:56:39.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-02-22 06:04:31.000000000","tz":-480},"subject":"Refactor public share test cases","message":"Refactor public share test cases\n\nThe RBAC policy defaults to create\na public share or to update an existing\nshare to public have changed since the\nStein release. [1]\n\nLet\u0027s stop assuming that regular users\ncan perform these actions, so that manila\ncan safely remove the policy rule deprecation\nsupport for these APIs.\n\n[1] https://review.opendev.org/634864\n\nChange-Id: I2a647fd5871ef6bb7d1ab45db893a44a560bed72\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0917334235a4bdd7e14b0a15c4ccae37d7c8e701"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0917334235a4bdd7e14b0a15c4ccae37d7c8e701"}]},"branch":"refs/heads/master"},"1db8e8557f86ad61328bf11031b312f6f1879805":{"kind":"REWORK","_number":2,"created":"2021-02-22 15:53:20.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/69/776869/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/69/776869/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/776869/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/776869/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/776869/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/776869/2"}}},"commit":{"parents":[{"commit":"8868f18b7c469ccdaf477d2247186be2884b5d17","subject":"Merge \"tox miniversion update to 3.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8868f18b7c469ccdaf477d2247186be2884b5d17"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-02-22 05:56:39.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-02-22 15:53:12.000000000","tz":-480},"subject":"Refactor public share test cases","message":"Refactor public share test cases\n\nThe RBAC policy defaults to create\na public share or to update an existing\nshare to public have changed since the\nStein release. [1]\n\nLet\u0027s stop assuming that regular users\ncan perform these actions, so that manila\ncan safely remove the policy rule deprecation\nsupport for these APIs.\n\n[1] https://review.opendev.org/634864\n\nChange-Id: I2a647fd5871ef6bb7d1ab45db893a44a560bed72\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1db8e8557f86ad61328bf11031b312f6f1879805"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1db8e8557f86ad61328bf11031b312f6f1879805"}]},"branch":"refs/heads/master"},"c1ed5e2a821dbea2e6cba645c07b68a5c2a2d010":{"kind":"REWORK","_number":3,"created":"2021-02-22 18:59:59.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/69/776869/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/69/776869/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/776869/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/776869/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/776869/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/776869/3"}}},"commit":{"parents":[{"commit":"8868f18b7c469ccdaf477d2247186be2884b5d17","subject":"Merge \"tox miniversion update to 3.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8868f18b7c469ccdaf477d2247186be2884b5d17"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-02-22 05:56:39.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-02-22 18:59:48.000000000","tz":-480},"subject":"Refactor public share test cases","message":"Refactor public share test cases\n\nThe RBAC policy defaults to create\na public share or to update an existing\nshare to public have changed since the\nStein release. [1]\n\nLet\u0027s stop assuming that regular users\ncan perform these actions, so that manila\ncan safely remove the policy rule deprecation\nsupport for these APIs.\n\n[1] https://review.opendev.org/634864\n\nChange-Id: I2a647fd5871ef6bb7d1ab45db893a44a560bed72\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c1ed5e2a821dbea2e6cba645c07b68a5c2a2d010"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c1ed5e2a821dbea2e6cba645c07b68a5c2a2d010"}]},"branch":"refs/heads/master"},"ff22629dd56a85c504ecf894c31f1731e8b5ce5e":{"kind":"REWORK","_number":4,"created":"2021-02-22 22:35:43.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/69/776869/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/69/776869/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/776869/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/776869/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/776869/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/776869/4"}}},"commit":{"parents":[{"commit":"8868f18b7c469ccdaf477d2247186be2884b5d17","subject":"Merge \"tox miniversion update to 3.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8868f18b7c469ccdaf477d2247186be2884b5d17"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-02-22 05:56:39.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-02-22 22:29:57.000000000","tz":-480},"subject":"Refactor public share test cases","message":"Refactor public share test cases\n\nThe RBAC policy defaults to create\na public share or to update an existing\nshare to public have changed since the\nStein release. [1]\n\nLet\u0027s stop assuming that regular users\ncan perform these actions, so that manila\ncan safely remove the policy rule deprecation\nsupport for these APIs.\n\n[1] https://review.opendev.org/634864\n\nChange-Id: I2a647fd5871ef6bb7d1ab45db893a44a560bed72\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ff22629dd56a85c504ecf894c31f1731e8b5ce5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ff22629dd56a85c504ecf894c31f1731e8b5ce5e"}]},"branch":"refs/heads/master"},"a3ed445fff5a0d90e64817aace08229069c37660":{"kind":"REWORK","_number":5,"created":"2021-02-23 15:13:45.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/69/776869/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/69/776869/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/776869/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/776869/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/776869/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/776869/5"}}},"commit":{"parents":[{"commit":"8868f18b7c469ccdaf477d2247186be2884b5d17","subject":"Merge \"tox miniversion update to 3.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8868f18b7c469ccdaf477d2247186be2884b5d17"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-02-22 05:56:39.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-02-23 15:13:18.000000000","tz":-480},"subject":"Refactor public share test cases","message":"Refactor public share test cases\n\nThe RBAC policy defaults to create\na public share or to update an existing\nshare to public have changed since the\nStein release. [1]\n\nLet\u0027s stop assuming that regular users\ncan perform these actions, so that manila\ncan safely remove the policy rule deprecation\nsupport for these APIs.\n\n[1] https://review.opendev.org/634864\nChange-Id: I2a647fd5871ef6bb7d1ab45db893a44a560bed72\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a3ed445fff5a0d90e64817aace08229069c37660"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a3ed445fff5a0d90e64817aace08229069c37660"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
