)]}'
{"id":"openstack%2Fmanila-tempest-plugin~778188","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ifb984a3cee219fa97a9768495f80ff0c8349830b","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bp/create-share-from-snapshot-cephfs","hashtags":[],"change_id":"Ifb984a3cee219fa97a9768495f80ff0c8349830b","subject":"Add create share from snapshot in CephFS","status":"MERGED","created":"2021-03-02 13:10:20.000000000","updated":"2021-03-25 18:05:08.000000000","submitted":"2021-03-25 17:41:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778188-bp/create-share-from-snapshot-cephfs","meta_rev_id":"7d951d370e4f71fe89f887219c1574f12ba8877f","_number":778188,"virtual_id_number":778188,"owner":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-25 17:40:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2021-03-25 05:34:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2021-03-25 10:33:41.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":2,"date":"2021-03-25 10:03:44.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":1,"date":"2021-03-25 16:30:02.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-02 14:29:22.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-03-02 15:20:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-02 17:53:56.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-02 17:57:08.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-21 21:27:45.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"CC"},{"updated":"2021-03-25 00:56:34.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-03-25 05:34:35.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-03-25 10:33:41.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"0cf44b6c17f0b245af3fe96a767c54b421c286d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-03-02 13:10:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2d66af69b966eef52117b8bb6d16b3d54d0b51d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-03-02 13:12:30.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ee9367eada99210c42799795aa9af4c3eac98552","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-02 14:29:22.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/88/778188/2/check/DellEMC_Unity_Manila_dhss_true/51b1e71/DellEMC_Unity_Manila_dhss_true/412 : FAILURE in 1h 15m 43s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/88/778188/2/check/DellEMC_Unity_Manila_dhss_false/535e396/DellEMC_Unity_Manila_dhss_false/396 : SUCCESS in 55m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"1a593c32facf36d5882cd9a21c4531ea822efb1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 15:20:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f646b5856f094f2f8fee377a01479f0e : SUCCESS in 3m 47s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8deb1d98e6d643918464dd071e2426b9 : SUCCESS in 31m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2a63654c16124b39b45f230f3bbbfad7 : SUCCESS in 31m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0583c108eddb477db294339bf6a28809 : SUCCESS in 1h 15m 10s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a60bc066196f4663a2193db39c6df9b2 : SUCCESS in 1h 19m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ac70f97441c043baac85b0e0aac153cd : FAILURE in 1h 00m 12s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e37d1db079584a04814b10fd33bbb3b8 : FAILURE in 1h 30m 30s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/41da904b19a64cc992cc83852347fd8a : SUCCESS in 48m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5326b84ac7af4553b78f692e8740f9f7 : FAILURE in 1h 52m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7dcfeaf6700d4462a7e8798dae2beda9 : FAILURE in 15m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a08375bc6625e08dd899a2a7e7a3d27d6ceceee6","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-02 17:53:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/88/778188/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/b77fce1/ : SUCCESS in 2h 32m 37s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/88/778188/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/85ac1cb/ : SUCCESS in 1h 44m 42s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"7793f8cfe21053fd4a5916b3668226c5b741df57","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-02 17:57:08.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/88/778188/2/check/DellEMC_VNX_Manila/d82cfef/DellEMC_VNX_Manila/231 : SUCCESS in 1h 28m 31s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"a1aaa5eee440bd4d68bb51c593fd7ff8e365e5b3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-02 18:33:02.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/88/778188/2/check/DellEMC_Unity_Manila_dhss_true/735f312/DellEMC_Unity_Manila_dhss_true/415 : FAILURE in 1h 26m 36s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/88/778188/2/check/DellEMC_Unity_Manila_dhss_false/884b6d4/DellEMC_Unity_Manila_dhss_false/399 : SUCCESS in 54m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"50bb71a65219e449f8c61201afc19f5a0cf60fb0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-03-09 12:03:24.000000000","message":"Abandoned\n\nWill submit a patch set rados client instead of the Ceph CLI","accounts_in_message":[],"_revision_number":2},{"id":"25c7c955f58477e083241b275f4ad54f91f7adec","tag":"autogenerated:gerrit:restore","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-18 21:08:14.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"ad4ad92b568d65c2536471bd5bbeea5d1e72d2ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-18 21:09:21.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"9557028b19488b25f0edfff973a3dd4dae8edcf6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-19 00:11:00.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/88/778188/3/check/DellEMC_Unity_Manila_dhss_true/57a5b13/DellEMC_Unity_Manila_dhss_true/489 : FAILURE in 55m 47s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/88/778188/3/check/DellEMC_Unity_Manila_dhss_false/30517f0/DellEMC_Unity_Manila_dhss_false/482 : FAILURE in 56m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"02777c88b6b93dd5bda48a80b37345cfd446135e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 01:36:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4150aedd46bf48d19f132ef6d038b08b : SUCCESS in 3m 52s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5085afe3bbae4ef4a2eea68c1351eec2 : SUCCESS in 28m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f83b62717c304200ae1bf87d26d94643 : SUCCESS in 37m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/93911d906b664aa485fcc5d27e4c0882 : SUCCESS in 1h 02m 45s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d3e03a463abf4b54ba04d7a39920d8b8 : FAILURE in 59m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/efe9f6eeaee74fc4b9e2693e94d88e84 : FAILURE in 1h 02m 16s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0beec113689144cb91b3d6d0a7e468d6 : FAILURE in 1h 17m 04s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fa25241ab64e4091be03880de759664c : SUCCESS in 49m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4d3ba60ff7c641a9b5fceded4f7142c9 : FAILURE in 1h 59m 37s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0b488cfc6e044250963c41b926da486c : SUCCESS in 24m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"89821b65c7e43c300f44c93b75486e6c31eb2b56","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-19 02:39:30.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/88/778188/3/check/DellEMC_Unity_Manila_dhss_true/acf572d/DellEMC_Unity_Manila_dhss_true/490 : FAILURE in 1h 02m 53s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/88/778188/3/check/DellEMC_Unity_Manila_dhss_false/49b5a64/DellEMC_Unity_Manila_dhss_false/483 : SUCCESS in 56m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"dc53a1cc5abd8c7f8026074ec74fac7d8bd180c3","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-19 02:53:57.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/88/778188/3/check/DellEMC_VNX_Manila/711cd5e/DellEMC_VNX_Manila/322 : SUCCESS in 1h 17m 20s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"48da56811e78d13c4f5afd9dc851263ada9c0aa8","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-19 04:08:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/88/778188/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/e4cd127/ : SUCCESS in 2h 32m 17s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/88/778188/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/036f971/ : SUCCESS in 1h 50m 34s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"4a1acfcf52ff0fca673a2cfbc3f3a2c42a66d9dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-19 12:41:44.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"6027100e61ac62857a004cb0ff73a86d558c9865","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 14:26:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82a8cb1067404737aff1d06a3110f339 : SUCCESS in 3m 44s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f85c27358f5a4380b74b54bc6701a4a5 : SUCCESS in 30m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d9a961ba7593414f95bf8a6af0c11634 : SUCCESS in 39m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9d1476d1e19d429fb5edd39fc7428d44 : SUCCESS in 1h 17m 36s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/dc0db4ae8bd7424ba1609c9208bebfb0 : SUCCESS in 1h 26m 14s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/39e69a7e1d794da190415ddf4d0493fb : FAILURE in 33m 55s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8335eb4d479e43b481799578def6b2a1 : FAILURE in 48m 46s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fecf8fada1834d68b6b9e66bff5d2f30 : SUCCESS in 47m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e452544b84384934939928414a37f091 : FAILURE in 1h 39m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4f9bd75c8031458486dd39221509c1e2 : FAILURE in 32m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c202f687fcf0448f7ad6c2770d4dd0c7d53c1789","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-19 15:31:40.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/88/778188/4/check/DellEMC_Unity_Manila_dhss_true/70208b8/DellEMC_Unity_Manila_dhss_true/493 : FAILURE in 58m 50s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/88/778188/4/check/DellEMC_Unity_Manila_dhss_false/3275a9a/DellEMC_Unity_Manila_dhss_false/486 : SUCCESS in 55m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"b9be6416f38ef00c9c4a916c0a8dd06bd0bf384c","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-19 16:01:56.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/88/778188/4/check/DellEMC_VNX_Manila/57bf3e1/DellEMC_VNX_Manila/323 : FAILURE in 1h 34m 53s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"a389bdb219c80a4c04d836b5697de732916bd3ac","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-19 17:04:27.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/88/778188/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/c42524d/ : SUCCESS in 2h 35m 31s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/88/778188/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/921e4bc/ : SUCCESS in 1h 49m 08s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"3c01232f120cd316e5360f1e5f5bcbedd29164d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-21 21:26:41.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"8ef7e5607d4421380062b7b3be20665a191e5703","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-21 21:27:45.000000000","message":"Patch Set 5:\n\nChanged the Depends-on temporarily to see how https://review.opendev.org/c/openstack/manila/+/781968 fares.  We will change it back :D","accounts_in_message":[],"_revision_number":5},{"id":"5def50a64aa08e16ad505fa378b6f451969a46ed","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-21 22:28:06.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/88/778188/5/check/DellEMC_Unity_Manila_dhss_true/6988558/DellEMC_Unity_Manila_dhss_true/511 : FAILURE in 1h 01m 06s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/88/778188/5/check/DellEMC_Unity_Manila_dhss_false/de25e75/DellEMC_Unity_Manila_dhss_false/504 : SUCCESS in 57m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"209d563c11d6ed92aa3f7cc75c2e6ca648bce5d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-21 23:18:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3def9ff1bacb4fdbb9135766bd19b1f5 : SUCCESS in 4m 24s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d03495e00bc4448a92372eb27ff0695c : SUCCESS in 40m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e2e2b5dcaabc49c7857e11d6f455c9de : SUCCESS in 42m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f9ae769ea2a045d288c5e669b268b330 : SUCCESS in 1h 45m 56s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b12745d95859408fa855f44ac064d808 : SUCCESS in 1h 04m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/4f3b771975184f37be692e5c817ff292 : FAILURE in 1h 01m 33s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1d5b49db8c7d4958bb0a2da3bf4a1614 : FAILURE in 1h 49m 54s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/eed9ac410d054e7db289b97556fa0045 : SUCCESS in 43m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0c2d0fe819874ba38695efc86938e440 : FAILURE in 1h 49m 40s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ace16f445a84462bb63628b1784f2606 : FAILURE in 30m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"93679e5a0a710ea43fc931bc8ffdb3c85a0c15f2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-22 00:27:15.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/88/778188/5/check/DellEMC_Unity_Manila_dhss_true/eea80a2/DellEMC_Unity_Manila_dhss_true/512 : FAILURE in 1h 08m 31s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/88/778188/5/check/DellEMC_Unity_Manila_dhss_false/7232239/DellEMC_Unity_Manila_dhss_false/505 : SUCCESS in 58m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"a3404b9d9c12c1d6c8451e396a825e79a57e6c6c","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-22 00:43:43.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/88/778188/5/check/DellEMC_VNX_Manila/954817f/DellEMC_VNX_Manila/331 : SUCCESS in 1h 24m 59s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"d50a40a47f3649ba5be8bc5bb617e70c001a999a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-22 01:51:44.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/88/778188/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/0ba6541/ : SUCCESS in 2h 33m 00s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/88/778188/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/2b99d19/ : SUCCESS in 1h 54m 38s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"6242cf2fbbe41eb25ef77df69465dc59ab4a4db9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-03-22 23:38:12.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"d22e500d3cc189f2d0ee80841ba4c6b90028aec0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 00:34:51.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acc2e1fa75594a08a8918d2de74e89f7 : SUCCESS in 7m 29s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2795707614e04a06a0fca29efc2368cc : FAILURE in 26m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/58c4c702725a4c22a022f8997264d82c : FAILURE in 5m 07s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/462703fa66514f6c915b00767d3e19b0 : FAILURE in 5m 06s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6022af91ab804bc0b3f989a4ca91234b : FAILURE in 4m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a26a7f2e5e1d4cac97806537effa5e35 : FAILURE in 9m 28s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8ed2a0592d1a4cf7816699527f7cf7a7 : FAILURE in 8m 40s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/85220c20fe1641a89ecf34379fc5157a : SUCCESS in 49m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bacf237b78214abcac2a5d61baae2746 : FAILURE in 5m 05s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2c502d3658384ae187ec9ed3f9ae7f63 : FAILURE in 5m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b74164a7065d4cf7eaa0faf88d8958b59b8f3cdf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-23 02:36:59.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/88/778188/6/check/DellEMC_Unity_Manila_dhss_true/347c330/DellEMC_Unity_Manila_dhss_true/522 : FAILURE in 1h 02m 24s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/88/778188/6/check/DellEMC_Unity_Manila_dhss_false/937e294/DellEMC_Unity_Manila_dhss_false/515 : SUCCESS in 51m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"6467cad96b17b4689c1f55eef6892cad3288cf91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-25 00:52:20.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"a7d08feb2e1642c4d370941dae310fc5e9450f29","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-25 00:56:34.000000000","message":"Patch Set 7: Code-Review+1\n\nWe need to get this to pass CI but we don\u0027t need the CephFS jobs to pass here -- we need to see them passing in the create-share-from-snapshot patch, which depends on this one (minus one testcase in native CephFS job, noted in that patch).  Here the CephFS jobs are non-voting, so the dependency direction really doesn\u0027t matter.","accounts_in_message":[],"_revision_number":7},{"id":"a9de333d1a9728f62d3cb34fe1b8ac622c0585d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 02:32:37.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3f7b46410434159a417f1852471451b : SUCCESS in 4m 14s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9e6abc65bf114cc294603c5f72aaea95 : SUCCESS in 30m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/72aa735816694bf1a0495d437a4ce6d1 : SUCCESS in 38m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e6417ad9aaa74ab1be509b2c79fe3ede : FAILURE in 1h 05m 58s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/48c223e24dfe48cf889c8a3ebbebe7c2 : SUCCESS in 1h 23m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/321914faf215400aaf8cd29c2128c91d : FAILURE in 1h 15m 55s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9b06c0efe6114ff4ad7d25adb065cb44 : FAILURE in 1h 37m 21s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4c2493d6c6f74479b16547ef8dfa203f : SUCCESS in 46m 11s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/97fd65be5dcf40aa9cc75001631aff93 : POST_FAILURE in 1h 09m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/168858f55b5842419bf1a2b0f83b8eff : SUCCESS in 24m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"dd2150341eb65f735f6b47dad149bba300b7f499","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-03-25 05:34:35.000000000","message":"Patch Set 7: Code-Review+2\n\nrecheck\n\nIf we merge this, the ceph-native job is broken. Not a fan of that, but we seem to have a chicken-and-egg situation - we don\u0027t *need* to see the create_share_from_snapshot test passing in the same patch. We did try turning off \"create_share_from_snapshot\" tests? \n\nI don\u0027t mind the job being broken in the meanwhile though","accounts_in_message":[],"_revision_number":7},{"id":"6f32c9ec70b2cee3f41d510ccc2c8734dfdbf8c9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-03-25 05:35:10.000000000","message":"Patch Set 7:\n\n\u003e I don\u0027t mind the job being broken in the meanwhile though\nuntil we ship RC - we aren\u0027t shipping manila-tempest-plugin yet","accounts_in_message":[],"_revision_number":7},{"id":"74fd9f7290d19fab11a7ae8b3ff64ebd9a4ec744","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-25 05:56:02.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/88/778188/7/check/DellEMC_Unity_Manila_dhss_true/e7cf355/DellEMC_Unity_Manila_dhss_true/535 : FAILURE in 54m 28s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/88/778188/7/check/DellEMC_Unity_Manila_dhss_false/aa7538c/DellEMC_Unity_Manila_dhss_false/531 : SUCCESS in 56m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"8c83a56771430a36b2dc9d17ef056e5e13e66e98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 07:42:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27d081d401e34773ac90781b8c894106 : SUCCESS in 2m 59s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/295b029abeca4c0ab91747dfabbecede : SUCCESS in 29m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/82c87e7d8d3342f1858883aa58a8da62 : SUCCESS in 34m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/394731da289f454889480a98b0a6a498 : SUCCESS in 1h 19m 07s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9afd2d6596ae4c0cbe4051e4c6a90ebc : FAILURE in 1h 18m 02s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/eb45dce0a00244b090a9b56f5588a3ba : FAILURE in 1h 04m 33s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/321ac43ead814d31978e6e724172d839 : FAILURE in 1h 27m 20s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3a27fdd17442414e90d093adf20c47eb : FAILURE in 22m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b8e9c2d46ffe443e82a587da6c345976 : FAILURE in 1h 56m 44s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c2c0913d5876484681097fab6faf227e : SUCCESS in 24m 43s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0959e56d89f363369c70e0cdc559090894b3cb78","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-25 08:46:34.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/88/778188/7/check/DellEMC_Unity_Manila_dhss_true/a82be0b/DellEMC_Unity_Manila_dhss_true/537 : FAILURE in 1h 03m 23s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/88/778188/7/check/DellEMC_Unity_Manila_dhss_false/db90f01/DellEMC_Unity_Manila_dhss_false/533 : SUCCESS in 55m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"410ba05ecf0488f247d381ac3ab209757603840a","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-25 08:51:39.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/88/778188/7/check/DellEMC_VNX_Manila/d124f50/DellEMC_VNX_Manila/353 : SUCCESS in 1h 08m 38s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"fdae6ad298c19617ef99bb81f39c44866851dbef","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-25 10:03:44.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"28b521b8dd09e9f1fbf7a46442a4e37719f1ea90","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-25 10:15:18.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/88/778188/7/upstream-check/manila-cDOT-manila-tempest-plugin-ss/deece84/ : FAILURE in 2h 20m 40s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/88/778188/7/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/b744350/ : FAILURE in 53s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"b66043afc7786af48f17a04e13cbf2931e736078","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-03-25 10:33:41.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"c11f4489ee864bdf45d0ceba291cad6d3bcd18c6","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-25 11:46:14.000000000","message":"Patch Set 7:\n\nI submitted this review [1] to skip the failing test case until it is fixed to support non \"ip\" access to shares.\n\n[1] https://review.opendev.org/c/openstack/manila-tempest-plugin/+/782947","accounts_in_message":[],"_revision_number":7},{"id":"d697e1b987d7a4322551da1aebf88991d42b75cc","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-25 16:30:02.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"67b23a4399a2e30e66dcf86e426e58a59933c29e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 16:30:30.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"1c969eebb84cb34de22e22a154ed0588a878233d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 17:40:43.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64b48209a9fa4041ae629ad0697b733a : SUCCESS in 4m 03s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8780c93956fb4872ae28034d09262f32 : SUCCESS in 30m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/153c6d27e6664b5a9782a43e894c6171 : SUCCESS in 35m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f2396bb31e254fd29e3b400708e92c41 : SUCCESS in 1h 03m 30s","accounts_in_message":[],"_revision_number":7},{"id":"9f755c0d03de10080bfcbe8a3ad61fa8db28958f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 17:41:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"7d951d370e4f71fe89f887219c1574f12ba8877f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-03-25 18:05:08.000000000","message":"Patch Set 7:\n\nrecheck\n\nUnrelated timeout hit in a scenario test","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"f083c71a9396a5a6f3300d3150b7fc1bee2c57c5","revisions":{"4c39640ea3010bc916dd0d1bdabf4d75459c1f5c":{"kind":"REWORK","_number":1,"created":"2021-03-02 13:10:20.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/88/778188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/778188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/1"}}},"commit":{"parents":[{"commit":"d2138765888fe16e68f713ed99121f41e01a3c51","subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-03-02 13:06:51.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-03-02 13:10:48.000000000","tz":0},"subject":"Add create share from snapshot in CephFS","message":"Add create share from snapshot in CephFS\n\nRemove capability_create_share_from_snapshot_support \u003d False\nfrom tempest config for both CephFS Native and CephFS NFS\n\nPartially-Implements: blueprint create-share-from-snapshot-cephfs\nChange-Id: Ifb984a3cee219fa97a9768495f80ff0c8349830b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4c39640ea3010bc916dd0d1bdabf4d75459c1f5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4c39640ea3010bc916dd0d1bdabf4d75459c1f5c"}]},"branch":"refs/heads/master"},"6130cc31a41460adc49f4f61b7ae7bc5ff3b5bc3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-02 13:12:30.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/88/778188/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/778188/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/2"}}},"commit":{"parents":[{"commit":"d2138765888fe16e68f713ed99121f41e01a3c51","subject":"Fix \u0027ss-type\u0027 conflicts in security service tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2138765888fe16e68f713ed99121f41e01a3c51"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-03-02 13:06:51.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-03-02 13:12:21.000000000","tz":0},"subject":"Add create share from snapshot in CephFS","message":"Add create share from snapshot in CephFS\n\nRemove capability_create_share_from_snapshot_support \u003d False\nfrom tempest config for both CephFS Native and CephFS NFS\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/772157\n\nPartially-Implements: blueprint create-share-from-snapshot-cephfs\nChange-Id: Ifb984a3cee219fa97a9768495f80ff0c8349830b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6130cc31a41460adc49f4f61b7ae7bc5ff3b5bc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6130cc31a41460adc49f4f61b7ae7bc5ff3b5bc3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1289bbd8f3bb60615b3af4c54b7f821a90e7aa3d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-03-18 21:09:21.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/88/778188/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/778188/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/3"}}},"commit":{"parents":[{"commit":"70685064383c5d23fdc9f87a93c4abeca6ed2276","subject":"Merge \"Harden manage negative tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/70685064383c5d23fdc9f87a93c4abeca6ed2276"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-03-02 13:06:51.000000000","tz":0},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-03-18 21:09:21.000000000","tz":0},"subject":"Add create share from snapshot in CephFS","message":"Add create share from snapshot in CephFS\n\nRemove capability_create_share_from_snapshot_support \u003d False\nfrom tempest config for both CephFS Native and CephFS NFS\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/772157\n\nPartially-Implements: blueprint create-share-from-snapshot-cephfs\nChange-Id: Ifb984a3cee219fa97a9768495f80ff0c8349830b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1289bbd8f3bb60615b3af4c54b7f821a90e7aa3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1289bbd8f3bb60615b3af4c54b7f821a90e7aa3d"}]},"branch":"refs/heads/master","description":"Rebase"},"753c6ed7bc292118ee6fbea321d8b2cabc95df3e":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-03-19 12:41:44.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/88/778188/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/778188/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/4"}}},"commit":{"parents":[{"commit":"70685064383c5d23fdc9f87a93c4abeca6ed2276","subject":"Merge \"Harden manage negative tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/70685064383c5d23fdc9f87a93c4abeca6ed2276"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-03-02 13:06:51.000000000","tz":0},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-03-19 12:41:29.000000000","tz":0},"subject":"Add create share from snapshot in CephFS","message":"Add create share from snapshot in CephFS\n\nRemove capability_create_share_from_snapshot_support \u003d False\nfrom tempest config for both CephFS Native and CephFS NFS\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/741560\n\nPartially-Implements: blueprint create-share-from-snapshot-cephfs\nChange-Id: Ifb984a3cee219fa97a9768495f80ff0c8349830b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/753c6ed7bc292118ee6fbea321d8b2cabc95df3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/753c6ed7bc292118ee6fbea321d8b2cabc95df3e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3d128f15a6abd2aee959d3aa0685466520c26834":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-03-21 21:26:41.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/88/778188/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/778188/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/5"}}},"commit":{"parents":[{"commit":"70685064383c5d23fdc9f87a93c4abeca6ed2276","subject":"Merge \"Harden manage negative tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/70685064383c5d23fdc9f87a93c4abeca6ed2276"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-03-02 13:06:51.000000000","tz":0},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-03-21 21:26:30.000000000","tz":0},"subject":"Add create share from snapshot in CephFS","message":"Add create share from snapshot in CephFS\n\nRemove capability_create_share_from_snapshot_support \u003d False\nfrom tempest config for both CephFS Native and CephFS NFS\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/781968\n\nPartially-Implements: blueprint create-share-from-snapshot-cephfs\nChange-Id: Ifb984a3cee219fa97a9768495f80ff0c8349830b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d128f15a6abd2aee959d3aa0685466520c26834"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d128f15a6abd2aee959d3aa0685466520c26834"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e41b82038520ece84fd1e1a34a85c2311f21bcf5":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2021-03-22 23:38:12.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/88/778188/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/778188/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/6"}}},"commit":{"parents":[{"commit":"70685064383c5d23fdc9f87a93c4abeca6ed2276","subject":"Merge \"Harden manage negative tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/70685064383c5d23fdc9f87a93c4abeca6ed2276"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-03-02 13:06:51.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-03-22 23:38:00.000000000","tz":0},"subject":"Add create share from snapshot in CephFS","message":"Add create share from snapshot in CephFS\n\nRemove capability_create_share_from_snapshot_support \u003d False\nfrom tempest config for both CephFS Native and CephFS NFS\n\nPartially-Implements: blueprint create-share-from-snapshot-cephfs\nChange-Id: Ifb984a3cee219fa97a9768495f80ff0c8349830b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e41b82038520ece84fd1e1a34a85c2311f21bcf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e41b82038520ece84fd1e1a34a85c2311f21bcf5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f083c71a9396a5a6f3300d3150b7fc1bee2c57c5":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-03-25 00:52:20.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/88/778188/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/778188/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/778188/7"}}},"commit":{"parents":[{"commit":"3b2b1b7282b17fb8dc9d4dd3a258858eead323c5","subject":"Merge \"Add tests for share size limited by share type.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3b2b1b7282b17fb8dc9d4dd3a258858eead323c5"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-03-02 13:06:51.000000000","tz":0},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-03-25 00:52:20.000000000","tz":0},"subject":"Add create share from snapshot in CephFS","message":"Add create share from snapshot in CephFS\n\nRemove capability_create_share_from_snapshot_support \u003d False\nfrom tempest config for both CephFS Native and CephFS NFS\n\nPartially-Implements: blueprint create-share-from-snapshot-cephfs\nChange-Id: Ifb984a3cee219fa97a9768495f80ff0c8349830b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f083c71a9396a5a6f3300d3150b7fc1bee2c57c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f083c71a9396a5a6f3300d3150b7fc1bee2c57c5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
