)]}'
{"id":"openstack%2Fmanila-tempest-plugin~781074","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I3ba1d5400e51fc475bbe4ea120f0c9f6fa7fa49c","project":"openstack/manila-tempest-plugin","branch":"master","topic":"old_microversion","hashtags":[],"change_id":"I3ba1d5400e51fc475bbe4ea120f0c9f6fa7fa49c","subject":"Stop supporting microversion 2.23 and less","status":"ABANDONED","created":"2021-03-17 10:40:56.000000000","updated":"2021-04-25 06:33:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0b2a027fa86c113f699fef3fa800c823f3571b4c","_number":781074,"virtual_id_number":781074,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-21 15:26:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-17 12:05:52.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-03-17 16:09:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-18 15:37:23.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2021-03-18 15:37:23.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-03-18 15:37:23.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-03-18 15:37:23.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-03-18 16:04:08.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-18 17:26:53.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"1a02a659221f18f9a92ebdf42f9d98f3d95759df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-03-17 10:40:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbbd3ce3def3c12d17c14d0369f166140280901e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-03-17 11:09:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f8292314d5c04c5272197fb8592076cd4696e752","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-17 12:05:52.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/74/781074/2/check/DellEMC_Unity_Manila_dhss_true/d9563e7/DellEMC_Unity_Manila_dhss_true/468 : FAILURE in 55m 44s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/74/781074/2/check/DellEMC_Unity_Manila_dhss_false/8fd8676/DellEMC_Unity_Manila_dhss_false/461 : FAILURE in 52m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"a8c4ef2a058a593015bcf5ea85b2945328980aa4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-03-17 12:17:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"856a8f36ce57d49071e840a9b0579666bc71b699","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-17 13:17:40.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/74/781074/3/check/DellEMC_Unity_Manila_dhss_true/9c56d97/DellEMC_Unity_Manila_dhss_true/469 : FAILURE in 59m 36s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/74/781074/3/check/DellEMC_Unity_Manila_dhss_false/e740942/DellEMC_Unity_Manila_dhss_false/462 : FAILURE in 53m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"595d3de25f032a51c7d0381f845c5a8da5d64b9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-03-17 13:38:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"000dce0b8cb8cd8fe00b69572769453aabc043a7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-17 14:35:38.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/74/781074/4/check/DellEMC_Unity_Manila_dhss_true/829be5c/DellEMC_Unity_Manila_dhss_true/470 : FAILURE in 56m 18s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/74/781074/4/check/DellEMC_Unity_Manila_dhss_false/3665c08/DellEMC_Unity_Manila_dhss_false/463 : FAILURE in 45m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"c7e73a9c0885d530603860ab6161b3385045918a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 16:09:53.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54a15bfc6edf4b3585f21f85c8287163 : SUCCESS in 4m 37s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/56fbedd5abdc497f859c0b5b6d08ea46 : FAILURE in 30m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5fb84775f9cd4f4b9ce22dc217340858 : FAILURE in 39m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/418a3b84b1064a54b3340c5993f46c83 : FAILURE in 1h 13m 58s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9452ca854d4641458c3ec760b51c3059 : FAILURE in 1h 12m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d4bc1cb168e74517903d0ee7eca73485 : FAILURE in 1h 12m 56s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/54a8de2a8ee34889ba1a008c93e3ab60 : FAILURE in 1h 44m 28s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/48a80d3044db4e6d8651d7fa564ece1b : FAILURE in 51m 58s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5fb8f717a5e747f4b14fb8bf0e59b884 : FAILURE in 2h 22m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ec1ddaf4b71a4952b3846037b91c3369 : FAILURE in 29m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ba938f04ce7813e64e850b5c2fd1554fcf51d586","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-03-18 08:53:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"63b430c9a7fdcba8266efae21bb525487b59ccb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-03-18 09:27:06.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"85a100d3a48f9f7e8223d257546e15cc192ad8c1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-18 10:28:14.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/74/781074/6/check/DellEMC_Unity_Manila_dhss_true/fb1c790/DellEMC_Unity_Manila_dhss_true/480 : FAILURE in 1h 00m 43s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/74/781074/6/check/DellEMC_Unity_Manila_dhss_false/34987ef/DellEMC_Unity_Manila_dhss_false/473 : FAILURE in 51m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"64792104e5eda3e4322903189a06fbc335712dfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-03-18 11:11:48.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e81e9f213c9c5f336748ecd7707f5ea1df05c7e0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-18 12:15:16.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/74/781074/7/check/DellEMC_Unity_Manila_dhss_true/88a35e6/DellEMC_Unity_Manila_dhss_true/481 : FAILURE in 1h 02m 50s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/74/781074/7/check/DellEMC_Unity_Manila_dhss_false/613b1c5/DellEMC_Unity_Manila_dhss_false/474 : FAILURE in 53m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"85adec02b60e4e2994ec1007a7c88703ade7d463","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-03-18 12:39:34.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"240ca427f703ab4afcff10f56f53038f85e81253","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-18 13:43:11.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/74/781074/8/check/DellEMC_Unity_Manila_dhss_true/10d0334/DellEMC_Unity_Manila_dhss_true/482 : FAILURE in 1h 02m 45s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/74/781074/8/check/DellEMC_Unity_Manila_dhss_false/10db19c/DellEMC_Unity_Manila_dhss_false/475 : SUCCESS in 48m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"6c18459452450c69d2b6362fb67400bbd21dd232","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 14:52:59.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a18a853f6390483895264db02c5fe68c : SUCCESS in 5m 06s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9ae82261d762442b806004ee3bd9bf09 : SUCCESS in 34m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f4f6b8b861144a54ad3f263e2f6fdd18 : SUCCESS in 44m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d6855fef917143408562ccbc30802971 : SUCCESS in 1h 18m 26s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/eb8fa2d3d05544eca477f1856addc252 : FAILURE in 1h 28m 19s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ddb9ee572b024985ae0a4c7ef3f4baf7 : SUCCESS in 1h 06m 23s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a672d2b642bc44d28218196ae330be17 : SUCCESS in 1h 24m 13s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/74a9ef7744d34a498c15979209ad80bf : SUCCESS in 41m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/420fd3e11d604eaf8f5821ff3dc5ea9e : FAILURE in 2h 03m 10s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/101877b218a646a5abbec10d27b8c270 : FAILURE in 26m 20s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9e1ad12fe553abeb5f713e0317d5797398edfd40","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-18 15:58:21.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/74/781074/8/check/DellEMC_Unity_Manila_dhss_true/3bbd923/DellEMC_Unity_Manila_dhss_true/483 : FAILURE in 1h 04m 52s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/74/781074/8/check/DellEMC_Unity_Manila_dhss_false/fd15765/DellEMC_Unity_Manila_dhss_false/476 : SUCCESS in 54m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"9ac7518550fe747c33cfb1a074e0e0c822210c42","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-18 16:04:08.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/74/781074/8/check/DellEMC_VNX_Manila/3f93fce/DellEMC_VNX_Manila/320 : SUCCESS in 1h 10m 34s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":8},{"id":"69e79dba15ac47002f0a8541c8ac726bca374e7e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-18 17:26:53.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/74/781074/8/upstream-check/manila-cDOT-manila-tempest-plugin-ss/5ae8cdc/ : SUCCESS in 2h 33m 06s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/74/781074/8/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/ca4bd53/ : SUCCESS in 1h 46m 05s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"f22b621eebac6e66ebecfa390179f6b2a2394dd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-21 13:25:36.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f8b24f13553c36273b631efa980378a7c4929e9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-21 13:32:57.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"bc1316270cc09780c89f99428a5c485b6622b7e4","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-21 13:35:20.000000000","message":"Patch Set 10:\n\nSorry I was confused with another patch.","accounts_in_message":[],"_revision_number":10},{"id":"5ff8886bf55864db08774c10e7821b375149ea5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 15:26:12.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28b97b5096e84b169d330ae3f7c9d355 : SUCCESS in 4m 05s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/503eb071dd7741eba9f79ea6e74a8bcb : SUCCESS in 36m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/339ef36f5d1244a6ac0a44d944f33c5a : SUCCESS in 40m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c49752a958d344cabbdc000ec6972805 : SUCCESS in 1h 20m 44s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9daec74a6bb74e31a253336d8d155c88 : FAILURE in 18m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e3a63f401aae4f5099394b8f818f7dac : SUCCESS in 1h 26m 19s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/60ad41560990449597f839283ea5712c : SUCCESS in 1h 43m 39s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7917fc9f7c0d4e7983991ae0d73906df : FAILURE in 52m 00s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1e918c267bc24dd2b7bcba5d49a1d7de : FAILURE in 1h 16m 17s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5944ff15dc1343c4beda884789fdebbf : SUCCESS in 29m 21s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e9f05487940aaf4657ab73c22ebc4030cc636348","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-21 16:39:56.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/74/781074/10/check/DellEMC_VNX_Manila/358811e/DellEMC_VNX_Manila/5 : SUCCESS in 1h 13m 06s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":10},{"id":"4e5991c46cb42bc03de9ac7bcca608b55398e624","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-21 17:18:44.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/74/781074/10/check/DellEMC_Unity_Manila_dhss_true/9071568/DellEMC_Unity_Manila_dhss_true/7 : FAILURE in 2h 03m 27s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/74/781074/10/check/DellEMC_Unity_Manila_dhss_false/dd19f88/DellEMC_Unity_Manila_dhss_false/7 : SUCCESS in 49m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"0b2a027fa86c113f699fef3fa800c823f3571b4c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-25 06:33:56.000000000","message":"Abandoned\n\nIt was decided not drop testing for API\u003c2.24","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"2f480de999dc0b3aa53339d5bd798fe0ff302546","revisions":{"b3d12258f127088cbaefda6380f2948eacf86f3d":{"kind":"REWORK","_number":1,"created":"2021-03-17 10:40:56.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/74/781074/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/74/781074/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/1"}}},"commit":{"parents":[{"commit":"9efbc94e963613a617631d27917ba278429566f5","subject":"Merge \"Remove a duplicated skipped functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9efbc94e963613a617631d27917ba278429566f5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-03-17 10:38:26.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-03-17 10:38:26.000000000","tz":120},"subject":"Stop supporting microversion 2.23 and less","message":"Stop supporting microversion 2.23 and less\n\nThis patch makes a cleanup for all stuff related\nto 2.23 microversion and less.\n\nChange-Id: I3ba1d5400e51fc475bbe4ea120f0c9f6fa7fa49c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b3d12258f127088cbaefda6380f2948eacf86f3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b3d12258f127088cbaefda6380f2948eacf86f3d"}]},"branch":"refs/heads/master"},"4d92605004bd47d8b552829177ac16b522bb3094":{"kind":"REWORK","_number":2,"created":"2021-03-17 11:09:50.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/74/781074/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/74/781074/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/2"}}},"commit":{"parents":[{"commit":"9efbc94e963613a617631d27917ba278429566f5","subject":"Merge \"Remove a duplicated skipped functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9efbc94e963613a617631d27917ba278429566f5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-03-17 10:38:26.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-03-17 11:09:31.000000000","tz":120},"subject":"Stop supporting microversion 2.23 and less","message":"Stop supporting microversion 2.23 and less\n\nThis patch makes a cleanup for all stuff related\nto 2.23 microversion and less.\n\nChange-Id: I3ba1d5400e51fc475bbe4ea120f0c9f6fa7fa49c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d92605004bd47d8b552829177ac16b522bb3094"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d92605004bd47d8b552829177ac16b522bb3094"}]},"branch":"refs/heads/master"},"107ff50901b05af6363a5622c04ed1ab6d9a8b74":{"kind":"REWORK","_number":3,"created":"2021-03-17 12:17:48.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/74/781074/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/74/781074/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/3"}}},"commit":{"parents":[{"commit":"9efbc94e963613a617631d27917ba278429566f5","subject":"Merge \"Remove a duplicated skipped functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9efbc94e963613a617631d27917ba278429566f5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-03-17 10:38:26.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-03-17 12:17:20.000000000","tz":120},"subject":"[WIP] Stop supporting microversion 2.23 and less","message":"[WIP] Stop supporting microversion 2.23 and less\n\nThis patch makes a cleanup for all stuff related\nto 2.23 microversion and less.\n\nChange-Id: I3ba1d5400e51fc475bbe4ea120f0c9f6fa7fa49c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/107ff50901b05af6363a5622c04ed1ab6d9a8b74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/107ff50901b05af6363a5622c04ed1ab6d9a8b74"}]},"branch":"refs/heads/master"},"ef62b24e67419b9a8d8bb882b59f824bab4128d2":{"kind":"REWORK","_number":4,"created":"2021-03-17 13:38:29.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/74/781074/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/74/781074/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/4"}}},"commit":{"parents":[{"commit":"9efbc94e963613a617631d27917ba278429566f5","subject":"Merge \"Remove a duplicated skipped functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9efbc94e963613a617631d27917ba278429566f5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-03-17 10:38:26.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-03-17 13:38:08.000000000","tz":120},"subject":"[WIP] Stop supporting microversion 2.23 and less","message":"[WIP] Stop supporting microversion 2.23 and less\n\nThis patch makes a cleanup for all stuff related\nto 2.23 microversion and less.\n\nChange-Id: I3ba1d5400e51fc475bbe4ea120f0c9f6fa7fa49c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ef62b24e67419b9a8d8bb882b59f824bab4128d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ef62b24e67419b9a8d8bb882b59f824bab4128d2"}]},"branch":"refs/heads/master"},"a369f741bf2d8e1390fcf1ee245799c17d8f071b":{"kind":"REWORK","_number":5,"created":"2021-03-18 08:53:37.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/74/781074/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/74/781074/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/5"}}},"commit":{"parents":[{"commit":"9efbc94e963613a617631d27917ba278429566f5","subject":"Merge \"Remove a duplicated skipped functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9efbc94e963613a617631d27917ba278429566f5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-03-17 10:38:26.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-03-18 08:53:10.000000000","tz":120},"subject":"Stop supporting microversion 2.23 and less","message":"Stop supporting microversion 2.23 and less\n\nThis patch makes a cleanup for all stuff related\nto 2.23 microversion and less.\n\nChange-Id: I3ba1d5400e51fc475bbe4ea120f0c9f6fa7fa49c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a369f741bf2d8e1390fcf1ee245799c17d8f071b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a369f741bf2d8e1390fcf1ee245799c17d8f071b"}]},"branch":"refs/heads/master"},"caa29780cfb71cf16a4ac8df2070e5131bf10aa0":{"kind":"REWORK","_number":6,"created":"2021-03-18 09:27:06.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/74/781074/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/74/781074/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/6"}}},"commit":{"parents":[{"commit":"9efbc94e963613a617631d27917ba278429566f5","subject":"Merge \"Remove a duplicated skipped functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9efbc94e963613a617631d27917ba278429566f5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-03-17 10:38:26.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-03-18 09:26:43.000000000","tz":120},"subject":"Stop supporting microversion 2.23 and less","message":"Stop supporting microversion 2.23 and less\n\nThis patch makes a cleanup for all stuff related\nto 2.23 microversion and less.\n\nChange-Id: I3ba1d5400e51fc475bbe4ea120f0c9f6fa7fa49c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/caa29780cfb71cf16a4ac8df2070e5131bf10aa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/caa29780cfb71cf16a4ac8df2070e5131bf10aa0"}]},"branch":"refs/heads/master"},"5a187948c61ed853c9d52403587149a4a2639199":{"kind":"REWORK","_number":7,"created":"2021-03-18 11:11:48.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/74/781074/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/74/781074/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/7"}}},"commit":{"parents":[{"commit":"9efbc94e963613a617631d27917ba278429566f5","subject":"Merge \"Remove a duplicated skipped functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9efbc94e963613a617631d27917ba278429566f5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-03-17 10:38:26.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-03-18 11:11:26.000000000","tz":120},"subject":"Stop supporting microversion 2.23 and less","message":"Stop supporting microversion 2.23 and less\n\nThis patch makes a cleanup for all stuff related\nto 2.23 microversion and less.\n\nChange-Id: I3ba1d5400e51fc475bbe4ea120f0c9f6fa7fa49c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5a187948c61ed853c9d52403587149a4a2639199"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5a187948c61ed853c9d52403587149a4a2639199"}]},"branch":"refs/heads/master"},"6c5b46d2e719995876f99df50dedd9b74dc8c57a":{"kind":"REWORK","_number":8,"created":"2021-03-18 12:39:34.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/74/781074/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/74/781074/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/8"}}},"commit":{"parents":[{"commit":"9efbc94e963613a617631d27917ba278429566f5","subject":"Merge \"Remove a duplicated skipped functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9efbc94e963613a617631d27917ba278429566f5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-03-17 10:38:26.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-03-18 12:39:09.000000000","tz":120},"subject":"Stop supporting microversion 2.23 and less","message":"Stop supporting microversion 2.23 and less\n\nThis patch makes a cleanup for all stuff related\nto 2.23 microversion and less.\n\nChange-Id: I3ba1d5400e51fc475bbe4ea120f0c9f6fa7fa49c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6c5b46d2e719995876f99df50dedd9b74dc8c57a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6c5b46d2e719995876f99df50dedd9b74dc8c57a"}]},"branch":"refs/heads/master"},"7199f4a566a4495ea80b3409bf1ece659b00d694":{"kind":"REWORK","_number":9,"created":"2021-04-21 13:25:36.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/74/781074/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/74/781074/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/9"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-03-17 10:38:26.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-21 13:25:13.000000000","tz":180},"subject":"Stop supporting microversion 2.23 and less","message":"Stop supporting microversion 2.23 and less\n\nThis patch makes a cleanup for all stuff related\nto 2.23 microversion and less.\n\nChange-Id: I3ba1d5400e51fc475bbe4ea120f0c9f6fa7fa49c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7199f4a566a4495ea80b3409bf1ece659b00d694"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7199f4a566a4495ea80b3409bf1ece659b00d694"}]},"branch":"refs/heads/master"},"2f480de999dc0b3aa53339d5bd798fe0ff302546":{"kind":"REWORK","_number":10,"created":"2021-04-21 13:32:57.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/74/781074/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/74/781074/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/781074/10"}}},"commit":{"parents":[{"commit":"9efbc94e963613a617631d27917ba278429566f5","subject":"Merge \"Remove a duplicated skipped functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9efbc94e963613a617631d27917ba278429566f5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-03-17 10:38:26.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-21 13:32:38.000000000","tz":180},"subject":"Stop supporting microversion 2.23 and less","message":"Stop supporting microversion 2.23 and less\n\nThis patch makes a cleanup for all stuff related\nto 2.23 microversion and less.\n\nChange-Id: I3ba1d5400e51fc475bbe4ea120f0c9f6fa7fa49c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2f480de999dc0b3aa53339d5bd798fe0ff302546"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2f480de999dc0b3aa53339d5bd798fe0ff302546"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
