)]}'
{"id":"openstack%2Fmanila-tempest-plugin~782947","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I806fd72b1b5e06e6d679ff71c6ee70ce88a4acd2","project":"openstack/manila-tempest-plugin","branch":"master","hashtags":[],"change_id":"I806fd72b1b5e06e6d679ff71c6ee70ce88a4acd2","subject":"skip write to share created from snap for CEPHFS","status":"ABANDONED","created":"2021-03-25 11:08:25.000000000","updated":"2021-04-19 11:12:21.000000000","total_comment_count":3,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"222b240e0cdc1592e76042858b59598a54ecfc83","_number":782947,"virtual_id_number":782947,"owner":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-05 16:40:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2021-04-05 14:31:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":1,"date":"2021-04-06 04:55:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-25 11:10:30.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-03-25 11:13:32.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-03-25 11:13:32.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2021-03-25 11:13:32.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2021-03-25 11:39:11.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-03-25 11:52:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-25 16:16:43.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-25 17:23:28.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"271c62cad9ae0a54d429abf38767ce209f43d3bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-25 11:08:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05c19979ec348892d40bca4854e39eebaadd2441","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-25 11:39:11.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/47/782947/1/check/DellEMC_Unity_Manila_dhss_true/1b4ab96/DellEMC_Unity_Manila_dhss_true/539 : FAILURE in 30m 27s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/47/782947/1/check/DellEMC_Unity_Manila_dhss_false/f55e821/DellEMC_Unity_Manila_dhss_false/535 : FAILURE in 27m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"02e9908ba73c52bfc57363aaf6efeba619729570","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 11:52:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2361506ef8474bfbb811cfa0a137d6fe : FAILURE in 5m 35s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e22752b34e334a55a37d5aa374a7253d : FAILURE in 25m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b12aac416afd4b9ab43074e97d78dc00 : FAILURE in 19m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0c0e9f5d120847b7b12ec0b48aad861e : FAILURE in 25m 51s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7204a5d5c54d4d79b6e494ff9711d527 : FAILURE in 31m 24s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3ab9a2ffbe4142099c3f3dd3a7e569fc : FAILURE in 27m 56s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/373af0ccc0d34e9085853352b2617695 : FAILURE in 33m 44s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6d25c03f60e24043924f433c648f4e97 : FAILURE in 33m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9c96ad6b6d7442168ffc08e9ef2029aa : FAILURE in 40m 17s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/193d10d03d9740289d743db90ae3ab6d : FAILURE in 20m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"29714635bb87f8a98a0c7dc0d643930d8da81654","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-25 12:54:59.000000000","message":"Patch Set 1:\n\napparently I had a paste error from uuidgen to code","accounts_in_message":[],"_revision_number":1},{"id":"d13c028333a758f7b88c07ab362e3d0eccc3b8c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-25 12:55:07.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"58554c92e37d4efe80aec3c216f52c5df48c8cd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 14:49:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca07c113daba4917b84d369ec8e69a24 : SUCCESS in 4m 02s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/78f982a657fb4f1cb25e51145104929b : SUCCESS in 30m 12s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/52ea4d3255734aa7a52d74610d9a1a86 : SUCCESS in 33m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e238b8e16dc94545bd9a62ad08e40a5f : SUCCESS in 1h 07m 09s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6b077e038af746f58f954c83b0b632f1 : FAILURE in 1h 11m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/cf119b5aeff3487b8c0141465a124315 : SUCCESS in 1h 25m 59s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/09c31c2c28224d47b4ea3fb6b0826343 : SUCCESS in 1h 30m 44s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/59c20bb972c240bdb770a8c956d6cfbb : SUCCESS in 46m 38s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3a904fc4efeb4bf088780af90613c16f : FAILURE in 1h 49m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3878e4b67cb64509b0cf410b7ad102fd : SUCCESS in 34m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4ee529cfc33e2cb2332a00d11dea57bba2540534","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-25 14:50:27.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/47/782947/2/check/DellEMC_Unity_Manila_dhss_true/0dc5ed8/DellEMC_Unity_Manila_dhss_true/542 : FAILURE in 1h 01m 15s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/47/782947/2/check/DellEMC_Unity_Manila_dhss_false/9764a49/DellEMC_Unity_Manila_dhss_false/538 : SUCCESS in 58m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"e6d4b0b89d8f72ad963dd1c6c5c84fb0440ab231","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-25 16:16:43.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/47/782947/2/check/DellEMC_VNX_Manila/6dfa0cb/DellEMC_VNX_Manila/355 : FAILURE in 1h 27m 08s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"04d5cf4a93d25c7db066f3141130554722a70c8b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-25 17:23:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/47/782947/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/6a9a744/ : SUCCESS in 2h 33m 51s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/47/782947/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/915dfb0/ : SUCCESS in 1h 45m 24s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"77808cef94779741f15651d86f1ab472ab337550","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-03-25 17:38:08.000000000","message":"Patch Set 2: Code-Review+2\n\nGood catch, thanks Tom","accounts_in_message":[],"_revision_number":2},{"id":"00c429bd59bc201767f2215458442c2358a9f4f1","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-03-25 18:56:03.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM, thanks.","accounts_in_message":[],"_revision_number":2},{"id":"7b8d50109259f222734ed18e7ba0fcc6d58a20ce","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-04 05:24:10.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e660cd71ef700680217433ae637f3d634bf70e3f","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-04-05 12:14:38.000000000","message":"Patch Set 2:\n\nRenamed the cephfs create share from snapshot test case as Liron suggested.","accounts_in_message":[],"_revision_number":2},{"id":"2151a1b37cdb1dca588726d091f7ef70f581da69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-04-05 12:14:41.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"fed49b33b8edcab7e55d8bc971e3cbde8c77579c","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-05 12:24:39.000000000","message":"Patch Set 3: Code-Review+1\n\nThanks Tom","accounts_in_message":[],"_revision_number":3},{"id":"36095e0a8a5cb134c0a833ecf2d4a44ccde5ceac","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-05 13:13:14.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/47/782947/3/check/DellEMC_Unity_Manila_dhss_true/5c835eb/DellEMC_Unity_Manila_dhss_true/577 : FAILURE in 58m 16s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/47/782947/3/check/DellEMC_Unity_Manila_dhss_false/c7fbbe5/DellEMC_Unity_Manila_dhss_false/581 : FAILURE in 48m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"a8e7a12424bb147e3f26ad538eee5e516915b23c","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-05 13:38:26.000000000","message":"Patch Set 3: -Code-Review\n\nOh sorry,\nThe name is too long 83 characters.\nSo I guess you can remove the word \"with\"\n\"test_write_data_to_share_created_from_snapshot_ceph_fuse_client\"","accounts_in_message":[],"_revision_number":3},{"id":"141eef916741e99cc2db6506487be7cc2d9ad844","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-04-05 13:42:33.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"6abe8e66e78d8bf0d89df144ef40b53c6d4c84c0","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-05 13:57:05.000000000","message":"Patch Set 4:\n\nI\u0027m so sorry for the rush I made.\nPlease ignore all my comments and revert to the first patch.\nSorry again :)","accounts_in_message":[],"_revision_number":4},{"id":"766d796631300b43d10e284c607387ae516c3936","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-04-05 14:24:32.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e I\u0027m so sorry for the rush I made.\n\u003e Please ignore all my comments and revert to the first patch.\n\u003e Sorry again :)\n\nNo problem, sir.  I\u0027m happy to name the test cases however they will make best sense for reporting, etc.","accounts_in_message":[],"_revision_number":4},{"id":"eaf351164efb10ee3122c2e2b7a3d1c4c4dec0d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-04-05 14:27:25.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"9a6b2aabb9be2d4b1cbb14d1f1d47e9bafa9ac8a","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-04-05 14:31:44.000000000","message":"Patch Set 5: Code-Review+1\n\nLooks good, thanks.","accounts_in_message":[],"_revision_number":5},{"id":"bfed5ae672a84f026ef3749c04b470051cbae413","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-05 15:26:10.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/47/782947/5/check/DellEMC_Unity_Manila_dhss_true/1c53bc6/DellEMC_Unity_Manila_dhss_true/579 : FAILURE in 54m 18s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/47/782947/5/check/DellEMC_Unity_Manila_dhss_false/bf2683f/DellEMC_Unity_Manila_dhss_false/584 : FAILURE in 32m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"75c33eacd18a4c9a5499953b845a7e97c4bc8b44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 16:40:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8349d6f0ef414028bf26e8076e7da2f3 : SUCCESS in 4m 05s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b95b841038f544e5889442f3cea44ac1 : SUCCESS in 29m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/acafc389900f4a1298cc1a290de8a0d1 : SUCCESS in 35m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1e098333f8294cf39978f70c50a90859 : SUCCESS in 1h 08m 10s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/35d7df46cff0405cb083e610af6418ee : SUCCESS in 1h 19m 41s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/725992edb36d4d3db21792a208405add : SUCCESS in 1h 03m 03s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/890e98e5e4884f0cb09b2a6ace54233c : POST_FAILURE in 1h 48m 00s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e7ee8775ff6c4cfdaccada03cc8244bd : FAILURE in 46m 24s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f75f813eabbc4075bc44c20ba301233e : FAILURE in 2h 11m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/441c30ad464a4970b303ba2c0c46134f : SUCCESS in 24m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dbe86bf444734b4696229e3cc3be8d1ac4e49b4e","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-04-05 17:09:43.000000000","message":"Patch Set 5:\n\nrecheck\n\npost failure on cephfs-nfs job","accounts_in_message":[],"_revision_number":5},{"id":"8bf77fbc5578ed763efd5591a3feb279c1efc6ce","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-05 17:34:09.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/47/782947/5/check/DellEMC_Unity_Manila_dhss_true/8cbc736/DellEMC_Unity_Manila_dhss_true/581 : FAILURE in 53m 17s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/47/782947/5/check/DellEMC_Unity_Manila_dhss_false/d9b51f0/DellEMC_Unity_Manila_dhss_false/586 : FAILURE in 37m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"81699b5fc8f5618313ba36ad888468d90b4133a0","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-05 18:24:53.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/47/782947/5/check/DellEMC_VNX_Manila/33c7e64/DellEMC_VNX_Manila/384 : SUCCESS in 1h 25m 52s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"74aedb389c1735c74dc037c5d84aa31332bc51ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 19:21:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4bbd8b421ea49759df1409e42038446 : SUCCESS in 3m 29s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e8be3ba034614dd1833b8e915e0c5393 : SUCCESS in 35m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/43ce6593976941518a6857d942327c34 : SUCCESS in 43m 07s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2d9a4f5da9ca4ae095baa7cd49d644a4 : SUCCESS in 1h 15m 52s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6e05718c61954b8c93d7a8ef25c33b37 : FAILURE in 1h 24m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b0d4a282bd654dc8b759350fa52f7cbd : SUCCESS in 1h 05m 00s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/50291bb1eb6c43cf82666de0449a7984 : SUCCESS in 1h 06m 54s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/07f6b630d9b647b583f6d39de5243818 : FAILURE in 52m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d5f0633ebf1b4907a298b98735af4c60 : FAILURE in 2h 05m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f548e634327f437a8b40e202b723a297 : FAILURE in 37m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f95754eb2e37768fbdd183de31ef851106585bd9","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-05 19:24:11.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/47/782947/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/2412db3/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/47/782947/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/1b2585e/ : FAILURE in 2m 32s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"68887498f61710f4e5e98b51bdf0c4690b44704c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-05 20:23:04.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/47/782947/5/check/DellEMC_Unity_Manila_dhss_true/97772ff/DellEMC_Unity_Manila_dhss_true/582 : FAILURE in 1h 01m 27s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/47/782947/5/check/DellEMC_Unity_Manila_dhss_false/1ea2aa5/DellEMC_Unity_Manila_dhss_false/587 : FAILURE in 39m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"0b45844cd57f99fb9335bd1081351990fc3f17c4","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-05 20:36:51.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/47/782947/5/check/DellEMC_VNX_Manila/1535c32/DellEMC_VNX_Manila/385 : FAILURE in 1h 15m 14s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"307c1cb5366579c690e0769f90530470a7dbf2db","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-06 04:55:47.000000000","message":"Patch Set 5: Code-Review+1\n\nThanks Tom","accounts_in_message":[],"_revision_number":5},{"id":"2aeae42e2fcb87fd0a78bcce8a3161ef7de29f40","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-04-09 06:24:11.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nThanks Tom, didn\u0027t act on this review sooner because I thought we can get these tests fixed up before we request a new release of manila-tempest-plugin. So instead of this skip, we can see if reorganizing the tests helps resolve this: https://review.opendev.org/c/openstack/manila-tempest-plugin/+/785558","accounts_in_message":[],"_revision_number":5},{"id":"222b240e0cdc1592e76042858b59598a54ecfc83","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-04-19 11:12:21.000000000","message":"Abandoned\n\nGoutham is fixing the tests so that we don\u0027t need to skip.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ccc637337f90a3514b47aaf4362ef647fe8ba75f","revisions":{"60142d36bc20aa420b7d61b08d34d8db18b07030":{"kind":"REWORK","_number":1,"created":"2021-03-25 11:08:25.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/47/782947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/47/782947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/782947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/782947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/782947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/782947/1"}}},"commit":{"parents":[{"commit":"3b2b1b7282b17fb8dc9d4dd3a258858eead323c5","subject":"Merge \"Add tests for share size limited by share type.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3b2b1b7282b17fb8dc9d4dd3a258858eead323c5"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-03-25 10:57:53.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-03-25 11:07:47.000000000","tz":-240},"subject":"skip write to share created from snap for CEPHFS","message":"skip write to share created from snap for CEPHFS\n\nThe test case in question uses only IP access for the VM\nthat mounts and writes to shares, which does not workk for\nnative CephFS shares.  Skip the test case until CEPHFS auth\ncan be used for this test case.\n\nAlso update the comment for the corresponding and already\nskipped CIFS test case to refer to our current bug tracking\nthe issue.\n\nRelated-Bug: #1921369\n\nChange-Id: I806fd72b1b5e06e6d679ff71c6ee70ce88a4acd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/60142d36bc20aa420b7d61b08d34d8db18b07030"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/60142d36bc20aa420b7d61b08d34d8db18b07030"}]},"branch":"refs/heads/master"},"ec5d8046ba0ea22c5a2ceb73988c5dfd4a8012b2":{"kind":"REWORK","_number":2,"created":"2021-03-25 12:55:07.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/47/782947/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/47/782947/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/782947/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/782947/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/782947/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/782947/2"}}},"commit":{"parents":[{"commit":"3b2b1b7282b17fb8dc9d4dd3a258858eead323c5","subject":"Merge \"Add tests for share size limited by share type.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3b2b1b7282b17fb8dc9d4dd3a258858eead323c5"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-03-25 10:57:53.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-03-25 12:54:28.000000000","tz":0},"subject":"skip write to share created from snap for CEPHFS","message":"skip write to share created from snap for CEPHFS\n\nThe test case in question uses only IP access for the VM\nthat mounts and writes to shares, which does not workk for\nnative CephFS shares.  Skip the test case until CEPHFS auth\ncan be used for this test case.\n\nAlso update the comment for the corresponding and already\nskipped CIFS test case to refer to our current bug tracking\nthe issue.\n\nRelated-Bug: #1921369\n\nChange-Id: I806fd72b1b5e06e6d679ff71c6ee70ce88a4acd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ec5d8046ba0ea22c5a2ceb73988c5dfd4a8012b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ec5d8046ba0ea22c5a2ceb73988c5dfd4a8012b2"}]},"branch":"refs/heads/master"},"49d69116f4cb522e6cfe502d0dc2b446de5c35a1":{"kind":"REWORK","_number":3,"created":"2021-04-05 12:14:41.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/47/782947/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/47/782947/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/782947/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/782947/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/782947/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/782947/3"}}},"commit":{"parents":[{"commit":"3b2b1b7282b17fb8dc9d4dd3a258858eead323c5","subject":"Merge \"Add tests for share size limited by share type.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3b2b1b7282b17fb8dc9d4dd3a258858eead323c5"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-03-25 10:57:53.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-04-05 12:13:34.000000000","tz":0},"subject":"skip write to share created from snap for CEPHFS","message":"skip write to share created from snap for CEPHFS\n\nThe test case in question uses only IP access for the VM\nthat mounts and writes to shares, which does not workk for\nnative CephFS shares.  Skip the test case until CEPHFS auth\ncan be used for this test case.\n\nAlso update the comment for the corresponding and already\nskipped CIFS test case to refer to our current bug tracking\nthe issue.\n\nRelated-Bug: #1921369\n\nChange-Id: I806fd72b1b5e06e6d679ff71c6ee70ce88a4acd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/49d69116f4cb522e6cfe502d0dc2b446de5c35a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/49d69116f4cb522e6cfe502d0dc2b446de5c35a1"}]},"branch":"refs/heads/master"},"96803f2fb5e46b5d35fecdff46a7035611d5ceff":{"kind":"REWORK","_number":4,"created":"2021-04-05 13:42:33.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/47/782947/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/47/782947/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/782947/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/782947/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/782947/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/782947/4"}}},"commit":{"parents":[{"commit":"3b2b1b7282b17fb8dc9d4dd3a258858eead323c5","subject":"Merge \"Add tests for share size limited by share type.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3b2b1b7282b17fb8dc9d4dd3a258858eead323c5"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-03-25 10:57:53.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-04-05 13:42:28.000000000","tz":0},"subject":"skip write to share created from snap for CEPHFS","message":"skip write to share created from snap for CEPHFS\n\nThe test case in question uses only IP access for the VM\nthat mounts and writes to shares, which does not workk for\nnative CephFS shares.  Skip the test case until CEPHFS auth\ncan be used for this test case.\n\nAlso update the comment for the corresponding and already\nskipped CIFS test case to refer to our current bug tracking\nthe issue.\n\nRelated-Bug: #1921369\n\nChange-Id: I806fd72b1b5e06e6d679ff71c6ee70ce88a4acd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/96803f2fb5e46b5d35fecdff46a7035611d5ceff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/96803f2fb5e46b5d35fecdff46a7035611d5ceff"}]},"branch":"refs/heads/master"},"ccc637337f90a3514b47aaf4362ef647fe8ba75f":{"kind":"REWORK","_number":5,"created":"2021-04-05 14:27:25.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/47/782947/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/47/782947/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/782947/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/782947/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/782947/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/782947/5"}}},"commit":{"parents":[{"commit":"3b2b1b7282b17fb8dc9d4dd3a258858eead323c5","subject":"Merge \"Add tests for share size limited by share type.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3b2b1b7282b17fb8dc9d4dd3a258858eead323c5"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-03-25 10:57:53.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-04-05 14:25:34.000000000","tz":0},"subject":"skip write to share created from snap for CEPHFS","message":"skip write to share created from snap for CEPHFS\n\nThe test case in question uses only IP access for the VM\nthat mounts and writes to shares, which does not workk for\nnative CephFS shares.  Skip the test case until CEPHFS auth\ncan be used for this test case.\n\nAlso update the comment for the corresponding and already\nskipped CIFS test case to refer to our current bug tracking\nthe issue.\n\nRelated-Bug: #1921369\n\nChange-Id: I806fd72b1b5e06e6d679ff71c6ee70ce88a4acd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ccc637337f90a3514b47aaf4362ef647fe8ba75f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ccc637337f90a3514b47aaf4362ef647fe8ba75f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
