)]}'
{"id":"openstack%2Fmanila-tempest-plugin~786177","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25","project":"openstack/manila-tempest-plugin","branch":"master","topic":"review-share-feature-enabled","attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-01-07 14:30:50.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-01-29 21:57:53.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2021-10-21 06:52:01.000000000","reason":"Liron Kuchlani replied on the change"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2026-01-08 13:59:58.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2026-02-05 16:34:21.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}},"hashtags":[],"change_id":"Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25","subject":"Add new configuration section for share features","status":"NEW","created":"2021-04-14 08:39:58.000000000","updated":"2026-02-05 16:34:21.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":130,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"337f4b79549e2b0c78965380fc1c9bc378239fbb","_number":786177,"virtual_id_number":786177,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-01-30 00:09:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2026-02-05 16:34:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-14 08:41:29.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2021-04-14 08:41:29.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-04-14 08:41:29.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-04-14 08:41:29.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-04-14 11:02:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-14 12:06:58.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-14 12:07:28.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-18 14:00:08.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-10-20 14:05:08.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2022-07-04 15:05:51.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-11-05 19:04:50.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2024-11-07 09:55:55.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-27 11:58:52.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"3b61acf2566875788acd11ddb28733da03a16a83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-14 08:39:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c60208fca7131240dcbdeacff443261acc56a94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-14 08:52:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"88960277ef8f12f716d6005e0f2d42088a674c69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 11:02:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b342cf8fd92344b194a959e1c1be72a9 : SUCCESS in 3m 27s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0262e48422214f98bb286e509f6f2d0c : SUCCESS in 33m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/05c8aa3b87a74257889b16feff6e0085 : SUCCESS in 38m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/198dae74522b470b9524dd071304093e : SUCCESS in 1h 09m 54s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e4d6a2badd9f4428933bfdf43cfd78da : FAILURE in 1h 18m 48s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/4806cff6e55446d19d62dcbb85550572 : FAILURE in 1h 15m 53s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e047cc2026d842f4b22cd7d9437aaf66 : SUCCESS in 1h 28m 58s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6ac908e290354f218d5dfe730b6133b7 : FAILURE in 46m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6fbf00d48f9c401eb95a8489d1ee0598 : FAILURE in 2h 05m 11s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d849072ee5274815b59b8e3bee0d38d6 : SUCCESS in 19m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"96a6bf48d9c8b829909c21cb2e712f01cf75ee9a","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-14 12:06:58.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/77/786177/2/check/DellEMC_VNX_Manila/cf24e26/DellEMC_VNX_Manila/408 : SUCCESS in 1h 04m 10s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"689af21249fd4c0699f6e1a6205231a58e14d16f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-14 12:07:28.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/77/786177/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/a820436/ : FAILURE in 1h 04m 36s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/77/786177/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/036ca40/ : FAILURE in 26m 08s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"2ac7f7815cc56771805483859ad15a0a21d3b885","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-18 12:55:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1d584b05f94e9edee1af197e225b6ffa6a5cba13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-18 13:15:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bcad56dc6b52a3d86b1c430c6eb59b7487c5bf51","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-18 14:00:08.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/77/786177/4/check/DellEMC_Unity_Manila_dhss_true/f36760f/DellEMC_Unity_Manila_dhss_true/615 : FAILURE in 44m 15s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/77/786177/4/check/DellEMC_Unity_Manila_dhss_false/e7bcdf4/DellEMC_Unity_Manila_dhss_false/620 : FAILURE in 33m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"3f4aaba147c8958bc1fde8aff74c49e08099c176","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-18 14:54:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5b150da9eab4988b19a583dc82f5244 : SUCCESS in 3m 54s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e4e7ecb4687547cba416dbd94e8fcaf7 : FAILURE in 18m 53s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f8bae3e9cf8841f58d2a1530f334a93e : FAILURE in 23m 08s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/aa64b50543c74ee498ccebd4314d4364 : FAILURE in 57m 41s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/82f39a0149d74a2386afa50b82fedbb1 : FAILURE in 51m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c273396603bb4a20ba51ae41979630cb : FAILURE in 1h 00m 02s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/47e5e1b903894063b961bb1caafe75e2 : FAILURE in 1h 12m 31s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/58a963f8faac4196afb45b94c7cbe18e : FAILURE in 34m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7b55f88c288c4fc6b71fd1bf366f70f8 : FAILURE in 1h 38m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/adf726a6ee194ea99193ba96b90ad0c3 : FAILURE in 19m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cbd51e9f4487fa08c482cceb02d7f9906c398334","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-19 04:32:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bd9f3f7305f30a14272aabcdcb48a57a9ca29d7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 06:08:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/999bff6df8a6431c81e20f9ce9f71cc3 : SUCCESS in 4m 44s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/77dc854f56194f6bb9f109b6999b19ce : FAILURE in 27m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/abcee3f2fe39456d8c2827ce96de0f86 : FAILURE in 24m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e32ea72e530f4e0b9a11eaa403c8fe6b : FAILURE in 1h 07m 19s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/751361d0295a4af7ac5faf985a0d1bbd : FAILURE in 55m 57s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9ee9271038df4735b9804d4bb9aa6c64 : FAILURE in 53m 49s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5cc878cc0c3748dab6216b8ab6578add : FAILURE in 1h 28m 57s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3b3beff1e6134e86bdbb38a77249ef13 : FAILURE in 48m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0e2087f4e28c43b49a758ad5e85d0589 : FAILURE in 1h 00m 40s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e855de63d04f462fb8e07b45fd24d09f : FAILURE in 26m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"93b9924e320846c39ae2264a403f3b38333d2568","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-19 06:26:22.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a2b5a17b30604675d5f6daeae4c3857f7c28b96c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 08:19:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc1016c1ff5a4b5198e4a4a00e7e81e8 : SUCCESS in 4m 46s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/156954b91c2546e5bd18953b96f6b81a : FAILURE in 25m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7bc604d7f2934973ac59bb3153ca7746 : FAILURE in 28m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/09c222b3334d4dd7a338205cd28011eb : FAILURE in 55m 04s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c30471b696524f48be3b75d53c81ce49 : FAILURE in 1h 03m 10s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/2b473e38a7b84ef49bf8810f298b9511 : FAILURE in 1h 01m 04s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/988c816235a246b4b324cfd8b9c50170 : FAILURE in 1h 14m 09s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6f08bfcef07e47419c0852339437639e : FAILURE in 29m 02s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7d781614e07b45d28890a02fd79e0f25 : FAILURE in 1h 47m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f4ecb6f7433b45708393631083d7467b : FAILURE in 16m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0962f1bb7dc39225498639c8152e74ac149ee7f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-19 09:40:53.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"cf0beed3dc7ba12b32875422712a08a408befab0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-19 10:37:26.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"55b14c94a4f67685657b47a7425b99ada6a00a70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-19 11:19:25.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2cdc513331f268eb91c02bd0d877160c6a51b258","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 12:47:32.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02b4f51a1717498a8d3b4258fb03f8f7 : SUCCESS in 4m 30s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a7d6bd7a2c9d4016961a1c59ea13cb00 : SUCCESS in 25m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9ef4a73627db4c02a98ab35cff22c1d9 : SUCCESS in 32m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f94ed2fb39214f75805a847cf37abd91 : SUCCESS in 1h 24m 55s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4a1bb6dc94e84aec81a6f126a3693211 : SUCCESS in 1h 05m 28s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/63376b34a26f407d822bab207777b838 : FAILURE in 57m 15s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b292d0108c52493ca328161117133124 : FAILURE in 1h 13m 48s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c88f089d2bbb4fb597ed91040a5416c0 : FAILURE in 39m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b0cacea2db4647cea81b197271abf8eb : FAILURE in 1h 25m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0065bff707aa45a18fb0ac745acce962 : SUCCESS in 25m 03s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"3a2e76b95ba68b2a633e9be7de31a287a4bab516","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-19 13:30:06.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/77/786177/9/check/DellEMC_VNX_Manila/df0a374/DellEMC_VNX_Manila/3 : FAILURE in 42m 16s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":9},{"id":"e2f8cefd74ab2af1338f8198160c550bbfa7ec8a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-19 15:15:43.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/77/786177/9/upstream-check/manila-cDOT-manila-tempest-plugin-ss/8b4838a/ : SUCCESS in 2h 27m 51s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/77/786177/9/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/2e61d42/ : SUCCESS in 1h 40m 25s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"8af91eeb2abb6b8aab6eb51bb3641f33163d7275","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-08-01 07:46:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":9},{"id":"0182d1ab57c6f05dcb5f5da041900e2af02481e4","tag":"autogenerated:gerrit:restore","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-08-01 07:47:26.000000000","message":"Restored","accounts_in_message":[],"_revision_number":9},{"id":"984810943a3db032e76ba289869bbdfa74b5beac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-01 07:47:41.000000000","message":"Patch Set 9: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"178e7b628ed571ea8e84c2990f740d8a0e9e9d3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-08-09 09:08:36.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"bd71c6afd8de7b341f0d7a7602b37dfddfc9bc45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-08-09 09:29:18.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"7ee74715584b9dd6571fa23a2554c95a19f0bf8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 10:13:58.000000000","message":"Patch Set 11: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad891eec402d4a12a32a1edd5bb71874 : FAILURE in 4m 42s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/397ffc5b879c47828c88fefd13b57f4d : FAILURE in 16m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/56a0a56c99f441dcb49b2ef04a0ba28a : FAILURE in 16m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6baaa33b8d9e474d81c8371691dc71b4 : FAILURE in 31m 28s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/62e8c0403c174ee298d11a9f893b8960 : FAILURE in 39m 29s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/1193707a2956402badf8904ccfc33a4e : FAILURE in 34m 24s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/87bb99c3e8594d20a8936eafacccfc45 : FAILURE in 34m 32s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9e9ee5675b314d34b961dad3ed68ef76 : FAILURE in 36m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d7352113c60a429681801cc65789618f : FAILURE in 33m 32s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/753fc7c7a4014ae290f819a1eac42310 : FAILURE in 32m 43s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/48c77df6189c40eb80604ae496769159 : FAILURE in 30m 00s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/097dc4ca89904c55b26b09c7d0abe372 : FAILURE in 33m 42s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6446c69b667a467184108329e5b71332 : FAILURE in 17m 52s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"684b6f87e9047bf9e04d8619fb7897943fcf1096","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-08-09 11:04:39.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"73919592e526acba82bf24c046ac600f1d21d261","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 12:29:40.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5902804f1d449fe942b8318b75c5a25 : SUCCESS in 4m 45s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/da4ec2df8f7e4a0ba12257d8f961b2df : FAILURE in 26m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2b29fe97023141589adb415f33d27979 : FAILURE in 26m 24s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f35c24fc9f4b44599510d3d4f9d61646 : FAILURE in 58m 13s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/6d441e2820ff4fce9ad26ca90d9a99d5 : FAILURE in 1h 03m 22s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/c50d03d2977d4c9ba2591c8d6a5a2f54 : FAILURE in 1h 09m 19s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/1dc5a763598549dd98723c94192447c9 : FAILURE in 53m 09s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b4e783ea6b6e4c7eadd1820f51d89db7 : FAILURE in 1h 04m 38s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b5da0af92efe4c34aeae98c739327697 : FAILURE in 52m 57s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/dc0b846c96f44bfdb42b55e9de51c721 : FAILURE in 25m 28s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f266f2b254d543dabdfad92bfd72d410 : FAILURE in 35m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b73f08284d154400b546edc05eb72158 : FAILURE in 1h 21m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/504cec271a9143a69450d9ae88f6c886 : FAILURE in 34m 05s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d9661abf2400418546ea4652c9c319eaf448fe44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-08-09 13:27:38.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"c9f1e06c2a0be72cad9e19d6ce25f367ea21ab9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 15:26:27.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b4f9185a4db484da6fe9823613353ee : SUCCESS in 4m 00s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2e22d48d466d42d4a002002f8f335481 : FAILURE in 23m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5d502137c08b4041bbadd4b1b19aa931 : FAILURE in 30m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/eaa91c83c09a4a37b3a4122a6e8f3f47 : FAILURE in 1h 01m 13s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/083e1c8cea7649c689cd5e178858ac5d : FAILURE in 1h 02m 01s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/86f4260173b143c3bf571db22d970b23 : FAILURE in 1h 00m 49s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/e0cdfd44932443aab8d406158af290ac : FAILURE in 1h 06m 10s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/576c380909584d8d93ff0981020f400e : FAILURE in 53m 38s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a6def07a30e54dda95b5b296d0dba143 : FAILURE in 1h 09m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/78c3799385e24cd6bb44674cd9c48714 : FAILURE in 1h 21m 42s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7d80b87022ec479787eb951387e9534f : FAILURE in 37m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/33f6b29479634242bd6a2b49ab1ab83b : FAILURE in 1h 55m 32s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3b8f38ffabae465ab0bb11c6269473ad : FAILURE in 22m 07s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"5dfa976d3f9c12c251afc7ac224c6743db0245bf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-08-09 16:34:55.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/77/786177/13/check/DellEMC_Unity_Manila_dhss_true/bc2d3f8/DellEMC_Unity_Manila_dhss_true/249 : FAILURE in 3h 06m 59s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/77/786177/13/check/DellEMC_Unity_Manila_dhss_false/e2370fe/DellEMC_Unity_Manila_dhss_false/247 : FAILURE in 2h 07m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"a1a609b577c67a5e0892529a402cd2944bdd363e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-08-10 03:40:25.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"490cd03c4ecf387aaa3fdc3e2ac4987d9472ea53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 05:22:17.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaad2272b86448d492fe0263dcc076e8 : SUCCESS in 3m 48s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8d74473e6b0446248768e13c1c5c401e : FAILURE in 24m 53s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f6e9cea09c234166ae6c09aae8768dc2 : FAILURE in 29m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/be4a9733b7464f619a291d2a4ff2e735 : FAILURE in 53m 00s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/efceeeecf7dc4822ace1a1a4a08ae34d : FAILURE in 56m 04s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/e4138f6ec3e54e118833a4f71c69e53b : FAILURE in 57m 21s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/bf15765219c1480d9d8838654402b5e4 : FAILURE in 56m 40s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/479a4d87a09b48deac95aa325b1d2b43 : FAILURE in 56m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/23e7c257d4bc43c4846878e0ecbdc987 : FAILURE in 1h 01m 30s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e08b3fba3af74cbba3c834df396e7072 : FAILURE in 1h 17m 50s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/27b6b542460c4237971f20d177b34eaf : FAILURE in 40m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/41952241b37a4e9c92ae621f0b7ad2be : FAILURE in 1h 37m 40s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2cab8e6b7f9e4aebb3ffb78fa3981efa : FAILURE in 24m 32s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"3d91617fe2c71b5f9c0bee5d836c5cf4745c0617","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-08-10 06:51:39.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/77/786177/14/check/DellEMC_Unity_Manila_dhss_true/cc94376/DellEMC_Unity_Manila_dhss_true/251 : FAILURE in 3h 11m 00s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/77/786177/14/check/DellEMC_Unity_Manila_dhss_false/47d2176/DellEMC_Unity_Manila_dhss_false/249 : FAILURE in 2h 07m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"48ddcbe74b59058b915538b7f8644bce0e3ff737","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-08-10 09:00:00.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"4f9c85b3ae562adbf72f6069aa4e42e947cd688a","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-08-10 09:49:25.000000000","message":"Patch Set 15:\n\nThanks Liron! Could you add more details on how this work? What are the benefits of having the features in a different configuration section?","accounts_in_message":[],"_revision_number":15},{"id":"390819b319d25a3fa7382d6af99e59abd1ded1c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 10:43:57.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1509764ea2914cd791d6ad2bc45d5e37 : SUCCESS in 5m 09s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/03b70b39a27c46378c979c08d7d7084f : FAILURE in 28m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/492681468f0740a2bf3df83613e5ebb2 : FAILURE in 36m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/afa895546301454dbaf47a005a75681b : FAILURE in 1h 05m 15s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/4dc85524d06c46449432c29103454d16 : FAILURE in 58m 55s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/3ee5d3323103414e8cba1f68d9b7ff25 : FAILURE in 1h 07m 05s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/027ebd8e16894e2194ffc2aafc4f0aa2 : FAILURE in 1h 37m 32s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/81a515c4767446cc86e6c6e937b11bd5 : FAILURE in 55m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/91e6e37a3f1f41d78dbef99770acf081 : SUCCESS in 1h 01m 59s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/668baeb5e7a544b1842a9f097e28f8ad : SUCCESS in 1h 14m 34s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/922de6f001fc46f2a44a8200703aa073 : SUCCESS in 45m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/19f8e16403054d01a23d06c0f08ebcd6 : FAILURE in 1h 20m 57s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5b4cf92f4a314751b24f83fbd4e2482c : SUCCESS in 27m 07s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"6c0a5859d74d0461a74905ff96f3b2a7ae41e59d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-08-10 11:43:46.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/77/786177/15/check/DellEMC_Unity_Manila_dhss_true/d60a532/DellEMC_Unity_Manila_dhss_true/252 : FAILURE in 2h 43m 32s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/77/786177/15/check/DellEMC_Unity_Manila_dhss_false/2ff81f3/DellEMC_Unity_Manila_dhss_false/250 : SUCCESS in 2h 17m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"e62dd11b050dc1e2fea681cf8f567e83a5eec032","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-08-10 11:59:37.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"13fd7641a9117e81a009ccc62bbae5e552009464","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-08-10 12:08:38.000000000","message":"Patch Set 16:\n\n\u003e Patch Set 15:\n\u003e \n\u003e Thanks Liron! Could you add more details on how this work? What are the benefits of having the features in a different configuration section?\n\nI hope it\u0027s clearer now.","accounts_in_message":[],"_revision_number":16},{"id":"450ac57c442fe2f2ae2ec0b68b2adc5aa8eec5cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-08-10 13:23:05.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"558bcbd0d6cf79dc89bc6ce856913e5db554285e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-08-10 13:25:36.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"113c557ec15c6555b96f6c9f28d5cbb53fd2aab2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-08-10 13:42:58.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/77/786177/18/check/DellEMC_Unity_Manila_dhss_true/3f36e37/DellEMC_Unity_Manila_dhss_true/255 : FAILURE in 14m 52s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/77/786177/18/check/DellEMC_Unity_Manila_dhss_false/936d6a1/DellEMC_Unity_Manila_dhss_false/253 : FAILURE in 17m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":18},{"id":"94d067c6cdff374158b4fb935c175591957a3953","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-08-10 14:39:13.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"87ea32004020770e9451167bde489fb28597085d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 16:14:53.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9edb1cf50cf45dfbee4a72241f68ab5 : SUCCESS in 5m 07s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9bdbce95e96c45bdba563e97ade8103d : SUCCESS in 32m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e7c4448789004bbd847ec1c83530715e : SUCCESS in 35m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7e2d6e5463574407a92577690438cdeb : SUCCESS in 1h 27m 23s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/b578f06aada240799ee59174b203b708 : SUCCESS in 1h 05m 25s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/121be8c86c5b4aa2b7081f7a1b5851b2 : SUCCESS in 1h 15m 41s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/68b13c3679c7433faefb8356dbebec21 : SUCCESS in 1h 17m 35s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5602425d15a44b1ea10cd93c8d228a26 : FAILURE in 1h 12m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/bd2c9f96bca44016a9f409212973abc4 : SUCCESS in 58m 04s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6ffe50d0119944c4a3e33da41fec1ad9 : FAILURE in 1h 23m 17s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7dc2a5565999487cb348acc564f60019 : SUCCESS in 41m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e267a53aece241a1b6a64b925bacd527 : FAILURE in 1h 33m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3d39f0d267d0466491dd199bf6a5ca61 : SUCCESS in 23m 54s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"339a685f3d19c47c23afa9e1fc66327bde6914c7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-10 16:15:14.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/77/786177/19/upstream-check/manila-cDOT-manila-tempest-plugin-ss/a7af39b/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/77/786177/19/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/b193ca5/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":19},{"id":"5e1fb9c19e129e4cfa9d48d2291f5ca1e07cbd68","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-08-10 18:03:32.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/77/786177/19/check/DellEMC_Unity_Manila_dhss_true/4e03443/DellEMC_Unity_Manila_dhss_true/256 : FAILURE in 3h 24m 05s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/77/786177/19/check/DellEMC_Unity_Manila_dhss_false/56150c9/DellEMC_Unity_Manila_dhss_false/254 : SUCCESS in 2h 09m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":19},{"id":"652cbcd502a941d7c989e0174d341837d099dd12","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-08-10 18:54:19.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/77/786177/19/check/DellEMC_VNX_Manila/ecaf049/DellEMC_VNX_Manila/125 : FAILURE in 2h 39m 09s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":19},{"id":"6f89fd3c205835e0a0c8ae826dc2de688a969d04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-09-21 03:05:44.000000000","message":"Patch Set 20: Patch Set 19 was rebased","accounts_in_message":[],"_revision_number":20},{"id":"3ee2227a508b2602f2c34be25674970e2e7ffef0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 04:47:50.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f06f5bae3cf446ba8ecc2b66ef5f1942 : SUCCESS in 4m 48s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e26bad79029e469990707f8dc6217a0e : SUCCESS in 43m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1f0b84ef388544ef987f67da8b2388e5 : SUCCESS in 31m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/72d254babedf447c9c01100aefaae71e : SUCCESS in 55m 08s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/ef5a8c062cfe491b870e23b1898f9990 : SUCCESS in 58m 47s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/a1f99c57db4f487387ce254790040cb1 : SUCCESS in 1h 04m 45s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/d7813b8be1e742aa8f3c663eb592caa2 : FAILURE in 1h 04m 26s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fe4cb01ef0ab4539bcb56c054aa05b12 : FAILURE in 57m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/8f41887b5b0a4754b84ee7bd4d6eeda0 : FAILURE in 54m 56s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2b8e16207b21494cb82d69e36359a941 : FAILURE in 1h 35m 01s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/689a1350b1cb49a9acaa8cc179a0a84a : SUCCESS in 44m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/035ec27cfc5a42929f810b9edc5ec25a : FAILURE in 1h 40m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c589a863ffb74b5cae17dadcb6ce5db2 : FAILURE in 22m 06s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"440114aacc32cbe7e43355b70a2869b155c2a62e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-09-21 04:50:46.000000000","message":"Patch Set 20:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/77/786177/20/check/DellEMC_Unity_Manila_dhss_true/415c614/DellEMC_Unity_Manila_dhss_true/370 : SUCCESS in 1h 44m 45s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/77/786177/20/check/DellEMC_Unity_Manila_dhss_false/744af9d/DellEMC_Unity_Manila_dhss_false/363 : SUCCESS in 1h 08m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":20},{"id":"5e01cde0827c91fa445d36d63171a6e610102acc","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-10-20 14:05:08.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"034628e85e9a63bd1b8ea54642862c73a0e87bf4","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-10-21 05:57:32.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"0ffe5e1c1c8acdfe60872a877834dd46ad6639a5","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-10-21 06:52:01.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"b13e3904894b31f048291e99148aff4ed2f0e831","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 07:52:18.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/919e9b62ce1043be82503e0547d51d37 : SUCCESS in 4m 30s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/620092cd50ce41c9bef11a12624ec0a0 : SUCCESS in 27m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2bfb78d2910c4b368f69c0cd51c72e7a : SUCCESS in 38m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9850642c818c4288a07f420883f6138c : SUCCESS in 1h 17m 58s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/177726cc65374046aafb13af9ef51715 : SUCCESS in 1h 18m 51s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/67f7ffeebbb2444b8532063b06634237 : SUCCESS in 1h 20m 29s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/3c542e86bef34a82aeb5232dad184fac : SUCCESS in 1h 10m 44s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5e0fdf5ccf764fbf9457d077f49be0c3 : FAILURE in 1h 10m 51s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/69c858b09efb434a9941c9e9be841071 : SUCCESS in 1h 16m 34s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/aa565995753d4669899a7f30889a1ff1 : FAILURE in 1h 31m 04s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/25bb35204ed240b48841b1183ad4b423 : SUCCESS in 43m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5b9e937317ad47389365493861be77ad : FAILURE in 1h 50m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2c75da2d54eb461f9c3971a5b80cfbe9 : SUCCESS in 23m 39s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"63d2b83d0caf9438533f2823e6675a1cd1c6ee8f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-10-21 07:52:39.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/77/786177/20/upstream-check/manila-cDOT-manila-tempest-plugin-ss/15a09c9/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/77/786177/20/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/b548640/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":20},{"id":"4f9bc19a72819d7420ac8c09c8a497e18c97e873","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-10-21 08:25:34.000000000","message":"Patch Set 20:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/77/786177/20/check/DellEMC_VNX_Manila/c399a99/DellEMC_VNX_Manila/13 : FAILURE in 33m 02s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":20},{"id":"7f0c44ac5c50b79905cb1da44854d94655ebaee0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-21 08:42:36.000000000","message":"Patch Set 20:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/77/786177/20/check/DellEMC_Unity_Manila_dhss_true/e40160b/DellEMC_Unity_Manila_dhss_true/29 : FAILURE in 50m 03s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/77/786177/20/check/DellEMC_Unity_Manila_dhss_false/ec8b6b0/DellEMC_Unity_Manila_dhss_false/32 : FAILURE in 14m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":20},{"id":"c0545fc195322b50f9c621317f040e9c4f6adb3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-02-17 12:18:37.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"a1d4559f1573792b90b87c95bcda41bc72c390f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-02-17 13:13:57.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"54e3892bac5a6283b4826946d5aa2beb946860d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 14:39:30.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27789d98802f45c7a4e4241a9006cb7b : SUCCESS in 3m 26s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/63ec09bbe7cb439d88c0b181c837c51d : SUCCESS in 24m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/965eda42b1f64ea6986d3613d57f187a : SUCCESS in 35m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3562fa98565c445aaa1bf4bfc0996577 : SUCCESS in 1h 01m 15s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/11e9000fb69a4f048c38f3af2cb4fce9 : SUCCESS in 1h 02m 22s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/8046708dd4e0424f80721b1e18d53384 : SUCCESS in 55m 21s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/31522920bf1749ea9fb5c82c8cef30fa : SUCCESS in 1h 04m 09s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/90cd3ca487a540079420e2574c881222 : SUCCESS in 53m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/72b8794757a84601b2cdb94e33c653b8 : FAILURE in 1h 11m 25s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/bffac3ebad38462088196b2b84afbcce : FAILURE in 1h 19m 34s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bff59add8a8c44bda4dbc98fc5f560d1 : FAILURE in 45m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/cf1fb9ad1dc3427bb0493087f2396f9c : FAILURE in 1h 18m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6501d255c2034bc195d05d0778ccd7dc : FAILURE in 21m 42s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"39daf337019e4680b1f6756426122fd58c6053cc","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-02-17 15:36:13.000000000","message":"Patch Set 22:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/77/786177/22/check/DellEMC_VNX_Manila/7bef17e/DellEMC_VNX_Manila/151 : FAILURE in 56m 29s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":22},{"id":"1c260412b50a968fd72113fa3a7b5767e76dcaf8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-02-17 16:49:50.000000000","message":"Patch Set 22:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/77/786177/22/check/DellEMC_Unity_Manila_dhss_true/859eb52/DellEMC_Unity_Manila_dhss_true/285 : SUCCESS in 3h 35m 35s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/77/786177/22/check/DellEMC_Unity_Manila_dhss_false/b056120/DellEMC_Unity_Manila_dhss_false/293 : FAILURE in 17m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":22},{"id":"751a0e8b363edf970f3c0dea9a0b82bac96a71e0","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-17 17:37:47.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/77/786177/22/upstream-check/manila-cDOT-manila-tempest-plugin-ss/001ba3a/ : POST_FAILURE in 2h 57m 59s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/77/786177/22/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/02ff635/ : POST_FAILURE in 1h 48m 06s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":22},{"id":"512a70a5f796ed9c14678410483d30023e0ca2bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-04-06 09:48:13.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"84d6643ba618923571c585c90c1f8023a237e531","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 12:22:47.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/735874b98bea41878c44f885f135cbcf : SUCCESS in 3m 19s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/342c7d484073428284cf04d2e818a4b1 : SUCCESS in 41m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/72995dc1e9264f5496100b411cfcce9e : SUCCESS in 41m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4bfc30c59a45425dafaffcc6c345f17d : SUCCESS in 1h 31m 38s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/1fdd9d433e534141b65f7b812047f6d7 : SUCCESS in 1h 17m 34s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/3c136742770f4f82ad64eb72065ff40f : SUCCESS in 1h 05m 28s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/48feada402cb4d7a881c5cace326d102 : SUCCESS in 1h 04m 52s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0dcf230ce9274ec89ab95e30a4c32665 : SUCCESS in 1h 27m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/8a84a3130d234cf2b7935869dcab2cae : SUCCESS in 1h 07m 50s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2843e287f4ea46ad9244c7725ab71129 : FAILURE in 1h 46m 11s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/962f3cedf09e43048cacb1c0e59164c4 : FAILURE in 52m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4ff35647f6364a2ab53a5f0577401c68 : FAILURE in 2h 26m 41s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/cab36285780a45eba612c38d5d5d9566 : FAILURE in 29m 45s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"2b4362a68c4501d394e549f989fc9ff369caa0ac","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-04-06 12:26:46.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/77/786177/23/upstream-check/manila-cDOT-manila-tempest-plugin-ss/b85542f/ : FAILURE in 2m 53s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/77/786177/23/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/dee8aa4/ : FAILURE in 2m 47s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":23},{"id":"5c8fd77bd98e8aba77c0fb9d13d4d5e75f1a500b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-07-04 10:03:13.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"0e74fe912cea211e4bb00ca1947be0ee02cdcc00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 12:43:35.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c84f1cb1f87147b989a90057868c6a4b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27f6dc45aa61439393cd1e340c02bba1 : SUCCESS in 3m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/df14a88701d94faa89fd72f10182bd18 : SUCCESS in 1h 12m 40s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5289400bcbed43c3b8660c9831156a5e : SUCCESS in 2h 34m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7b4d6f71924945baabb85e83470dde52 : SUCCESS in 1h 05m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1bb98f7f0b9143aca86257c9369315d1 : SUCCESS in 1h 25m 34s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2544669815a94fa69083f3ab61f8ba3c : FAILURE in 1h 45m 29s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4ef582e386b8450aa0099eaf9ab70f38 : SUCCESS in 59m 46s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b8388c137d874763939789f291c86d87 : SUCCESS in 35m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/79eb9e19391d4f0d82b2a3a4b8ff5d14 : SUCCESS in 29m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d32c5ff9ebf940868775405eff7204d0 : FAILURE in 42m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/94baab4290e347bb98d89e3dae0f1d74 : FAILURE in 26m 53s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"953212a720061c9fe2e48969e0f3d81e8d7da271","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-04 14:26:18.000000000","message":"Patch Set 24:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/77/786177/24/check/DellEMC_Unity_Manila_dhss_true/a637720/DellEMC_Unity_Manila_dhss_true/601 : FAILURE in 3h 57m 15s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/77/786177/24/check/DellEMC_Unity_Manila_dhss_false/daf3553/DellEMC_Unity_Manila_dhss_false/628 : SUCCESS in 1h 00m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":24},{"id":"6c0a9c577276a0d390f389c382f6f11d810ecbac","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-04 15:05:51.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/77/786177/24/upstream-check/manila-tempest-plugin-ontap-dhss/30ed401/ : SUCCESS in 2h 20m 48s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/77/786177/24/upstream-check/manila-tempest-plugin-ontap-no-dhss/192dc0e/ : SUCCESS in 1h 12m 14s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":24},{"id":"fcbcf532e08a3d7208b71c02bfe185af5a80839c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-03-10 12:29:52.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"2a8b181f348c5a5e70ba28102d6fcea7242cf3d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-10 15:42:51.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/503f018cab9949b599e2c955a65016e3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f9485d6fd424cf9ba5cccfa46e5136a : SUCCESS in 2m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b9fb2e048dcf4803892c661a7499d93e : SUCCESS in 1h 22m 10s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d1a66de9cdaf463c907cf007f2ca62a6 : SUCCESS in 2h 03m 07s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/500d78f240f4427aaab0ae83c8fe7784 : TIMED_OUT in 3h 03m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d9cd67b549b44dd8b86543a191e9166c : SUCCESS in 45m 05s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d1a3d8e4de7d43ecbdd3e927ca458762 : SUCCESS in 1h 11m 31s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c2197d0becc042d1b94e828767c96755 : SUCCESS in 1h 21m 25s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6cd94e9f2ac74a8f8d7d6c961541e676 : SUCCESS in 26m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/71042fbe57ef4930ab4d96011a244446 : SUCCESS in 27m 53s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/4c98be9c6a8d427f82812a9a6d10b6f7 : SUCCESS in 59m 36s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/ecd98f40034b41a88e562d0f45b52969 : SUCCESS in 1h 00m 06s\n- manila-tempest-plugin-lvm-zed https://zuul.opendev.org/t/openstack/build/bb6a508714574480b2b6753edec4dcea : SUCCESS in 1h 20m 47s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3d787be2eef7480192997d2de3365848 : SUCCESS in 23m 36s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/87255c26ba3a4f8494892e3a538f2569 : FAILURE in 35m 47s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ad46d6f2719a47e584c2251770844f7b : FAILURE in 10m 46s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"ad83355666b883a22daedf67bcf932a21d7d87b3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-10 17:58:37.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/77/786177/25/upstream-check/manila-tempest-plugin-ontap-dhss/b10a6a2/ : FAILURE in 1h 16m 56s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/77/786177/25/upstream-check/manila-tempest-plugin-ontap-no-dhss/e397352/ : SUCCESS in 2h 09m 32s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":25},{"id":"e03ac840f42026d224a6edffd2e9fa57703cf28e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-06-27 23:46:12.000000000","message":"Patch Set 25:\n\n(4 comments)","accounts_in_message":[],"_revision_number":25},{"id":"917a1ba5e8bfdd1431be5fa1f0ceffebe0f88626","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-11-05 19:04:50.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"3a1875d40b93dda4ad56d4caa2525fd0ddaa9c87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-11-07 09:10:43.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"16ee9d1c6f39da4bc4fd20f9522c0dc8e2621709","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-11-07 09:25:01.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"d891e0c838298bd3b29cd9128e534532662f4067","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-11-07 09:27:25.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"7e07da17d76308db7106e999b29fffd3d2825a09","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-11-07 09:27:44.000000000","message":"Patch Set 26:\n\n(3 comments)","accounts_in_message":[],"_revision_number":26},{"id":"e3a1f7e1e02214ac8f6f68d26453102a4f84936d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-07 09:55:55.000000000","message":"Patch Set 27:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/77/786177/27/check/DellEMC_PowerFlex_Manila_dhss_false/a7cc32f/DellEMC_PowerFlex_Manila_dhss_false/1985 : FAILURE in 28m 11s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":27},{"id":"84e387cef4607984519c441afa4807aee2f61364","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-07 09:59:12.000000000","message":"Patch Set 27:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/77/786177/27/check/DellEMC_Unity_Manila_dhss_true/17833c4/DellEMC_Unity_Manila_dhss_true/1968 : FAILURE in 13m 50s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/77/786177/27/check/DellEMC_Unity_Manila_dhss_false/4e7cd9a/DellEMC_Unity_Manila_dhss_false/2072 : FAILURE in 31m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":27},{"id":"c58884a0c538f2984fc9ab436541ca70f490cc9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-11-07 11:12:32.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"d1e22edd0e6cc27194665708dec8cf9bf617604d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-07 11:44:50.000000000","message":"Patch Set 28:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/77/786177/28/check/DellEMC_PowerFlex_Manila_dhss_false/08e5ca2/DellEMC_PowerFlex_Manila_dhss_false/1986 : FAILURE in 32m 03s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":28},{"id":"2d009af2c30f75c806dd7e360619004de8cbd867","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-07 11:46:03.000000000","message":"Patch Set 28:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/77/786177/28/check/DellEMC_Unity_Manila_dhss_true/bff011d/DellEMC_Unity_Manila_dhss_true/1969 : FAILURE in 13m 36s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/77/786177/28/check/DellEMC_Unity_Manila_dhss_false/c648168/DellEMC_Unity_Manila_dhss_false/2073 : FAILURE in 33m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":28},{"id":"081b0e98960a8d17a0b8f8f6547c4e91ed96058d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 13:54:30.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c243286b365a47e7988fc01ddcfe1b27\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bad9b8e9e4ff4ea3b01ae4380991d496 : SUCCESS in 2m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5742a5ef37114805a0a4f4d319c2edc5 : SUCCESS in 1h 05m 54s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2fefd17c82dd41d4ac6062576a506540 : FAILURE in 2h 40m 58s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/cd289c080aec44fcad48b2fece9008a5 : FAILURE in 1h 54m 49s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/72428b61d64d4fb8a807761a97388097 : SUCCESS in 54m 46s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d6e2cbb8917f4ea4aa36ffd93b501556 : FAILURE in 57m 03s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/dee710ac48334e47b51dee5ff7395b84 : SUCCESS in 1h 10m 46s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/788e171f05054ebdb8132df393030991 : SUCCESS in 40m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7081f44bf2524e6fab242e5a665eb29c : SUCCESS in 40m 08s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/37d6f7a907d642508eb9e17879b5b6eb : FAILURE in 48m 05s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/343690fa81be462f962e3a36bf00c4a4 : SUCCESS in 54m 15s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/64f269f666064c2cb0a9489da0d9855d : SUCCESS in 1h 02m 27s\n- manila-tempest-plugin-lvm-2023.1 https://zuul.opendev.org/t/openstack/build/fc15cd05f352413e878bebddbf2f43de : SUCCESS in 1h 20m 34s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d15cbcbd051c4fb2b9a593a2d0683a35 : SUCCESS in 22m 33s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2c296624dd0544569041d13b9e807efb : FAILURE in 28m 28s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"21e755db190b000906ebe829d8d26663cad1f445","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-12-09 13:01:31.000000000","message":"Uploaded patch set 29: Patch Set 28 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":29},{"id":"fd27d421e0589e7979feec6d2e8655b62df4d611","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-09 13:34:45.000000000","message":"Patch Set 29:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/77/786177/29/check/DellEMC_PowerFlex_Manila_dhss_false/1beea5c/DellEMC_PowerFlex_Manila_dhss_false/2197 : FAILURE in 32m 58s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":29},{"id":"c4b5377c0257091970fc8f8cc65a608ce6d80c48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 16:20:09.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ab391461e8049aca71f11b79bb730f5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a6bcb4b16a447dcaa6263a7f6e78139 : SUCCESS in 3m 33s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/86d398c7edf0443a9cebad2fd8eb8f00 : SUCCESS in 1h 14m 54s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/a69130f0156a4dafbfec5ee1cbd0d432 : SUCCESS in 1h 16m 25s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5649fbe7fdf442ab9a878613f4a25b6d : SUCCESS in 1h 19m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/405b87ad85024f22a66f2d50527e980a : FAILURE in 1h 08m 55s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/97b4ddbaa0114fd898f25cc2a0af229d : SUCCESS in 42m 11s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/7de59aa6642f42958d47591b3b309eef : FAILURE in 10m 31s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1c3df23de77245e68ef4592c8c0ea614 : SUCCESS in 1h 20m 15s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/48391685784348fb919ea9b184e26e5f : SUCCESS in 37m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5c239dea0d4d48579b7d3b9708ca533c : SUCCESS in 39m 06s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/5c6aada4cf5d45909c2da793b9b8ee26 : SUCCESS in 1h 07m 25s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/8d5adae025f9414bbfef74664f634966 : SUCCESS in 1h 28m 46s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/8d589d1175cf49b69afdc544140a94c9 : SUCCESS in 1h 15m 55s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/0f10117582b14da6b98235b8e2804b0a : SUCCESS in 29m 42s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5efdef7874a743f7a346f43891408ef4 : FAILURE in 1h 07m 55s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"ef21e03f5cc78e522912a7da1f99d68a14e747fe","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-12-09 16:31:57.000000000","message":"Patch Set 29:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss finger://managesf.services.openstack.netapp.com:7979/423eafa2dea14a6bb90a6910844e4a07 : RETRY_LIMIT in 10m 54s\n- manila-tempest-plugin-ontap-no-dhss finger://managesf.services.openstack.netapp.com:7979/851786cbe09f4e49bd6aa5ffdc8b1053 : RETRY_LIMIT in 10m 46s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":29},{"id":"106308f252f57bdf2b53786998cdc026f9328f8a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-09 16:54:53.000000000","message":"Patch Set 29:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/77/786177/29/check/DellEMC_PowerFlex_Manila_dhss_false/7cfecff/DellEMC_PowerFlex_Manila_dhss_false/2198 : FAILURE in 34m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":29},{"id":"ffaf61edb887813e600f25d41877f777db606cb8","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-03-26 19:49:30.000000000","message":"Patch Set 29:\n\n(4 comments)","accounts_in_message":[],"_revision_number":29},{"id":"519a62507709ecca112045d4b448033500c3d501","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-03-27 09:05:17.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":30},{"id":"3632ee02537024448207c86edc5bfaf751ab3a50","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-03-27 09:07:35.000000000","message":"Patch Set 29:\n\n(4 comments)","accounts_in_message":[],"_revision_number":29},{"id":"6f48e272245593db30a885694befe343303030b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 11:40:55.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b51a407ad2d4bac8bc749f7d32ab823\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fb6a2f96c56481cbfb32b9a45bfb8e8 : SUCCESS in 2m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b96eb080414b4f948613b92f6698ad1b : FAILURE in 34m 30s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/7eb0593f46994ea3b187ef8490d32015 : SUCCESS in 1h 13m 06s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c4bbf7c414644b0e8f535cfb2d6beca2 : SUCCESS in 2h 21m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/382d8b4065f14415bbd67e06995f4673 : FAILURE in 1h 03m 38s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ca8f2635f78b4caab7be32d2f6542dcc : SUCCESS in 58m 13s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b8d6abf90105429d9ca2a1d02d55225e : FAILURE in 1h 19m 57s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f11a18f8738348a6a73a404856350e6c : SUCCESS in 1h 16m 27s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ff7cbf84216c4cce89213fff130ed5b8 : SUCCESS in 25m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b57114ff29b843f9a0e8d1f0217e0d08 : SUCCESS in 31m 39s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/3ef7e2da2b0849648afeddde8c4fb07c : SUCCESS in 1h 03m 24s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/4445079d434e46158e9eefab4733d3b7 : FAILURE in 22m 52s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/5f2d521f99aa4c7da232d09da659dd67 : SUCCESS in 1h 01m 56s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c621bdba8bd54afd9c20d0d40e6bcab5 : SUCCESS in 28m 32s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/07d857cb9b9347bb8f158cb6468220f9 : FAILURE in 40m 53s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"bfbb38e7b5ac9cefb27a376210d2993e79b5c838","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-03-27 11:44:28.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"441237452f1ae54d3d7c525076b632ab24863698","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-03-27 11:58:44.000000000","message":"Patch Set 30:\n\n(2 comments)","accounts_in_message":[],"_revision_number":30},{"id":"b504682ccc45a2bf97329626c012ef90e73b3ee9","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-03-27 11:58:52.000000000","message":"Patch Set 30: Code-Review+1","accounts_in_message":[],"_revision_number":30},{"id":"e4a74ff5a79bca5f6fb7b4cb586ee01d2b386a0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 13:34:48.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aacb0b24fd56447db65ef350c4d2c85b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f31f76774af47f8925bf68eb7ea6242 : SUCCESS in 2m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/444471da74a64f3b9c859e14ffffd35a : SUCCESS in 43m 18s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/fe293ca849fb4f4d86e45c7966f7178d : SUCCESS in 38m 01s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/20d9cb5b2b3c4706b179e79dfd451cc1 : SUCCESS in 1h 02m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c2e545dd50004e34a5db8e7e50652d4d : FAILURE in 51m 36s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e41d1a3989524bd39dbef09e556781c6 : SUCCESS in 49m 54s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/64ed488aa5fe436687bf5f7ef340d372 : FAILURE in 1h 30m 25s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5edce85e1c8f4ee98d9e8eeb8475d3cc : SUCCESS in 1h 19m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0d4499cd3841447699b46d58aab68429 : SUCCESS in 30m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/339fa6a1338a4fe2913db77b50f4984a : SUCCESS in 32m 02s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/a93f250b7ebb4452864ec8e7f2ffaf6b : SUCCESS in 1h 08m 20s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/42e866c5ef61469ba206eb48c46ddfe6 : SUCCESS in 1h 23m 42s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/37ac9a72005d479ba96617417f340ccb : SUCCESS in 1h 19m 06s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/fd32dd615d204f17a873f25aa0140bb1 : SUCCESS in 29m 22s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/56569cbf450a4da7b03c3e3e46e5c32e : FAILURE in 45m 07s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"48e751b43e085ff8a489f367638851183cfdc358","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-03-27 13:51:58.000000000","message":"Patch Set 30: Code-Review+2","accounts_in_message":[],"_revision_number":30},{"id":"8591fa9007058f32e37c931931c79c5b95c1cda0","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-27 16:36:12.000000000","message":"Patch Set 30:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":30},{"id":"185787b8b4f5fcefdf1404c90b3013cc0128250c","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-04-22 09:17:50.000000000","message":"Patch Set 30: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":30},{"id":"5854ab4306cd92dd3f79e9bc8598f13c8ffea613","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-04-22 10:11:07.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":31},{"id":"6300a0436df9bafce20ef6fabc184c30585bdd4d","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-04-22 10:11:16.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"d3e6954c59e8328db0787d77cdafde263ac52d38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 11:30:36.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4abeec18c84e495ba747e4e912f379ff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57900870589e45a39a71942c35407f1e : SUCCESS in 2m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/49741c3f172b4f1486215931bac4999f : SUCCESS in 1h 14m 06s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5d90f71237f449278cc35913bb2c9563 : SUCCESS in 1h 02m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f0f879684f7e42fb8d970ab021aa8c10 : FAILURE in 56m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f7e05c7c1b7040d1a8518b42771c3655 : FAILURE in 26m 57s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8424adc6ca6f423d960a4bd923c04697 : FAILURE in 1h 00m 33s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b49c1812ba42445994a65a6de3347ded : SUCCESS in 54m 24s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4573a361ee484f0d82b08c9f4257015b : SUCCESS in 26m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/49cbbd2470e246658f8ba14d028f4740 : SUCCESS in 21m 42s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/3f5035a3aff8487bb9beab73ec31e2a9 : SUCCESS in 1h 00m 44s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/0bd44ae89d9e4524934beee0c120c5a2 : SUCCESS in 1h 06m 23s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/84ff28a069ca4cb299119941a27999ba : SUCCESS in 1h 07m 21s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/c4960474461c4cf9827438100a2144a7 : SUCCESS in 1h 01m 11s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/8974ed73f71b425987ec7064a23f5b88 : SUCCESS in 27m 49s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fecef458a9d64848be7784b9589db16e : FAILURE in 46m 28s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"8a65c29fa0d224a49e244e0ff1e06c196d8aadf7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-22 14:32:00.000000000","message":"Patch Set 31:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":31},{"id":"6d8f506158af54b4bf9c933e7dbb81d0da0b1a3e","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-05-01 11:29:36.000000000","message":"Patch Set 31: Code-Review+2","accounts_in_message":[],"_revision_number":31},{"id":"e79db640c57f8fc12a5c533926004ead37d01462","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-01-06 15:46:01.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":32},{"id":"347200959ea50c2d833b3e1295ab4fe9828b03ce","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-01-06 15:46:30.000000000","message":"Patch Set 32:\n\n(2 comments)","accounts_in_message":[],"_revision_number":32},{"id":"eab7083575ff25c7a436b0dd7bc635217e9b72a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 17:20:36.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63dcccbfe5fd4fbbb2d11cd37ffdc391\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e39c9e1c36747c7abd7e022ece9e551 : SUCCESS in 3m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/62bb27ab34bb41a39787a99aa4ce6b7c : SUCCESS in 1h 08m 47s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/eddb3eb0f6284f30910a5aec0a6dcc43 : SUCCESS in 1h 32m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/80059e49776f4482bd3441584a7d1e00 : FAILURE in 1h 07m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a120831e226e4ae6a6d0ebe86925c3ac : SUCCESS in 56m 45s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d87c0381af714b18856973d975bd267f : FAILURE in 42m 36s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fbb5f897f0fa4c91be57ceb4eae96c9f : SUCCESS in 1h 17m 16s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2b0c306411474e399c9c92dca0590c55 : SUCCESS in 35m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e5db0d2f758c450fabe52bab4117be6d : SUCCESS in 30m 19s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/12533790a2ab45dbb1edeac7ec106e45 : SUCCESS in 56m 40s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/a1326480cfbb4e78bdce4cb601972113 : SUCCESS in 53m 06s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/707a7d62add0432595b4ecceaac58f35 : SUCCESS in 50m 15s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/6bd4f8d1e4c544f2aadada7aa8e208b0 : SUCCESS in 30m 14s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fafe1b2acf1d4e1495d1381a96947f3a : FAILURE in 47m 52s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"3ca40f0ea57fbae0927ebfcf8e4f13c8fd369fa5","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2026-01-07 14:30:50.000000000","message":"Patch Set 32: Code-Review+2","accounts_in_message":[],"_revision_number":32},{"id":"619ebd4e500384b8fdfa10ba3bde048bc4d28792","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-07 20:23:51.000000000","message":"Patch Set 32: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":32},{"id":"463f4d5efecbcbbe979555db5d2f9051ed367a79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2026-01-08 10:53:04.000000000","message":"Uploaded patch set 33.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":33},{"id":"b0c26f6f5ebc08e3289f5885452429888b1d843c","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2026-01-08 10:56:39.000000000","message":"Patch Set 32:\n\n(6 comments)","accounts_in_message":[],"_revision_number":32},{"id":"9bd76bd35c4f34e17172dcfebfe944d07ee9f22c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2026-01-08 11:59:30.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"b3ea161fd6a9ed31efb1a02535513f12aa71be4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 13:59:58.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/287e92e8c17f48359449a01e8073cd34\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1147530db0664f139618dface394a0de : SUCCESS in 2m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/580bc6a17e094ed9a17e901ac84483ae : FAILURE in 38m 42s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/dbda6796bc714227803621ec47870ec6 : FAILURE in 1h 53m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c47a75cbc2004c1984c23ee1e57df1e8 : FAILURE in 1h 15m 38s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b69388314b854584b9bc045090eb9137 : FAILURE in 36m 48s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/39c6abbfd08647d1978df57bdc424220 : FAILURE in 1h 10m 33s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/68f5133cc3274c8db3d3bbbeac86965c : FAILURE in 1h 02m 53s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6474ca843c8c499e99c878cdd646db45 : FAILURE in 29m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a200e09adedc443791d58b05bb64359c : FAILURE in 35m 39s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/5ad237f829e5496e962605d87679af28 : FAILURE in 1h 02m 04s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/309f16d1d8f44e0b8a00502cc61a322e : FAILURE in 43m 39s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/84c7518df48848cd9c3fb9c13157338a : FAILURE in 47m 15s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ef1a7d6f207949d0a64d2943ad12b44e : SUCCESS in 30m 24s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2f0293ce9c504a829e356d974754c38e : FAILURE in 57m 19s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"5be2e40891f9f9449324788df56f81321f5194f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2026-01-08 14:10:32.000000000","message":"Uploaded patch set 35.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":35},{"id":"1c3c13df6bb15b520380af79bab33a8fe5dd2163","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 16:31:33.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2ce889b04514c7c88ab2687b30c487c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a0abd8938a7463997cfbdd6efc5c5ac : SUCCESS in 3m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bad863942f09424f99fa44e68f2c8968 : FAILURE in 57m 47s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/fb7a523738184b9db7ba07d00eeef542 : SUCCESS in 2h 01m 56s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b5ea225f23eb409fb27aa1c4a6ca289e : POST_FAILURE in 1h 13m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d05b05dc8b6846bfb9905c78b4a9d590 : FAILURE in 24m 30s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/69933527ec404091a7ce0cdc3d10d767 : FAILURE in 44m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/37cbd352e99c40558ff99e28d38f6c81 : POST_FAILURE in 1h 05m 16s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6770a0fd7ae54192b218937405ced48e : FAILURE in 19m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2d230fb945884147b57ec556a3f9883c : FAILURE in 30m 09s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/3e53ce2d388b4a08955762aceeab2630 : FAILURE in 51m 39s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/5a926bb99f684bad9f7aa3a16de6e67b : FAILURE in 42m 10s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/92fb0276b20b4147996b2f4159ae6ab7 : FAILURE in 53m 36s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/afebb0c3488a4cb19beea68d230f8c83 : SUCCESS in 29m 01s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/240612f7e51344c28d2a0a8be71813fc : FAILURE in 1h 04m 42s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"7716a312d9f3bd0d4cc7716d30eea2117a4f48eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2026-01-11 13:16:12.000000000","message":"Uploaded patch set 36.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":36},{"id":"9d1d3ecfc82dadd1b7b2a1633d4151df4bd93595","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-11 15:03:02.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8fa63ac0f3744b848535f5681e6937e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb1276ed369f491595bce3fcc74e107f : SUCCESS in 2m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8e7467ca4d87487cbf8a6a22a9625289 : FAILURE in 44m 42s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/46b47fe893564de7a9d7d1ca25011413 : FAILURE in 1h 42m 42s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a410b8a35d884d2382d1e60235dfc1d4 : FAILURE in 51m 16s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6aa1bc0156ce48bf94d255c729670e9e : SUCCESS in 52m 15s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ab732da81c2440708c2620a732d87dc3 : SUCCESS in 1h 07m 16s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cb1285a5944f41cab7535a9e479d4193 : FAILURE in 1h 01m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cb206e3eaf13406791a952e86a770a1f : FAILURE in 35m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1af8b174b239492981acda6411728a91 : FAILURE in 39m 40s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/d0ad432e10b541248b33057399f8cbb2 : FAILURE in 1h 02m 56s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/e768fbaa365847b9b402e740bd1311d0 : FAILURE in 59m 37s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/afe6a7060ef64f948d238d638dc1b03e : FAILURE in 41m 11s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/cf90bc0eac8b43239e99b96f498fdbdb : SUCCESS in 28m 42s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c0f72d21260441b1936c1dd0baa60a56 : FAILURE in 56m 21s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"321d3e8a52871e06e93428cfd7d972330f9e7692","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2026-01-14 10:35:39.000000000","message":"Uploaded patch set 37.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":37},{"id":"7c886e18cf26d9ece1814e2fc8b13152fd222f3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 13:09:50.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6210b4ad26d341d6a311e2475d08d420\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e7d93a5f5d5446a848f3e6e787780af : SUCCESS in 3m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6501974b4faf409a8846ee21ceec6aa3 : SUCCESS in 1h 11m 37s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/0535e18393cf4f80bdea61186b55ba8a : FAILURE in 2h 28m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9d856ea1ad754327a73aa270ed65be2f : FAILURE in 47m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8dcda635b986455abad870bb410fe424 : SUCCESS in 26m 21s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/7a387c3ef7d64ff28b1137c9a6076938 : SUCCESS in 45m 59s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/19569eed190449adae28ecf7f54994e3 : FAILURE in 1h 15m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8a72c6a82faa4465be4c1d90c0eca4fb : SUCCESS in 30m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b96ecf293911405eafa36ffddee6a762 : SUCCESS in 37m 20s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/4c4a8a0273564f01b40a76cabd5bc082 : SUCCESS in 39m 56s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/8e4be9f3daff4cdb8445f57b6da6c7f3 : SUCCESS in 1h 13m 21s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/cf57d1cdd0a54ea8903fb86c1168bdb2 : SUCCESS in 1h 10m 34s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/e9c5e1b6ef924bf88e01e7bcdaa0deed : SUCCESS in 35m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f086f5388b9b404d93a1a68d470346ad : FAILURE in 46m 03s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"024604b99cd57a00905a05995374e8252ff531d3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-29 18:58:59.000000000","message":"Patch Set 37:\n\n(43 comments)","accounts_in_message":[],"_revision_number":37},{"id":"95e63800cbed39142950310ce21ec892ffae66bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-29 19:10:29.000000000","message":"Uploaded patch set 38.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":38},{"id":"ec4efdd9bdb7076af3f224b1c1e342a47df6110f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-29 19:11:27.000000000","message":"Patch Set 37:\n\n(3 comments)","accounts_in_message":[],"_revision_number":37},{"id":"95c3c447cffd8206c79b178c4f90c33b8a5eb98c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-29 19:12:22.000000000","message":"Patch Set 38: Code-Review+1\n\n(43 comments)","accounts_in_message":[],"_revision_number":38},{"id":"7735be54ee83b37de6d73c19628a2b7e9104a0eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 20:52:03.000000000","message":"Patch Set 38: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b8581d3cfb347c9a92bc43a1c3d2819\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9702e049fe774e768aa74d05645dcb0d : FAILURE in 2m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8c4391a2ef1f4b1ba72b2ab1b436a82c : SUCCESS in 1h 16m 09s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5967d3e51b4647ef8f9f59b24718e06b : SUCCESS in 1h 28m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bb101aa67bcf4045982096e7a4d01e9b : FAILURE in 50m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6fae45c5e97d4aaa94e0ec135b161aee : SUCCESS in 1h 01m 58s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/1d534fd0181c4a5a8192453d5317b503 : SUCCESS in 1h 35m 55s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/747faf0945614af0a1de48bd5fa8f4a4 : FAILURE in 53m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ca48361744e74ba9a6ae92a070fda003 : SUCCESS in 34m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fe763f9a9fcc486da0a7d12a40e7e2a5 : SUCCESS in 43m 07s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/d0e2c47ddbb642f2bd2b39cfb6714f44 : SUCCESS in 1h 15m 06s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/79ddc9f6fb314c7583ef1f2f91be0cfa : SUCCESS in 48m 34s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/29ef7e554e5641f29c944dc145baab85 : SUCCESS in 1h 10m 47s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/533f67a0bdc342bb8d1d1e66607f200b : SUCCESS in 29m 33s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/34c2d95506f541f1af5393801a85b8cf : FAILURE in 59m 08s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"fbb5919e58f6ee74efaf5381fbce64272b147e79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-29 21:57:53.000000000","message":"Uploaded patch set 39.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":39},{"id":"a654a073a030c705b689ef493b1b8c689d1407f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 00:09:29.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30c9e24b48e045f8bbf328cfa20f0600\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/957ebd5ecbd44bdab6eff17421b2f2a1 : SUCCESS in 2m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/eceec45789e34746b99333280ce50898 : SUCCESS in 40m 14s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/385b93b15ba246369967078a572edcf3 : SUCCESS in 2h 05m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/663c3a9ac85a47de8b795d1eabe690bb : FAILURE in 54m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d1963d8664794a1abd2bbbda678e053a : SUCCESS in 1h 10m 51s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a945e960951d4c99b2e8cf7f592b8c72 : SUCCESS in 39m 29s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/49cd61ca740f492f9995b814134edbbb : FAILURE in 1h 17m 39s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/662f9c43f8334766ab951f6d38568fbb : SUCCESS in 29m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7f88d81fc2374682bf80f3edf4f79660 : SUCCESS in 42m 29s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/0946c74718934019b248644a92d5809b : SUCCESS in 35m 31s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/886dc770042d4b92aae884bd81742dc6 : SUCCESS in 1h 08m 29s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/a70aeb864ea2480cba349351093d01de : SUCCESS in 1h 11m 30s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/b8459ae97af34273bdd3e27498043739 : SUCCESS in 20m 56s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9b37023816da4f95ae1358b250753294 : FAILURE in 1h 30m 22s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"337f4b79549e2b0c78965380fc1c9bc378239fbb","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2026-02-05 16:34:21.000000000","message":"Patch Set 39: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":39}],"current_revision_number":39,"current_revision":"00026c86fa3e3c415b5aef309a5cdb11ebd14bf0","revisions":{"2cb85fd63f9866b96ee5ecbe92a21622d55837e2":{"kind":"REWORK","_number":1,"created":"2021-04-14 08:39:58.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/1"}}},"commit":{"parents":[{"commit":"f083c71a9396a5a6f3300d3150b7fc1bee2c57c5","subject":"Add create share from snapshot in CephFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f083c71a9396a5a6f3300d3150b7fc1bee2c57c5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:39:38.000000000","tz":180},"subject":"[WIP] Add new configuration section for share features","message":"[WIP] Add new configuration section for share features\n\nLike other openstack services, let\u0027s add a new section in\ntempest.conf for manila features called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2cb85fd63f9866b96ee5ecbe92a21622d55837e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2cb85fd63f9866b96ee5ecbe92a21622d55837e2"}]},"branch":"refs/heads/master"},"44824aaa8a435138b5406d31261f8eedfa617fd7":{"kind":"REWORK","_number":2,"created":"2021-04-14 08:52:23.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/2"}}},"commit":{"parents":[{"commit":"f083c71a9396a5a6f3300d3150b7fc1bee2c57c5","subject":"Add create share from snapshot in CephFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f083c71a9396a5a6f3300d3150b7fc1bee2c57c5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:52:03.000000000","tz":180},"subject":"[WIP] Add new configuration section for share features","message":"[WIP] Add new configuration section for share features\n\nLike other openstack services, let\u0027s add a new section in\ntempest.conf for manila features called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/44824aaa8a435138b5406d31261f8eedfa617fd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/44824aaa8a435138b5406d31261f8eedfa617fd7"}]},"branch":"refs/heads/master"},"21019608995e1636b738f4c8ca0f67e1df082d6a":{"kind":"REWORK","_number":3,"created":"2021-04-18 12:55:07.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/3"}}},"commit":{"parents":[{"commit":"f083c71a9396a5a6f3300d3150b7fc1bee2c57c5","subject":"Add create share from snapshot in CephFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f083c71a9396a5a6f3300d3150b7fc1bee2c57c5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-18 12:54:47.000000000","tz":180},"subject":"[WIP] Add new configuration section for share features","message":"[WIP] Add new configuration section for share features\n\nLike other openstack services, let\u0027s add a new section in\ntempest.conf for manila features called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/21019608995e1636b738f4c8ca0f67e1df082d6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/21019608995e1636b738f4c8ca0f67e1df082d6a"}]},"branch":"refs/heads/master"},"f884e90f16646f64388b6fc80ec862590bf1a43a":{"kind":"REWORK","_number":4,"created":"2021-04-18 13:15:35.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/4"}}},"commit":{"parents":[{"commit":"f083c71a9396a5a6f3300d3150b7fc1bee2c57c5","subject":"Add create share from snapshot in CephFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f083c71a9396a5a6f3300d3150b7fc1bee2c57c5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-18 13:15:15.000000000","tz":180},"subject":"[WIP] Add new configuration section for share features","message":"[WIP] Add new configuration section for share features\n\nLike other openstack services, let\u0027s add a new section in\ntempest.conf for manila features called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f884e90f16646f64388b6fc80ec862590bf1a43a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f884e90f16646f64388b6fc80ec862590bf1a43a"}]},"branch":"refs/heads/master"},"a16a497311843db7bd35c182266bc77dbec86371":{"kind":"REWORK","_number":5,"created":"2021-04-19 04:32:50.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/5"}}},"commit":{"parents":[{"commit":"f083c71a9396a5a6f3300d3150b7fc1bee2c57c5","subject":"Add create share from snapshot in CephFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f083c71a9396a5a6f3300d3150b7fc1bee2c57c5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-19 04:32:31.000000000","tz":180},"subject":"[WIP] Add new configuration section for share features","message":"[WIP] Add new configuration section for share features\n\nLike other openstack services, let\u0027s add a new section in\ntempest.conf for manila features called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a16a497311843db7bd35c182266bc77dbec86371"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a16a497311843db7bd35c182266bc77dbec86371"}]},"branch":"refs/heads/master"},"bb55777c44c290f7f5350cd1db40717b5b34eb5c":{"kind":"REWORK","_number":6,"created":"2021-04-19 06:26:22.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/6"}}},"commit":{"parents":[{"commit":"f083c71a9396a5a6f3300d3150b7fc1bee2c57c5","subject":"Add create share from snapshot in CephFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f083c71a9396a5a6f3300d3150b7fc1bee2c57c5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-19 06:26:02.000000000","tz":180},"subject":"[WIP] Add new configuration section for share features","message":"[WIP] Add new configuration section for share features\n\nLike other openstack services, let\u0027s add a new section in\ntempest.conf for manila features called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bb55777c44c290f7f5350cd1db40717b5b34eb5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bb55777c44c290f7f5350cd1db40717b5b34eb5c"}]},"branch":"refs/heads/master"},"1651549c89e66738f1bb8dedee13882f8656c16f":{"kind":"REWORK","_number":7,"created":"2021-04-19 09:40:53.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/7"}}},"commit":{"parents":[{"commit":"f083c71a9396a5a6f3300d3150b7fc1bee2c57c5","subject":"Add create share from snapshot in CephFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f083c71a9396a5a6f3300d3150b7fc1bee2c57c5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-19 09:40:30.000000000","tz":180},"subject":"[WIP] Add new configuration section for share features","message":"[WIP] Add new configuration section for share features\n\nLike other openstack services, let\u0027s add a new section in\ntempest.conf for manila features called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1651549c89e66738f1bb8dedee13882f8656c16f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1651549c89e66738f1bb8dedee13882f8656c16f"}]},"branch":"refs/heads/master"},"acde622b29796922e9ed64538ea36d20a3194114":{"kind":"REWORK","_number":8,"created":"2021-04-19 10:37:26.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/8"}}},"commit":{"parents":[{"commit":"f083c71a9396a5a6f3300d3150b7fc1bee2c57c5","subject":"Add create share from snapshot in CephFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f083c71a9396a5a6f3300d3150b7fc1bee2c57c5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-19 10:37:07.000000000","tz":180},"subject":"[WIP] Add new configuration section for share features","message":"[WIP] Add new configuration section for share features\n\nLike other openstack services, let\u0027s add a new section in\ntempest.conf for manila features called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/acde622b29796922e9ed64538ea36d20a3194114"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/acde622b29796922e9ed64538ea36d20a3194114"}]},"branch":"refs/heads/master"},"9fcbb846f44d3de0c982b90b927525ac65dabd8a":{"kind":"REWORK","_number":9,"created":"2021-04-19 11:19:25.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/9"}}},"commit":{"parents":[{"commit":"f083c71a9396a5a6f3300d3150b7fc1bee2c57c5","subject":"Add create share from snapshot in CephFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f083c71a9396a5a6f3300d3150b7fc1bee2c57c5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-19 11:19:04.000000000","tz":180},"subject":"[WIP] Add new configuration section for share features","message":"[WIP] Add new configuration section for share features\n\nLike other openstack services, let\u0027s add a new section in\ntempest.conf for manila features called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9fcbb846f44d3de0c982b90b927525ac65dabd8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9fcbb846f44d3de0c982b90b927525ac65dabd8a"}]},"branch":"refs/heads/master"},"4340d265a9b7f44ebf87280e36378ef3e9c5e4b4":{"kind":"REWORK","_number":10,"created":"2021-08-09 09:08:36.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/10"}}},"commit":{"parents":[{"commit":"da415b4f1c3107ca05245ed9abd53a3ab67d83e3","subject":"Fix parsing of share group type json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da415b4f1c3107ca05245ed9abd53a3ab67d83e3"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-08-09 09:08:15.000000000","tz":180},"subject":"[WIP] Add new configuration section for share features","message":"[WIP] Add new configuration section for share features\n\nLike other openstack services, let\u0027s add a new section in\ntempest.conf for manila features called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4340d265a9b7f44ebf87280e36378ef3e9c5e4b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4340d265a9b7f44ebf87280e36378ef3e9c5e4b4"}]},"branch":"refs/heads/master"},"0abd83a7090c924585e5fab05bd95ea856fdf950":{"kind":"REWORK","_number":11,"created":"2021-08-09 09:29:18.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/11"}}},"commit":{"parents":[{"commit":"da415b4f1c3107ca05245ed9abd53a3ab67d83e3","subject":"Fix parsing of share group type json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da415b4f1c3107ca05245ed9abd53a3ab67d83e3"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-08-09 09:29:01.000000000","tz":180},"subject":"[WIP] Add new configuration section for share features","message":"[WIP] Add new configuration section for share features\n\nLike other openstack services, let\u0027s add a new section in\ntempest.conf for manila features called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0abd83a7090c924585e5fab05bd95ea856fdf950"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0abd83a7090c924585e5fab05bd95ea856fdf950"}]},"branch":"refs/heads/master"},"6c157fc4c64661c83a6a5bf91c0fca8b2c493119":{"kind":"REWORK","_number":12,"created":"2021-08-09 11:04:39.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/12"}}},"commit":{"parents":[{"commit":"da415b4f1c3107ca05245ed9abd53a3ab67d83e3","subject":"Fix parsing of share group type json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da415b4f1c3107ca05245ed9abd53a3ab67d83e3"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-08-09 11:04:18.000000000","tz":180},"subject":"[WIP] Add new configuration section for share features","message":"[WIP] Add new configuration section for share features\n\nLike other openstack services, let\u0027s add a new section in\ntempest.conf for manila features called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6c157fc4c64661c83a6a5bf91c0fca8b2c493119"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6c157fc4c64661c83a6a5bf91c0fca8b2c493119"}]},"branch":"refs/heads/master"},"9fabe54cfd01beebcb08e65bdc9f85c2fbb269a9":{"kind":"REWORK","_number":13,"created":"2021-08-09 13:27:38.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/13"}}},"commit":{"parents":[{"commit":"da415b4f1c3107ca05245ed9abd53a3ab67d83e3","subject":"Fix parsing of share group type json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da415b4f1c3107ca05245ed9abd53a3ab67d83e3"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-08-09 13:27:19.000000000","tz":180},"subject":"[WIP] Add new configuration section for share features","message":"[WIP] Add new configuration section for share features\n\nLike other openstack services, let\u0027s add a new section in\ntempest.conf for manila features called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9fabe54cfd01beebcb08e65bdc9f85c2fbb269a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9fabe54cfd01beebcb08e65bdc9f85c2fbb269a9"}]},"branch":"refs/heads/master"},"afe036f4700b154aa84d4ad81c52922aa39c2450":{"kind":"REWORK","_number":14,"created":"2021-08-10 03:40:25.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/14"}}},"commit":{"parents":[{"commit":"da415b4f1c3107ca05245ed9abd53a3ab67d83e3","subject":"Fix parsing of share group type json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da415b4f1c3107ca05245ed9abd53a3ab67d83e3"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-08-10 03:40:04.000000000","tz":180},"subject":"[WIP] Add new configuration section for share features","message":"[WIP] Add new configuration section for share features\n\nLike other openstack services, let\u0027s add a new section in\ntempest.conf for manila features called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/afe036f4700b154aa84d4ad81c52922aa39c2450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/afe036f4700b154aa84d4ad81c52922aa39c2450"}]},"branch":"refs/heads/master"},"1a2aa6d28621744e96a98ac0ae2199236132e836":{"kind":"REWORK","_number":15,"created":"2021-08-10 09:00:00.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/15"}}},"commit":{"parents":[{"commit":"da415b4f1c3107ca05245ed9abd53a3ab67d83e3","subject":"Fix parsing of share group type json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da415b4f1c3107ca05245ed9abd53a3ab67d83e3"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-08-10 08:59:31.000000000","tz":180},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nLike other openstack services, let\u0027s add a new section in\ntempest.conf for manila features called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1a2aa6d28621744e96a98ac0ae2199236132e836"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1a2aa6d28621744e96a98ac0ae2199236132e836"}]},"branch":"refs/heads/master"},"730b629e0aaeee255faf44c1aff139d369e5c73c":{"kind":"REWORK","_number":16,"created":"2021-08-10 11:59:37.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/16"}}},"commit":{"parents":[{"commit":"da415b4f1c3107ca05245ed9abd53a3ab67d83e3","subject":"Fix parsing of share group type json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da415b4f1c3107ca05245ed9abd53a3ab67d83e3"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-08-10 11:59:31.000000000","tz":180},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nWe are currently skipping feature tests in a confusing way,\nfor example regarding snapshot feature we use two configuration\noptions: ``capability_snapshot_support``, ``run_snapshot_tests``\nSo in order to run a snapshot test we need enable these two\nconfiguration options.\n\nThis patch suggests using a new single configuration option,\nlike other openstack services, called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch is part of the patches that will be uploaded later.\nThis patch removes the configuration options below:\n``capability_snapshot_support``\n``capability_create_share_from_snapshot_support``\n``run_snapshot_tests``\n``run_manage_unmanage_snapshot_tests``\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/730b629e0aaeee255faf44c1aff139d369e5c73c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/730b629e0aaeee255faf44c1aff139d369e5c73c"}]},"branch":"refs/heads/master"},"f466dd6932713434eaec30db56750c9f3aebf235":{"kind":"REWORK","_number":17,"created":"2021-08-10 13:23:05.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/17"}}},"commit":{"parents":[{"commit":"da415b4f1c3107ca05245ed9abd53a3ab67d83e3","subject":"Fix parsing of share group type json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da415b4f1c3107ca05245ed9abd53a3ab67d83e3"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-08-10 13:22:45.000000000","tz":180},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nLike other openstack services, let\u0027s add a new section in\ntempest.conf for manila features called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f466dd6932713434eaec30db56750c9f3aebf235"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f466dd6932713434eaec30db56750c9f3aebf235"}]},"branch":"refs/heads/master"},"56a3c90d25529379f80a89be083a17c69404700b":{"kind":"REWORK","_number":18,"created":"2021-08-10 13:25:36.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/18"}}},"commit":{"parents":[{"commit":"da415b4f1c3107ca05245ed9abd53a3ab67d83e3","subject":"Fix parsing of share group type json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da415b4f1c3107ca05245ed9abd53a3ab67d83e3"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-08-10 13:25:14.000000000","tz":180},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nWe are currently skipping feature tests in a confusing way,\nfor example regarding snapshot feature we use two configuration\noptions: ``capability_snapshot_support``, ``run_snapshot_tests``\nSo in order to run a snapshot test we need enable these two\nconfiguration options.\n\nThis patch suggests using a new single configuration option,\nlike other openstack services, called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch is part of the patches that will be uploaded later.\nThis patch removes the configuration options below:\n``capability_snapshot_support``\n``capability_create_share_from_snapshot_support``\n``run_snapshot_tests``\n``run_manage_unmanage_snapshot_tests``\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/56a3c90d25529379f80a89be083a17c69404700b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/56a3c90d25529379f80a89be083a17c69404700b"}]},"branch":"refs/heads/master"},"db2a36f17ab69b46f1574744cf1a5f7484c0c235":{"kind":"REWORK","_number":19,"created":"2021-08-10 14:39:13.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/19"}}},"commit":{"parents":[{"commit":"da415b4f1c3107ca05245ed9abd53a3ab67d83e3","subject":"Fix parsing of share group type json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da415b4f1c3107ca05245ed9abd53a3ab67d83e3"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-08-10 14:38:53.000000000","tz":180},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nWe are currently skipping feature tests in a confusing way,\nfor example regarding snapshot feature we use two configuration\noptions: ``capability_snapshot_support``, ``run_snapshot_tests``\nSo in order to run a snapshot test we need enable these two\nconfiguration options.\n\nThis patch suggests using a new single configuration option,\nlike other openstack services, called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch is part of the patches that will be uploaded later.\nThis patch removes the configuration options below:\n``capability_snapshot_support``\n``capability_create_share_from_snapshot_support``\n``run_snapshot_tests``\n``run_manage_unmanage_snapshot_tests``\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/db2a36f17ab69b46f1574744cf1a5f7484c0c235"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/db2a36f17ab69b46f1574744cf1a5f7484c0c235"}]},"branch":"refs/heads/master"},"b0259ae2c6deb8263e5b9edf00804e7baaac0607":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2021-09-21 03:05:44.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/77/786177/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/20"}}},"commit":{"parents":[{"commit":"a4fab9a60e4c0c59fa26891a65873cec070960c5","subject":"Merge \"Use local variable instead of instance variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a4fab9a60e4c0c59fa26891a65873cec070960c5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-09-21 03:05:44.000000000","tz":0},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nWe are currently skipping feature tests in a confusing way,\nfor example regarding snapshot feature we use two configuration\noptions: ``capability_snapshot_support``, ``run_snapshot_tests``\nSo in order to run a snapshot test we need enable these two\nconfiguration options.\n\nThis patch suggests using a new single configuration option,\nlike other openstack services, called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch is part of the patches that will be uploaded later.\nThis patch removes the configuration options below:\n``capability_snapshot_support``\n``capability_create_share_from_snapshot_support``\n``run_snapshot_tests``\n``run_manage_unmanage_snapshot_tests``\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0259ae2c6deb8263e5b9edf00804e7baaac0607"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0259ae2c6deb8263e5b9edf00804e7baaac0607"}]},"branch":"refs/heads/master","description":"Rebase"},"7e1014d552c118cb02202b7c108b5f663425bd5a":{"kind":"REWORK","_number":21,"created":"2022-02-17 12:18:37.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/21"}}},"commit":{"parents":[{"commit":"73e0b6fdc450b7539bf6a97349f5774177ffb24a","subject":"Added code to show share network neutron_net_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/73e0b6fdc450b7539bf6a97349f5774177ffb24a"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-02-17 12:18:16.000000000","tz":120},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nWe are currently skipping feature tests in a confusing way,\nfor example regarding snapshot feature we use two configuration\noptions: ``capability_snapshot_support``, ``run_snapshot_tests``\nSo in order to run a snapshot test we need enable these two\nconfiguration options.\n\nThis patch suggests using a new single configuration option,\nlike other openstack services, called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch is part of the patches that will be uploaded later.\nThis patch removes the configuration options below:\n``capability_snapshot_support``\n``capability_create_share_from_snapshot_support``\n``run_snapshot_tests``\n``run_manage_unmanage_snapshot_tests``\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7e1014d552c118cb02202b7c108b5f663425bd5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7e1014d552c118cb02202b7c108b5f663425bd5a"}]},"branch":"refs/heads/master"},"c250fc5a2b0d606f69bc86357acfd1a3713f86f6":{"kind":"REWORK","_number":22,"created":"2022-02-17 13:13:57.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/22"}}},"commit":{"parents":[{"commit":"73e0b6fdc450b7539bf6a97349f5774177ffb24a","subject":"Added code to show share network neutron_net_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/73e0b6fdc450b7539bf6a97349f5774177ffb24a"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-02-17 13:13:37.000000000","tz":120},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nWe are currently skipping feature tests in a confusing way,\nfor example regarding snapshot feature we use two configuration\noptions: ``capability_snapshot_support``, ``run_snapshot_tests``\nSo in order to run a snapshot test we need enable these two\nconfiguration options.\n\nThis patch suggests using a new single configuration option,\nlike other openstack services, called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch is part of the patches that will be uploaded later.\nThis patch removes the configuration options below:\n``capability_snapshot_support``\n``capability_create_share_from_snapshot_support``\n``run_snapshot_tests``\n``run_manage_unmanage_snapshot_tests``\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c250fc5a2b0d606f69bc86357acfd1a3713f86f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c250fc5a2b0d606f69bc86357acfd1a3713f86f6"}]},"branch":"refs/heads/master"},"1e87cba0d1e4dafe065ad067a1e6f37e66eeabb5":{"kind":"REWORK","_number":23,"created":"2022-04-06 09:48:13.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/23"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-04-06 09:47:47.000000000","tz":180},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nWe are currently skipping feature tests in a confusing way,\nfor example regarding snapshot feature we use two configuration\noptions: ``capability_snapshot_support``, ``run_snapshot_tests``\nSo in order to run a snapshot test we need enable these two\nconfiguration options.\n\nThis patch suggests using a new single configuration option,\nlike other openstack services, called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch is part of the patches that will be uploaded later.\nThis patch removes the configuration options below:\n``capability_snapshot_support``\n``capability_create_share_from_snapshot_support``\n``run_snapshot_tests``\n``run_manage_unmanage_snapshot_tests``\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1e87cba0d1e4dafe065ad067a1e6f37e66eeabb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1e87cba0d1e4dafe065ad067a1e6f37e66eeabb5"}]},"branch":"refs/heads/master"},"b4964408d07845b78d339aa97c0cc54736d02a14":{"kind":"REWORK","_number":24,"created":"2022-07-04 10:03:13.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/24"}}},"commit":{"parents":[{"commit":"d12285b25b40031ec90743c45b095d1bbcba4ca4","subject":"Follow up for CI jobs refactoring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d12285b25b40031ec90743c45b095d1bbcba4ca4"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-07-04 10:03:04.000000000","tz":180},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nWe are currently skipping feature tests in a confusing way,\nfor example regarding snapshot feature we use two configuration\noptions: ``capability_snapshot_support``, ``run_snapshot_tests``\nSo in order to run a snapshot test we need enable these two\nconfiguration options.\n\nThis patch suggests using a new single configuration option,\nlike other openstack services, called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch is part of the patches that will be uploaded later.\nThis patch removes the configuration options below:\n``capability_snapshot_support``\n``capability_create_share_from_snapshot_support``\n``run_snapshot_tests``\n``run_manage_unmanage_snapshot_tests``\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b4964408d07845b78d339aa97c0cc54736d02a14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b4964408d07845b78d339aa97c0cc54736d02a14"}]},"branch":"refs/heads/master"},"1dfcc82e131e48328978c7ddab510591ddf60297":{"kind":"REWORK","_number":25,"created":"2024-03-10 12:29:52.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/25"}}},"commit":{"parents":[{"commit":"cbab9fe0c72337704e1a364ce1db1d78a3c13826","subject":"Fix backup build erro exception class name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cbab9fe0c72337704e1a364ce1db1d78a3c13826"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-03-10 12:29:39.000000000","tz":120},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nWe are currently skipping feature tests in a confusing way,\nfor example regarding snapshot feature we use two configuration\noptions: ``capability_snapshot_support``, ``run_snapshot_tests``\nSo in order to run a snapshot test we need enable these two\nconfiguration options.\n\nThis patch suggests using a new single configuration option,\nlike other openstack services, called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch is part of the patches that will be uploaded later.\nThis patch removes the configuration options below:\n``capability_snapshot_support``\n``capability_create_share_from_snapshot_support``\n``run_snapshot_tests``\n``run_manage_unmanage_snapshot_tests``\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1dfcc82e131e48328978c7ddab510591ddf60297"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1dfcc82e131e48328978c7ddab510591ddf60297"}]},"branch":"refs/heads/master"},"a4e972142d494bbaa37e0c23bdfb87f2b421dd62":{"kind":"REWORK","_number":26,"created":"2024-11-07 09:10:43.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/26"}}},"commit":{"parents":[{"commit":"31df75e18b2cfad59322c20575a608fdbb592de1","subject":"Merge \"Support py3.12 and drop py3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/31df75e18b2cfad59322c20575a608fdbb592de1"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-11-07 09:10:28.000000000","tz":120},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nWe are currently skipping feature tests in a confusing way,\nfor example regarding snapshot feature we use two configuration\noptions: ``capability_snapshot_support``, ``run_snapshot_tests``\nSo in order to run a snapshot test we need enable these two\nconfiguration options.\n\nThis patch suggests using a new single configuration option,\nlike other openstack services, called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch is part of the patches that will be uploaded later.\nThis patch removes the configuration options below:\n``capability_snapshot_support``\n``capability_create_share_from_snapshot_support``\n``run_snapshot_tests``\n``run_manage_unmanage_snapshot_tests``\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a4e972142d494bbaa37e0c23bdfb87f2b421dd62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a4e972142d494bbaa37e0c23bdfb87f2b421dd62"}]},"branch":"refs/heads/master"},"3306eb2ac348b6407b64fed52f9f3d4535089bb9":{"kind":"REWORK","_number":27,"created":"2024-11-07 09:27:25.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/27"}}},"commit":{"parents":[{"commit":"31df75e18b2cfad59322c20575a608fdbb592de1","subject":"Merge \"Support py3.12 and drop py3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/31df75e18b2cfad59322c20575a608fdbb592de1"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-11-07 09:26:54.000000000","tz":120},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nWe are currently skipping feature tests in a confusing way,\nfor example regarding snapshot feature we use two configuration\noptions: ``capability_snapshot_support``, ``run_snapshot_tests``\nSo in order to run a snapshot test we need enable these two\nconfiguration options.\n\nThis patch suggests using a new single configuration option,\nlike other openstack services, called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch is part of the patches that will be uploaded later.\nThis patch removes the configuration options below:\n``capability_snapshot_support``\n``capability_create_share_from_snapshot_support``\n``run_snapshot_tests``\n``run_manage_unmanage_snapshot_tests``\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3306eb2ac348b6407b64fed52f9f3d4535089bb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3306eb2ac348b6407b64fed52f9f3d4535089bb9"}]},"branch":"refs/heads/master"},"2c7d9e4f81ecdb1a53669150802d6e06694e71af":{"kind":"REWORK","_number":28,"created":"2024-11-07 11:12:32.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/28"}}},"commit":{"parents":[{"commit":"31df75e18b2cfad59322c20575a608fdbb592de1","subject":"Merge \"Support py3.12 and drop py3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/31df75e18b2cfad59322c20575a608fdbb592de1"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-11-07 11:12:21.000000000","tz":120},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nWe are currently skipping feature tests in a confusing way,\nfor example regarding snapshot feature we use two configuration\noptions: ``capability_snapshot_support``, ``run_snapshot_tests``\nSo in order to run a snapshot test we need enable these two\nconfiguration options.\n\nThis patch suggests using a new single configuration option,\nlike other openstack services, called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch is part of the patches that will be uploaded later.\nThis patch removes the configuration options below:\n``capability_snapshot_support``\n``capability_create_share_from_snapshot_support``\n``run_snapshot_tests``\n``run_manage_unmanage_snapshot_tests``\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2c7d9e4f81ecdb1a53669150802d6e06694e71af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2c7d9e4f81ecdb1a53669150802d6e06694e71af"}]},"branch":"refs/heads/master"},"4d18a1486809ed0696ac7f267981388f2f95ec56":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2024-12-09 13:01:31.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/29"}}},"commit":{"parents":[{"commit":"ee3079db74bd8ce6be5115a99263907fd3958da3","subject":"Merge \"Add nfs_versions option for NFS protocol version selection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ee3079db74bd8ce6be5115a99263907fd3958da3"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-12-09 13:01:19.000000000","tz":120},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nWe are currently skipping feature tests in a confusing way,\nfor example regarding snapshot feature we use two configuration\noptions: ``capability_snapshot_support``, ``run_snapshot_tests``\nSo in order to run a snapshot test we need enable these two\nconfiguration options.\n\nThis patch suggests using a new single configuration option,\nlike other openstack services, called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch is part of the patches that will be uploaded later.\nThis patch removes the configuration options below:\n``capability_snapshot_support``\n``capability_create_share_from_snapshot_support``\n``run_snapshot_tests``\n``run_manage_unmanage_snapshot_tests``\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d18a1486809ed0696ac7f267981388f2f95ec56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d18a1486809ed0696ac7f267981388f2f95ec56"}]},"branch":"refs/heads/master"},"0751ed5c542e3b0000b2988a3162fee1bf8de3cc":{"kind":"REWORK","_number":30,"created":"2025-03-27 09:05:17.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/30"}}},"commit":{"parents":[{"commit":"58903bdd47ad33cc82fd29bd1066f4c4f1245143","subject":"Ensure Manila shares are created on the same backend host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/58903bdd47ad33cc82fd29bd1066f4c4f1245143"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2025-03-27 09:05:07.000000000","tz":120},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nWe are currently skipping feature tests in a confusing way,\nfor example regarding snapshot feature we use two configuration\noptions: ``capability_snapshot_support``, ``run_snapshot_tests``\nSo in order to run a snapshot test we need enable these two\nconfiguration options.\n\nThis patch suggests using a new single configuration option,\nlike other openstack services, called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch is part of the patches that will be uploaded later.\nThis patch removes the configuration options below:\n``capability_snapshot_support``\n``capability_create_share_from_snapshot_support``\n``run_snapshot_tests``\n``run_manage_unmanage_snapshot_tests``\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0751ed5c542e3b0000b2988a3162fee1bf8de3cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0751ed5c542e3b0000b2988a3162fee1bf8de3cc"}]},"branch":"refs/heads/master"},"86bbeb54c77e763e4ebbe1738e548927f29a5a8c":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":31,"created":"2025-04-22 10:11:07.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/31"}}},"commit":{"parents":[{"commit":"43708a26d3f7a049bb22bc07d8e589c79c4fead0","subject":"Merge \"Stop testing the LVM driver with Jammy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/43708a26d3f7a049bb22bc07d8e589c79c4fead0"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2025-04-22 10:09:06.000000000","tz":180},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nWe are currently skipping feature tests in a confusing way,\nfor example regarding snapshot feature we use two configuration\noptions: ``capability_snapshot_support``, ``run_snapshot_tests``\nSo in order to run a snapshot test we need enable these two\nconfiguration options.\n\nThis patch suggests using a new single configuration option,\nlike other openstack services, called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch is part of the patches that will be uploaded later.\nThis patch removes the configuration options below:\n``capability_snapshot_support``\n``capability_create_share_from_snapshot_support``\n``run_snapshot_tests``\n``run_manage_unmanage_tests``\n``run_manage_unmanage_snapshot_tests``\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/86bbeb54c77e763e4ebbe1738e548927f29a5a8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/86bbeb54c77e763e4ebbe1738e548927f29a5a8c"}]},"branch":"refs/heads/master"},"2e65ad565cf080fe8b0963a95000ae907984614b":{"kind":"REWORK","_number":32,"created":"2026-01-06 15:46:01.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/77/786177/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/32"}}},"commit":{"parents":[{"commit":"a19b4387a9a52e8856c20fe39d906195c27bcc48","subject":"Merge \"Drop testing with PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a19b4387a9a52e8856c20fe39d906195c27bcc48"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-01-06 15:45:45.000000000","tz":-180},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nWe are currently skipping feature tests in a confusing way,\nfor example regarding snapshot feature we use two configuration\noptions: ``capability_snapshot_support``, ``run_snapshot_tests``\nSo in order to run a snapshot test we need enable these two\nconfiguration options.\n\nThis patch suggests using a new single configuration option,\nlike other openstack services, called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch is part of the patches that will be uploaded later.\nThis patch removes the configuration options below:\n``capability_snapshot_support``\n``capability_create_share_from_snapshot_support``\n``run_snapshot_tests``\n``run_manage_unmanage_tests``\n``run_manage_unmanage_snapshot_tests``\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e65ad565cf080fe8b0963a95000ae907984614b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e65ad565cf080fe8b0963a95000ae907984614b"}]},"branch":"refs/heads/master"},"6f17fda4b0a38d4256a30f4b6a098953e87b8a6a":{"kind":"REWORK","_number":33,"created":"2026-01-08 10:53:04.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/33"}}},"commit":{"parents":[{"commit":"3f76739525905fc34792d9c2490f3546c3af75b2","subject":"Merge \"Refactor stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3f76739525905fc34792d9c2490f3546c3af75b2"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2026-01-08 10:52:49.000000000","tz":120},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nWe are currently skipping feature tests in a confusing way,\nfor example regarding snapshot feature we use two configuration\noptions: ``capability_snapshot_support``, ``run_snapshot_tests``\nSo in order to run a snapshot test we need enable these two\nconfiguration options.\n\nThis patch suggests using a new single configuration option,\nlike other openstack services, called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch is part of the patches that will be uploaded later.\nThis patch removes the configuration options below:\n``capability_snapshot_support``\n``capability_create_share_from_snapshot_support``\n``run_snapshot_tests``\n``run_manage_unmanage_tests``\n``run_manage_unmanage_snapshot_tests``\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\nSigned-off-by: lkuchlan \u003clkuchlan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6f17fda4b0a38d4256a30f4b6a098953e87b8a6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6f17fda4b0a38d4256a30f4b6a098953e87b8a6a"}]},"branch":"refs/heads/master"},"c3e418c44dad7ca60c3b7c0b8494c2e0ebc3e0ce":{"kind":"REWORK","_number":34,"created":"2026-01-08 11:59:30.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/34"}}},"commit":{"parents":[{"commit":"3f76739525905fc34792d9c2490f3546c3af75b2","subject":"Merge \"Refactor stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3f76739525905fc34792d9c2490f3546c3af75b2"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2026-01-08 11:58:58.000000000","tz":120},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nWe are currently skipping feature tests in a confusing way,\nfor example regarding snapshot feature we use two configuration\noptions: ``capability_snapshot_support``, ``run_snapshot_tests``\nSo in order to run a snapshot test we need enable these two\nconfiguration options.\n\nThis patch suggests using a new single configuration option,\nlike other openstack services, called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch is part of the patches that will be uploaded later.\nThis patch removes the configuration options below:\n``capability_snapshot_support``\n``capability_create_share_from_snapshot_support``\n``run_snapshot_tests``\n``run_manage_unmanage_tests``\n``run_manage_unmanage_snapshot_tests``\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\nSigned-off-by: lkuchlan \u003clkuchlan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c3e418c44dad7ca60c3b7c0b8494c2e0ebc3e0ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c3e418c44dad7ca60c3b7c0b8494c2e0ebc3e0ce"}]},"branch":"refs/heads/master"},"97194d80b847ddbd31660059a5517c106aca088b":{"kind":"REWORK","_number":35,"created":"2026-01-08 14:10:32.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/35"}}},"commit":{"parents":[{"commit":"3f76739525905fc34792d9c2490f3546c3af75b2","subject":"Merge \"Refactor stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3f76739525905fc34792d9c2490f3546c3af75b2"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2026-01-08 14:10:18.000000000","tz":120},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nWe are currently skipping feature tests in a confusing way,\nfor example regarding snapshot feature we use two configuration\noptions: ``capability_snapshot_support``, ``run_snapshot_tests``\nSo in order to run a snapshot test we need enable these two\nconfiguration options.\n\nThis patch suggests using a new single configuration option,\nlike other openstack services, called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch is part of the patches that will be uploaded later.\nThis patch removes the configuration options below:\n``capability_snapshot_support``\n``capability_create_share_from_snapshot_support``\n``run_snapshot_tests``\n``run_manage_unmanage_tests``\n``run_manage_unmanage_snapshot_tests``\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\nSigned-off-by: lkuchlan \u003clkuchlan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/97194d80b847ddbd31660059a5517c106aca088b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/97194d80b847ddbd31660059a5517c106aca088b"}]},"branch":"refs/heads/master"},"c9688eb341942c90697d67460e50eec3a76f6801":{"kind":"REWORK","_number":36,"created":"2026-01-11 13:16:12.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/36"}}},"commit":{"parents":[{"commit":"3f76739525905fc34792d9c2490f3546c3af75b2","subject":"Merge \"Refactor stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3f76739525905fc34792d9c2490f3546c3af75b2"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2026-01-11 13:15:56.000000000","tz":120},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nWe are currently skipping feature tests in a confusing way,\nfor example regarding snapshot feature we use two configuration\noptions: ``capability_snapshot_support``, ``run_snapshot_tests``\nSo in order to run a snapshot test we need enable these two\nconfiguration options.\n\nThis patch suggests using a new single configuration option,\nlike other openstack services, called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch is part of the patches that will be uploaded later.\nThis patch removes the configuration options below:\n``capability_snapshot_support``\n``capability_create_share_from_snapshot_support``\n``run_snapshot_tests``\n``run_manage_unmanage_tests``\n``run_manage_unmanage_snapshot_tests``\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\nSigned-off-by: lkuchlan \u003clkuchlan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c9688eb341942c90697d67460e50eec3a76f6801"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c9688eb341942c90697d67460e50eec3a76f6801"}]},"branch":"refs/heads/master"},"1c43cd9c6db931bea30e5397008c3974da16c9f0":{"kind":"REWORK","_number":37,"created":"2026-01-14 10:35:39.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/786177/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/37"}}},"commit":{"parents":[{"commit":"3f76739525905fc34792d9c2490f3546c3af75b2","subject":"Merge \"Refactor stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3f76739525905fc34792d9c2490f3546c3af75b2"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2026-01-14 10:35:24.000000000","tz":120},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nWe are currently skipping feature tests in a confusing way,\nfor example regarding snapshot feature we use two configuration\noptions: ``capability_snapshot_support``, ``run_snapshot_tests``\nSo in order to run a snapshot test we need enable these two\nconfiguration options.\n\nThis patch suggests using a new single configuration option,\nlike other openstack services, called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch is part of the patches that will be uploaded later.\nThis patch removes the configuration options below:\n``capability_snapshot_support``\n``capability_create_share_from_snapshot_support``\n``run_snapshot_tests``\n``run_manage_unmanage_tests``\n``run_manage_unmanage_snapshot_tests``\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\nSigned-off-by: lkuchlan \u003clkuchlan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1c43cd9c6db931bea30e5397008c3974da16c9f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1c43cd9c6db931bea30e5397008c3974da16c9f0"}]},"branch":"refs/heads/master"},"c74ad76d1dd6f154c26d9a184a719b9a4c75e9fc":{"kind":"REWORK","_number":38,"created":"2026-01-29 19:10:29.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/77/786177/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/38"}}},"commit":{"parents":[{"commit":"3f76739525905fc34792d9c2490f3546c3af75b2","subject":"Merge \"Refactor stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3f76739525905fc34792d9c2490f3546c3af75b2"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-01-29 19:09:36.000000000","tz":-480},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nWe are currently skipping feature tests in a confusing way,\nfor example regarding snapshot feature we use two configuration\noptions: ``capability_snapshot_support``, ``run_snapshot_tests``\nSo in order to run a snapshot test we need enable these two\nconfiguration options.\n\nThis patch suggests using a new single configuration option,\nlike other openstack services, called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch is part of the patches that will be uploaded later.\nThis patch removes the configuration options below:\n``capability_snapshot_support``\n``capability_create_share_from_snapshot_support``\n``run_snapshot_tests``\n``run_manage_unmanage_tests``\n``run_manage_unmanage_snapshot_tests``\n``run_revert_to_snapshot_tests``\n``run_mount_snapshot_tests``\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\nSigned-off-by: lkuchlan \u003clkuchlan@redhat.com\u003e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c74ad76d1dd6f154c26d9a184a719b9a4c75e9fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c74ad76d1dd6f154c26d9a184a719b9a4c75e9fc"}]},"branch":"refs/heads/master"},"00026c86fa3e3c415b5aef309a5cdb11ebd14bf0":{"kind":"REWORK","_number":39,"created":"2026-01-29 21:57:53.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/77/786177/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/786177/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/786177/39"}}},"commit":{"parents":[{"commit":"3f76739525905fc34792d9c2490f3546c3af75b2","subject":"Merge \"Refactor stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3f76739525905fc34792d9c2490f3546c3af75b2"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 08:16:41.000000000","tz":180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-01-29 21:57:45.000000000","tz":-480},"subject":"Add new configuration section for share features","message":"Add new configuration section for share features\n\nWe are currently skipping feature tests in a confusing way,\nfor example regarding snapshot feature we use two configuration\noptions: ``capability_snapshot_support``, ``run_snapshot_tests``\nSo in order to run a snapshot test we need enable these two\nconfiguration options.\n\nThis patch suggests using a new single configuration option,\nlike other openstack services, called \u0027share-feature-enabled\u0027.\nThis way we can skip tests depending on the features that the\nenvironment supports.\n\nThis patch is part of the patches that will be uploaded later.\nThis patch removes the configuration options below:\n``capability_snapshot_support``\n``capability_create_share_from_snapshot_support``\n``run_snapshot_tests``\n``run_manage_unmanage_tests``\n``run_manage_unmanage_snapshot_tests``\n``run_revert_to_snapshot_tests``\n``run_mount_snapshot_tests``\n\nThis patch adds a new decorator used to set an attr for each\nshare feature and skips a test in case the environment does\nnot support the feature.\n\nChange-Id: Ia41761352f1c3fb5004eeeda2d789a3ea6a8fc25\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\nSigned-off-by: lkuchlan \u003clkuchlan@redhat.com\u003e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/00026c86fa3e3c415b5aef309a5cdb11ebd14bf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/00026c86fa3e3c415b5aef309a5cdb11ebd14bf0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
