)]}'
{"id":"openstack%2Fmanila-tempest-plugin~788248","triplet_id":"openstack%2Fmanila-tempest-plugin~master~If0e39afb635c469a25919770a869087bf5def561","project":"openstack/manila-tempest-plugin","branch":"master","topic":"rest_client","hashtags":[],"change_id":"If0e39afb635c469a25919770a869087bf5def561","subject":"Enable http response extraction","status":"MERGED","created":"2021-04-27 11:33:07.000000000","updated":"2021-07-21 18:58:21.000000000","submitted":"2021-07-21 18:58:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"788248-rest_client","meta_rev_id":"a03f69d5769fda610af85f2a00f3c693a5784a97","_number":788248,"virtual_id_number":788248,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-21 18:58:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2021-07-21 17:14:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2021-07-21 17:20:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2021-07-21 17:53:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"reviewers":{"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-27 12:55:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-27 13:04:50.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-04-27 20:26:04.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-04-29 19:03:52.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-29 20:02:50.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-02 03:49:41.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2021-05-02 03:49:41.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-05-02 03:49:41.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"e6649d77a2d7a42456b22720910091e32651d3c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-27 11:33:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c72fecf8ea9f97705c3378011b8d78f7ac8779ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-27 11:46:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e83c43c04da8441dc3d62393af7009867fb4a4cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 12:55:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a5ca46033974f9a8d04ac0af05b2fbd : SUCCESS in 4m 01s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/eb7e86c9c9ba422fa5530377baa9117d : FAILURE in 21m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/805715f5a85c4b728fcd67cf75ff275f : FAILURE in 27m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f508343fc56a4aefa125e24e018b9282 : FAILURE in 24m 06s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b49c72676fe547958ef2b86e91575025 : FAILURE in 13m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/aa860da97e1743c4bb64b1d01cfe8397 : FAILURE in 28m 47s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a9a88ebe6cf140b1bda74b994a8324e4 : FAILURE in 35m 35s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e7dfa60a55b84d39913ee85df8a6c996 : FAILURE in 34m 19s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a666eef629ca42bc92293eeb82e878a7 : FAILURE in 1h 01m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4f9da311117f4ca5a4d2d881a3256fec : FAILURE in 24m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"59360715156820f9bde736ffa78a2fd93efee6a7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-27 13:04:50.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/48/788248/2/check/DellEMC_Unity_Manila_dhss_true/82a8acd/DellEMC_Unity_Manila_dhss_true/25 : FAILURE in 42m 37s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/48/788248/2/check/DellEMC_Unity_Manila_dhss_false/0217645/DellEMC_Unity_Manila_dhss_false/26 : FAILURE in 41m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"7456aa69b764f1f335748b1ab5d30f430573ce17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-04-27 20:26:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c7bc9da8fc2d7aa330dab30beed9729ceda7644c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-27 21:07:32.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/48/788248/3/check/DellEMC_Unity_Manila_dhss_true/cce2561/DellEMC_Unity_Manila_dhss_true/27 : FAILURE in 40m 39s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/48/788248/3/check/DellEMC_Unity_Manila_dhss_false/579c6cf/DellEMC_Unity_Manila_dhss_false/28 : FAILURE in 40m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"23091d64632dfd93bb74cb4fa0dd4d5e9faf1ac2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 21:55:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ab997cc7dac45f8aed67c1c7d62d16a : SUCCESS in 5m 11s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4534b39f26b441ae805422de320f4707 : FAILURE in 25m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/48b73b0e68b14874afd38d642e2ac127 : FAILURE in 27m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/845e1fcc19c74607a47e6073afeda972 : FAILURE in 39m 06s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/40ad5468cb6a40658fa273655611d9c6 : FAILURE in 41m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/579772b833194e35af60370ac9f8cd27 : FAILURE in 41m 35s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7d6baec5378d48248290e323b5ba9fcd : FAILURE in 48m 42s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2890d28f85054511ae3fd79e010818a4 : FAILURE in 39m 19s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/71435bd3c280446db77c04e59ec42401 : FAILURE in 56m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/fa13cb7e266e4f889b1db6407ac74b15 : FAILURE in 24m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9e3f0f839306554e0fa7e389b84d0f035bf5e2ee","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-04-27 22:02:32.000000000","message":"Patch Set 3:\n\nPatchset #3 fixed test_ensure_share_server_creation_when_dhss_enabled failure in #2","accounts_in_message":[],"_revision_number":3},{"id":"e2db7818646a4a769f9a23bf33a5a251d96472e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-04-28 03:46:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cfac3b632415cb79d1f12db2008b728f66b0a95c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-28 04:30:10.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/48/788248/4/check/DellEMC_Unity_Manila_dhss_true/fc2423e/DellEMC_Unity_Manila_dhss_true/28 : FAILURE in 43m 28s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/48/788248/4/check/DellEMC_Unity_Manila_dhss_false/299c27b/DellEMC_Unity_Manila_dhss_false/29 : FAILURE in 41m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"1f42927cf85e6c8fc6415182823f64744948433e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 05:05:01.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/205e2cace871483da71ac29fc3f2b2c7 : SUCCESS in 3m 24s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/53cb2fba2456440da9a311dcc4ecd052 : FAILURE in 24m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/85b6246f32bb4b089c60807f6650fc3d : FAILURE in 27m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a2b63fae52ae40cdb0f23d9944741cc2 : FAILURE in 35m 24s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/910b2c66857f4b73adcc60934d711baa : FAILURE in 34m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/12a6f30ab5a240878eac6739868c2e5a : FAILURE in 44m 31s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/19f38ce7a37041a7b70bb4143aa3ffca : FAILURE in 45m 38s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1166e4e0cae340aca7005abb542734a6 : FAILURE in 33m 42s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/46da43449bb247c6bf246ff5eaf8cff6 : FAILURE in 1h 12m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/669c210431e0478389833a9b01983db0 : FAILURE in 20m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c726ea2f2b1df210cd4752ad047734f18cb56ab7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-28 06:48:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"465f55b991a1ba8f6a45832e248d4d4fc12a5909","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-28 07:41:27.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/48/788248/5/check/DellEMC_Unity_Manila_dhss_true/501309c/DellEMC_Unity_Manila_dhss_true/29 : FAILURE in 52m 52s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/48/788248/5/check/DellEMC_Unity_Manila_dhss_false/d75c32f/DellEMC_Unity_Manila_dhss_false/30 : FAILURE in 48m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"058340c1c9449bc849eac3c8060b8648a0f0dcd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 08:03:57.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/412debe8ba4b4464923585bd33df0cd9 : SUCCESS in 4m 50s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/756db206e74b49b5b8afb0eda0dc6e6d : FAILURE in 30m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7defebe04bd647ff9bbcc35bdec54cf4 : FAILURE in 25m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/14a2c0cd7db54de7b249dfdf1ef0e491 : FAILURE in 39m 53s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/31718dd707984c539b51d2d91fa40d91 : FAILURE in 36m 12s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a77309d06a8449f8bdccc11b412a6e28 : FAILURE in 39m 33s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/606499167c0a4372808d02f3fe7bbdaf : FAILURE in 58m 42s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d0719d28ad584d00a4bfc5975a5934ce : FAILURE in 38m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7ee681c749104e80a134a96805b1e502 : FAILURE in 1h 10m 28s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a0e13faf442442d4aa3650952cb5eb3a : FAILURE in 27m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0b3b393c7134075a0d0770feec931ce8e1a6d78b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-28 09:45:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c04844cae756918aeba08c909ac5f9af9daabd23","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-28 10:42:30.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/48/788248/6/check/DellEMC_Unity_Manila_dhss_true/f054ff9/DellEMC_Unity_Manila_dhss_true/30 : FAILURE in 57m 01s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/48/788248/6/check/DellEMC_Unity_Manila_dhss_false/4b3b2a0/DellEMC_Unity_Manila_dhss_false/31 : FAILURE in 50m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"b3991f6ab8370b47ae7391860505bc3ce7f9544d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 12:04:55.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a12715aee6ff44d5b4192123a7a3f2c4 : SUCCESS in 3m 39s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c082c0af05cd4ccab0b94bc775633afd : FAILURE in 18m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b96f74b06f984277bccf575f552e705b : FAILURE in 29m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ebdf2083956f4397bcfb43b6b2f78699 : FAILURE in 26m 50s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/45d6dfdfcbd9475bacf9694ff1db937c : FAILURE in 38m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a1201cf993e1461a874dfe6c05993484 : FAILURE in 43m 54s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/fb54f0f6ff8945bf800df8b0374bc910 : FAILURE in 48m 53s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/84c2aed5373c4614a56966c4cec55c05 : FAILURE in 42m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4dbeb008ca524b95a7e63fc5a503d777 : FAILURE in 2h 17m 28s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/acd66cc2177c4126b2bfecac12b5ffb0 : FAILURE in 28m 59s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"40ed4d8eb31cd0ca83166c5e16fa63609f4fb7cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-28 14:01:49.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0c8e15ce77e15ac2b74de8fcc5285cb53dff6396","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-28 14:59:20.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/48/788248/7/check/DellEMC_Unity_Manila_dhss_true/7413e65/DellEMC_Unity_Manila_dhss_true/32 : FAILURE in 55m 18s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/48/788248/7/check/DellEMC_Unity_Manila_dhss_false/d612320/DellEMC_Unity_Manila_dhss_false/33 : FAILURE in 57m 13s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"e3a51508182bae9a4c412056d93e2cf744c16310","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-28 16:16:56.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"50407a236ce1b002fa9c5d8fabfe8096d935f2d2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-28 17:40:57.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/48/788248/8/check/DellEMC_Unity_Manila_dhss_true/1ddb18c/DellEMC_Unity_Manila_dhss_true/33 : FAILURE in 53m 46s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/48/788248/8/check/DellEMC_Unity_Manila_dhss_false/4cdff95/DellEMC_Unity_Manila_dhss_false/34 : FAILURE in 1h 23m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"efcd3bb5072101c23380d318b990586053f7746e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 18:13:11.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c20be73fff9460782f2acf448bdfac4 : SUCCESS in 3m 51s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bbd648ef9e444de1b3d7791cbe5db018 : FAILURE in 33m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/97ffdfadd6214ddcbc78b4ce569a3129 : FAILURE in 41m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1c3f06521a32412aa2364d393055d443 : FAILURE in 42m 48s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/000352b0d8bb451fb642cc539bb75980 : FAILURE in 1h 10m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1a902ce82cad491d98aff01be7f59168 : FAILURE in 48m 10s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/16ad005e25c44199b61140ebebc312da : FAILURE in 40m 51s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0d5b969d33d84a368f22f46a77ad0c77 : FAILURE in 43m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6282b90477d2400a849cbaee9357efb6 : FAILURE in 1h 48m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7f2c5ede9ab24b4bb9ba53ab9d7e2779 : FAILURE in 20m 17s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"bc033c1d6b4d594bd18b4659e5941f085a36caf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-28 19:38:36.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"172d19dd163d741e89cbe189984261d675cdc800","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-28 20:38:59.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/48/788248/9/check/DellEMC_Unity_Manila_dhss_true/2253210/DellEMC_Unity_Manila_dhss_true/34 : FAILURE in 1h 00m 02s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/48/788248/9/check/DellEMC_Unity_Manila_dhss_false/9434e19/DellEMC_Unity_Manila_dhss_false/35 : FAILURE in 55m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"ccf356692a3ad34b304c59283e9bce3a2ce56d1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 21:42:32.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/035ea76413ed4654ab5b814ddeebd4df : SUCCESS in 4m 18s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ac16eab299d94ca787d920a98b3a7339 : FAILURE in 28m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2c832b8087f24d3db843b42b2e9fa39c : FAILURE in 30m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f8a6fc297f964a4680b83f091627a6d4 : FAILURE in 36m 09s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1a8d55837f874bfbb141e9f9b6af2213 : FAILURE in 39m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6100d2a5253d4bd68b497cb9bab2678e : FAILURE in 55m 14s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3530e93cdaaa4020ba4bafd044fabb26 : FAILURE in 48m 18s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/77fd38c3936d4f0fbc6f2d93d2877aa8 : FAILURE in 47m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8aec9df9a7d84ceaaa31e98c944e0c6e : FAILURE in 2h 01m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/fb087a0254e0461fb97f61924ab7cf0f : FAILURE in 31m 08s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a6a73577183708984877d9fcda001c2676fee6ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-04-29 02:09:24.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"bf1da36733fe4983ce7fedc285218e8d320a89e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 04:15:35.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/300a20180ecf457fb65463cba85c9c0d : SUCCESS in 3m 07s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d69a2d29a55341f1b53c1fb5a7576006 : FAILURE in 24m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6fdad81a533646239313bc111e1cde1b : FAILURE in 29m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a881a90c0b9247459178b30e1b219491 : FAILURE in 38m 15s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f855dd9860504ce39e9e22f0ea76399c : FAILURE in 36m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6982f91fb82344329c2928d068560133 : FAILURE in 1h 34m 56s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9989dfc8042343f3901ecc01760a27f9 : FAILURE in 1h 09m 45s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/006bada087f04a0eb6245ed8ae14143e : FAILURE in 37m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/55cdb6644e484d84bf3b877c8ff39459 : FAILURE in 2h 01m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1af1b4d94fd54b7ab285a15be41359aa : FAILURE in 16m 59s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"137abca465918ba8e005edffbae44fa13d06a311","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-29 06:24:59.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"71448dcf534f3b9052b03123c79167ad5113a226","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-29 07:25:46.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"199dc3e790ab675c356b4364495ab486d2ee3d2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 09:18:40.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26b92e355724408ebdd6b46d59c72352 : SUCCESS in 3m 37s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b7603dd7f4b540f3a96a94f20d583cf9 : FAILURE in 25m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0719570c7723437399d2cb046a9dd7ae : FAILURE in 32m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/264a68ef2b16484c862772490cc9a190 : FAILURE in 33m 55s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a9ac5aeff482482d9b233b11ce6f432e : FAILURE in 46m 51s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/72f95459067242eabf39f169ecb4e4ff : FAILURE in 50m 55s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3803277eee134805add881f14c0f8545 : FAILURE in 36m 45s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/176424ab81284a78a6cc8bc398e91f2a : FAILURE in 32m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2bf65472105841f89e77eb3d4d53c8c1 : FAILURE in 1h 50m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f638b86ff37e467b9d45a76e88548976 : FAILURE in 18m 07s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"a0e67519b56a6097df143144d951fb5eb0f766f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-29 09:47:13.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"dc0ddca25096cb1cd0f45fbe417656d5648d3f5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-29 11:23:04.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"c583b7c22d57b1d006b50c2382ad47113afced18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-29 13:00:04.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"225e94dc272747cf9f85dcd33cab3681b621a67c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-29 14:05:34.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"55e53171e174de239a216dabda174c1ff294f0df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-29 14:53:48.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"da0561a2f24ac9e7526ca2d62ad39c512d136193","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-04-29 17:07:34.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"209d8e976c929a9634e80d9687ee0f278168dae3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 19:03:28.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e85bf4a02042433c8114d30789de998f : SUCCESS in 3m 26s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/31b4f3c81258436688423a997c9ca98b : SUCCESS in 37m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/95d4dae76535499b818278a4ddc76273 : SUCCESS in 37m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c43b5214e2324b4080ca8e059b6aa7e3 : SUCCESS in 1h 20m 06s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f94dbf6a8b2a4e0f8d77b475aefb727d : SUCCESS in 1h 10m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/cf27fba4c66e4c968327a57aea80fe2b : FAILURE in 1h 24m 33s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/be461a4512ff4aba86f44badcbe9f900 : SUCCESS in 1h 49m 39s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/236084976a634cc692025813aa29dc48 : FAILURE in 49m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/79b6c4282436444cba0e68065f651262 : FAILURE in 1h 23m 54s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1b6228d49a5242a19468cd7bb3dbe5df : FAILURE in 33m 14s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"e2450190146e5a97f460c97d7e9c4ceb54ad8167","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-29 19:03:52.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/48/788248/18/upstream-check/manila-cDOT-manila-tempest-plugin-ss/c6644ad/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/48/788248/18/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/a183b28/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":18},{"id":"e35db444decbe2eef96f0001843e56804b6b75fb","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-29 20:02:50.000000000","message":"Patch Set 18:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/48/788248/18/check/DellEMC_VNX_Manila/9de5ae9/DellEMC_VNX_Manila/19 : SUCCESS in 58m 55s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":18},{"id":"07e1e9d9cff52d9acb17896905e136ba7c2228cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-05-03 07:01:51.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"f8fe86100d9d57ecf4e83746385317d8ffec271b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 08:44:05.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4424eeb93f374a02a9c21d84ae30b9bb : SUCCESS in 3m 35s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/450e387bd51b4a07bd3bb3652ddc7ae0 : SUCCESS in 35m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/dbd62f2271ed4b8887ec9879530ab336 : FAILURE in 30m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b77f0559bcce4ee29807bdfcff677d8a : SUCCESS in 1h 16m 26s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3af1099496b846d4bcbd6f6dd8aa5244 : FAILURE in 1h 02m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/4862444b24aa425da9183f157255693b : FAILURE in 1h 23m 50s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e8b0a17926a1440a835f35e3c5d51a80 : SUCCESS in 1h 21m 41s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3de773edbb594038a96980c6c89f1515 : FAILURE in 49m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/adadc34b5dd449c9b251f7c0ca3c2726 : FAILURE in 1h 41m 02s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3d494c4eeb0442d8b8ce46a801d19c1b : SUCCESS in 19m 57s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"a03ff0acd690eabee0476bbcaf11b5c7fc0a2798","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-03 08:44:43.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/48/788248/19/check/DellEMC_Unity_Manila_dhss_true/4da0361/DellEMC_Unity_Manila_dhss_true/36 : FAILURE in 1h 42m 32s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/48/788248/19/check/DellEMC_Unity_Manila_dhss_false/2042361/DellEMC_Unity_Manila_dhss_false/37 : SUCCESS in 50m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":19},{"id":"d8b5c8bbe20cdf7e219e6d145f2936839c0e1137","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-05-03 08:50:28.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"70c359e787319dd6e1a2d534cccba826e91bb2cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 10:09:34.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7afe43999cf145c4974044621182cbf2 : SUCCESS in 4m 20s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b2bffe07576c49928e0b9941839652f5 : SUCCESS in 29m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/155e13cf7a404578a7f124c4389e9f0b : SUCCESS in 30m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/32cf1896afbe4857815bf62124b25c6e : SUCCESS in 1h 10m 39s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5452989d691c45f094badab985d9c357 : FAILURE in 1h 06m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/438242f3e89b4ed19c60582f6a23e4f6 : FAILURE in 1h 01m 25s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/57826ad1a56244b3b8ef86340dfc5766 : SUCCESS in 1h 17m 42s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/258959fc73b6409da35a4d26a92d06fd : FAILURE in 40m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/79db62aa661c4d08aa05f4c2902c93e2 : FAILURE in 1h 12m 49s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ffd5fe76b2984c088d05a8e2fc72fdba : SUCCESS in 21m 48s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"5016e845f6d387628ae0fe328f6fab5f578a8779","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-03 10:09:55.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/48/788248/20/upstream-check/manila-cDOT-manila-tempest-plugin-ss/8a44c9e/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/48/788248/20/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/66a4a9d/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":20},{"id":"da23c98fc9cf869b9d620bc79d7bfae0f994a258","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-03 10:11:07.000000000","message":"Patch Set 20:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/48/788248/20/check/DellEMC_Unity_Manila_dhss_true/e09b906/DellEMC_Unity_Manila_dhss_true/37 : FAILURE in 1h 20m 20s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/48/788248/20/check/DellEMC_Unity_Manila_dhss_false/f638fab/DellEMC_Unity_Manila_dhss_false/38 : SUCCESS in 49m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":20},{"id":"7ce8054ac4222f1e36ca634629aa2bd56503c7d4","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-05-03 11:09:01.000000000","message":"Patch Set 20:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/48/788248/20/check/DellEMC_VNX_Manila/11fd5d2/DellEMC_VNX_Manila/21 : SUCCESS in 59m 07s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":20},{"id":"04b7eb4409a61534c9f99e4c97a21ec4f83e5837","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-05-03 15:54:54.000000000","message":"Patch Set 20: Code-Review+1\n\nNice job with required changes. Thanks Liron.","accounts_in_message":[],"_revision_number":20},{"id":"d01219008fd4bc4b7e9db0be0e0cf5ee005fcc93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-05-11 07:32:55.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"92984363a6ec16c42e3caf6cb354b2bfb94dbb5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-05-11 08:11:18.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"87221b4ff4b2560e73f5fb29abc8480332103927","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 10:06:35.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee71122ae4324d228e140adfe28a9c10 : SUCCESS in 4m 37s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/79923862dd2c4593ba63f80a4c05d6f3 : SUCCESS in 32m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a09095f79f7d4bf6ba9b190e66ae1b4e : SUCCESS in 39m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5718f5afa4674e75a869ca7e25f7f254 : SUCCESS in 1h 17m 14s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5db5bb5b30084073bdf0d22831635a4c : SUCCESS in 1h 15m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/4e485a9826b748e0809a4ecdebc980eb : SUCCESS in 1h 27m 23s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b7ef3eacdd0645c79cd4c251aea6d590 : SUCCESS in 1h 29m 33s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4b89528eb95748c19cf28f5577f7eb0e : FAILURE in 46m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3da29eb2fca849ffa4df83cdc96ba7c9 : FAILURE in 1h 49m 03s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/046b64a063d74885a865366ff9ada7e4 : SUCCESS in 27m 19s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"eb622ea913d366b209760951bb164844569ffec5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-11 10:27:23.000000000","message":"Patch Set 22:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/48/788248/22/check/DellEMC_Unity_Manila_dhss_true/4479b37/DellEMC_Unity_Manila_dhss_true/60 : FAILURE in 2h 15m 40s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/48/788248/22/check/DellEMC_Unity_Manila_dhss_false/fb0ff26/DellEMC_Unity_Manila_dhss_false/60 : SUCCESS in 54m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":22},{"id":"4582efd729575bc08e43beba56c7d03575547016","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-05-11 11:04:33.000000000","message":"Patch Set 22:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/48/788248/22/check/DellEMC_VNX_Manila/685ad78/DellEMC_VNX_Manila/33 : SUCCESS in 57m 34s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":22},{"id":"dbc720df16a0e7c264d1a0f0509221b6823e640f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-11 12:37:02.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/48/788248/22/upstream-check/manila-cDOT-manila-tempest-plugin-ss/d3e83b0/ : FAILURE in 2h 30m 02s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/48/788248/22/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/ebcc738/ : FAILURE in 1h 43m 08s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":22},{"id":"19503e3c7720a48ea4a6fb61b84039fe75fb4222","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-06-28 13:30:27.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"46533ec2ec6dc0af8777428bb04d950851329e58","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-28 15:10:05.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/48/788248/23/check/DellEMC_Unity_Manila_dhss_true/0b2d1bb/DellEMC_Unity_Manila_dhss_true/165 : FAILURE in 1h 12m 45s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/48/788248/23/check/DellEMC_Unity_Manila_dhss_false/a90a52c/DellEMC_Unity_Manila_dhss_false/163 : FAILURE in 1h 05m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":23},{"id":"4ec056259ea20dd4956dfb51c7fc3859d3c6c468","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 15:21:11.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d85c6b8b89514900b5fda4fa640a203a : SUCCESS in 7m 10s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d979b4f5c5ee42f09c14a43338c56af8 : FAILURE in 30m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/54108b361f2c44359ec6e5546187b5b1 : FAILURE in 37m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4901afed52564092952413915a32f20b : SUCCESS in 1h 08m 39s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/b93e9e9f09a742969541cb673168ab24 : SUCCESS in 1h 05m 32s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/baac15dab37f48398c4b6f1d890ce443 : SUCCESS in 1h 12m 48s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/cd21e139dbf948f580d8f6ce6d2f8ae0 : SUCCESS in 1h 02m 02s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4f604641ae18494399bd17ef5a1adb73 : SUCCESS in 1h 08m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9dd9d1ddffaf4c4e997fd5e1802104c7 : FAILURE in 1h 24m 43s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/84d36d2c71434bd9b31a4708749d6d8a : FAILURE in 21m 54s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/785eeaa34ec248eeb88bdbda1afabc94 : FAILURE in 43m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/59f78cd35c6348938e63d28b9c0154f8 : FAILURE in 1h 45m 14s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1c3267e7834b453d8622f34618162708 : FAILURE in 25m 40s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"bbf4e46f45372bb2ff50f2d4b3ad67558905cc38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-06-29 04:03:24.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"07e6c4f8e88f8048592abadccc76c12b292733a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 05:50:51.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b71a0c6ce1a400f81651fb3be648b6d : SUCCESS in 5m 07s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/587ce8b7e86f48cda9eefc2e6bab332a : SUCCESS in 32m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c298af4de60041bbaaec2df4b7751fe8 : SUCCESS in 43m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/332ccce1f4d3498380fd0760990742b6 : SUCCESS in 1h 12m 28s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d57555fd18d4431eaef987d7f40ab01d : SUCCESS in 1h 26m 55s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/c6ebec231c424c409cb361eb9a5be46d : SUCCESS in 1h 23m 57s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/b55329d497a54e1c88decce37dee07d0 : SUCCESS in 1h 11m 51s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6fd41116d75a4a5db3e4a1c22d4bd02f : FAILURE in 1h 29m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/386ce8ade6dc4402828cdd48a63ae058 : SUCCESS in 1h 28m 49s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/99a8e1e05d4948178dd6dcc8e08ea262 : FAILURE in 28m 45s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0c2bdf0cb9ed4700b22b999645c8c1f8 : SUCCESS in 40m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/08c06196179e409196e86c02d913bb12 : FAILURE in 1h 46m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9d740ea319684f15bd203e36a4fa23b2 : FAILURE in 36m 21s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"55a9ffd1fbd427bcab2fae60c3d75762d11b9bba","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-29 05:51:10.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/48/788248/24/upstream-check/manila-cDOT-manila-tempest-plugin-ss/59fda46/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/48/788248/24/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/8aebbf0/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":24},{"id":"aa86d109b2221ffb87bb088596e18eb1c7eb26b1","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-06-29 06:54:32.000000000","message":"Patch Set 24:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/48/788248/24/check/DellEMC_VNX_Manila/cba642f/DellEMC_VNX_Manila/98 : FAILURE in 1h 03m 20s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":24},{"id":"faab749dbf703734454acc7378a24eef71681c0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-07-20 12:19:17.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"6a0f3c56e9d78dee93e0f3499a77fd1fc9c647b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 15:10:54.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd1e4b089901452e816d36f603fe9f32 : SUCCESS in 3m 50s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/63a22b87b3ee4f69908034c8d99100cc : FAILURE in 34m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e384ee04720b456ba69ec0a1793fd98c : FAILURE in 39m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ffeca9d6b8de4b789e31a67a9becfd3e : SUCCESS in 1h 02m 00s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/169aa60976b243379d172066639da399 : SUCCESS in 1h 28m 35s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/ee0181c54a0b431f9c7f6507cca37eef : SUCCESS in 1h 35m 15s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/5e2029d013cd4ab890c3826f3da33442 : SUCCESS in 1h 04m 11s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0d52d53d06984032be654bf99f396fdd : SUCCESS in 1h 08m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/cac34f51c686484f845530f1035a2c31 : FAILURE in 1h 11m 26s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/eb735a1729014edb9aa2599bef6d7509 : FAILURE in 1h 37m 48s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6fe459cd3d2e47d1b0bbca8e4cbe5909 : FAILURE in 46m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1019d91d91b940e5b1dff2e0f8488ab7 : FAILURE in 2h 45m 41s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/851bf6ca174b4f98bb7b1a4afd6ef1eb : FAILURE in 26m 02s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"82870ea62228fb3bab0e0be75f9e771d804debc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-07-20 19:04:19.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"903e5769be32484dc79940b13e9ececf40d889b4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-07-20 19:18:26.000000000","message":"Patch Set 25: Code-Review-1\n\n(10 comments)\n\nThis is a huge change, thanks for working through these. I\u0027ve left some comments in a prior patch. Please take a look","accounts_in_message":[],"_revision_number":25},{"id":"2a786aa12454281b492be6fc9c38ba6289c90f98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 23:50:41.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bc824aa44fb40d2b37b70f6d806cbe3 : SUCCESS in 4m 42s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a506a4e1165647eb9d469f1031f6fd94 : SUCCESS in 27m 39s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f0e033dbc3b04248b1f14e53cf23a767 : SUCCESS in 41m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a3f218a1af4f4305ab9fa72402d09e86 : SUCCESS in 58m 27s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/c8753cf5819f40afa6bcd9df04392493 : SUCCESS in 1h 08m 18s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/6204fb77f3ca48b0bb7b04f1bbe200ed : SUCCESS in 1h 19m 36s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/a2ae498b32534d178b7eabfe98c02052 : SUCCESS in 1h 14m 00s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/34d713afe25d4f5189fb6f8bab37aa1e : SUCCESS in 53m 36s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0487e92360c446c2ab5b56bd0b8041ba : FAILURE in 55m 16s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/cd3bd2b144794d08a8e892b28385bb1a : SUCCESS in 1h 42m 04s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4696c8d08adb435e91ace338ef417df3 : SUCCESS in 36m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c23cecd1c89640ec8fd969d5a8fc2d55 : FAILURE in 1h 46m 11s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/568f9319abd04a42bcd21adfa542a189 : FAILURE in 23m 51s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"45c02308a24f154bbd0829b689673074531b0098","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-20 23:51:00.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/48/788248/26/upstream-check/manila-cDOT-manila-tempest-plugin-ss/39825f4/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/48/788248/26/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/264b692/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":26},{"id":"29e02fe270686f2ad026efdf75a0b2fb72f08b91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-07-21 07:31:36.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"775cec8e4cd305d4ce19de066bb41adf043c0185","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-07-21 07:38:24.000000000","message":"Patch Set 27:\n\n(10 comments)\n\nGoutham, thank you so much for your comments!","accounts_in_message":[],"_revision_number":27},{"id":"ff2be6a6a7f5ac49814b1866416b0b4e59012352","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-07-21 08:55:28.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"d95a0ed1607b9fc5fca7d18e8a6d6526e664d986","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-07-21 09:45:30.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"23d543fafb0cce0e929dc7397245655812fa0c7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 11:39:28.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b7cd638501348978888c09a57c9d1c1 : SUCCESS in 3m 39s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/39f520741a8f47d29afc1cebdd7a786c : SUCCESS in 26m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4db9ead7c6d945a0b9a72b8c6f3e0b54 : SUCCESS in 33m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fb25b2f6084749869c833e401a8ee276 : SUCCESS in 1h 01m 58s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/99aafc6705b84e41bd5cb12f4c86a713 : SUCCESS in 1h 15m 25s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/5349f459f00c476ebd443d933fe1d6b8 : SUCCESS in 1h 10m 01s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/92867b35980b4e4cabef10b4a3a3b531 : SUCCESS in 1h 10m 37s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b8253104dc644c3faf611590c58d4c9a : SUCCESS in 1h 08m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0fe26a4f3a9e4298909203caa9210845 : SUCCESS in 1h 25m 17s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/93bd090e69df4acb83e87f68436e379c : SUCCESS in 1h 48m 19s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/010e1e0210a846ccbcc3c760846c7298 : SUCCESS in 42m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4e571018540c4243a0bbe59543b9dfa4 : FAILURE in 1h 41m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5e724e88bb3245a08d7ad4bc403c05d4 : FAILURE in 22m 21s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"39e37e1947e376bbe78b1f76abab4f17d8ca29f8","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-21 11:39:44.000000000","message":"Patch Set 29:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/48/788248/29/upstream-check/manila-cDOT-manila-tempest-plugin-ss/2442406/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/48/788248/29/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/1fe031e/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":29},{"id":"7b307b0a9e30fc8d5c8d8a4bfd2192a609c98e67","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-07-21 17:14:57.000000000","message":"Patch Set 29: Code-Review+2\n\n(2 comments)\n\nLGTM, thank you","accounts_in_message":[],"_revision_number":29},{"id":"d9b05c7df897893221288c06e2c2499127b1bfa9","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-07-21 17:20:27.000000000","message":"Patch Set 29: Code-Review+2\n\nLooks good, thanks Liron.","accounts_in_message":[],"_revision_number":29},{"id":"2a13e8a9068594a9808fea2662e12ce568ce4ef6","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-07-21 17:53:06.000000000","message":"Patch Set 29: Workflow+1\n\nOur CI jobs passed here; lets get this in and see if there are issues we need to work through after a wider gamut of CIs run this via the manila repo","accounts_in_message":[],"_revision_number":29},{"id":"96a109c80371473ac9f5d7a1a79c397ea6cf783f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 17:53:36.000000000","message":"Patch Set 29: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":29},{"id":"70400abd9467da384ae68660391745d99d0e3ab8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 18:58:20.000000000","message":"Patch Set 29: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c266c9cd8ea64657bd2df26d69228fa4 : SUCCESS in 3m 36s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/eb354bf447a24bfa91ba66d2b963b310 : SUCCESS in 27m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/06faf66f0b784d0cbe2692a81b62bd5e : SUCCESS in 34m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0ad92ae9c0774141a43dcc6f7da492eb : SUCCESS in 1h 04m 06s","accounts_in_message":[],"_revision_number":29},{"id":"a03f69d5769fda610af85f2a00f3c693a5784a97","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 18:58:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":29}],"current_revision_number":29,"current_revision":"86f2432476ba9128bb2c26a78ae3b3f53327d4bd","revisions":{"e19c984b7c5776ff95a673ea6a5aad08c8162e87":{"kind":"REWORK","_number":1,"created":"2021-04-27 11:33:07.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/1"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"subject":"[WIP] Enable http response extraction","message":"[WIP] Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e19c984b7c5776ff95a673ea6a5aad08c8162e87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e19c984b7c5776ff95a673ea6a5aad08c8162e87"}]},"branch":"refs/heads/master"},"c7373e671490146ebebceda4b69389c8d82860ed":{"kind":"REWORK","_number":2,"created":"2021-04-27 11:46:14.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/2"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:45:53.000000000","tz":180},"subject":"[WIP] Enable http response extraction","message":"[WIP] Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c7373e671490146ebebceda4b69389c8d82860ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c7373e671490146ebebceda4b69389c8d82860ed"}]},"branch":"refs/heads/master"},"fc0e6d367e716d03757d876474b5f803f3374f78":{"kind":"REWORK","_number":3,"created":"2021-04-27 20:26:04.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/48/788248/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/3"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2021-04-27 20:25:21.000000000","tz":-240},"subject":"[WIP] Enable http response extraction","message":"[WIP] Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fc0e6d367e716d03757d876474b5f803f3374f78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fc0e6d367e716d03757d876474b5f803f3374f78"}]},"branch":"refs/heads/master"},"72128110b6575b37f0d41e7a646b2450a4101462":{"kind":"REWORK","_number":4,"created":"2021-04-28 03:46:26.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/48/788248/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/4"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2021-04-28 03:44:28.000000000","tz":-240},"subject":"[WIP] Enable http response extraction","message":"[WIP] Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/72128110b6575b37f0d41e7a646b2450a4101462"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/72128110b6575b37f0d41e7a646b2450a4101462"}]},"branch":"refs/heads/master"},"c871ffcfa7e88ba90ed684cdfd6b44c6b236738e":{"kind":"REWORK","_number":5,"created":"2021-04-28 06:48:17.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/5"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-28 06:47:55.000000000","tz":180},"subject":"[WIP] Enable http response extraction","message":"[WIP] Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c871ffcfa7e88ba90ed684cdfd6b44c6b236738e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c871ffcfa7e88ba90ed684cdfd6b44c6b236738e"}]},"branch":"refs/heads/master"},"08fc096ca2c47e0263b5ec711cf8f7605b14a434":{"kind":"REWORK","_number":6,"created":"2021-04-28 09:45:12.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/6"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-28 09:44:53.000000000","tz":180},"subject":"[WIP] Enable http response extraction","message":"[WIP] Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/08fc096ca2c47e0263b5ec711cf8f7605b14a434"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/08fc096ca2c47e0263b5ec711cf8f7605b14a434"}]},"branch":"refs/heads/master"},"50459b395b75c3d75a5df66f419f40da85c80d0e":{"kind":"REWORK","_number":7,"created":"2021-04-28 14:01:49.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/7"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-28 14:01:29.000000000","tz":180},"subject":"[WIP] Enable http response extraction","message":"[WIP] Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/50459b395b75c3d75a5df66f419f40da85c80d0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/50459b395b75c3d75a5df66f419f40da85c80d0e"}]},"branch":"refs/heads/master"},"1e3a309f3f7a6408b9d5ff4d25658e7caeb0d026":{"kind":"REWORK","_number":8,"created":"2021-04-28 16:16:56.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/8"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-28 16:16:35.000000000","tz":180},"subject":"[WIP] Enable http response extraction","message":"[WIP] Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1e3a309f3f7a6408b9d5ff4d25658e7caeb0d026"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1e3a309f3f7a6408b9d5ff4d25658e7caeb0d026"}]},"branch":"refs/heads/master"},"6168e240a9edabb0be787523540fcc6c47a564f3":{"kind":"REWORK","_number":9,"created":"2021-04-28 19:38:36.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/9"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-28 19:38:15.000000000","tz":180},"subject":"[WIP] Enable http response extraction","message":"[WIP] Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6168e240a9edabb0be787523540fcc6c47a564f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6168e240a9edabb0be787523540fcc6c47a564f3"}]},"branch":"refs/heads/master"},"1afe49c17c0fbf7f7b22fdd6038ab0b591b6013c":{"kind":"REWORK","_number":10,"created":"2021-04-29 02:09:24.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/48/788248/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/10"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2021-04-29 02:08:49.000000000","tz":-240},"subject":"[WIP] Enable http response extraction","message":"[WIP] Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1afe49c17c0fbf7f7b22fdd6038ab0b591b6013c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1afe49c17c0fbf7f7b22fdd6038ab0b591b6013c"}]},"branch":"refs/heads/master"},"9793297a0ec57f630a8499996485528acecd494d":{"kind":"REWORK","_number":11,"created":"2021-04-29 06:24:59.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/11"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-29 06:24:38.000000000","tz":180},"subject":"[WIP] Enable http response extraction","message":"[WIP] Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9793297a0ec57f630a8499996485528acecd494d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9793297a0ec57f630a8499996485528acecd494d"}]},"branch":"refs/heads/master"},"329862e89bee86c7fbd7b2dbe0203cf60a41a750":{"kind":"REWORK","_number":12,"created":"2021-04-29 07:25:46.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/12"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-29 07:25:25.000000000","tz":180},"subject":"[WIP] Enable http response extraction","message":"[WIP] Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/329862e89bee86c7fbd7b2dbe0203cf60a41a750"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/329862e89bee86c7fbd7b2dbe0203cf60a41a750"}]},"branch":"refs/heads/master"},"e268cb034df732b6c666ac45cc88b89c5ac64ddb":{"kind":"REWORK","_number":13,"created":"2021-04-29 09:47:13.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/13"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-29 09:46:52.000000000","tz":180},"subject":"[WIP] Enable http response extraction","message":"[WIP] Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e268cb034df732b6c666ac45cc88b89c5ac64ddb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e268cb034df732b6c666ac45cc88b89c5ac64ddb"}]},"branch":"refs/heads/master"},"d0ef523c2d5d2cc70a1403b2eb28645af60e33ce":{"kind":"REWORK","_number":14,"created":"2021-04-29 11:23:04.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/14"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-29 11:22:41.000000000","tz":180},"subject":"Enable http response extraction","message":"Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d0ef523c2d5d2cc70a1403b2eb28645af60e33ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d0ef523c2d5d2cc70a1403b2eb28645af60e33ce"}]},"branch":"refs/heads/master"},"6d2241f4d90e222900b9ed8f05b2e0365565b8e7":{"kind":"REWORK","_number":15,"created":"2021-04-29 13:00:04.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/15"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-29 12:59:45.000000000","tz":180},"subject":"Enable http response extraction","message":"Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6d2241f4d90e222900b9ed8f05b2e0365565b8e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6d2241f4d90e222900b9ed8f05b2e0365565b8e7"}]},"branch":"refs/heads/master"},"7bb818613096f17c9fb1138b29a793a616bf733b":{"kind":"REWORK","_number":16,"created":"2021-04-29 14:05:34.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/16"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-29 14:05:13.000000000","tz":180},"subject":"Enable http response extraction","message":"Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7bb818613096f17c9fb1138b29a793a616bf733b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7bb818613096f17c9fb1138b29a793a616bf733b"}]},"branch":"refs/heads/master"},"f57ac4828550c12f4d00e54adfe9a828915a3e17":{"kind":"REWORK","_number":17,"created":"2021-04-29 14:53:48.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/17"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-29 14:53:28.000000000","tz":180},"subject":"Enable http response extraction","message":"Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f57ac4828550c12f4d00e54adfe9a828915a3e17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f57ac4828550c12f4d00e54adfe9a828915a3e17"}]},"branch":"refs/heads/master"},"514cd896ae1aac88551de5dda33e411bde840911":{"kind":"REWORK","_number":18,"created":"2021-04-29 17:07:34.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/18"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-29 17:07:15.000000000","tz":180},"subject":"Enable http response extraction","message":"Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/514cd896ae1aac88551de5dda33e411bde840911"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/514cd896ae1aac88551de5dda33e411bde840911"}]},"branch":"refs/heads/master"},"7cf40968647649e06d82dee223c6201777c61bac":{"kind":"REWORK","_number":19,"created":"2021-05-03 07:01:51.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/19"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-05-03 07:01:31.000000000","tz":180},"subject":"Enable http response extraction","message":"Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7cf40968647649e06d82dee223c6201777c61bac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7cf40968647649e06d82dee223c6201777c61bac"}]},"branch":"refs/heads/master"},"34fecc3ac9d105b16a7456233d28eedadad790f9":{"kind":"REWORK","_number":20,"created":"2021-05-03 08:50:28.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/20"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-05-03 08:50:07.000000000","tz":180},"subject":"Enable http response extraction","message":"Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/34fecc3ac9d105b16a7456233d28eedadad790f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/34fecc3ac9d105b16a7456233d28eedadad790f9"}]},"branch":"refs/heads/master"},"1b131774d4e8c9f6587f6b4717bce9526ee2b2d0":{"kind":"REWORK","_number":21,"created":"2021-05-11 07:32:55.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/21"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-05-11 07:32:34.000000000","tz":180},"subject":"Enable http response extraction","message":"Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1b131774d4e8c9f6587f6b4717bce9526ee2b2d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1b131774d4e8c9f6587f6b4717bce9526ee2b2d0"}]},"branch":"refs/heads/master"},"cc06b1ab3bab2378a73d9a88fbf05c6b4a44811f":{"kind":"REWORK","_number":22,"created":"2021-05-11 08:11:18.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/22"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-05-11 08:10:53.000000000","tz":180},"subject":"Enable http response extraction","message":"Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc06b1ab3bab2378a73d9a88fbf05c6b4a44811f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc06b1ab3bab2378a73d9a88fbf05c6b4a44811f"}]},"branch":"refs/heads/master"},"f3b95d68f1bc8659b9efb831af6336335a348859":{"kind":"REWORK","_number":23,"created":"2021-06-28 13:30:27.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/23"}}},"commit":{"parents":[{"commit":"b10b9d6dd8ffda07268f572d4350009e8807616f","subject":"Merge \"Add stable branch jobs on the plugins master gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b10b9d6dd8ffda07268f572d4350009e8807616f"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-06-28 13:29:47.000000000","tz":180},"subject":"Enable http response extraction","message":"Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f3b95d68f1bc8659b9efb831af6336335a348859"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f3b95d68f1bc8659b9efb831af6336335a348859"}]},"branch":"refs/heads/master"},"b6fd739f1145740bf87e03f99aeb2ebe16264ff2":{"kind":"REWORK","_number":24,"created":"2021-06-29 04:03:24.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/24"}}},"commit":{"parents":[{"commit":"b10b9d6dd8ffda07268f572d4350009e8807616f","subject":"Merge \"Add stable branch jobs on the plugins master gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b10b9d6dd8ffda07268f572d4350009e8807616f"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-06-29 04:02:59.000000000","tz":180},"subject":"Enable http response extraction","message":"Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b6fd739f1145740bf87e03f99aeb2ebe16264ff2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b6fd739f1145740bf87e03f99aeb2ebe16264ff2"}]},"branch":"refs/heads/master"},"1abd40193c86e2c2e51e3798733224e8893da023":{"kind":"REWORK","_number":25,"created":"2021-07-20 12:19:17.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/25"}}},"commit":{"parents":[{"commit":"149acc51adf0c17b22fe690a2e9018596bcc5bbd","subject":"Merge \"Fix class inheritances in base.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/149acc51adf0c17b22fe690a2e9018596bcc5bbd"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-07-20 12:18:56.000000000","tz":180},"subject":"Enable http response extraction","message":"Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1abd40193c86e2c2e51e3798733224e8893da023"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1abd40193c86e2c2e51e3798733224e8893da023"}]},"branch":"refs/heads/master"},"b3a712925d69d639fe5935e670caff0a185ce48c":{"kind":"REWORK","_number":26,"created":"2021-07-20 19:04:19.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/26"}}},"commit":{"parents":[{"commit":"149acc51adf0c17b22fe690a2e9018596bcc5bbd","subject":"Merge \"Fix class inheritances in base.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/149acc51adf0c17b22fe690a2e9018596bcc5bbd"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-07-20 19:03:53.000000000","tz":180},"subject":"Enable http response extraction","message":"Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b3a712925d69d639fe5935e670caff0a185ce48c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b3a712925d69d639fe5935e670caff0a185ce48c"}]},"branch":"refs/heads/master"},"0a3a675476ce21d5509fe47ea264936ad0199662":{"kind":"REWORK","_number":27,"created":"2021-07-21 07:31:36.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/27"}}},"commit":{"parents":[{"commit":"149acc51adf0c17b22fe690a2e9018596bcc5bbd","subject":"Merge \"Fix class inheritances in base.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/149acc51adf0c17b22fe690a2e9018596bcc5bbd"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-07-21 07:31:12.000000000","tz":180},"subject":"Enable http response extraction","message":"Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0a3a675476ce21d5509fe47ea264936ad0199662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0a3a675476ce21d5509fe47ea264936ad0199662"}]},"branch":"refs/heads/master"},"b37653d599aacbb4817d2c77a5f132209f4e9a6f":{"kind":"REWORK","_number":28,"created":"2021-07-21 08:55:28.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/28"}}},"commit":{"parents":[{"commit":"149acc51adf0c17b22fe690a2e9018596bcc5bbd","subject":"Merge \"Fix class inheritances in base.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/149acc51adf0c17b22fe690a2e9018596bcc5bbd"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-07-21 08:55:06.000000000","tz":180},"subject":"Enable http response extraction","message":"Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b37653d599aacbb4817d2c77a5f132209f4e9a6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b37653d599aacbb4817d2c77a5f132209f4e9a6f"}]},"branch":"refs/heads/master"},"86f2432476ba9128bb2c26a78ae3b3f53327d4bd":{"kind":"REWORK","_number":29,"created":"2021-07-21 09:45:30.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/48/788248/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/788248/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/788248/29"}}},"commit":{"parents":[{"commit":"149acc51adf0c17b22fe690a2e9018596bcc5bbd","subject":"Merge \"Fix class inheritances in base.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/149acc51adf0c17b22fe690a2e9018596bcc5bbd"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-27 11:23:05.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-07-21 09:45:09.000000000","tz":180},"subject":"Enable http response extraction","message":"Enable http response extraction\n\nThis patch is a preparation for adding RBAC tests.\nCurrently, we only return the body without the response\nin manila REST client. The response is necessary for\nthe testing, because we need to check the returned code\nstatus and to compare it with the expected status.\nThis way we will check if the user has the right\npermissions for the action.\n\nChange-Id: If0e39afb635c469a25919770a869087bf5def561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/86f2432476ba9128bb2c26a78ae3b3f53327d4bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/86f2432476ba9128bb2c26a78ae3b3f53327d4bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
