)]}'
{"id":"openstack%2Fmanila-tempest-plugin~792026","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I2eceb1d5fe42b1b779e115a0144a5d9639d3753f","project":"openstack/manila-tempest-plugin","branch":"master","topic":"fix-test-skips","hashtags":[],"change_id":"I2eceb1d5fe42b1b779e115a0144a5d9639d3753f","subject":"Fix API version test skip logic","status":"MERGED","created":"2021-05-18 18:04:34.000000000","updated":"2021-06-23 08:06:29.000000000","submitted":"2021-05-29 19:15:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"792026-fix-test-skips","meta_rev_id":"6b94bab59dffa364e0ec5aa30f80713cfe14ca7e","_number":792026,"virtual_id_number":792026,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-29 19:15:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2021-05-29 17:55:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2021-05-29 17:05:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2021-05-29 17:55:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"reviewers":{"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-18 19:20:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-18 19:35:59.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-18 20:55:33.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-20 19:06:31.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-05-28 18:29:48.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2021-05-29 17:05:23.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2021-05-29 17:55:16.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"1e1493088ea8e63ef2be4082815f11a5e401414c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-18 18:04:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c7680355aae887a3c78c76e8375efe2b327ceff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 19:20:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/540f7c94ea7345f980a118867207a49a : SUCCESS in 3m 16s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/21ced4f1a5fc467c99bf3aaaf1c6ff6b : SUCCESS in 26m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/77d8d4afd8034dccb0b3561599c63a58 : SUCCESS in 31m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3c95d099ce1346a1b8b973162e81e3ca : SUCCESS in 1h 05m 35s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/06edc9b4524c4fc9a1ec028acb20430c : SUCCESS in 1h 06m 49s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/daf340ee675c4d4d8179039f1967e4a7 : SUCCESS in 1h 01m 39s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9f4cd390721d4904bd596e11675a07f2 : SUCCESS in 1h 09m 54s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8f10404705834216ae980ae815ecdf05 : FAILURE in 42m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9d1f0ea58946443cb9d88669067fe335 : FAILURE in 1h 08m 24s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/80feae254c9e4e7c94c0b1df3cb990b2 : SUCCESS in 26m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ef30167d3838782e95d8eb646fbf95bfb5afb0a7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-18 19:35:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/26/792026/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/1044e91/ : FAILURE in 10m 37s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/26/792026/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/56829b5/ : FAILURE in 15m 34s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"c8b1894ee272e2bcd5beafac24a6d5960c43b617","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-05-18 20:55:33.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/26/792026/1/check/DellEMC_VNX_Manila/fd60a81/DellEMC_VNX_Manila/41 : FAILURE in 1h 35m 07s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"052ab76a6427527aec87e8fdb75f884d3473029c","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-05-20 19:06:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5c18042a9fc36346b7e0f52c5d0b4f6eff610f14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-28 02:58:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"98a42aee874edd19af4e6f57b075e3a14f54c683","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 03:43:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63433db7c77a42dda89c0901c1e93a4c : FAILURE in 4m 00s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/81c8985eada0479a8dea4ccf6ca37a21 : FAILURE in 21m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3a908b31c0d94c9e848b8d32c4146b0b : FAILURE in 21m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d333c3756d5047b6ac32ed96fc8fe9be : FAILURE in 33m 33s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f85e686b1fb246148510fd4bb4903925 : FAILURE in 33m 24s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7e12308d8c37458f8aefdf1dc92819fe : FAILURE in 36m 00s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0a1d5001770a4abb8bb0edd0151958a2 : FAILURE in 37m 20s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3894216141ea4bc6b6241c134b6820a2 : FAILURE in 35m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/202ff0c1f72c4d109472cc94a06f4598 : FAILURE in 44m 59s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e80e8cf9b9b541bf836ba89e8b1c3121 : FAILURE in 22m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a6afb6b29a958c44eb78879f3b368d50d2626bd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-28 07:58:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2489912c7039685ecc98cb2455e5962cf502d712","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 11:01:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3546794c31ab49c886f5766254644c81 : SUCCESS in 4m 03s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/606412f1cbb546df83ce72c9e65573cf : SUCCESS in 33m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/67bcb7afc8c04f2b973c6ee14b57168a : SUCCESS in 36m 33s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d11d97e70fee4749afe1ff02885d26bb : SUCCESS in 1h 15m 08s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/72c37105294249958d42e70135e7863f : FAILURE in 1h 22m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/4c1042d0f7e44edeaf31d7256684e1d7 : SUCCESS in 1h 33m 25s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2a62a3411a954018b7231ee49788be06 : SUCCESS in 1h 46m 47s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2d2f7731bf434969aafdab8283db52c7 : SUCCESS in 59m 56s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e8e5eae44ae64d91914daa0cabfe8f2d : FAILURE in 2h 57m 11s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a7dc7530c6bf406aa0157b39b59a9586 : FAILURE in 25m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bd72e9a41f4dc6ae2370eb09ea628ae05a45e3d0","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-05-29 17:05:23.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks for refactoring this, LGTM!","accounts_in_message":[],"_revision_number":3},{"id":"deeb00a1819d952e1169950676c7bbde8f27c685","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-05-29 17:55:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThank you for picking this up. LGTM. Thanks!","accounts_in_message":[],"_revision_number":3},{"id":"18aad85188860d18c03713e5177c81c42e108bb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-29 17:55:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"513c6f1914c61d133c611b73bfb50292c00e4ca4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-29 19:15:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8946d2d1a1574998bfda9e47cf181277 : SUCCESS in 3m 25s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a5b4fe8140064843bc385103879bf018 : SUCCESS in 34m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/361975df68b74c8b8d33d09766955d05 : SUCCESS in 41m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/03d445161bd1405291349374abc88d8d : SUCCESS in 1h 14m 31s","accounts_in_message":[],"_revision_number":3},{"id":"a67cb082412d3d88bbb2c5ea7ce86d936def08e6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-29 19:15:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"42599932529e449ff33b7fdc461a439d0ed4a5ba","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-06-23 08:05:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6b94bab59dffa364e0ec5aa30f80713cfe14ca7e","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-06-23 08:06:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a0acf25969daf37f3b1f527809f297e6ba227cc9","revisions":{"61e60304cab7f35acce7d5674e51e0d7b72b8047":{"kind":"REWORK","_number":1,"created":"2021-05-18 18:04:34.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/26/792026/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/26/792026/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/792026/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/792026/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/792026/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/792026/1"}}},"commit":{"parents":[{"commit":"0031ba11387091fcc830047cd265d7ed9f52c129","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0031ba11387091fcc830047cd265d7ed9f52c129"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-05-18 18:04:26.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-05-18 18:04:26.000000000","tz":-420},"subject":"Skip tests by raising skipException directly","message":"Skip tests by raising skipException directly\n\nWe\u0027re using a deprecated method [1] from testtools\ncurrently, we could rather just directly raise\nskipException like we do in many other places\nin the repository.\n\nTrivialFix\n\n[1] https://github.com/testing-cabal/testtools/blob/f38af6765d4e412b7a7ed1c77ddc9d68320330aa/testtools/testcase.py#L334-L339\n\nChange-Id: I2eceb1d5fe42b1b779e115a0144a5d9639d3753f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/61e60304cab7f35acce7d5674e51e0d7b72b8047"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/61e60304cab7f35acce7d5674e51e0d7b72b8047"}]},"branch":"refs/heads/master"},"62ed2e82dfefc2491e1bd5d8adb2903a1273ec6a":{"kind":"REWORK","_number":2,"created":"2021-05-28 02:58:35.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/26/792026/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/26/792026/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/792026/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/792026/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/792026/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/792026/2"}}},"commit":{"parents":[{"commit":"0031ba11387091fcc830047cd265d7ed9f52c129","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0031ba11387091fcc830047cd265d7ed9f52c129"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-05-28 02:57:55.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-05-28 02:58:19.000000000","tz":-420},"subject":"Fix API version test skip logic","message":"Fix API version test skip logic\n\nAPI tests must be evaluated by min and max\nversions, however, we\u0027ve multiple places\nwhere we don\u0027t seem to evaluate the min_version;\nthis leads to incorrect behavior when test\nrunners set a different min_version in their\ntest configuration.\n\nWe can also consolidate the decorators and\ncallables that we have. We attempted this in\nthe recent past [1], however, we replaced the\ncallables with calling a decorator method, which\nresulted in many tests not being skipped.\n\nThis bugfix is currently necessary to get\ntests working against stable/ussuri and\nstable/train branches.\n\n[1] I0c35aff993e67c1ef44c27580f8c2c829a6275f7\nChange-Id: I2eceb1d5fe42b1b779e115a0144a5d9639d3753f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/62ed2e82dfefc2491e1bd5d8adb2903a1273ec6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/62ed2e82dfefc2491e1bd5d8adb2903a1273ec6a"}]},"branch":"refs/heads/master"},"a0acf25969daf37f3b1f527809f297e6ba227cc9":{"kind":"REWORK","_number":3,"created":"2021-05-28 07:58:22.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/26/792026/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/26/792026/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/792026/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/792026/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/792026/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/792026/3"}}},"commit":{"parents":[{"commit":"0031ba11387091fcc830047cd265d7ed9f52c129","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0031ba11387091fcc830047cd265d7ed9f52c129"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-05-28 02:57:55.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-05-28 07:57:45.000000000","tz":-420},"subject":"Fix API version test skip logic","message":"Fix API version test skip logic\n\nAPI tests must be evaluated by min and max\nversions, however, we\u0027ve multiple places\nwhere we don\u0027t seem to evaluate the min_version;\nthis leads to incorrect behavior when test\nrunners set a different min_version in their\ntest configuration.\n\nWe can also consolidate the decorators and\ncallables that we have. We attempted this in\nthe recent past [1], however, we replaced the\ncallables with calling a decorator method, which\nresulted in many tests not being skipped.\n\nThis bugfix is currently necessary to get\ntests working against stable/ussuri and\nstable/train branches.\n\n[1] I0c35aff993e67c1ef44c27580f8c2c829a6275f7\nChange-Id: I2eceb1d5fe42b1b779e115a0144a5d9639d3753f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a0acf25969daf37f3b1f527809f297e6ba227cc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a0acf25969daf37f3b1f527809f297e6ba227cc9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
