)]}'
{"id":"openstack%2Fmanila-tempest-plugin~792217","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Id312005dc0cc9e039a9c6a8a58476965e8d6954d","project":"openstack/manila-tempest-plugin","branch":"master","topic":"devstack-plugin","hashtags":[],"change_id":"Id312005dc0cc9e039a9c6a8a58476965e8d6954d","subject":"[ci] Don\u0027t install manila-tempest-plugin system-wide","status":"MERGED","created":"2021-05-19 17:31:36.000000000","updated":"2021-06-01 13:37:09.000000000","submitted":"2021-06-01 13:37:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"792217-devstack-plugin","meta_rev_id":"081e135ec41e20d458f46fab7850ffa142531038","_number":792217,"virtual_id_number":792217,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-01 13:37:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"recommended":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":1,"date":"2021-05-24 18:47:02.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-01 11:05:13.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":2,"date":"2021-05-31 05:53:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"all":[{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-01 11:05:13.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-19 18:13:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-20 19:29:49.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"CC"},{"updated":"2021-05-20 22:15:32.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-05-21 22:44:37.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-21 23:59:20.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-24 18:47:02.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-05-27 14:10:24.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2021-06-01 11:05:13.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"}],"messages":[{"id":"e555ea00fd090579d02b4c505037df41a0b0e351","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-19 17:31:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77df415f3201d727aeb21d3111c372a4809a6cde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 18:13:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eba8013ecc0485c8680aa5a5c029f65 : SUCCESS in 4m 01s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cf7aea4747fa458cbe684c86ad05abd4 : SUCCESS in 32m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/168c9a2ab0e64a01aac2a86dff1fe760 : SUCCESS in 40m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a269b3ac322648449d3a87b5212b6d28 : FAILURE in 33m 05s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/513ebb0882074daab15434dcd8238a73 : FAILURE in 29m 10s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/91cd665f10ad4a03809245ffd8536ba0 : FAILURE in 31m 14s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/cc820152a2de46edae87db7053bb5b3c : FAILURE in 33m 42s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/27da435a357140e58acefbe517263483 : FAILURE in 31m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e8345c8c43064c8e9755b7b619259097 : FAILURE in 35m 40s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b8053d0791f84f968a0e855111de9062 : FAILURE in 26m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf1bada34a03a9d7d69e8c544ffdf101a5aa3e31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-20 19:29:25.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1d891ad21554b237f6f6ca4815ef4ed823cc26a2","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-05-20 19:29:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"394d9d796e3dbf05481e1c959286217b647fa8a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 21:32:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27cdb889ed874b9e86ff4236fd6e62ff : SUCCESS in 3m 36s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cbdd49eaa11a4539af9be453086315e7 : TIMED_OUT in 1h 03m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/dee1ae399d684c38957a59f929628b9f : SUCCESS in 37m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f62d74ee9c524884b1b8d31a80e1efe9 : SUCCESS in 1h 02m 56s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0ed6fd3170cf4621bd33f7833a7f0f0f : FAILURE in 1h 19m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e4f7769780eb4b00a8ded11d2973dc85 : SUCCESS in 56m 27s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/16b2b6785cef41c0a20c2b3b1cb29dbd : FAILURE in 1h 11m 56s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bbb799d71636442cabc7a149929a6084 : FAILURE in 36m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8844f5f73d694380b00a5b147b225a03 : FAILURE in 2h 00m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/52d8e938aca94236934e8e0836897593 : FAILURE in 24m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"abb6fda73b20fff079828043bc4a6cea2d1e628c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-20 22:15:32.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/17/792217/2/check/DellEMC_Unity_Manila_dhss_true/93986e8/DellEMC_Unity_Manila_dhss_true/68 : FAILURE in 2h 13m 43s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/17/792217/2/check/DellEMC_Unity_Manila_dhss_false/ac653f0/DellEMC_Unity_Manila_dhss_false/68 : SUCCESS in 1h 10m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"8641b901eff26e70df03ec159657d8f0fdef3d3d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-20 23:08:00.000000000","message":"Patch Set 2:\n\nrecheck\n\nChange works - failures seem unrelated.","accounts_in_message":[],"_revision_number":2},{"id":"1efb29d0e1fd00cbf3ca7cff714b30eeef17af45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 00:34:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d63cfc1541f746c8a94f523c19ee38a7 : SUCCESS in 3m 40s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7db44022cf8f4885b2539587cee5411d : FAILURE in 12m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/dbf84baba66a43b0a6f76996def34546 : SUCCESS in 32m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/23837894be2f4b4586553ac5f96998c4 : SUCCESS in 52m 38s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/78299b0b041f4ed7baea20092f4e7657 : SUCCESS in 55m 36s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/291682f845e0400db3e1ac319f1a902a : FAILURE in 1h 06m 36s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/05c889ee9dc24319b8ce1d90f38fe859 : SUCCESS in 1h 24m 29s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5f8afec946c4437d8ada66c2734f3b94 : FAILURE in 38m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/cc1c0cd83c874d7682aa232d21934300 : FAILURE in 1h 17m 00s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/76fb62b44a5b4a2681d73720130a459f : SUCCESS in 21m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a2ea33a2b8da69ca415dfab80c5b8efe62c6614b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-21 19:58:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5cf711ad30e3a8449a9a9341f2fd71c7181fcf3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 22:24:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/548f15fad0b14221b9065f7506f0f72c : SUCCESS in 3m 36s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/23e90b764cc54e52bb0ae66306c3e562 : SUCCESS in 32m 47s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7b6a8abacf0e4adb96fe8f090341b24d : SUCCESS in 32m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9e6d804680dd4071be0b68e157ae28bd : SUCCESS in 1h 05m 43s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/556c03c3b0324d6e8f6bf69491ed0009 : FAILURE in 47m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/bb31b6daa4144377801d793d5fbc4df0 : SUCCESS in 1h 05m 42s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b09ecd1513ef4d96898339f1a712fc32 : SUCCESS in 1h 31m 40s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7e8c34bd8b8849ef8247e8a9562cfb88 : SUCCESS in 32m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/45bfd34585f1486dacbe5d04e349f672 : FAILURE in 2h 24m 42s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a50c64f98e1f4f41a64c1fe4cb88a337 : SUCCESS in 21m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"13012fee2c90c2a1ff8e937955675e89392d8d35","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-21 22:44:37.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/17/792217/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/439eca7/ : FAILURE in 11m 10s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/17/792217/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/247e8fb/ : FAILURE in 19m 59s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"ab8683c028433d6121b138bff623a3ac2b264385","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-05-21 23:59:20.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/17/792217/2/check/DellEMC_VNX_Manila/f101bf7/DellEMC_VNX_Manila/48 : FAILURE in 1h 34m 46s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"6e4989de3019b864ddaf6275727fbd4314094650","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-22 00:46:45.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/17/792217/2/check/DellEMC_Unity_Manila_dhss_true/27ffc0c/DellEMC_Unity_Manila_dhss_true/73 : FAILURE in 2h 22m 13s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/17/792217/2/check/DellEMC_Unity_Manila_dhss_false/7e21ae0/DellEMC_Unity_Manila_dhss_false/73 : SUCCESS in 53m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"f3f51d6a7337802780f2b1f17eec5e315f34f428","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-05-24 18:47:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3cd48a01b478d533c560ceda2e79f2fb68c0f5da","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-05-31 05:53:12.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM!","accounts_in_message":[],"_revision_number":2},{"id":"07c9fa426d647fc707de13b6d163365374d40e14","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-06-01 11:05:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLooks good, nice catch, thanks Goutham!","accounts_in_message":[],"_revision_number":2},{"id":"ce95984c26a8ac413b712024a0d5cb964e4f598f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 11:05:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b4346d8ce4961d95417d2279eed2d94fe3b5e204","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 13:37:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1cc43ecdf894fb8ab5973636ab04ed8 : SUCCESS in 3m 57s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8cbfacff65214f07a83f1f77193f04cd : SUCCESS in 27m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ac6cea754d044f2cbf3e604f85b21a50 : SUCCESS in 36m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b231cf32275441338b7501a9288fa7e0 : SUCCESS in 51m 13s","accounts_in_message":[],"_revision_number":2},{"id":"081e135ec41e20d458f46fab7850ffa142531038","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 13:37:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"71865020f86f8a42354bfe47cee64a5e5b88e505","revisions":{"5a92a5c68ad79c9fcb14d5a965a767ec56d613bb":{"kind":"REWORK","_number":1,"created":"2021-05-19 17:31:36.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/17/792217/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/17/792217/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/792217/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/792217/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/792217/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/792217/1"}}},"commit":{"parents":[{"commit":"0031ba11387091fcc830047cd265d7ed9f52c129","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0031ba11387091fcc830047cd265d7ed9f52c129"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-05-19 17:31:25.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-05-19 17:31:25.000000000","tz":-420},"subject":"[ci] Don\u0027t install manila-tempest-plugin system-wide","message":"[ci] Don\u0027t install manila-tempest-plugin system-wide\n\nSystem-wide installation via the tempest plugin\u0027s\ndevstack-plugin is unnecessary since we set the\n\"TEMPEST_PLUGINS\" devstack localrc variable with\nmanila-tempest-plugin, and that should handle installing\nthe plugin alongside other tempest requirements within\na test virtual environment.\n\nChange-Id: Id312005dc0cc9e039a9c6a8a58476965e8d6954d\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5a92a5c68ad79c9fcb14d5a965a767ec56d613bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5a92a5c68ad79c9fcb14d5a965a767ec56d613bb"}]},"branch":"refs/heads/master"},"71865020f86f8a42354bfe47cee64a5e5b88e505":{"kind":"REWORK","_number":2,"created":"2021-05-20 19:29:25.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/17/792217/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/17/792217/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/792217/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/792217/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/792217/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/792217/2"}}},"commit":{"parents":[{"commit":"0031ba11387091fcc830047cd265d7ed9f52c129","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0031ba11387091fcc830047cd265d7ed9f52c129"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-05-19 17:31:25.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-05-20 19:29:17.000000000","tz":0},"subject":"[ci] Don\u0027t install manila-tempest-plugin system-wide","message":"[ci] Don\u0027t install manila-tempest-plugin system-wide\n\nSystem-wide installation via the tempest plugin\u0027s\ndevstack-plugin is unnecessary since we set the\n\"TEMPEST_PLUGINS\" devstack localrc variable with\nmanila-tempest-plugin, and that should handle installing\nthe plugin alongside other tempest requirements within\na test virtual environment.\n\nChange-Id: Id312005dc0cc9e039a9c6a8a58476965e8d6954d\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/71865020f86f8a42354bfe47cee64a5e5b88e505"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/71865020f86f8a42354bfe47cee64a5e5b88e505"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
