)]}'
{"id":"openstack%2Fmanila-tempest-plugin~794977","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Idd2c2285e266f0b733301ccfae5c423340de2511","project":"openstack/manila-tempest-plugin","branch":"master","topic":"class_inheritances","hashtags":[],"change_id":"Idd2c2285e266f0b733301ccfae5c423340de2511","subject":"Fix class inheritances in base.py","status":"MERGED","created":"2021-06-06 07:58:27.000000000","updated":"2021-07-14 07:44:42.000000000","submitted":"2021-07-14 07:44:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"794977-class_inheritances","meta_rev_id":"0a470f01bad6e314e5c7bfe182230b6781e99322","_number":794977,"virtual_id_number":794977,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-14 07:44:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2021-07-14 06:38:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2021-06-28 17:56:12.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-29 11:34:38.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2021-07-14 06:38:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-06 09:43:12.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-06-06 09:50:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-06 13:05:16.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"CC"},{"updated":"2021-06-06 13:05:16.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"CC"},{"updated":"2021-06-06 13:05:16.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"CC"},{"updated":"2021-06-06 13:05:16.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2021-06-06 19:08:04.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-06 20:24:11.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-28 17:56:12.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-06-29 11:34:38.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2021-07-14 06:38:03.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"aad17c48723626ea0e2d4e08e9d5b0830c9c76e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-06-06 07:58:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e972c2921deab5f2ae2c77b29c19024461bb8a5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-06 09:43:12.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/77/794977/1/check/DellEMC_Unity_Manila_dhss_true/56aa700/DellEMC_Unity_Manila_dhss_true/85 : FAILURE in 1h 44m 25s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/77/794977/1/check/DellEMC_Unity_Manila_dhss_false/e5193dc/DellEMC_Unity_Manila_dhss_false/84 : FAILURE in 50m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"b1ebbb9cdca3b9326e6f38d329276834ef40ee5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-06 09:50:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2b77c355ef64509a0b77144b2787fac : SUCCESS in 3m 43s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/418a0d26fa6b4fec9dd84957f8b77a10 : FAILURE in 27m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/43a5a7817a7c489d9de2e2afe15c37df : FAILURE in 30m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/de3e0aa1c76c47478b0baf7f58353761 : FAILURE in 1h 00m 31s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5968f7e01f104cf4b3edd8e560de2347 : FAILURE in 1h 05m 53s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d7bddc20db914f84bced556331e9e2b0 : FAILURE in 1h 08m 47s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2d98296c3b1641e88726790d0afdd736 : FAILURE in 1h 43m 24s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a4613a46f52d4619bda95bef544a7773 : FAILURE in 55m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/affff986274e4cb8af54a92454e85198 : FAILURE in 1h 50m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/43efe1982bb346bbaa1382d0108e093a : FAILURE in 22m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae40c7ad4094bbcd2ce64e5f5382c00263ae73f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-06-06 10:06:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b9292f023b7bb5a5f1429dbe8ae5bf0dd373c622","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-06-06 11:24:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"efe54fa978fef193de3e87e9727b530d32b980fb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-06 13:17:39.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/77/794977/3/check/DellEMC_Unity_Manila_dhss_true/3a15607/DellEMC_Unity_Manila_dhss_true/87 : FAILURE in 1h 52m 54s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/77/794977/3/check/DellEMC_Unity_Manila_dhss_false/de8fcae/DellEMC_Unity_Manila_dhss_false/86 : SUCCESS in 59m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"90029cf75545f565d9d6b5a082d77baaf7cb946e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-06 19:07:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/576f8f2856b442c58da0f2798eb5011e : SUCCESS in 3m 58s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6907cd8890944025bd44ea1dbd3e306a : SUCCESS in 27m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fc36a330e95642c7b14fd5d35647dcda : SUCCESS in 44m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/068470ffda78418f839f0de4c363089c : SUCCESS in 55m 58s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/27a2d2999f3f4cf4b70cf36a952c593f : SUCCESS in 57m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e40e62bb8b83409797c09e33403cad94 : SUCCESS in 1h 00m 49s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5d9edc43dda046b593ae63ddf975667c : SUCCESS in 1h 31m 35s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9da2d41997e24eb4bae9650ff8bb5e3a : SUCCESS in 50m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fa30930616424b538ced407fa5505c4b : FAILURE in 1h 17m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/66cf434100c84316aec1034e094b01fa : SUCCESS in 20m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e9dc2f22ac5b699921b7c7d183e1a37439c882a5","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-06 19:08:04.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/77/794977/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/ab48989/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/77/794977/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/45def22/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"0b458c5a0d90c2723f959e1057e2528c8910b7a9","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-06-06 20:24:11.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/77/794977/3/check/DellEMC_VNX_Manila/fd5ac45/DellEMC_VNX_Manila/53 : FAILURE in 1h 16m 07s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"8b9f423bc34e3f57b20dcbab38ba1834dc495dea","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-06 21:13:02.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/77/794977/3/check/DellEMC_Unity_Manila_dhss_true/85f3e76/DellEMC_Unity_Manila_dhss_true/88 : FAILURE in 2h 05m 00s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/77/794977/3/check/DellEMC_Unity_Manila_dhss_false/c9a0d02/DellEMC_Unity_Manila_dhss_false/87 : SUCCESS in 1h 01m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"3195bb57abfdff2bf0f0c8cc0bae67e0e93c7b83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-06-21 18:14:25.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"bfdcb3b9365b233a2982fb352c3b65786686c64f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 20:43:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edfa9c93674049499391e4600aa0c34b : SUCCESS in 3m 30s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0daf88a2751949df9d8dd38496ca64a6 : SUCCESS in 39m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/323f76624aac4b158db0ebdaca0bff83 : SUCCESS in 39m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ee8febb92af84243ac17a9e8b523e26c : SUCCESS in 1h 25m 39s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/977956000800472ca714e9115cdebb55 : SUCCESS in 1h 28m 02s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a9777414fb0a4067b2895384d12703f6 : SUCCESS in 1h 12m 10s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/80a2eae101c74917b2a807a67d465183 : FAILURE in 30m 00s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/daa6eb4ec2e248788b2bf5f2539914b3 : SUCCESS in 43m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2bfe4905c830492386da9bc8cb0ad3e3 : FAILURE in 2h 23m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a7b08c1a467f4d02bc60504036dde71e : SUCCESS in 26m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"00abe607915a6c45cababf27b0c2073c2f195007","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-21 20:43:47.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/77/794977/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/5cb0e6b/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/77/794977/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/70cbebf/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"37dfa8c5fa177ecf800700efe9f9451275dfbdd9","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-06-21 22:03:42.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/77/794977/4/check/DellEMC_VNX_Manila/70237ab/DellEMC_VNX_Manila/72 : SUCCESS in 1h 19m 56s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"802ff60297200bf4fe90707e4e8d16d7a24d45e9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-06-24 18:19:16.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"2b745c562967c2e89490b09a5d37fc0e60a4e958","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 21:13:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63f4861051c84a3a8fd7e3b353439ba8 : SUCCESS in 6m 30s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c41a5b61943e47dcb87cd74dd6864ca4 : SUCCESS in 29m 39s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e2681f65fe7d44ed83de2f58f71452d8 : SUCCESS in 35m 22s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/535157b5d9dd42fea25220c9ae957120 : SUCCESS in 1h 06m 16s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/016bf2feec53439aa90755bddb49c345 : SUCCESS in 1h 09m 24s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/c00cab2787c04b7e86e35f9985a90264 : FAILURE in 1h 15m 42s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/74659afa3f2d43d8b69b3d951df88ea3 : SUCCESS in 1h 09m 48s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/18ea56f944a94dfc85fd9528b9170fb6 : SUCCESS in 1h 10m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/bced5659763c4433b27f22d332af1850 : SUCCESS in 1h 09m 34s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/78042b8c259d49b69c656c6be0fb5f5f : FAILURE in 23m 50s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1c35d5780da6404cbf65a5d0cce876f9 : SUCCESS in 56m 38s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2b4840b7f0cd43e49715048f7a85247e : FAILURE in 2h 46m 59s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e50eafb9d1a44b67b325c20a432cd2f2 : SUCCESS in 22m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8f26db6137ab49625c8ac384a33ef5a9ebe36fdf","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-06-27 04:04:06.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1f9d3a9f63c27b095c2e5ce3038de93e3ac10570","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-27 05:52:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32bc20b90efc419c8f895a098b774d68 : SUCCESS in 5m 24s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bd287c030e674c94879f4b4b1fbdf334 : SUCCESS in 41m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6ae20b626a1b410c926b196b089be037 : SUCCESS in 41m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cb68c7ccf94f44bb92ecaada65451dcf : SUCCESS in 1h 13m 24s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/52e92b25ab474893a6087601ac8abf88 : SUCCESS in 1h 13m 02s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/72d4dee7d51f43b99a1ed11a325280c9 : SUCCESS in 1h 19m 32s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/0d9a1e1c0aae4771b909fe8d8bcbcec9 : SUCCESS in 1h 44m 18s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ba89776586f34b77a4455de4c4cc85c8 : SUCCESS in 1h 07m 10s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/fcb70a3af09c49f78569331769bc3b16 : SUCCESS in 1h 24m 00s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/222afebb19dc43cd8319339923ecba7d : FAILURE in 30m 42s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d04157c6b4fe42c99e4deb71b2c349c1 : SUCCESS in 42m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d0993a45421844069d3c8d92a0c37c70 : FAILURE in 1h 42m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/98733416eacc40dd9521bc5b986c68a6 : SUCCESS in 28m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2c0b92e1f3a82013f9893dc0ec22fa3a4802f3e1","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-27 05:52:56.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/77/794977/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/ddde9be/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/77/794977/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/55b1d2c/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"7c30a41be70522e7a92a266f051908085fcd0a3b","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-06-27 06:46:42.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/77/794977/4/check/DellEMC_VNX_Manila/2b89e88/DellEMC_VNX_Manila/97 : FAILURE in 53m 46s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"33f5e931dddd1a41038cafa948ab663f6f7f9233","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-06-28 17:56:12.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"85524019e89369073fd7c68d652c41e1828352bc","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-06-29 11:34:38.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)\n\nGreat job Liron! Sorry it took so long to review. I think it\u0027s good to go, just a minor nit that can be addressed in a follow up patch.","accounts_in_message":[],"_revision_number":4},{"id":"8b79ba8b5fac526cfacc4db26e8d1460788165bb","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-06-29 11:58:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5594c822b79c0dcb41cde6eab1111e492a9112ef","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-07-14 06:38:03.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nThanks! Lets get this enhancement in and see if we\u0027ve any problems","accounts_in_message":[],"_revision_number":4},{"id":"3c9f572ce236e14d9b4f71d757dd0b41668c349e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 06:38:29.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e0ce152ddc463e3a9d05f2d4265c8f997fdef4c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 07:44:41.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af44e49f2aa942ac8098f0a207be8dba : SUCCESS in 4m 04s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/93aba4401d3e4520b2dc85ed65084c1c : SUCCESS in 35m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/20209bb2bce84d77a1bde36509154fdd : SUCCESS in 34m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b8477790290745bda39f631a37c12481 : SUCCESS in 1h 03m 41s","accounts_in_message":[],"_revision_number":4},{"id":"0a470f01bad6e314e5c7bfe182230b6781e99322","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 07:44:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3af822bf9653ca4468481efd289c94b2e1cb7c48","revisions":{"5970364287afdb3cbeea38bd65d3130328068307":{"kind":"REWORK","_number":1,"created":"2021-06-06 07:58:27.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/794977/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/794977/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/794977/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/794977/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/794977/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/794977/1"}}},"commit":{"parents":[{"commit":"669c6e72041a8c659a1789bb1e5684c7d1da01c2","subject":"Merge \"Setup snapshot extra specs only if required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/669c6e72041a8c659a1789bb1e5684c7d1da01c2"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-06-06 07:35:30.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-06-06 07:58:07.000000000","tz":180},"subject":"Fix class inheritances in base.py","message":"Fix class inheritances in base.py\n\nChange of inheritance to \"BaseSharesMixedTest\" class since it\nrequires admin privileges.\nThis change will allow removal of duplicate code.\n\nBeyond that, \"create_share_type\" method is under \"BaseSharesTest\",\nwhich is incorrect since it requires admin privileges.\n\nChange-Id: Idd2c2285e266f0b733301ccfae5c423340de2511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5970364287afdb3cbeea38bd65d3130328068307"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5970364287afdb3cbeea38bd65d3130328068307"}]},"branch":"refs/heads/master"},"6dc8559dcdce27b9ccf0305f27e73e1ca12230a1":{"kind":"REWORK","_number":2,"created":"2021-06-06 10:06:07.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/794977/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/794977/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/794977/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/794977/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/794977/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/794977/2"}}},"commit":{"parents":[{"commit":"669c6e72041a8c659a1789bb1e5684c7d1da01c2","subject":"Merge \"Setup snapshot extra specs only if required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/669c6e72041a8c659a1789bb1e5684c7d1da01c2"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-06-06 07:35:30.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-06-06 10:05:42.000000000","tz":180},"subject":"Fix class inheritances in base.py","message":"Fix class inheritances in base.py\n\nChange of inheritance to \"BaseSharesMixedTest\" class since it\nrequires admin privileges.\nThis change will allow removal of duplicate code.\n\nBeyond that, \"create_share_type\" method is under \"BaseSharesTest\",\nwhich is incorrect since it requires admin privileges.\n\nChange-Id: Idd2c2285e266f0b733301ccfae5c423340de2511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6dc8559dcdce27b9ccf0305f27e73e1ca12230a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6dc8559dcdce27b9ccf0305f27e73e1ca12230a1"}]},"branch":"refs/heads/master"},"2f2aad64e93b83715f2ee8315be7abf68890bd45":{"kind":"REWORK","_number":3,"created":"2021-06-06 11:24:22.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/77/794977/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/794977/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/794977/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/794977/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/794977/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/794977/3"}}},"commit":{"parents":[{"commit":"669c6e72041a8c659a1789bb1e5684c7d1da01c2","subject":"Merge \"Setup snapshot extra specs only if required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/669c6e72041a8c659a1789bb1e5684c7d1da01c2"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-06-06 07:35:30.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-06-06 11:24:00.000000000","tz":180},"subject":"Fix class inheritances in base.py","message":"Fix class inheritances in base.py\n\nChange of inheritance to \"BaseSharesMixedTest\" class since it\nrequires admin privileges.\nThis change will allow removal of duplicate code.\n\nBeyond that, \"create_share_type\" method is under \"BaseSharesTest\",\nwhich is incorrect since it requires admin privileges.\n\nChange-Id: Idd2c2285e266f0b733301ccfae5c423340de2511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2f2aad64e93b83715f2ee8315be7abf68890bd45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2f2aad64e93b83715f2ee8315be7abf68890bd45"}]},"branch":"refs/heads/master"},"3af822bf9653ca4468481efd289c94b2e1cb7c48":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-06-21 18:14:25.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/77/794977/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/794977/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/794977/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/794977/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/794977/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/794977/4"}}},"commit":{"parents":[{"commit":"ac53004eddfa5be7b95b2264cb181dbe2d97ec8c","subject":"Add a validation check that the share has been deleted","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ac53004eddfa5be7b95b2264cb181dbe2d97ec8c"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-06-06 07:35:30.000000000","tz":180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-06-21 18:14:25.000000000","tz":0},"subject":"Fix class inheritances in base.py","message":"Fix class inheritances in base.py\n\nChange of inheritance to \"BaseSharesMixedTest\" class since it\nrequires admin privileges.\nThis change will allow removal of duplicate code.\n\nBeyond that, \"create_share_type\" method is under \"BaseSharesTest\",\nwhich is incorrect since it requires admin privileges.\n\nChange-Id: Idd2c2285e266f0b733301ccfae5c423340de2511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3af822bf9653ca4468481efd289c94b2e1cb7c48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3af822bf9653ca4468481efd289c94b2e1cb7c48"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
