)]}'
{"id":"openstack%2Fmanila-tempest-plugin~796306","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I7f40922d6d38d3c7388ba8faf51a248e46e05a29","project":"openstack/manila-tempest-plugin","branch":"master","topic":"test-nautilus","hashtags":[],"change_id":"I7f40922d6d38d3c7388ba8faf51a248e46e05a29","subject":"[DNM] Test Ceph Nautilus","status":"ABANDONED","created":"2021-06-14 17:08:56.000000000","updated":"2023-09-11 14:13:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"946ada8129bdc7a50a741cccae759c7f4b607fca","_number":796306,"virtual_id_number":796306,"owner":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-06-14 18:28:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-14 17:09:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-14 18:44:10.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"d67a6509836d8553d1757ef82bdd3ac6eb4c2f44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-06-14 17:08:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32ddcb478608b0976cd3fb456f13cb4d10f97d27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 17:09:15.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-image-elements on branch master.  The error was:\n\n  Job manila-tempest-plugin-container not defined\n\nThe error appears in the following job stanza:\n\n  job:\n      name: manila-image-elements-container-scenario\n      parent: manila-tempest-plugin-container\n      pre-run: playbooks/manila-buildimage-docker/run.yaml\n      required-projects:\n        - openstack/manila-image-elements\n        - openstack/manila\n        - openstack/manila-tempest-plugin\n        - openstack/python-manilaclient\n        - openstack/tempest\n      vars:\n        # NOTE(gouthamr): Not running scenario tests yet, need to test them\n        # against the container driver first\n        tempest_test_regex: \u0027(^manila_tempest_tests.tests)(?\u003d.*\\[.*\\bbackend\\b.*\\])\u0027\n        devstack_localrc:\n          # \"docker build\" uploads the image we need to the local registry\n          # manila\u0027s devstack plugin just needs the name of the image to look up\n          MANILA_DOCKER_IMAGE_URL: \"file:///home/zuul/src/opendev.org/openstack/manila-image-elements/manila-docker-container.tar.gz\"\n\n  in \"openstack/manila-image-elements/zuul.d/manila-image-elements-jobs.yaml@master\", line 34, column 3","accounts_in_message":[],"_revision_number":1},{"id":"eb47f097eefa00f03d0b17548d6928f33b6f04f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-06-14 17:10:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"daa55ffd48cc7a4195e507bece0bc704854bed52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 17:10:33.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-image-elements on branch master.  The error was:\n\n  Job manila-tempest-plugin-container not defined\n\nThe error appears in the following job stanza:\n\n  job:\n      name: manila-image-elements-container-scenario\n      parent: manila-tempest-plugin-container\n      pre-run: playbooks/manila-buildimage-docker/run.yaml\n      required-projects:\n        - openstack/manila-image-elements\n        - openstack/manila\n        - openstack/manila-tempest-plugin\n        - openstack/python-manilaclient\n        - openstack/tempest\n      vars:\n        # NOTE(gouthamr): Not running scenario tests yet, need to test them\n        # against the container driver first\n        tempest_test_regex: \u0027(^manila_tempest_tests.tests)(?\u003d.*\\[.*\\bbackend\\b.*\\])\u0027\n        devstack_localrc:\n          # \"docker build\" uploads the image we need to the local registry\n          # manila\u0027s devstack plugin just needs the name of the image to look up\n          MANILA_DOCKER_IMAGE_URL: \"file:///home/zuul/src/opendev.org/openstack/manila-image-elements/manila-docker-container.tar.gz\"\n\n  in \"openstack/manila-image-elements/zuul.d/manila-image-elements-jobs.yaml@master\", line 34, column 3","accounts_in_message":[],"_revision_number":2},{"id":"833a1de9cdd5da2369a91765246784f9834b9f8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-06-14 17:40:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2c93cf47ab6b50172d7f2074b157b4831a0b56b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 18:28:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/040c5bf4fb06404d9afa8771932c7448 : FAILURE in 43m 53s\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/14def2afa2034f998360aa1d115f5e89 : FAILURE in 15m 29s","accounts_in_message":[],"_revision_number":3},{"id":"f27fd790962d24c40dc676a84fb181f08114bf98","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-14 18:44:10.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/06/796306/3/check/DellEMC_Unity_Manila_dhss_true/cb62b00/DellEMC_Unity_Manila_dhss_true/122 : FAILURE in 19m 09s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/06/796306/3/check/DellEMC_Unity_Manila_dhss_false/fba659b/DellEMC_Unity_Manila_dhss_false/120 : SUCCESS in 1h 03m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"a631700d438b1efa5574e47b5884d35394c02c43","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-06-15 12:14:12.000000000","message":"Patch Set 3:\n\nThis will always fail since there are no packages for ceph in the pulled repo. Check out https://lists.ceph.io/hyperkitty/list/ceph-users@ceph.io/thread/3NQQRCOB2YJPOMMSCHMJE2QD3NSHLQQW/ for details. Will need to test this with other Ubuntu version or with other distro.","accounts_in_message":[],"_revision_number":3},{"id":"946ada8129bdc7a50a741cccae759c7f4b607fca","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-11 14:13:09.000000000","message":"Abandoned\n\nNautilus is now EOL :)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6cd336478da1e752ef013215cfd8da7a22361baa","revisions":{"98594493f137a4c8ff9d00bf0bb163f56d23d66c":{"kind":"REWORK","_number":1,"created":"2021-06-14 17:08:56.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/06/796306/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/06/796306/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/796306/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/796306/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/796306/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/796306/1"}}},"commit":{"parents":[{"commit":"72869f90fc9566d1efe7ffb8b37474a9621c91a7","subject":"Merge \"[ci] Don\u0027t install manila-tempest-plugin system-wide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/72869f90fc9566d1efe7ffb8b37474a9621c91a7"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-06-14 17:09:12.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-06-14 17:09:58.000000000","tz":0},"subject":"[DNM] Test Ceph Nautilus","message":"[DNM] Test Ceph Nautilus\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/794275\nChange-Id: I7f40922d6d38d3c7388ba8faf51a248e46e05a29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/98594493f137a4c8ff9d00bf0bb163f56d23d66c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/98594493f137a4c8ff9d00bf0bb163f56d23d66c"}]},"branch":"refs/heads/master"},"d1b1925ba82deb5a277ae53335ec3025fc2ec265":{"kind":"REWORK","_number":2,"created":"2021-06-14 17:10:14.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/06/796306/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/06/796306/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/796306/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/796306/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/796306/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/796306/2"}}},"commit":{"parents":[{"commit":"72869f90fc9566d1efe7ffb8b37474a9621c91a7","subject":"Merge \"[ci] Don\u0027t install manila-tempest-plugin system-wide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/72869f90fc9566d1efe7ffb8b37474a9621c91a7"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-06-14 17:09:12.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-06-14 17:11:11.000000000","tz":0},"subject":"[DNM] Test Ceph Nautilus","message":"[DNM] Test Ceph Nautilus\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/794275\nChange-Id: I7f40922d6d38d3c7388ba8faf51a248e46e05a29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d1b1925ba82deb5a277ae53335ec3025fc2ec265"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d1b1925ba82deb5a277ae53335ec3025fc2ec265"}]},"branch":"refs/heads/master"},"6cd336478da1e752ef013215cfd8da7a22361baa":{"kind":"REWORK","_number":3,"created":"2021-06-14 17:40:39.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/06/796306/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/06/796306/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/796306/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/796306/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/796306/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/796306/3"}}},"commit":{"parents":[{"commit":"72869f90fc9566d1efe7ffb8b37474a9621c91a7","subject":"Merge \"[ci] Don\u0027t install manila-tempest-plugin system-wide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/72869f90fc9566d1efe7ffb8b37474a9621c91a7"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-06-14 17:09:12.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-06-14 17:41:36.000000000","tz":0},"subject":"[DNM] Test Ceph Nautilus","message":"[DNM] Test Ceph Nautilus\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/794275\nChange-Id: I7f40922d6d38d3c7388ba8faf51a248e46e05a29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6cd336478da1e752ef013215cfd8da7a22361baa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6cd336478da1e752ef013215cfd8da7a22361baa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
