)]}'
{"id":"openstack%2Fmanila-tempest-plugin~796838","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I70979bb5ca913c6e5bddd22795012b3fbcf3c8c3","project":"openstack/manila-tempest-plugin","branch":"master","topic":"DNM","hashtags":[],"change_id":"I70979bb5ca913c6e5bddd22795012b3fbcf3c8c3","subject":"DNM","status":"ABANDONED","created":"2021-06-17 13:21:51.000000000","updated":"2021-06-23 12:47:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e3c8566b8ae096af7e12c158f9911ecc6a9a6338","_number":796838,"virtual_id_number":796838,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-17 15:06:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-17 15:06:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-17 15:06:37.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"0fc2ed1c1ac374a995da155cf97d3edd05690745","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-06-17 13:21:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"227fcb5ced3accb3ab021b8c15f648570612ad49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 15:06:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1038f87fa28f4ed58a3f7c4478f6adf2 : SUCCESS in 4m 08s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/277f718b8cd84157b4edd11c4aa42bcb : SUCCESS in 30m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6970118702fb4a64b132c046b60616b5 : SUCCESS in 35m 23s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a2e7afcff2ba4e20853d208d07277317 : SUCCESS in 1h 28m 43s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/992254c04a934f8587be23155046b04c : FAILURE in 1h 14m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/55ea87faebf64df4990e14e33bd8483e : SUCCESS in 1h 04m 26s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/08b20143bba84914a5cb822da2d75e6a : FAILURE in 26m 22s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/49d60033f6654e9e94a15fb4c264d7eb : SUCCESS in 44m 02s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f7cdc965412b4c65a2bd0e6e1125fd7a : FAILURE in 1h 37m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/837afc24512c46d78d7cd73c8a3871cc : SUCCESS in 23m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"908b45a91879aea00e5c49ec5ec69e4dafad244f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-17 15:06:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/38/796838/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/beb21a2/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/38/796838/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/80039ea/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"e3c8566b8ae096af7e12c158f9911ecc6a9a6338","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-06-23 12:47:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b8f3ee8e87145ee782e2562d1eba2d86a5f9d791","revisions":{"b8f3ee8e87145ee782e2562d1eba2d86a5f9d791":{"kind":"REWORK","_number":1,"created":"2021-06-17 13:21:51.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/796838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/796838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/796838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/796838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/796838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/796838/1"}}},"commit":{"parents":[{"commit":"ac53004eddfa5be7b95b2264cb181dbe2d97ec8c","subject":"Add a validation check that the share has been deleted","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ac53004eddfa5be7b95b2264cb181dbe2d97ec8c"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-06-17 13:21:32.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-06-17 13:21:32.000000000","tz":180},"subject":"DNM","message":"DNM\n\nChange-Id: I70979bb5ca913c6e5bddd22795012b3fbcf3c8c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b8f3ee8e87145ee782e2562d1eba2d86a5f9d791"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b8f3ee8e87145ee782e2562d1eba2d86a5f9d791"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
