)]}'
{"id":"openstack%2Fmanila-tempest-plugin~797127","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ie90854f7f7a962263af0475aa505e6f6def1354c","project":"openstack/manila-tempest-plugin","branch":"master","topic":"cephfs-centos8-stream-job","hashtags":[],"change_id":"Ie90854f7f7a962263af0475aa505e6f6def1354c","subject":"[DNM] Test Manila CephFS on CentOS 8 Stream","status":"ABANDONED","created":"2021-06-18 15:28:13.000000000","updated":"2023-08-22 18:20:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0bca4b3c5b1a4e00684d6b427b3debb43997f20d","_number":797127,"virtual_id_number":797127,"owner":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-10-08 15:03:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-18 15:29:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-18 23:51:08.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-21 18:12:41.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2021-06-21 18:13:02.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-06-21 18:59:47.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-13 20:45:54.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"63e9daa1bcd90071a05f6766e7e1963ceed59b2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-06-18 15:28:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79317eddc059da73979e635bdb47d5605617fcc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 15:29:13.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"160f2cc105c7a66a11b4ab445a866214f1249e67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-06-18 23:07:09.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b5746fc1934fc54370b3adc5cba827ff30d3be08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 23:50:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- manila-tempest-plugin-cephfs-native-8-stream https://zuul.opendev.org/t/openstack/build/e191978b13284c2890652ac7850eecd2 : RETRY_LIMIT in 13m 28s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-8-stream https://zuul.opendev.org/t/openstack/build/c0dd4040d3474b1384ac665a5ff14671 : RETRY_LIMIT in 6m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cac063b7c42f6bff8dd283b04ba541a6496198c0","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-18 23:51:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/27/797127/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/dc2da06/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/27/797127/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/90778ed/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"2dad4fe2a6b6b424104e688a4348ea9f26ecf1c5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-06-21 18:12:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"195d0106ee730f45c3edfa142f3c7e9ff41fed72","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-06-21 18:13:02.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"dc5a80db0f20ed1cfac13f6448cf42bb08b46f6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 18:48:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- manila-tempest-plugin-cephfs-native-8-stream https://zuul.opendev.org/t/openstack/build/48ada3ead11943db8fdceedae706ec5c : RETRY_LIMIT in 8m 31s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-8-stream https://zuul.opendev.org/t/openstack/build/ddd59da7a782420bb5f778890d2c06be : RETRY_LIMIT in 14m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dcd3b1281a69bc15d839f0df4838239002129601","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-21 18:49:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/27/797127/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/2fe8ef3/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/27/797127/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/a59acff/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"8e219ce59ea354231baf2666ae015a29866f1c02","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-06-21 18:59:47.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/27/797127/2/check/DellEMC_VNX_Manila/5cb0bb7/DellEMC_VNX_Manila/70 : FAILURE in 10m 38s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"cfafecc8e40cbd5acf725c0b84b3688f10ce5ddc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-07-13 19:12:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fe3d98b8b6dbebfbd037cb00a4a9942bcb083794","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-07-13 19:13:04.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"03aff1dcc190c2004ecf9776f98657a7c1087dd6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-13 20:45:54.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/27/797127/3/check/DellEMC_Unity_Manila_dhss_true/45c37a9/DellEMC_Unity_Manila_dhss_true/187 : SUCCESS in 1h 32m 41s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/27/797127/3/check/DellEMC_Unity_Manila_dhss_false/d383853/DellEMC_Unity_Manila_dhss_false/184 : SUCCESS in 59m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"01a62c3d64fd04cc56852943ab585ba67863cb20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 21:55:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b00311d298a94a10930b879e0eec05c8 : SUCCESS in 5m 28s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/60b0da1dba5b43d1b7612268d01d3031 : SUCCESS in 46m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ad5c33b18b9a4827a3bbccd913eaaa77 : SUCCESS in 43m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/06f6745a0d6747d49cac7316fa37a7df : SUCCESS in 1h 16m 16s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/e14c727236c5464f8f65857d0a2442de : SUCCESS in 1h 26m 01s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/82c3fbced2f54e72b1cbc7c390b2bb6c : SUCCESS in 1h 20m 04s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/c338c93b0b0c4a6c86729bc8c9219a96 : FAILURE in 1h 26m 38s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3acd5f05932b4e60aa974d782b1226ac : FAILURE in 1h 23m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/62151692b85542c8a71d58488dec8101 : SUCCESS in 1h 31m 54s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/71b6b9dbd2704e5b89cc04603ffb5fa9 : SUCCESS in 1h 47m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native-8-stream https://zuul.opendev.org/t/openstack/build/71a343680b0d496b8b019f0bcded685b : FAILURE in 1h 47m 26s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-8-stream https://zuul.opendev.org/t/openstack/build/5005797ad6e74af8b1c5f7a5d04b1ad2 : FAILURE in 26m 00s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a77ce3edd0f54f9796d8d9d2de6b6fbf : SUCCESS in 54m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/13a493ef0178461ca50a90383718b852 : FAILURE in 1h 18m 09s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6c61ffe97d0746dd93677b4da6a922e7 : FAILURE in 31m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c781c8471ba536e581f6438e5e72faa39217114f","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-07-14 09:34:58.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d67938c9e01493a9b1f386a04b2945da464fd010","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 11:16:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34a074ed56874eee99a3a7f32ff519c0 : SUCCESS in 3m 48s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/dffa2729e17d4007b2f991af85832510 : SUCCESS in 26m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/42828bdd6bcb4451b9cd59b2093001b8 : SUCCESS in 30m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/09c2da1d7d6e4405a6bce52f45d3a422 : SUCCESS in 1h 05m 24s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/5c5a63044f3b4fb98bb5c8f43123a9bf : SUCCESS in 1h 05m 34s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/e77dd915fc2245eb9f59d298d95ea90f : SUCCESS in 1h 06m 15s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/1eba52f3f56d4d20837c7b9def95796f : SUCCESS in 59m 45s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/541669514b6a4ce2a61cfb27e2ef67ed : SUCCESS in 1h 05m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/085cbc160b2249128c9d010c8bad3cbb : SUCCESS in 1h 12m 36s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8ff1ba7af866441faf44eaf91c82ea5e : FAILURE in 1h 40m 14s (non-voting)\n- manila-tempest-plugin-cephfs-native-8-stream https://zuul.opendev.org/t/openstack/build/1766aa5ce93841d48e9718ef791be804 : FAILURE in 1h 20m 53s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-8-stream https://zuul.opendev.org/t/openstack/build/cddf35932a674ca4af3713f00228cfdd : FAILURE in 20m 51s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/314654a001424701bc313be37bf1079e : SUCCESS in 44m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ecff659068d74e4bbe2e63c49231e8fa : FAILURE in 1h 20m 42s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e57e38980a9241e4a0e3b3bfe200bffd : SUCCESS in 22m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1afcb1f31cc037f99eb8977871672108ef206020","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-14 11:17:23.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/27/797127/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/27bd73f/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/27/797127/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/bb22293/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"b5f12f9b369e07358e9f77026fde22ad0be47456","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-14 15:58:12.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/27/797127/3/check/DellEMC_Unity_Manila_dhss_true/f58aa4f/DellEMC_Unity_Manila_dhss_true/199 : SUCCESS in 1h 38m 15s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/27/797127/3/check/DellEMC_Unity_Manila_dhss_false/3556f6d/DellEMC_Unity_Manila_dhss_false/196 : SUCCESS in 1h 11m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"d185046cbe700f429b60230ded0068ba4ec08ded","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-07-19 17:52:47.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"dcd64a5c8d534a88833647c647b8fe044c60891b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 19:37:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b905e4990ac34a6aac357f0f3d03fc7c : SUCCESS in 4m 35s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cff0a1b5b1544c7ca88ed6ca90a9b444 : SUCCESS in 28m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4387d795e115401890efc36e48fed86a : SUCCESS in 42m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ac0d62c63cf641379b674bc91522c3b5 : SUCCESS in 1h 03m 20s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/ca86a54b14e44f36a5f3614082754277 : SUCCESS in 1h 04m 17s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/8ee7503a383a4a2990beee08bafcd1b9 : SUCCESS in 1h 09m 19s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/b15fd2243e0d486ba9bfc446c5502db5 : SUCCESS in 1h 14m 04s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/18886eab27da4434b1bf04f9c3525217 : SUCCESS in 1h 04m 02s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a8e2463d9b934530bb0cbca6dc8e5e15 : SUCCESS in 1h 10m 06s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/157f227176554200a26be682ffda48e8 : SUCCESS in 1h 23m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native-8-stream https://zuul.opendev.org/t/openstack/build/4bd0d79c1bbe443b971d181c2b777402 : FAILURE in 1h 33m 41s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-8-stream https://zuul.opendev.org/t/openstack/build/5dd8242f9cc64110ae0d1d156dac7452 : FAILURE in 21m 04s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c320a7d18922498bb98ff6b7f7b643ba : SUCCESS in 48m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/30b2b71d698f49729466148bdfda74c7 : FAILURE in 1h 40m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/23195ac42bf04f0e8043bdd537d9cbe1 : FAILURE in 32m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7aacd5582687a846e1c48e61313db345993bc9cd","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-19 19:37:31.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/27/797127/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/cde7d7c/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/27/797127/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/b8ea0cf/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"f677e8994c687c4474d396dbe6809dc53599fd8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-10-08 13:15:48.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"f156db7ca996de23c77565e4d779fc6c45c300a5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-08 14:21:18.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/27/797127/4/check/DellEMC_Unity_Manila_dhss_true/ca80ac6/DellEMC_Unity_Manila_dhss_true/1 : FAILURE in 1h 05m 16s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/27/797127/4/check/DellEMC_Unity_Manila_dhss_false/ea92543/DellEMC_Unity_Manila_dhss_false/1 : FAILURE in 54m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"7531f42d35b644e82f0c1da9aed733884b419fb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 15:03:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a0ad898ea334f88ae2cafe170b745ff : SUCCESS in 4m 52s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bc37ac903f344c9aa7c4ae07ad8ed18c : SUCCESS in 30m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ad134a103c7344ef8cad3f579c03cc77 : SUCCESS in 37m 08s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6700a46f216d492aa539a2b9561009eb : SUCCESS in 1h 10m 45s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/19d7a2fd6a164baeaa6eb82b08817029 : SUCCESS in 53m 41s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/fe47404c2b194b9eb271d03f85e237f6 : SUCCESS in 1h 40m 41s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/f1805a49eead48ecb110c1b84bcff2a4 : SUCCESS in 56m 52s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fe2a588887394b2b983b5d8e4f0cfd10 : SUCCESS in 1h 06m 35s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3414fe55413941d5965ddf2e4e706437 : SUCCESS in 1h 16m 19s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9770bde135ae41e282136ff3d8f39601 : FAILURE in 1h 32m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native-8-stream https://zuul.opendev.org/t/openstack/build/f283ee1666e845db90b470a4b23661d1 : FAILURE in 1h 35m 29s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-8-stream https://zuul.opendev.org/t/openstack/build/8ab71cfa8ef44ffdbb31deb7315214c0 : FAILURE in 10m 16s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b237214edc0046b6bc5d8398e46833ee : SUCCESS in 44m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ff259ef3b8304ad3adaeef4ad8db7d26 : FAILURE in 1h 33m 32s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/317cc2bd19e644488a53e624cdac9380 : FAILURE in 21m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ff9f6e84b2720bc6293cec1b1e7e627dcea02e7c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-10-08 15:03:34.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/27/797127/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/5496bba/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/27/797127/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/d106895/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"6d853337506bf7bfc6cba01f75305f8eb7538f66","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-08 16:03:19.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/27/797127/4/check/DellEMC_Unity_Manila_dhss_true/da13c6b/DellEMC_Unity_Manila_dhss_true/2 : FAILURE in 1h 00m 06s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/27/797127/4/check/DellEMC_Unity_Manila_dhss_false/5d9a339/DellEMC_Unity_Manila_dhss_false/2 : FAILURE in 52m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"12f1d85a14b14481bb9498d4052480e6ce0d42cf","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-10-08 16:20:26.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/27/797127/4/check/DellEMC_VNX_Manila/fcafed7/DellEMC_VNX_Manila/1 : SUCCESS in 1h 17m 13s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"0bca4b3c5b1a4e00684d6b427b3debb43997f20d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-22 18:20:19.000000000","message":"Abandoned\n\njobs have been enabled with 9-stream now","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"89e908c392e106293716a741765337389f2cdb0f","revisions":{"e663377d59a6827e0f9c50b3eb15c291dba1c830":{"kind":"REWORK","_number":1,"created":"2021-06-18 15:28:13.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/27/797127/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/27/797127/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/797127/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/797127/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/797127/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/797127/1"}}},"commit":{"parents":[{"commit":"ac53004eddfa5be7b95b2264cb181dbe2d97ec8c","subject":"Add a validation check that the share has been deleted","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ac53004eddfa5be7b95b2264cb181dbe2d97ec8c"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-06-18 15:27:33.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-06-18 15:27:33.000000000","tz":0},"subject":"Test Manila CephFS on CentOS 8 Stream","message":"Test Manila CephFS on CentOS 8 Stream\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/797124\nChange-Id: Ie90854f7f7a962263af0475aa505e6f6def1354c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e663377d59a6827e0f9c50b3eb15c291dba1c830"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e663377d59a6827e0f9c50b3eb15c291dba1c830"}]},"branch":"refs/heads/master"},"f9305fb1857e5da87dd57c2ff8c19c79d5469277":{"kind":"REWORK","_number":2,"created":"2021-06-18 23:07:09.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/27/797127/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/27/797127/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/797127/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/797127/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/797127/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/797127/2"}}},"commit":{"parents":[{"commit":"ac53004eddfa5be7b95b2264cb181dbe2d97ec8c","subject":"Add a validation check that the share has been deleted","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ac53004eddfa5be7b95b2264cb181dbe2d97ec8c"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-06-18 15:27:33.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-06-18 23:07:04.000000000","tz":0},"subject":"Test Manila CephFS on CentOS 8 Stream","message":"Test Manila CephFS on CentOS 8 Stream\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/797124\nChange-Id: Ie90854f7f7a962263af0475aa505e6f6def1354c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f9305fb1857e5da87dd57c2ff8c19c79d5469277"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f9305fb1857e5da87dd57c2ff8c19c79d5469277"}]},"branch":"refs/heads/master"},"817fa33bbff8b838787b33220c4b64f3f4baa3af":{"kind":"REWORK","_number":3,"created":"2021-07-13 19:12:52.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/27/797127/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/27/797127/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/797127/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/797127/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/797127/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/797127/3"}}},"commit":{"parents":[{"commit":"8a53946b54477a56e32d80d5897f0d017b187f16","subject":"Merge \"Add testcase for share metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8a53946b54477a56e32d80d5897f0d017b187f16"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-06-18 15:27:33.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-07-13 19:13:44.000000000","tz":0},"subject":"[DNM] Test Manila CephFS on CentOS 8 Stream","message":"[DNM] Test Manila CephFS on CentOS 8 Stream\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/797124\nChange-Id: Ie90854f7f7a962263af0475aa505e6f6def1354c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/817fa33bbff8b838787b33220c4b64f3f4baa3af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/817fa33bbff8b838787b33220c4b64f3f4baa3af"}]},"branch":"refs/heads/master"},"89e908c392e106293716a741765337389f2cdb0f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-10-08 13:15:48.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/27/797127/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/27/797127/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/797127/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/797127/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/797127/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/797127/4"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-06-18 15:27:33.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-10-08 13:15:48.000000000","tz":0},"subject":"[DNM] Test Manila CephFS on CentOS 8 Stream","message":"[DNM] Test Manila CephFS on CentOS 8 Stream\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/797124\nChange-Id: Ie90854f7f7a962263af0475aa505e6f6def1354c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/89e908c392e106293716a741765337389f2cdb0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/89e908c392e106293716a741765337389f2cdb0f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
