)]}'
{"id":"openstack%2Fmanila-tempest-plugin~802644","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I1eaaa0eda12690b097a7e65170064c0935259f4c","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bp/manila-share-support-recycle-bin","attention_set":{},"removed_from_attention_set":{"30998":{"account":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"last_update":"2022-02-22 01:36:42.000000000","reason":"Change was submitted"},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2022-02-22 01:36:42.000000000","reason":"Change was submitted"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2022-02-22 01:36:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1eaaa0eda12690b097a7e65170064c0935259f4c","subject":"API2.69,Add tests for manila recycle bin.","status":"MERGED","created":"2021-07-28 01:55:44.000000000","updated":"2022-02-22 01:36:42.000000000","submitted":"2022-02-22 01:36:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":40,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"802644-bp/manila-share-support-recycle-bin","meta_rev_id":"3cab287a57727684f5fff9386b81bbdcbbb2a9f6","_number":802644,"virtual_id_number":802644,"owner":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-22 01:36:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2022-02-22 00:11:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2022-02-16 21:06:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},{"value":2,"date":"2022-02-09 21:36:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-08 08:59:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2022-02-22 00:11:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-28 04:42:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-28 06:14:32.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-07-29 14:49:15.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-29 15:55:19.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-03 07:39:58.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2021-09-14 09:03:36.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"CC"},{"updated":"2022-02-08 08:59:36.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2022-02-09 21:36:55.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2022-02-16 20:42:45.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2022-02-16 21:06:37.000000000","updated_by":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"reviewer":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"state":"REVIEWER"},{"updated":"2022-02-17 14:20:10.000000000","updated_by":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"reviewer":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"state":"CC"},{"updated":"2022-02-22 00:11:30.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"5ba0641d40e9a82ed14a1a3eaad1ab2d1787aff1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-07-28 01:55:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d535392bd6f36f892ac382142ce682f8e96a25a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-07-28 01:57:16.000000000","message":"Topic set to bp/manila-share-support-recycle-bin","accounts_in_message":[],"_revision_number":1},{"id":"91f58c2f8570ed97148ad831b425ec703f5bf781","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-07-28 02:57:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5fc52ee4738bdfb176739eec6d351ae65f31b165","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 04:42:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e157c732e30b42688734790ff9271656 : SUCCESS in 4m 06s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e58901cd093043d7824b918d5d6fb1b1 : FAILURE in 26m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/24bd981fcf76450cba9d872db3c546f8 : FAILURE in 32m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dceda5d91bba4795a305900cd348788b : SUCCESS in 1h 15m 19s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/0d5ba247618549e8b863feb75c58a89a : SUCCESS in 58m 16s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/612cc19733aa43888e6399231cc73d16 : SUCCESS in 1h 14m 23s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/446d0ab9b2304d3c9866672949413a20 : SUCCESS in 1h 00m 04s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3e0304c74a7a4a109cd51923a66553ad : FAILURE in 1h 05m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/45bc745972194dbda332b52bb74ea4a6 : SUCCESS in 1h 09m 19s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/766315937123415dbc97d9a3953b50c9 : SUCCESS in 1h 43m 06s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4e637934a57045019d1d8650c8ff5283 : FAILURE in 45m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d803e5994842407bbab6cb780dbdf3fb : FAILURE in 1h 38m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5e7bbe7be9674a20be24b5fe9f14bdbd : SUCCESS in 21m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"012d63266a58e968845a057fa30869f74791cb4d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-28 06:14:32.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/44/802644/2/check/DellEMC_Unity_Manila_dhss_true/5290605/DellEMC_Unity_Manila_dhss_true/221 : FAILURE in 1h 13m 43s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/44/802644/2/check/DellEMC_Unity_Manila_dhss_false/367e6ec/DellEMC_Unity_Manila_dhss_false/217 : SUCCESS in 1h 20m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"36bc92c1a2db1a731c4c6674fb0283ec3ff52260","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-07-28 07:16:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5d37054fb69b0060eec3a7d1146db42183517a0e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-28 08:49:35.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/44/802644/3/check/DellEMC_Unity_Manila_dhss_true/65404e4/DellEMC_Unity_Manila_dhss_true/223 : FAILURE in 1h 23m 10s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/44/802644/3/check/DellEMC_Unity_Manila_dhss_false/fd4b070/DellEMC_Unity_Manila_dhss_false/219 : SUCCESS in 1h 05m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"d79228f05354d466e285a001d193791be39a76d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 11:56:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66f6830a40f54aa98a3cb0022abb97cf : SUCCESS in 4m 27s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fece31ab7c75407e9f533aceda40a16c : FAILURE in 24m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a0fdfeeff7ff45ce8de50069f03383c4 : FAILURE in 30m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5112df2ea0c04bc4adadac31f3b63d5a : SUCCESS in 1h 04m 10s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/85f1a793e72b4b1db6c380f3bae2e105 : SUCCESS in 1h 03m 58s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/9533ab126178428d8ac159d120313efd : SUCCESS in 56m 25s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/3252c692c6bc405ca3d4ec177c0bdfe1 : SUCCESS in 1h 13m 04s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6cfc48c9fe76464da3d1d09c08dd4c74 : FAILURE in 1h 00m 36s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9a2f5ae864ed4192b3846a0f27991356 : SUCCESS in 1h 10m 31s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5c593620e1f14420aada7c7095091b98 : SUCCESS in 1h 31m 27s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f7797f5f63bb4a139850543bd79f271a : FAILURE in 48m 38s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a4bf5a1f0623469ca53f4437157e3b76 : FAILURE in 19m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4eca353b15274c32bebbd723766c7304 : SUCCESS in 25m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9aa5a6cdd0f766efbb524e24489495d2e02ae749","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-07-29 08:35:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a0cb04ae11c1daead626cccf808f5b4e43110575","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-29 09:09:38.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"f4e3ac158ca8602c5a9e155b88f97a4e60a7362f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 09:09:39.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"030c847d4fd1fbcfd3aa6c7dc3271c12593b1bdd","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-07-29 12:04:50.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c202df69f526cba56c9dad0418273c21c6bab8b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 14:48:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05ff841bedeb4f91bedadba1eb6d8ab2 : SUCCESS in 3m 42s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/47d4a0ea9bae443f991dc85d661bce31 : SUCCESS in 32m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/71d7a2e97b3a46399d5484e35df50d59 : SUCCESS in 30m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4327510b50ff4c4aa2d713b48bc2bfc5 : SUCCESS in 1h 04m 39s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/a6acce12f6154f22ac9e10c7c51c677c : SUCCESS in 1h 09m 30s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/86adf3d0396a4aecb1fc621fa66f23ea : SUCCESS in 1h 08m 13s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/1b1e8c08410b4a0ba241a17e7cb910e2 : SUCCESS in 1h 13m 39s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f10a6d941bc241839f1d1e2609dbd750 : FAILURE in 1h 17m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6754319394da43bd9158c9ad40f5c9b9 : SUCCESS in 1h 08m 23s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f20af735c64449b39c31d51b366693d0 : SUCCESS in 1h 45m 53s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1211a23cb8a144b4849d0e2238439a5e : FAILURE in 40m 36s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2332819a87f94b14a831640acaa38477 : FAILURE in 2h 37m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/04787c5e3d1245ba97d13a54934a3dc0 : SUCCESS in 23m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e6f62ca1e52e10a4043f1a8456dca260210623ea","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-29 14:49:15.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/44/802644/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/6cb72bf/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/44/802644/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/9fc56ee/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"9962cb9bec1a7621634bfdb2e6741f2276ec8bc1","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-07-29 15:55:19.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/44/802644/4/check/DellEMC_VNX_Manila/83e79af/DellEMC_VNX_Manila/122 : SUCCESS in 1h 06m 15s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"802abb369c0ce44c4946fd4a0ca21a9c2e611b73","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-29 18:20:28.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/44/802644/4/check/DellEMC_Unity_Manila_dhss_true/61ab4fa/DellEMC_Unity_Manila_dhss_true/230 : FAILURE in 1h 40m 28s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/44/802644/4/check/DellEMC_Unity_Manila_dhss_false/4295d37/DellEMC_Unity_Manila_dhss_false/227 : SUCCESS in 1h 06m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"1f1475a5f3de62e08301ef110a5034b8a9f510bc","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-09-03 07:39:58.000000000","message":"Patch Set 4:\n\n(3 comments)\n\nThanks for adding these; could you please update the API version?","accounts_in_message":[],"_revision_number":4},{"id":"841772d7d7cdba40caa81133ce87197df7a56dd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-09-03 09:35:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7c71cb6dce66de3ada76c18faeedcb71dba287da","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-09-03 09:37:40.000000000","message":"Patch Set 5:\n\n(3 comments)\n\n\u003e Patch Set 4:\n\u003e \n\u003e (3 comments)\n\u003e \n\u003e Thanks for adding these; could you please update the API version?\n\ni have update micversion to 2.66, to avoid conflict.","accounts_in_message":[],"_revision_number":5},{"id":"c351e92f4b600a9ce75c91ca90709d221b0f3731","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 11:10:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/776cf5c843e748cc81798fec96d69f88 : SUCCESS in 3m 51s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7236115d4a804af4b3cddcdc66f1f866 : SUCCESS in 35m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/04785294d1e9419cb09d4e5876ff8d4d : SUCCESS in 41m 24s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1a04b0af9f2143b78d54e36003ab1645 : SUCCESS in 1h 14m 51s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/9bb787d9e00d489f9127caf7a596b130 : SUCCESS in 1h 06m 48s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/84e5e6cba9174ec19d968ee899c1f4b6 : SUCCESS in 1h 02m 15s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/1186f6bffedd44b9a6e113442d41a620 : SUCCESS in 1h 29m 13s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8e694b6a4f7e483a84f151edee73a0f3 : FAILURE in 1h 02m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/2fabd1d0c15a44838b54aa64f6d6d14d : SUCCESS in 1h 23m 52s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a2b8b1001cfc4288acfa6e9d44f31841 : FAILURE in 1h 21m 11s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/58d7cadc7772479989b4efecdb7e5721 : FAILURE in 51m 03s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c9ceb59370cf4414b2c1fdffc94a8cd0 : FAILURE in 1h 30m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/213dc4f477d84911a37bbe31274fe8d3 : SUCCESS in 26m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b80e92c11e1eca9b62d599c7f8b055af4138aad2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-09-03 11:11:16.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/44/802644/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/ec57610/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/44/802644/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/35f4417/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"9afbbef82f7b79da2c38a7d6b9b5517cff0f615f","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-09-14 09:03:36.000000000","message":"Patch Set 5:\n\n(7 comments)\n\nThanks Haixin,\nPlease see the comments.","accounts_in_message":[],"_revision_number":5},{"id":"58f154a3e620549f96b4d49c9674ba7b56de10c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-09-15 03:21:13.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2eb3a3bed2eb3e1d1f82068a3b12363c5b14e7e1","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-09-15 03:21:49.000000000","message":"Patch Set 6:\n\n(7 comments)\n\nhi, Liron Kuchlani.\nthanks for your review.","accounts_in_message":[],"_revision_number":6},{"id":"c0d74e0eacf480e78502eccdcec572f1cbe01168","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-09-15 04:50:23.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/44/802644/6/check/DellEMC_Unity_Manila_dhss_true/c2e109b/DellEMC_Unity_Manila_dhss_true/346 : FAILURE in 1h 28m 49s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/44/802644/6/check/DellEMC_Unity_Manila_dhss_false/0fb5acb/DellEMC_Unity_Manila_dhss_false/342 : FAILURE in 1h 26m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"79d053006f5a62a0dc57c6865d0eb12b3f1340a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 05:05:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf5d67bd1db544028adcee21529735e6 : SUCCESS in 4m 28s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f19f8f451e914eac89091cdf619ab5a9 : FAILURE in 27m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/379a873a58924803a20ed9e28fcafd74 : FAILURE in 30m 08s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/743f84b738b24c639cf1568ae0528ecf : SUCCESS in 59m 00s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/8a33106706b14761af02e262f77cd502 : SUCCESS in 1h 05m 01s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/6c96fccc9b1f400bb893a03e03f34ef7 : SUCCESS in 1h 00m 41s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/b73775fba42244b19d86fecff168fba3 : SUCCESS in 1h 05m 39s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ff36518e4746482da4648648bf1e8775 : FAILURE in 1h 03m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3d3103d6dd68436cade90720ef6195b2 : FAILURE in 1h 08m 00s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/58888b1de1564c37aab65cd4740a56bd : FAILURE in 1h 40m 16s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6150e509c3e74544af887be2d5ee6b40 : FAILURE in 49m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/73b70ffb90704441b44b0df35550c198 : FAILURE in 1h 40m 52s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/bcd23fbddfc04f38a0bd5847bc3f5178 : FAILURE in 21m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1c79d0ece1cf652ca1aac53d782eee40719f89ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-09-15 05:55:58.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c09036357a4a8354b8b05d8c6328231d6506aaca","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-09-15 07:32:18.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/44/802644/7/check/DellEMC_Unity_Manila_dhss_true/f7aed6e/DellEMC_Unity_Manila_dhss_true/347 : FAILURE in 1h 33m 57s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/44/802644/7/check/DellEMC_Unity_Manila_dhss_false/61d563d/DellEMC_Unity_Manila_dhss_false/343 : SUCCESS in 1h 36m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"48184abdc1b423755e23af55169a01746f06980e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 07:53:54.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10715d56384346cf88a7dc25faacd1ca : SUCCESS in 4m 16s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1498a741883e4439899c3d0e841474fc : SUCCESS in 35m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c3742731b47047f4b27f8dc254ced213 : SUCCESS in 38m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5c3e11ace1514d72a13268f89e385437 : SUCCESS in 1h 05m 03s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/f0c16d80845747e987c8cbaf20e3c525 : SUCCESS in 1h 14m 18s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/9c4308ec887c481d985fe7d662c3a429 : SUCCESS in 1h 15m 58s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/013aa7c183cc45f69832f83c147f19a1 : SUCCESS in 1h 54m 00s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c4b68ccc45194cdab4c19e10e66acf49 : FAILURE in 1h 07m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/8f32b4eeb54e4fcf9db2253d1169f7ba : SUCCESS in 1h 14m 50s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b7e0ede7cc7e453da9b83f847170954d : FAILURE in 1h 27m 24s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/515006af936846cfbfd3107454a3f9d8 : FAILURE in 1h 01m 24s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ed7ca463a3bb4a798bbadd0406c14ade : FAILURE in 1h 27m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a5641170a1be4b7685cf587f3eecac04 : SUCCESS in 20m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"cbe471739aa51976d4b27ecfef562d8af29b9d23","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-09-15 07:54:25.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/44/802644/7/upstream-check/manila-cDOT-manila-tempest-plugin-ss/aacabf1/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/44/802644/7/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/d77dcce/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"5d63a09e25919e6447a0b3fbddb2f8aa84ca0f05","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-09-15 08:51:57.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/44/802644/7/check/DellEMC_VNX_Manila/2da750e/DellEMC_VNX_Manila/180 : FAILURE in 57m 47s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"b8e5371264aaeb629c5add681ea39286be46ad3e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-09-15 11:02:28.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/44/802644/7/check/DellEMC_Unity_Manila_dhss_true/29efcb9/DellEMC_Unity_Manila_dhss_true/349 : FAILURE in 1h 17m 13s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/44/802644/7/check/DellEMC_Unity_Manila_dhss_false/892e5a8/DellEMC_Unity_Manila_dhss_false/345 : FAILURE in 1h 34m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"387206d515a82b9856aba62255da57359fb79d98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-10-26 02:16:59.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e723790948a3114f63226eba9cd99c019af37bbe","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-26 02:36:19.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/44/802644/8/check/DellEMC_Unity_Manila_dhss_true/5e26aec/DellEMC_Unity_Manila_dhss_true/37 : FAILURE in 14m 32s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/44/802644/8/check/DellEMC_Unity_Manila_dhss_false/56c8a4e/DellEMC_Unity_Manila_dhss_false/40 : FAILURE in 13m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"721393385390efc1e901b61a070252b0a8eff723","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 04:47:21.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8030e11816774f10ae2929dea623141a : SUCCESS in 5m 10s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/09d110f74f3e4b80b168c89e4064d21d : SUCCESS in 35m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/af7d307be45b44e9a8c3cba70a008ca5 : SUCCESS in 35m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/554f3c1667784c6c8f19886fa0bae816 : SUCCESS in 1h 17m 17s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/77dabf1dbce645698253a27671f2952b : SUCCESS in 1h 16m 32s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/6f8f4798e59f42f4a08acb45fb71f33c : SUCCESS in 1h 14m 06s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/1e3b786293ca417180da3ec4238d28b4 : SUCCESS in 1h 05m 49s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8c29e510e3be4e2286809804a09f2506 : FAILURE in 1h 12m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/632fe3a88a214302a3c4d500b7c4795f : SUCCESS in 1h 23m 19s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f141905e866e41498f70375d9305e7a6 : FAILURE in 1h 36m 55s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f0ddc3397b734e3d9032cf2197da4263 : FAILURE in 53m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4e774901feb14cf1bb0c6cba7f34ae89 : FAILURE in 2h 07m 40s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/437846a18c1041e7b35fc89f4105b761 : FAILURE in 29m 07s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0825ea02aee004875af0b62171364e07a9aacecb","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-10-26 04:47:53.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/44/802644/8/upstream-check/manila-cDOT-manila-tempest-plugin-ss/e70ebfd/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/44/802644/8/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/be3e55e/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"38dee63a7170a94272ff9bc36b1222d21df3d695","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-10-26 05:01:58.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/44/802644/8/check/DellEMC_VNX_Manila/1218bdb/DellEMC_VNX_Manila/18 : FAILURE in 14m 20s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":8},{"id":"76f7cd99e96f4a3e611811c179ff7931a1ee60eb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-26 05:03:08.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/44/802644/8/check/DellEMC_Unity_Manila_dhss_true/1f9c10a/DellEMC_Unity_Manila_dhss_true/38 : FAILURE in 15m 23s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/44/802644/8/check/DellEMC_Unity_Manila_dhss_false/371a4ae/DellEMC_Unity_Manila_dhss_false/41 : FAILURE in 15m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"7d2a5dc92a663536b37511e10285a6a32d28eee1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-12-22 02:13:32.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"75ec716a6bb4e4dd80ad0758f4770bce35623c7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 04:18:12.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2fd220801574eb1b205e4d8ebe4d29b : SUCCESS in 4m 40s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/704fc46851f84b82b524eeaf66eca7ac : SUCCESS in 33m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/823d14b73a3e414ea09750adf512b5fb : SUCCESS in 34m 24s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b2578249405e4ba0999554e144ddd8c1 : SUCCESS in 1h 20m 12s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/0e233396e9e3471d8e1898641a239c35 : FAILURE in 1h 14m 50s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/1df8ef04477f45dea94afa533f02ca32 : SUCCESS in 1h 04m 56s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/140188b5bc8c44e7aafe9f8b3289c4f7 : SUCCESS in 1h 06m 36s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3a781257e099435b96620597dce6491f : FAILURE in 1h 06m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0b2a5f5bb0f44dcb8eccf64100c4ae04 : SUCCESS in 1h 00m 35s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/328dfe456c254959a812fe776c983583 : FAILURE in 1h 29m 53s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3ae60b7942b948f69dfe7cca906db001 : FAILURE in 40m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bdd02c1317304fe0a0066ff2d2ba7806 : FAILURE in 1h 39m 47s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6301b0e22c8448c68a83a568109148e4 : SUCCESS in 21m 23s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"189b49548c3e559dca9e33170ef685cd6a0aae9c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-12-22 05:54:09.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/44/802644/9/check/DellEMC_Unity_Manila_dhss_true/137c1bf/DellEMC_Unity_Manila_dhss_true/181 : FAILURE in 1h 15m 49s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/44/802644/9/check/DellEMC_Unity_Manila_dhss_false/6e692a8/DellEMC_Unity_Manila_dhss_false/174 : FAILURE in 14m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"3877950d582f85c84b551bbaed4cd2fa2a796298","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-01-10 01:51:48.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5bb8fe1c49d3f4d595d550b83081496c8240f34d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 03:46:13.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f34e298da2c49e4be0bda3177150554 : SUCCESS in 3m 44s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/856c5374e8da48a2918700122dbbf2a2 : SUCCESS in 31m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/35bd3d8bebe742739c11eb1f68e69d9e : SUCCESS in 33m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/be307f0ec75f4a55bd8bc7ef661a2532 : SUCCESS in 1h 11m 11s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/59a3ed8332f34d25a7175affa5dc5317 : SUCCESS in 1h 04m 02s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/35ee2061d97745de8dbe36ea6ba54205 : FAILURE in 1h 03m 21s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/fd341d0704dd4a85a5c2ceac4c5a294d : SUCCESS in 1h 04m 10s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fd2631afe4924b23abd9853004728c4d : FAILURE in 1h 13m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7c9d7a8e828548789dad89f72a47c9b2 : SUCCESS in 1h 16m 12s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/78118e7e1aba4cf3af61545b2818c7cf : FAILURE in 1h 42m 53s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8adc9ae9f95e47bb8d0e513ce81cdd09 : FAILURE in 50m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/68ec7eef141a40079d6839ebb0ba8cda : FAILURE in 1h 49m 06s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e5e7b252cc814cecb43641c53abdc133 : SUCCESS in 22m 38s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7b70ff89ad54a07d9ff65036c8d4d53f8702f1c7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-03 15:14:47.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c253643095b778bb3387a7b300a9c911464f398a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 17:10:11.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a11a4fb9fc046248b95dc6712acbd24 : SUCCESS in 4m 20s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4d69b3c6fbb44ed792e641a6969af5d6 : FAILURE in 33m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/315d22de4411472fa21839eef8e234da : FAILURE in 37m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f6023693a03e44b08d32a1c909bf09f8 : SUCCESS in 1h 20m 54s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/31ff6c6bc4b8496a8ed7f34dfd0e8f66 : SUCCESS in 1h 23m 27s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/38060ac9eadc4f678b28b6cfdc7861ee : SUCCESS in 1h 22m 30s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/8c88760477c143cdb484c25046128411 : FAILURE in 1h 03m 11s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/249496ca68044de5bb037a692bb36555 : FAILURE in 1h 07m 53s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/09c27f56aaa241968876990c657d09a6 : FAILURE in 1h 06m 02s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1046e1b6abcf4b12be4944b1dedcb846 : FAILURE in 1h 25m 21s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/86776217a5ae4c1f872a054505ffe46a : FAILURE in 46m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b2963fc090ae49a383261ff8deeabee6 : FAILURE in 1h 50m 28s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ec343b51a7244683980bcc539b98ac0e : FAILURE in 28m 45s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"bfc8735d6be8409c29e0e413dc08cda85b3b4ee7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-02-08 06:10:13.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"15b20530f43b400f711dcaa46ab6da3f0d2ddf78","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-02-08 06:19:04.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"9424dee4a78a0380e7ebf25d7feb2fcea2899878","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 08:05:10.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96cc1e1708fb49e399e086c84ed01ca1 : SUCCESS in 3m 29s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/156fae8e46c845fab5a0eb322f404e61 : SUCCESS in 23m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b4f2d4aac53948e7a9d7845062000f30 : SUCCESS in 31m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a06db31d1b644fbf91cde2201f13c962 : SUCCESS in 1h 05m 34s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/223b73950a094d07955e7beba3c09713 : SUCCESS in 57m 09s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/29ca62e814134818bd2da0f4db3491d9 : SUCCESS in 1h 02m 52s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/a2d420ae2d7e47af81e549746e2c6430 : SUCCESS in 59m 33s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fe70b9c1e9744edd98a9e4ab86c0c5fa : FAILURE in 59m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e1317e09a69e4507862afe8f419c937e : SUCCESS in 1h 01m 21s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/07be283e0e1940c689ddfcab55fb8bb6 : FAILURE in 1h 40m 25s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/200d5212bfa04d40aa3b9c645e1797d4 : FAILURE in 47m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6f04c680b045425688238033fa6b3e80 : FAILURE in 1h 54m 01s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/59d1fc2dda624e25810fd3eaeb6dce47 : SUCCESS in 23m 07s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f6fe5a8bd21971f87a51f81678cc754cd4cd211c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-08 08:06:34.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/44/802644/10/upstream-check/manila-cDOT-manila-tempest-plugin-ss/586321d/ : FAILURE in 50s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/44/802644/10/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/bf1104f/ : FAILURE in 50s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"693d220c6fecb9e4509f36cc56b3423705bc61fd","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-02-08 08:59:36.000000000","message":"Patch Set 10: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"2a889771611391666df94f6b04c389fcb9bbfaff","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-02-09 21:36:55.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7822c2d1c26ab1a81bfca907029be50425893ccd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-16 20:34:33.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4ad7ba3e1f3f6c06f6eea671f3b5b5c02d6c4921","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-02-16 20:42:45.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"c8d06f4298d7331c0f049fa9ae80dccb4caf5e34","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2022-02-16 21:06:37.000000000","message":"Patch Set 10: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"a9debccacdbd8c532ba9653d8ba3b410827e49d9","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-16 22:30:36.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/44/802644/10/upstream-check/manila-cDOT-manila-tempest-plugin-ss/25c98ab/ : POST_FAILURE in 9m 39s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/44/802644/10/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/ed1dc56/ : POST_FAILURE in 1h 55m 35s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"abf7d689a02297b7b30982cb4dc29893e5a560d1","author":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"date":"2022-02-17 14:20:10.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"eb785890e111e2a2d650dcb2462c35b1341edaa4","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2022-02-18 13:21:00.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4438857335c16c0ceebf85ec4b122e9e28be63c0","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-18 16:22:42.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/44/802644/10/upstream-check/manila-cDOT-manila-tempest-plugin-ss/e660734/ : POST_FAILURE in 3h 01m 14s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/44/802644/10/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/e4f2bad/ : POST_FAILURE in 2h 03m 03s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"effce1841d3a4f60f46077cdb92ff34d693d23eb","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2022-02-20 12:32:29.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"2a5012374be91c5f44eb8c00525189064cda3d9c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-22 00:11:30.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"98dbf05e7417d462b0f0bb8cf15f56385769e571","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 00:14:03.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"729f195a559403b4ae74fa65b6f8a1e8ef91f982","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 01:36:42.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6e961ba75624891a68ab893fb8fc863 : SUCCESS in 4m 10s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1e589a04eaab4d14a9cc7b7b56aee6cc : SUCCESS in 30m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d91545214a334f0cb555011f8f9173d4 : SUCCESS in 32m 19s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d9e7cefab2db478291510b78e50caef6 : SUCCESS in 1h 10m 48s","accounts_in_message":[],"_revision_number":10},{"id":"3cab287a57727684f5fff9386b81bbdcbbb2a9f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 01:36:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"1c5e68da5ebbd4e544d6a01b31c7dbe23fcecaec","revisions":{"ed53120e63629f3ea9b4c42374e4b5aed1717e6b":{"kind":"REWORK","_number":1,"created":"2021-07-28 01:55:44.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/44/802644/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/44/802644/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/1"}}},"commit":{"parents":[{"commit":"86f2432476ba9128bb2c26a78ae3b3f53327d4bd","subject":"Enable http response extraction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/86f2432476ba9128bb2c26a78ae3b3f53327d4bd"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-07-28 01:54:34.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-07-28 01:54:47.000000000","tz":480},"subject":"Add tests for manila recycle bin","message":"Add tests for manila recycle bin\n\nDepends-On: Ic838eec5fea890be6513514053329b1d2d86b3ba\nPartially-Implements: blueprint manila-share-support-recycle-bin\nChange-Id: I1eaaa0eda12690b097a7e65170064c0935259f4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ed53120e63629f3ea9b4c42374e4b5aed1717e6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ed53120e63629f3ea9b4c42374e4b5aed1717e6b"}]},"branch":"refs/heads/master"},"5857232c88f4c78a7d3cd91ec0072b92b5feb4b3":{"kind":"REWORK","_number":2,"created":"2021-07-28 02:57:22.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/44/802644/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/44/802644/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/2"}}},"commit":{"parents":[{"commit":"86f2432476ba9128bb2c26a78ae3b3f53327d4bd","subject":"Enable http response extraction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/86f2432476ba9128bb2c26a78ae3b3f53327d4bd"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-07-28 01:54:34.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-07-28 02:56:53.000000000","tz":480},"subject":"Add tests for manila recycle bin","message":"Add tests for manila recycle bin\n\nDepends-On: Ic838eec5fea890be6513514053329b1d2d86b3ba\nPartially-Implements: blueprint manila-share-support-recycle-bin\nChange-Id: I1eaaa0eda12690b097a7e65170064c0935259f4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5857232c88f4c78a7d3cd91ec0072b92b5feb4b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5857232c88f4c78a7d3cd91ec0072b92b5feb4b3"}]},"branch":"refs/heads/master"},"37295a3ed9a33a56cc6875f9a624adbc7f15a85b":{"kind":"REWORK","_number":3,"created":"2021-07-28 07:16:22.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/44/802644/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/44/802644/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/3"}}},"commit":{"parents":[{"commit":"86f2432476ba9128bb2c26a78ae3b3f53327d4bd","subject":"Enable http response extraction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/86f2432476ba9128bb2c26a78ae3b3f53327d4bd"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-07-28 01:54:34.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-07-28 07:15:35.000000000","tz":480},"subject":"Add tests for manila recycle bin","message":"Add tests for manila recycle bin\n\nDepends-On: Ic838eec5fea890be6513514053329b1d2d86b3ba\nPartially-Implements: blueprint manila-share-support-recycle-bin\nChange-Id: I1eaaa0eda12690b097a7e65170064c0935259f4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/37295a3ed9a33a56cc6875f9a624adbc7f15a85b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/37295a3ed9a33a56cc6875f9a624adbc7f15a85b"}]},"branch":"refs/heads/master"},"e0552c4c463d025268f21f249b06b774f0dea0e4":{"kind":"REWORK","_number":4,"created":"2021-07-29 08:35:40.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/44/802644/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/44/802644/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/4"}}},"commit":{"parents":[{"commit":"86f2432476ba9128bb2c26a78ae3b3f53327d4bd","subject":"Enable http response extraction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/86f2432476ba9128bb2c26a78ae3b3f53327d4bd"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-07-28 01:54:34.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-07-29 08:35:12.000000000","tz":480},"subject":"Add tests for manila recycle bin","message":"Add tests for manila recycle bin\n\nDepends-On: Ic838eec5fea890be6513514053329b1d2d86b3ba\nPartially-Implements: blueprint manila-share-support-recycle-bin\nChange-Id: I1eaaa0eda12690b097a7e65170064c0935259f4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e0552c4c463d025268f21f249b06b774f0dea0e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e0552c4c463d025268f21f249b06b774f0dea0e4"}]},"branch":"refs/heads/master"},"ac8d454c9cdcd1f03aa09553eb9438242b85ad8d":{"kind":"REWORK","_number":5,"created":"2021-09-03 09:35:16.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/44/802644/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/44/802644/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/5"}}},"commit":{"parents":[{"commit":"86f2432476ba9128bb2c26a78ae3b3f53327d4bd","subject":"Enable http response extraction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/86f2432476ba9128bb2c26a78ae3b3f53327d4bd"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-07-28 01:54:34.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-09-03 09:34:28.000000000","tz":480},"subject":"Add tests for manila recycle bin","message":"Add tests for manila recycle bin\n\nupdate micversion to 2.66\n\nDepends-On: Ic838eec5fea890be6513514053329b1d2d86b3ba\nPartially-Implements: blueprint manila-share-support-recycle-bin\nChange-Id: I1eaaa0eda12690b097a7e65170064c0935259f4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ac8d454c9cdcd1f03aa09553eb9438242b85ad8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ac8d454c9cdcd1f03aa09553eb9438242b85ad8d"}]},"branch":"refs/heads/master"},"66592e5bc0e7d1336ba69592ef1b67b96e3488f7":{"kind":"REWORK","_number":6,"created":"2021-09-15 03:21:13.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/44/802644/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/44/802644/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/6"}}},"commit":{"parents":[{"commit":"86f2432476ba9128bb2c26a78ae3b3f53327d4bd","subject":"Enable http response extraction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/86f2432476ba9128bb2c26a78ae3b3f53327d4bd"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-07-28 01:54:34.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-09-15 03:20:44.000000000","tz":480},"subject":"Add tests for manila recycle bin","message":"Add tests for manila recycle bin\n\nupdate micversion to 2.66\n\nDepends-On: Ic838eec5fea890be6513514053329b1d2d86b3ba\nPartially-Implements: blueprint manila-share-support-recycle-bin\nChange-Id: I1eaaa0eda12690b097a7e65170064c0935259f4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/66592e5bc0e7d1336ba69592ef1b67b96e3488f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/66592e5bc0e7d1336ba69592ef1b67b96e3488f7"}]},"branch":"refs/heads/master"},"30fa34d9afc6fe8d4796a347917bdb637d74f297":{"kind":"REWORK","_number":7,"created":"2021-09-15 05:55:58.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/44/802644/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/44/802644/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/7"}}},"commit":{"parents":[{"commit":"86f2432476ba9128bb2c26a78ae3b3f53327d4bd","subject":"Enable http response extraction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/86f2432476ba9128bb2c26a78ae3b3f53327d4bd"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-07-28 01:54:34.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-09-15 05:55:33.000000000","tz":480},"subject":"Add tests for manila recycle bin","message":"Add tests for manila recycle bin\n\nupdate micversion to 2.66\n\nDepends-On: Ic838eec5fea890be6513514053329b1d2d86b3ba\nPartially-Implements: blueprint manila-share-support-recycle-bin\nChange-Id: I1eaaa0eda12690b097a7e65170064c0935259f4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/30fa34d9afc6fe8d4796a347917bdb637d74f297"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/30fa34d9afc6fe8d4796a347917bdb637d74f297"}]},"branch":"refs/heads/master"},"5f29fe50af610cd855a10eeae240549bb02fa4df":{"kind":"REWORK","_number":8,"created":"2021-10-26 02:16:59.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/44/802644/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/44/802644/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/8"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-07-28 01:54:34.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2021-10-26 02:16:38.000000000","tz":480},"subject":"Add tests for manila recycle bin","message":"Add tests for manila recycle bin\n\nupdate micversion to 2.66\n\nDepends-On: Ic838eec5fea890be6513514053329b1d2d86b3ba\nPartially-Implements: blueprint manila-share-support-recycle-bin\nChange-Id: I1eaaa0eda12690b097a7e65170064c0935259f4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5f29fe50af610cd855a10eeae240549bb02fa4df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5f29fe50af610cd855a10eeae240549bb02fa4df"}]},"branch":"refs/heads/master"},"ef85a7c61870791a782b850bec8f611010bd1acb":{"kind":"REWORK","_number":9,"created":"2021-12-22 02:13:32.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/44/802644/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/44/802644/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/9"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-07-28 01:54:34.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2021-12-22 02:13:03.000000000","tz":480},"subject":"API2.67,Add tests for manila recycle bin.","message":"API2.67,Add tests for manila recycle bin.\n\nupdate micversion to 2.67\n\nDepends-On: Ic838eec5fea890be6513514053329b1d2d86b3ba\nPartially-Implements: blueprint manila-share-support-recycle-bin\nChange-Id: I1eaaa0eda12690b097a7e65170064c0935259f4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ef85a7c61870791a782b850bec8f611010bd1acb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ef85a7c61870791a782b850bec8f611010bd1acb"}]},"branch":"refs/heads/master"},"1c5e68da5ebbd4e544d6a01b31c7dbe23fcecaec":{"kind":"REWORK","_number":10,"created":"2022-02-08 06:10:13.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/44/802644/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/44/802644/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/802644/10"}}},"commit":{"parents":[{"commit":"73e0b6fdc450b7539bf6a97349f5774177ffb24a","subject":"Added code to show share network neutron_net_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/73e0b6fdc450b7539bf6a97349f5774177ffb24a"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-07-28 01:54:34.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-02-08 06:09:43.000000000","tz":480},"subject":"API2.69,Add tests for manila recycle bin.","message":"API2.69,Add tests for manila recycle bin.\n\nupdate micversion to 2.69\n\nDepends-On: Ic838eec5fea890be6513514053329b1d2d86b3ba\nPartially-Implements: blueprint manila-share-support-recycle-bin\nChange-Id: I1eaaa0eda12690b097a7e65170064c0935259f4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1c5e68da5ebbd4e544d6a01b31c7dbe23fcecaec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1c5e68da5ebbd4e544d6a01b31c7dbe23fcecaec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
