)]}'
{"id":"openstack%2Fmanila-tempest-plugin~805319","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ib39cb2bda862758f575262647e068dd9deaf1459","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1940324","hashtags":[],"change_id":"Ib39cb2bda862758f575262647e068dd9deaf1459","subject":"[CI] Temporarily disable IPv6 scenario tests","status":"MERGED","created":"2021-08-20 00:01:32.000000000","updated":"2021-08-24 15:17:17.000000000","submitted":"2021-08-24 15:17:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"805319-bug/1940324","meta_rev_id":"db26be3cdf2ac37330c118da7e1fa10472b012fa","_number":805319,"virtual_id_number":805319,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-24 15:17:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2021-08-24 14:10:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2021-08-24 02:45:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-24 04:18:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2021-08-24 14:10:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"reviewers":{"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-20 00:01:51.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-20 00:01:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-20 02:26:54.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-08-23 23:03:24.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-24 00:15:48.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2021-08-24 00:15:48.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2021-08-24 00:15:48.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-08-24 00:15:48.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2021-08-24 00:16:03.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2021-08-24 00:30:39.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"8ceb40ef8c366ddcfd1fd51bc301b3e07ed4f116","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-08-20 00:01:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9177fd5645b9e58b89d5a324997f72d9f986802b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-08-20 00:01:51.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 63, column 32","accounts_in_message":[],"_revision_number":1},{"id":"1c2801bae7e6610c5694ed82d2f0d0de7f1c2411","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 00:01:52.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 63, column 32","accounts_in_message":[],"_revision_number":1},{"id":"64b6cf3f774b1329936860cd196f2c4981ffabbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-08-20 00:07:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"298bf7ae5b18368be5cce6282f5a50ec793a44ef","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-08-20 00:12:09.000000000","message":"Topic set to bug/1940324","accounts_in_message":[],"_revision_number":2},{"id":"1f3a0b80b606e4034faa773b6142d475b23ce8fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 02:11:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b88ab4fd04d94a0ea4ab0456ec7a9163 : SUCCESS in 4m 16s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4f566856e9a04662870a087f5852bd99 : SUCCESS in 26m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/47ce8dd7fdc34fce91122178eccd1ea8 : SUCCESS in 41m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/142b5240051f4085aab0b0b37a6bacbb : FAILURE in 59m 34s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/205364fef422419a894dbcc306307eba : FAILURE in 1h 54m 03s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/f6fceb233783448baf93175cec177f85 : SUCCESS in 1h 16m 39s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/8b7e9cd6a3444f0e9f623643ab4170b4 : SUCCESS in 1h 09m 30s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1fa89bcc838e46c8b2cac803506b5594 : FAILURE in 1h 06m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0c3f2de8212d4777a51ef17f87b4b605 : SUCCESS in 59m 11s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c8f6e1c1418845979cec71e5a55ae125 : FAILURE in 1h 23m 21s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3573b6994d9e4cc6b97465c111e46e0b : SUCCESS in 45m 34s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/aaa4d1ea455443e98cad77fe9580e379 : FAILURE in 1h 46m 04s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/22abdb4a1d394ccbb52cb36baa2c654f : FAILURE in 27m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0116e519f661356a023909c49fda4e0051dad993","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-08-20 02:26:54.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/19/805319/2/check/DellEMC_Unity_Manila_dhss_true/d49fc4d/DellEMC_Unity_Manila_dhss_true/280 : FAILURE in 2h 18m 56s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/19/805319/2/check/DellEMC_Unity_Manila_dhss_false/2e19842/DellEMC_Unity_Manila_dhss_false/277 : SUCCESS in 1h 34m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"174f495e15490a7bb377016f9c918500ec262c4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-08-20 05:43:11.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"7e1dc4c005e68fbde6e7bd4e8fba48f61585ec19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 07:28:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce3744e374344ee3a3b497e47f27138d : SUCCESS in 3m 55s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/662e58caf4044c2eb0c854f77eb6b9ea : SUCCESS in 33m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/55c3024e48ee47ff9de03560bce421cd : SUCCESS in 38m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7b16ba1e04384efb921ddd4899994db6 : FAILURE in 42m 54s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/b3e3a2d442af4a33b90ef340344fb21c : SUCCESS in 1h 37m 13s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/9973cda52b094022aac6da5ccda32dd4 : SUCCESS in 1h 10m 08s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/88db9841645e46f8841ae7e1fe819e2b : SUCCESS in 1h 14m 07s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ec8f45eae2ef4140aac6d42fd37f9a1c : SUCCESS in 1h 10m 12s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b7bc3439b25e46a4bf8cc55ad571ddaa : SUCCESS in 1h 38m 31s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/dbc37dd5f4cc4274b949adad667914cc : FAILURE in 1h 39m 32s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8cfe06fd45184852b4c2e3816d6a896a : SUCCESS in 46m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/697367568eb1455988b6ce499ffa745f : FAILURE in 1h 40m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/68b9cb3ce8a1470ea84b388d678f0f9f : FAILURE in 23m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d17534492ea529e316f7b3aeebd594288bf6ed9e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-08-20 12:22:04.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/19/805319/3/check/DellEMC_Unity_Manila_dhss_true/deeafc1/DellEMC_Unity_Manila_dhss_true/283 : FAILURE in 2h 41m 55s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/19/805319/3/check/DellEMC_Unity_Manila_dhss_false/90a5661/DellEMC_Unity_Manila_dhss_false/280 : FAILURE in 2h 10m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"8b36a95bc80937b2e303adf545767cd505ca2c9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-08-20 14:04:09.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"edeb17aadd1dd07fb26afa9e0445ffd34aae8d8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 15:59:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aff5df005704d039085057c82ea5fd1 : SUCCESS in 4m 09s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5bd9da7cb4ac49abb8488335f37635fe : SUCCESS in 33m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b86322db456b477d8ead43579ac23573 : SUCCESS in 35m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6ab9b72bb8564cbf897df9beeb4eaf52 : FAILURE in 39m 02s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/c24911ab21c74bb8b6781216991d66b6 : SUCCESS in 1h 14m 34s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/e7f8df056e7942d289f0abc67b1b8952 : SUCCESS in 1h 08m 40s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/fe8394423bfb448eb3be4284211cf8c9 : SUCCESS in 1h 41m 46s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7b534eb498914b54bfdd1c91515f4d29 : SUCCESS in 1h 08m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/25f433f615504bedbfeb43674b4bd5df : SUCCESS in 1h 18m 59s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/308e1fa46e824670a666f613553e6c62 : FAILURE in 1h 35m 46s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/49b58189c6ab47bbbda9cb6791ecba00 : SUCCESS in 49m 19s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fd9d5dab000b4a798b850f05ffe7ba05 : FAILURE in 1h 50m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/bc9737c1bc394eaf8ae4297c5498ab1b : SUCCESS in 28m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b880a8064dcafdc7a1a5050c720fefc3f1c35c5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-08-20 17:25:04.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"40e8d4472f238af1e68f4378ec02fcd0eba547fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 19:12:44.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e7ef1f5159d4ccfa447e334e0c3f585 : SUCCESS in 4m 13s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/70e91b0af9f34b6a8284f33f93a3c012 : SUCCESS in 27m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f64652c48e3240099a9c58d696f404e2 : SUCCESS in 35m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4420fcc520c44220acd6a3f0ec95a891 : FAILURE in 1h 17m 12s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/0ba9e3924dc143c0b741347676f6cb41 : SUCCESS in 1h 08m 19s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/d6b29d9b48ed4f0d98d078cc40743d1e : SUCCESS in 1h 08m 05s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/03729c67c40c44d780fc012057bf7559 : SUCCESS in 1h 30m 35s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0a6cefbf79fe4f68a8e4fc0b131351a2 : SUCCESS in 1h 02m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3c994fed9ef24ca5a44adc942c0fc630 : SUCCESS in 1h 09m 57s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/441c30870f754c899dc2fd52f52decb8 : FAILURE in 1h 42m 29s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f4a3288033e44b16bff1b294ca2ababd : SUCCESS in 44m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/13197177f2754a19836b89900a2cd524 : FAILURE in 1h 42m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/240fe66c1f6a4058be175b0a3ca7904b : FAILURE in 26m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4dc3877de8a18365b597eefee71fe0d6dbfd6ffb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-08-20 19:56:11.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/19/805319/5/check/DellEMC_Unity_Manila_dhss_true/86d7845/DellEMC_Unity_Manila_dhss_true/286 : FAILURE in 2h 30m 47s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/19/805319/5/check/DellEMC_Unity_Manila_dhss_false/20f4478/DellEMC_Unity_Manila_dhss_false/283 : FAILURE in 1h 51m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"a272053395ee7b74ff9e7f53b0712598a91e66e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-08-23 19:54:57.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"6ac7e664adb2f483bdf51d83de1f827bc76f24fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-08-23 21:17:59.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"cac23e7fc67587e394bc668289947ac072932571","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 23:03:05.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c5f29bee07945ceadd5b70d46d946d6 : SUCCESS in 5m 07s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/37e75fc0aef54bd38e1b607c6480bccc : SUCCESS in 26m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0cd6664e39a4455dbf962f3b73e21046 : SUCCESS in 36m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fb69a67f827541bda9424d09054dccae : SUCCESS in 1h 00m 55s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/be9e7b2ed6af4e71807af1979dbe8b49 : SUCCESS in 1h 14m 29s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/b081fed0cbba4e1abab1fd8ba24cc0ed : SUCCESS in 1h 01m 54s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/95b2283ca0454d289921363db2ca2464 : SUCCESS in 1h 08m 29s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fe33cc6ada044bb4b8da12529854b07d : FAILURE in 1h 01m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/bc11dc4cf3b64115bfc89f2e653043ac : SUCCESS in 1h 09m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/05a2bec6b5334d829d417fa80d7b5ae9 : FAILURE in 1h 43m 52s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c1ac412255e24b69a1d03263ebd0e6c1 : SUCCESS in 45m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4c29a4343b2e44149c749ab2043ebd52 : FAILURE in 1h 39m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/488d1a99b2024ea7b9dc147dfcb52114 : SUCCESS in 25m 07s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ec67074c24d6060e134fe221466dfbd9428d8329","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-23 23:03:24.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/19/805319/7/upstream-check/manila-cDOT-manila-tempest-plugin-ss/82d2989/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/19/805319/7/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/3410624/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"d54b3762de707f011c2417b8b3ba2529cbfe0949","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-08-24 00:30:39.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/19/805319/7/check/DellEMC_VNX_Manila/5c8dd85/DellEMC_VNX_Manila/130 : SUCCESS in 1h 27m 18s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"8af28cff0df573d36915bdacfde4386db952665a","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-08-24 02:45:04.000000000","message":"Patch Set 7: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":7},{"id":"4ce9a06836e430a1975c8adace3a6dcd0e8a207a","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-08-24 04:18:43.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)\n\nThanks Goutham!","accounts_in_message":[],"_revision_number":7},{"id":"23f02e1511126a993010e06e9aab897b5a004241","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-08-24 14:10:09.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nLGTM. Thank you for disabling this tests.","accounts_in_message":[],"_revision_number":7},{"id":"727a688253f8452f71b6da9cfce0577d78ff6f4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 14:10:46.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"88df1d5222b8802ca82d64b6eea5d62bdade97bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 15:17:17.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4c4a526d6bc444880839f9e00e3360e : SUCCESS in 3m 56s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/eff000130b2b470e97c397ca774781ca : SUCCESS in 31m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d56db9b7db5d4889b595e1d6a9852163 : SUCCESS in 36m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c23b162e2f3649b6949d994871f4b831 : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":7},{"id":"db26be3cdf2ac37330c118da7e1fa10472b012fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 15:17:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"8ccbbebe6d424265ff717197b9a323917518e8c5","revisions":{"16cdd3d3339543a42e940d55c8b5f70738a440d7":{"kind":"REWORK","_number":1,"created":"2021-08-20 00:01:32.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/19/805319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/805319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/1"}}},"commit":{"parents":[{"commit":"da415b4f1c3107ca05245ed9abd53a3ab67d83e3","subject":"Fix parsing of share group type json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da415b4f1c3107ca05245ed9abd53a3ab67d83e3"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-08-20 00:01:21.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-08-20 00:01:21.000000000","tz":-420},"subject":"[CI] Temporarily disable IPv6 scenario tests","message":"[CI] Temporarily disable IPv6 scenario tests\n\nWe\u0027re experiencing failures to SSH into test\nVMs over IPv6 IP addresses from private tenant\nnetworks in the LVM driver job when testing\nmain branch code. It\u0027s possible this is a regression\nin nova, neutron, neutron dynamic routing or bgp. We\u0027ll\ndrop this test exclusion when the bug has been\nidentified and fixed.\n\nChange-Id: Ib39cb2bda862758f575262647e068dd9deaf1459\nPartial-Bug: #1940324\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/16cdd3d3339543a42e940d55c8b5f70738a440d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/16cdd3d3339543a42e940d55c8b5f70738a440d7"}]},"branch":"refs/heads/master"},"b736a44dc53e581d945545aec0c247211e8ebd6c":{"kind":"REWORK","_number":2,"created":"2021-08-20 00:07:43.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/19/805319/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/805319/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/2"}}},"commit":{"parents":[{"commit":"da415b4f1c3107ca05245ed9abd53a3ab67d83e3","subject":"Fix parsing of share group type json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da415b4f1c3107ca05245ed9abd53a3ab67d83e3"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-08-20 00:01:21.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-08-20 00:07:37.000000000","tz":-420},"subject":"[CI] Temporarily disable IPv6 scenario tests","message":"[CI] Temporarily disable IPv6 scenario tests\n\nWe\u0027re experiencing failures to SSH into test\nVMs over IPv6 IP addresses from private tenant\nnetworks in the LVM driver job when testing\nmain branch code. It\u0027s possible this is a regression\nin nova, neutron, neutron dynamic routing or bgp. We\u0027ll\ndrop this test exclusion when the bug has been\nidentified and fixed.\n\nChange-Id: Ib39cb2bda862758f575262647e068dd9deaf1459\nPartial-Bug: #1940324\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b736a44dc53e581d945545aec0c247211e8ebd6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b736a44dc53e581d945545aec0c247211e8ebd6c"}]},"branch":"refs/heads/master"},"82108dc15d2c7d2b144102fe4627dc87dbf8ca9c":{"kind":"REWORK","_number":3,"created":"2021-08-20 05:43:11.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/19/805319/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/805319/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/3"}}},"commit":{"parents":[{"commit":"da415b4f1c3107ca05245ed9abd53a3ab67d83e3","subject":"Fix parsing of share group type json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da415b4f1c3107ca05245ed9abd53a3ab67d83e3"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-08-20 00:01:21.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-08-20 05:43:07.000000000","tz":0},"subject":"[CI] Temporarily disable IPv6 scenario tests","message":"[CI] Temporarily disable IPv6 scenario tests\n\nWe\u0027re experiencing failures to SSH into test\nVMs over IPv6 IP addresses from private tenant\nnetworks in the LVM driver job when testing\nmain branch code. It\u0027s possible this is a regression\nin nova, neutron, neutron dynamic routing or bgp. We\u0027ll\ndrop this test exclusion when the bug has been\nidentified and fixed.\n\nChange-Id: Ib39cb2bda862758f575262647e068dd9deaf1459\nPartial-Bug: #1940324\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/82108dc15d2c7d2b144102fe4627dc87dbf8ca9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/82108dc15d2c7d2b144102fe4627dc87dbf8ca9c"}]},"branch":"refs/heads/master"},"52c9ce73776ddb5b6ced7d2ca400636196f82581":{"kind":"REWORK","_number":4,"created":"2021-08-20 14:04:09.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/19/805319/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/805319/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/4"}}},"commit":{"parents":[{"commit":"da415b4f1c3107ca05245ed9abd53a3ab67d83e3","subject":"Fix parsing of share group type json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da415b4f1c3107ca05245ed9abd53a3ab67d83e3"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-08-20 00:01:21.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-08-20 14:04:05.000000000","tz":0},"subject":"[CI] Temporarily disable IPv6 scenario tests","message":"[CI] Temporarily disable IPv6 scenario tests\n\nWe\u0027re experiencing failures to SSH into test\nVMs over IPv6 IP addresses from private tenant\nnetworks in the LVM driver job when testing\nmain branch code. It\u0027s possible this is a regression\nin nova, neutron, neutron dynamic routing or bgp. We\u0027ll\ndrop this test exclusion when the bug has been\nidentified and fixed.\n\nChange-Id: Ib39cb2bda862758f575262647e068dd9deaf1459\nPartial-Bug: #1940324\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/52c9ce73776ddb5b6ced7d2ca400636196f82581"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/52c9ce73776ddb5b6ced7d2ca400636196f82581"}]},"branch":"refs/heads/master"},"dc60ee4d17ec45a865ef14a4338713fb6c047a60":{"kind":"REWORK","_number":5,"created":"2021-08-20 17:25:04.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/19/805319/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/805319/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/5"}}},"commit":{"parents":[{"commit":"da415b4f1c3107ca05245ed9abd53a3ab67d83e3","subject":"Fix parsing of share group type json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da415b4f1c3107ca05245ed9abd53a3ab67d83e3"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-08-20 00:01:21.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-08-20 17:25:01.000000000","tz":0},"subject":"[CI] Temporarily disable IPv6 scenario tests","message":"[CI] Temporarily disable IPv6 scenario tests\n\nWe\u0027re experiencing failures to SSH into test\nVMs over IPv6 IP addresses from private tenant\nnetworks in the LVM driver job when testing\nmain branch code. It\u0027s possible this is a regression\nin nova, neutron, neutron dynamic routing or bgp. We\u0027ll\ndrop this test exclusion when the bug has been\nidentified and fixed.\n\nChange-Id: Ib39cb2bda862758f575262647e068dd9deaf1459\nPartial-Bug: #1940324\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dc60ee4d17ec45a865ef14a4338713fb6c047a60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dc60ee4d17ec45a865ef14a4338713fb6c047a60"}]},"branch":"refs/heads/master"},"256189770730d2995a4be4daf76174ed0e8fb3f5":{"kind":"REWORK","_number":6,"created":"2021-08-23 19:54:57.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/19/805319/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/805319/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/6"}}},"commit":{"parents":[{"commit":"da415b4f1c3107ca05245ed9abd53a3ab67d83e3","subject":"Fix parsing of share group type json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da415b4f1c3107ca05245ed9abd53a3ab67d83e3"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-08-20 00:01:21.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-08-23 19:54:53.000000000","tz":0},"subject":"[CI] Temporarily disable IPv6 scenario tests","message":"[CI] Temporarily disable IPv6 scenario tests\n\nWe\u0027re experiencing failures to SSH into test\nVMs over IPv6 IP addresses from private tenant\nnetworks in the LVM driver job when testing\nmain branch code. It\u0027s possible this is a regression\nin nova, neutron, neutron dynamic routing or bgp. We\u0027ll\ndrop this test exclusion when the bug has been\nidentified and fixed.\n\nChange-Id: Ib39cb2bda862758f575262647e068dd9deaf1459\nPartial-Bug: #1940324\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/256189770730d2995a4be4daf76174ed0e8fb3f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/256189770730d2995a4be4daf76174ed0e8fb3f5"}]},"branch":"refs/heads/master"},"8ccbbebe6d424265ff717197b9a323917518e8c5":{"kind":"REWORK","_number":7,"created":"2021-08-23 21:17:59.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/19/805319/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/19/805319/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/19/805319/7"}}},"commit":{"parents":[{"commit":"da415b4f1c3107ca05245ed9abd53a3ab67d83e3","subject":"Fix parsing of share group type json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da415b4f1c3107ca05245ed9abd53a3ab67d83e3"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-08-20 00:01:21.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-08-23 21:17:54.000000000","tz":0},"subject":"[CI] Temporarily disable IPv6 scenario tests","message":"[CI] Temporarily disable IPv6 scenario tests\n\nWe\u0027re experiencing failures to SSH into test\nVMs over IPv6 IP addresses from private tenant\nnetworks in the LVM driver job when testing\nmain branch code. It\u0027s possible this is a regression\nin nova, neutron, neutron dynamic routing or bgp. We\u0027ll\ndrop this test exclusion when the bug has been\nidentified and fixed.\n\nChange-Id: Ib39cb2bda862758f575262647e068dd9deaf1459\nPartial-Bug: #1940324\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8ccbbebe6d424265ff717197b9a323917518e8c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8ccbbebe6d424265ff717197b9a323917518e8c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
